)]}'
{"id":"openstack%2Fpython-tripleoclient~663384","triplet_id":"openstack%2Fpython-tripleoclient~master~Ia86f6f71eddcd4aa8505601ba51b4723e05cf488","project":"openstack/python-tripleoclient","branch":"master","topic":"tox/sphinx","hashtags":[],"change_id":"Ia86f6f71eddcd4aa8505601ba51b4723e05cf488","subject":"Change source for upper constraints","status":"MERGED","created":"2019-06-05 17:08:44.000000000","updated":"2019-06-06 16:41:50.000000000","submitted":"2019-06-06 16:41:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663384-1559839310270-589de8e3","meta_rev_id":"9e388412bd0744e337e88bdd85b4451ea5be0466","_number":663384,"virtual_id_number":663384,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-06 16:41:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2019-06-06 02:09:12.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2019-06-06 13:23:26.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2019-06-06 13:23:26.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 02:09:12.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2019-06-06 13:23:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-06 16:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8dc7838345bea202e9aa1a69ab93b9b9630b222f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-05 17:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"deedd36610970be99bca87f685b86e8c4f4459a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:29:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663384/1/check/openstack-tox-lower-constraints/dac37dd/ : SUCCESS in 5m 15s\n- openstackclient-check-plugins http://logs.openstack.org/84/663384/1/check/openstackclient-check-plugins/b01afa7/ : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/663384/1/check/openstack-tox-pep8/494027b/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/84/663384/1/check/openstack-tox-py27/d3e4c35/ : SUCCESS in 2m 56s\n- openstack-tox-py35 http://logs.openstack.org/84/663384/1/check/openstack-tox-py35/73e92ec/ : SUCCESS in 3m 47s\n- openstack-tox-py36 http://logs.openstack.org/84/663384/1/check/openstack-tox-py36/0ea19dd/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/84/663384/1/check/openstack-tox-py37/4811e9b/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/84/663384/1/check/openstack-tox-docs/592dedc/html/ : SUCCESS in 2m 48s\n- build-openstack-releasenotes http://logs.openstack.org/84/663384/1/check/build-openstack-releasenotes/34e391c/html/ : SUCCESS in 2m 47s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/84/663384/1/check/tripleo-buildimage-overcloud-full-centos-7/823adcd/ : SUCCESS in 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2741423a9799bc40636eb9fd8e850361881b5ed1","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-06-06 02:09:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb7a212fe7e979c880617d12e71dc94dfefd37cb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 13:23:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthe new url just 302\u0027s to the other one, so this doesn\u0027t really change anything","accounts_in_message":[],"_revision_number":1},{"id":"6e1bd7247d8591c367cc67e6993c88e1ee37fcfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 13:23:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f67de1bde1100cc0d4b75e8d059aec827616cd4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:41:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/84/663384/1/gate/openstack-tox-lower-constraints/11aece6/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/84/663384/1/gate/openstack-tox-pep8/d0605ff/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/84/663384/1/gate/openstack-tox-py27/d6489a7/ : SUCCESS in 2m 59s\n- openstack-tox-py35 http://logs.openstack.org/84/663384/1/gate/openstack-tox-py35/cd36b5f/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/84/663384/1/gate/openstack-tox-py36/aa7d4ec/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/84/663384/1/gate/openstack-tox-py37/1b0e12a/ : SUCCESS in 3m 13s\n- openstack-tox-docs http://logs.openstack.org/84/663384/1/gate/openstack-tox-docs/4d584ea/html/ : SUCCESS in 3m 17s\n- build-openstack-releasenotes http://logs.openstack.org/84/663384/1/gate/build-openstack-releasenotes/413cb71/html/ : SUCCESS in 3m 05s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/84/663384/1/gate/tripleo-buildimage-overcloud-full-centos-7/384b29f/ : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9cd41ea510492b84a8445433d0fdd16e6ecbf0cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:41:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","revisions":{"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5":{"kind":"REWORK","_number":1,"created":"2019-06-05 17:08:44.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/84/663384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/84/663384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/84/663384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/84/663384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/84/663384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/84/663384/1"}}},"commit":{"parents":[{"commit":"e11b687c2aa75ed34be0dec1e5daca06f385c265","subject":"Merge \"Remove tripleo-ui items from the undercloud install\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e11b687c2aa75ed34be0dec1e5daca06f385c265"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 17:07:59.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 17:07:59.000000000","tz":-240},"subject":"Change source for upper constraints","message":"Change source for upper constraints\n\npy27 is currently broken due to a Sphinx failure\nthat is resolved if you properly include the upper constraints.\n\nChange-Id: Ia86f6f71eddcd4aa8505601ba51b4723e05cf488\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
