)]}'
{"id":"openstack%2Fpython-tripleoclient~663425","triplet_id":"openstack%2Fpython-tripleoclient~master~I1e39fa5c220390fcd658147d9f47ae2196eefe56","project":"openstack/python-tripleoclient","branch":"master","topic":"sphinx","hashtags":[],"change_id":"I1e39fa5c220390fcd658147d9f47ae2196eefe56","subject":"doc/requirements: fix sphinx requirement lines","status":"MERGED","created":"2019-06-05 20:41:42.000000000","updated":"2019-06-15 22:06:27.000000000","submitted":"2019-06-15 22:06:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663425-1560636387274-b1f8f8f4","meta_rev_id":"3f92cef8472e277ba744e1ba255a91b7a98d4d18","_number":663425,"virtual_id_number":663425,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-06-15 22:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-15 18:29:39.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-06-13 13:32:28.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-14 14:27:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-06-13 13:32:28.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 13:32:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-14 14:27:02.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-15 18:29:39.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-15 22:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c90310dee9aa18c534efcf1f48fcc44f31d8c03f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-05 20:41:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfafc6a96eb564df1dfc0d4a80723681ee1e85f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 21:43:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/1/check/requirements-check/aa5cf9a/ : FAILURE in 1m 50s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/1/check/openstackclient-check-plugins/84ddfa2/ : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/1/check/openstack-tox-pep8/eea10fb/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/25/663425/1/check/openstack-tox-docs/b807321/html/ : SUCCESS in 3m 03s\n- build-openstack-releasenotes http://logs.openstack.org/25/663425/1/check/build-openstack-releasenotes/4ac8245/html/ : SUCCESS in 2m 51s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/1/check/tripleo-buildimage-overcloud-full-centos-7/aab5507/ : SUCCESS in 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5665408ee22836326e85533b000ca2168449242c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 13:24:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfb54774715981db89cc832c80f608fdccdbbabd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 13:25:21.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c631a14723dba312f7603296d29481aeb0eaf0d7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 13:25:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4454ce0099a29709ba51b235756428419fdf34b5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 13:43:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"11a98f3fabcf8412293a66c9399e1486c2728e63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 18:11:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/3/check/requirements-check/4c74f8c/ : FAILURE in 3m 06s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/3/check/openstackclient-check-plugins/d68b886/ : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/3/check/openstack-tox-pep8/0c39652/ : SUCCESS in 2m 52s\n- openstack-tox-docs http://logs.openstack.org/25/663425/3/check/openstack-tox-docs/cf19b74/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/25/663425/3/check/build-openstack-releasenotes/13fad20/html/ : SUCCESS in 2m 31s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/3/check/tripleo-buildimage-overcloud-full-centos-7/c51bc5b/ : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":3},{"id":"f6359a68f93eeda82a3e574f61150784a6fa0cd1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 18:51:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a6eb7578873c7575696b16c71fcf1f725cfcb08f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-06 19:22:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"542e72d82dce44a95a02aa48ab7a5fdce214ae5f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 21:17:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3281e90/ : SUCCESS in 2h 16m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/062c957/ : SUCCESS in 2h 24m 22s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/807f82c/ : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":4},{"id":"b4684bfc2da9500a91a99d6d8c679832ae85354a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-07 02:15:12.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"63dd729495321d28cc7363c63e057cdd2777b69b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 02:31:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/4/check/requirements-check/b9222f6/ : FAILURE in 2m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/4/check/openstack-tox-lower-constraints/326ca6f/ : SUCCESS in 5m 00s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/4/check/openstackclient-check-plugins/0ae13ad/ : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/4/check/openstack-tox-pep8/8046bfd/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/4/check/openstack-tox-py27/442f2ab/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/4/check/openstack-tox-py35/41b22aa/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/4/check/openstack-tox-py36/b760e35/ : SUCCESS in 3m 23s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/4/check/openstack-tox-py37/31bc93e/ : SUCCESS in 3m 44s\n- openstack-tox-docs http://logs.openstack.org/25/663425/4/check/openstack-tox-docs/ec0b8c4/html/ : SUCCESS in 3m 07s\n- build-openstack-releasenotes http://logs.openstack.org/25/663425/4/check/build-openstack-releasenotes/97c1c43/html/ : SUCCESS in 3m 15s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/4/check/tripleo-buildimage-overcloud-full-centos-7/63cfb3c/ : SUCCESS in 20m 04s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/4/check/tripleo-build-containers-centos-7/55aa3a8/ : SUCCESS in 43m 18s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/4/check/tripleo-build-containers-centos-7-stein/9b74506/ : SUCCESS in 45m 23s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/4/check/tripleo-build-containers-centos-7-rocky/97ff51a/ : SUCCESS in 38m 21s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/4/check/tripleo-build-containers-centos-7-buildah/22765ea/ : SUCCESS in 1h 06m 57s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/4/check/tripleo-build-containers-centos-7-buildah-stein/7336675/ : SUCCESS in 59m 45s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-containers-multinode/b6e7971/ : SUCCESS in 2h 13m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-undercloud-containers/9dde2c2/ : SUCCESS in 1h 31m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/302ca89/ : SUCCESS in 2h 16m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a76d622/ : SUCCESS in 2h 19m 41s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-standalone/5441dc0/ : SUCCESS in 2h 00m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/4/check/tripleo-ci-fedora-28-standalone/2385d29/ : SUCCESS in 1h 44m 39s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-standalone-upgrade/5425ac5/ : SUCCESS in 1h 52m 51s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/cbb2e1d/ : SUCCESS in 1h 57m 32s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d599514/ : FAILURE in 1h 28m 52s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/050e915/ : SUCCESS in 2h 17m 45s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/7249e91/ : FAILURE in 1h 19m 36s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/333f13b/ : SUCCESS in 2h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ed16ad3faf776f7c86c076a753c108dd352482e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-10 02:26:15.000000000","message":"Patch Set 4: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"7881428911172213fea14986726daee0a2eab4c6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:25:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8a23a3d5a1f58c0ed7bfdd3a79b72dd78278ea02","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 16:06:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bf7d3be/ : SUCCESS in 2h 39m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a7f5c68/ : SUCCESS in 2h 26m 23s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/83b95d4/ : SUCCESS in 1h 13m 42s","accounts_in_message":[],"_revision_number":5},{"id":"b20c2769ee3c64cfa5421d538562948c750696af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:31:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/5/check/requirements-check/8096b7e/ : FAILURE in 2m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/5/check/openstack-tox-lower-constraints/eac445d/ : SUCCESS in 4m 54s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/5/check/openstackclient-check-plugins/f826b3a/ : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/5/check/openstack-tox-pep8/d35f409/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/5/check/openstack-tox-py27/fa0bb75/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/5/check/openstack-tox-py35/26b1b1d/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/5/check/openstack-tox-py36/c2b2de0/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/5/check/openstack-tox-py37/7354597/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/25/663425/5/check/openstack-tox-docs/597b2f8/html/ : SUCCESS in 2m 55s\n- build-openstack-releasenotes http://logs.openstack.org/25/663425/5/check/build-openstack-releasenotes/1768667/html/ : SUCCESS in 3m 42s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/5/check/tripleo-buildimage-overcloud-full-centos-7/928a5ec/ : SUCCESS in 19m 36s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/5/check/tripleo-build-containers-centos-7/8b967a8/ : SUCCESS in 39m 30s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/5/check/tripleo-build-containers-centos-7-stein/0c8caf0/ : SUCCESS in 55m 44s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/5/check/tripleo-build-containers-centos-7-rocky/cd39814/ : SUCCESS in 38m 01s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/5/check/tripleo-build-containers-centos-7-buildah/be83255/ : SUCCESS in 1h 00m 08s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/5/check/tripleo-build-containers-centos-7-buildah-stein/3614979/ : SUCCESS in 55m 05s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-containers-multinode/13a09c1/ : SUCCESS in 2h 17m 38s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-undercloud-containers/311f334/ : SUCCESS in 1h 50m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/87ff9f7/ : SUCCESS in 2h 16m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2f4cd59/ : SUCCESS in 2h 34m 37s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-standalone/55859f7/ : SUCCESS in 2h 05m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/5/check/tripleo-ci-fedora-28-standalone/bdd2983/ : SUCCESS in 1h 55m 25s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-standalone-upgrade/bdeca92/ : SUCCESS in 2h 03m 39s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/4beb6b3/ : SUCCESS in 2h 20m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c05d4bd/ : FAILURE in 1h 31m 28s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/57ac673/ : SUCCESS in 2h 23m 02s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/728ff68/ : FAILURE in 1h 34m 31s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/90a019c/ : SUCCESS in 2h 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"216a6962ca0fb0fa7b547ff39245387d69045e78","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 16:45:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"77c77d50e22522879d82f2042a6eb38ed184c45d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-10 19:25:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9df916c/ : SUCCESS in 2h 31m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/412c882/ : SUCCESS in 2h 38m 47s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/d1d0ffd/ : SUCCESS in 1h 21m 36s","accounts_in_message":[],"_revision_number":6},{"id":"36b450fde86abdeba5def012dc5f5271e72e4260","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 00:01:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/6/check/requirements-check/d7b1895/ : SUCCESS in 2m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/6/check/openstack-tox-lower-constraints/c83ac06/ : SUCCESS in 5m 07s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/6/check/openstackclient-check-plugins/43181a5/ : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/6/check/openstack-tox-pep8/d846aca/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/6/check/openstack-tox-py27/a2d1e1c/ : SUCCESS in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/6/check/openstack-tox-py35/ed4ae7b/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/6/check/openstack-tox-py36/d83040e/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/6/check/openstack-tox-py37/829f8a4/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/25/663425/6/check/openstack-tox-docs/8165290/html/ : SUCCESS in 3m 32s\n- build-openstack-releasenotes http://logs.openstack.org/25/663425/6/check/build-openstack-releasenotes/3ccfd10/html/ : SUCCESS in 3m 49s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/6/check/tripleo-buildimage-overcloud-full-centos-7/1969317/ : SUCCESS in 22m 22s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7/675249e/ : SUCCESS in 52m 01s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-stein/adab6d7/ : SUCCESS in 37m 47s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-rocky/c978283/ : SUCCESS in 45m 13s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-buildah/75f4b78/ : SUCCESS in 55m 21s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-buildah-stein/7cdfb78/ : SUCCESS in 51m 30s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-containers-multinode/8566a10/ : FAILURE in 58m 45s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-undercloud-containers/bca6198/ : FAILURE in 57m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6362b7a/ : FAILURE in 54m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a82dc24/ : FAILURE in 50m 29s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-standalone/0f66050/ : FAILURE in 50m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/6/check/tripleo-ci-fedora-28-standalone/1285769/ : FAILURE in 54m 00s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-standalone-upgrade/d894112/ : FAILURE in 1h 33m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/3e60abd/ : SUCCESS in 2h 06m 45s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ad6bb4d/ : FAILURE in 1h 10m 26s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/de7a854/ : FAILURE in 57m 55s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/701b41f/ : FAILURE in 57m 49s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/2efbc39/ : FAILURE in 57m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2ffd68985e128ed3093aad8063f2143eb50e97a5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-11 13:23:55.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e3bcbeed09ab9e4e2b39ba4cbdc4ae3ab0967eae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-11 16:28:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ebfde1f/ : SUCCESS in 2h 42m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f4e6277/ : FAILURE in 3h 02m 42s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/5e7b562/ : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":6},{"id":"068bd4bb26dc868fd65fcf2177c51c3b42c345a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:47:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/25/663425/6/check/requirements-check/3d2d4ac/ : SUCCESS in 2m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/6/check/openstack-tox-lower-constraints/7fbe905/ : SUCCESS in 5m 49s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/6/check/openstackclient-check-plugins/5b08b3b/ : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/6/check/openstack-tox-pep8/5a1873e/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/6/check/openstack-tox-py27/77351f0/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/6/check/openstack-tox-py35/83fdfa6/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/6/check/openstack-tox-py36/034f7cb/ : SUCCESS in 3m 00s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/6/check/openstack-tox-py37/80f39b1/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/25/663425/6/check/openstack-tox-docs/d817ad4/html/ : SUCCESS in 4m 35s\n- build-openstack-releasenotes http://logs.openstack.org/25/663425/6/check/build-openstack-releasenotes/c71d1d8/html/ : SUCCESS in 3m 00s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/6/check/tripleo-buildimage-overcloud-full-centos-7/4bb2dc1/ : SUCCESS in 22m 48s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7/9da3529/ : SUCCESS in 49m 57s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-stein/709d4c3/ : SUCCESS in 44m 04s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-rocky/921eb3d/ : SUCCESS in 41m 36s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-buildah/9d0843c/ : SUCCESS in 51m 47s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/6/check/tripleo-build-containers-centos-7-buildah-stein/a892d0a/ : SUCCESS in 52m 50s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-containers-multinode/66c437f/ : SUCCESS in 2h 15m 02s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-undercloud-containers/1168a52/ : SUCCESS in 1h 48m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/55285db/ : SUCCESS in 2h 25m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1f2e7b9/ : SUCCESS in 2h 43m 53s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-standalone/4b9d75a/ : SUCCESS in 2h 04m 04s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/6/check/tripleo-ci-fedora-28-standalone/4075bb5/ : SUCCESS in 1h 38m 43s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-standalone-upgrade/8648927/ : SUCCESS in 2h 14m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/82f4c59/ : SUCCESS in 2h 25m 33s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7e5051e/ : FAILURE in 1h 29m 02s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3cd0966/ : SUCCESS in 2h 29m 15s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/e6668fe/ : FAILURE in 1h 24m 33s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/6/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/9a1d091/ : SUCCESS in 2h 14m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"131dc0b4daa93a3f54267eeaab1d8131d5e5808d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 01:03:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"aabf2553acf313de6fa433375ffd2b805dd61483","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 08:26:36.000000000","message":"Patch Set 6: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"21ad1a1933a3750d768018343058ba835f37d21f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 14:03:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"227b7b35e00b5dbd0123fa35adbb0c056c549cbf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 17:22:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f14a2f4/ : RETRY_LIMIT in 4m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/51a40f2/ : SUCCESS in 2h 48m 47s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/783d923/ : SUCCESS in 1h 27m 06s","accounts_in_message":[],"_revision_number":7},{"id":"9d5e1d13063dcc203cce7fa598ffca60d9c85cce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 02:14:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/25/663425/7/check/requirements-check/119b7a7/ : SUCCESS in 2m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/7/check/openstack-tox-lower-constraints/5f1dd5c/ : SUCCESS in 5m 13s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/7/check/openstackclient-check-plugins/2602d6c/ : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/7/check/openstack-tox-pep8/340fc23/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/7/check/openstack-tox-py27/1a3efb7/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/7/check/openstack-tox-py35/806fe2a/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/7/check/openstack-tox-py36/7ec18a9/ : SUCCESS in 4m 04s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/7/check/openstack-tox-py37/adc424c/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/25/663425/7/check/openstack-tox-docs/09857f4/html/ : SUCCESS in 3m 32s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-buildimage-overcloud-full-centos-7/4af34ce/ : SUCCESS in 23m 30s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7/b12b504/ : SUCCESS in 41m 39s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-stein/f6f7b25/ : SUCCESS in 40m 26s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-rocky/2777015/ : SUCCESS in 42m 21s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah/188a72d/ : SUCCESS in 55m 26s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah-stein/3421128/ : SUCCESS in 52m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containers-multinode/c3f13c2/ : SUCCESS in 2h 13m 30s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-undercloud-containers/25a401a/ : SUCCESS in 1h 39m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2014a17/ : SUCCESS in 2h 25m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/167e451/ : SUCCESS in 2h 39m 09s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone/4768190/ : SUCCESS in 1h 46m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-fedora-28-standalone/e467ba9/ : SUCCESS in 2h 02m 14s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade/7b3e00c/ : SUCCESS in 2h 22m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/3bb447d/ : SUCCESS in 2h 22m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ecbc3e8/ : FAILURE in 2h 09m 09s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/5037c2a/ : SUCCESS in 2h 15m 40s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/e09e5f4/ : FAILURE in 1h 25m 08s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/34858ef/ : SUCCESS in 2h 08m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"599f3c5657c8bf9a6f92049261a1a2d3c5adc6e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-13 13:32:28.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nwe really should exclude the tripleo jobs from requirements changes","accounts_in_message":[],"_revision_number":7},{"id":"51f1816dbd1315185e2f1e57c3fc242e1dd7a50f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 13:32:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"faaaf38490c8128c13dba55cff4e3b45d1580177","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 16:00:02.000000000","message":"Patch Set 7:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e9a32ebd4556ec57c999f07cf199009e6a9c1375","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 18:51:05.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/7/gate/requirements-check/58a1813/ : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/7/gate/openstack-tox-lower-constraints/d0e1fc1/ : SUCCESS in 6m 42s\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/7/gate/openstack-tox-pep8/df66f33/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py27/01a43b5/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py35/e27341f/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py36/bfe12a2/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py37/bda6e61/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/25/663425/7/gate/openstack-tox-docs/b1aba2f/html/ : SUCCESS in 5m 16s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/7/gate/tripleo-buildimage-overcloud-full-centos-7/44fd3e6/ : SUCCESS in 27m 49s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7/c72ff4a/ : SUCCESS in 47m 04s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-stein/2ec96ae/ : SUCCESS in 53m 42s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-rocky/eeb1a17/ : SUCCESS in 49m 11s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-buildah/9901976/ : SUCCESS in 57m 16s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-buildah-stein/6ba270b/ : SUCCESS in 57m 10s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-containers-multinode/2c74d73/ : SUCCESS in 2h 17m 20s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-standalone/4f9e25c/ : SUCCESS in 2h 08m 10s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/7/gate/tripleo-ci-fedora-28-standalone/12d3a13/ : FAILURE in 2h 02m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/43e8188/ : SUCCESS in 2h 37m 12s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-standalone-upgrade-stein/5de1da6/ : SUCCESS in 2h 08m 24s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-undercloud-containers/3937dd5/ : SUCCESS in 1h 58m 07s","accounts_in_message":[],"_revision_number":7},{"id":"1e984fc091f29947e7f872ce006242bc1d909610","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-13 19:59:50.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"55c474c0fdf612394da7dff53b60ec1f7bc9c66a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 22:41:19.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ebd4d12/ : FAILURE in 1h 25m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8783ceb/ : SUCCESS in 2h 40m 11s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/eb18080/ : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":7},{"id":"4ee20785d7a88b2cf6640bea1c38e54f946f69e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 02:52:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/7/check/requirements-check/74694e0/ : SUCCESS in 2m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/7/check/openstack-tox-lower-constraints/4f1520f/ : SUCCESS in 4m 54s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/7/check/openstackclient-check-plugins/c6a8403/ : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/7/check/openstack-tox-pep8/31db6ff/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/7/check/openstack-tox-py27/1968a39/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/7/check/openstack-tox-py35/3fd1628/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/7/check/openstack-tox-py36/30250eb/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/7/check/openstack-tox-py37/e720945/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/25/663425/7/check/openstack-tox-docs/7fda162/html/ : SUCCESS in 3m 02s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-buildimage-overcloud-full-centos-7/0e51ec4/ : SUCCESS in 24m 26s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7/5ec85ca/ : SUCCESS in 52m 16s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-stein/db885ab/ : SUCCESS in 46m 59s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-rocky/7a1548c/ : SUCCESS in 45m 13s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah/5cdfc49/ : SUCCESS in 59m 31s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah-stein/1cdf8fa/ : SUCCESS in 52m 22s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containers-multinode/e8c5dd7/ : SUCCESS in 2h 08m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-undercloud-containers/719c2ab/ : SUCCESS in 1h 44m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/028fce5/ : SUCCESS in 2h 11m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d4cd240/ : SUCCESS in 2h 41m 42s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone/4ad189b/ : FAILURE in 2h 17m 37s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-fedora-28-standalone/afd075a/ : SUCCESS in 1h 56m 48s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade/0b4d7cb/ : SUCCESS in 2h 15m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/8093677/ : SUCCESS in 2h 05m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0338f79/ : FAILURE in 2h 20m 39s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/5cac0d1/ : SUCCESS in 2h 37m 48s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/37dfb27/ : FAILURE in 1h 26m 44s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/5bf4bba/ : SUCCESS in 2h 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0dd8be8d6eddf8e1a22ff138e8d14a7f5d439338","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-14 13:12:17.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"72f048712646ca9d6125ca978aa5ed535479a14e","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:27:02.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e2a501caea881dbe61b6f58d90da485d3e1a63d9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-14 15:54:10.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d77d1ab/ : FAILURE in 1h 50m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3275dc4/ : SUCCESS in 2h 37m 15s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/42b7cdc/ : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":7},{"id":"d70cb825448a74091308bb162968500630eef8d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:51:25.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/7/check/requirements-check/5813007/ : SUCCESS in 2m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/7/check/openstack-tox-lower-constraints/553a6dd/ : SUCCESS in 4m 49s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/7/check/openstackclient-check-plugins/228c708/ : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/7/check/openstack-tox-pep8/e57314e/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/7/check/openstack-tox-py27/05e95a6/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/7/check/openstack-tox-py35/cf17a8c/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/7/check/openstack-tox-py36/e47a8a4/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/7/check/openstack-tox-py37/2626e48/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/25/663425/7/check/openstack-tox-docs/51173f2/html/ : SUCCESS in 3m 26s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-buildimage-overcloud-full-centos-7/ef36889/ : SUCCESS in 24m 19s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7/740d2eb/ : SUCCESS in 54m 43s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-stein/8564fc5/ : SUCCESS in 48m 53s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-rocky/b32c048/ : SUCCESS in 40m 58s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah/b960b42/ : SUCCESS in 1h 05m 40s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah-stein/c32ff95/ : SUCCESS in 1h 01m 41s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containers-multinode/7c84d0b/ : SUCCESS in 2h 01m 36s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-undercloud-containers/48ad45e/ : FAILURE in 24m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/30d2a27/ : SUCCESS in 2h 19m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/373d336/ : SUCCESS in 2h 21m 39s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone/7012c1e/ : FAILURE in 2h 14m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-fedora-28-standalone/2cce422/ : SUCCESS in 1h 50m 54s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade/9e51668/ : FAILURE in 1h 39m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/b95ec12/ : SUCCESS in 2h 34m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/81008e5/ : FAILURE in 1h 54m 29s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/970a4d2/ : SUCCESS in 2h 14m 16s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/8118782/ : SUCCESS in 2h 06m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cac169b310c6a32bf322dc0903bd861ecd8551e2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-15 02:02:40.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"f538e126360bfd8a4ab2c0690d97259150fc0812","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-15 04:53:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6d6d6ba/ : SUCCESS in 2h 41m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8c2c246/ : SUCCESS in 2h 48m 48s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b4e2d09/ : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":7},{"id":"ddaf72393bea5322858a4da9c3a0046737026724","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 10:56:36.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/25/663425/7/check/requirements-check/a22d187/ : SUCCESS in 2m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/7/check/openstack-tox-lower-constraints/cd8375a/ : SUCCESS in 4m 41s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/7/check/openstackclient-check-plugins/cbd80f6/ : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/7/check/openstack-tox-pep8/786b149/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/7/check/openstack-tox-py27/ef7e316/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/7/check/openstack-tox-py35/1aa90e3/ : SUCCESS in 3m 47s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/7/check/openstack-tox-py36/223ade9/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/7/check/openstack-tox-py37/f8e3319/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/25/663425/7/check/openstack-tox-docs/f8807a1/html/ : SUCCESS in 3m 44s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-buildimage-overcloud-full-centos-7/e0f6398/ : SUCCESS in 22m 25s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7/414a25d/ : SUCCESS in 53m 23s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-stein/6795414/ : SUCCESS in 58m 26s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-rocky/8fc9dd3/ : SUCCESS in 49m 12s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah/4d1750f/ : SUCCESS in 45m 46s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah-stein/5be1d61/ : SUCCESS in 56m 17s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containers-multinode/b042174/ : TIMED_OUT in 3h 13m 54s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-undercloud-containers/c8d2c7f/ : TIMED_OUT in 3h 13m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/05c5ac7/ : SUCCESS in 2h 19m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7e17ed6/ : SUCCESS in 2h 23m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone/b509eea/ : SUCCESS in 1h 58m 57s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-fedora-28-standalone/eae2398/ : SUCCESS in 1h 48m 11s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade/9527da9/ : TIMED_OUT in 3h 11m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/37eafe1/ : TIMED_OUT in 3h 11m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f567f5a/ : FAILURE in 1h 37m 05s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/393086d/ : TIMED_OUT in 3h 11m 58s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4d766b4/ : SUCCESS in 2h 06m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ab6c84c6e5e3b139a5cca7e68b6a5e86e6c3fd42","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-15 14:39:25.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e8007cd3cd919e9c80257a647f07f91ecab9e897","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 17:35:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/25/663425/7/check/requirements-check/ec443b6/ : SUCCESS in 1m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/7/check/openstack-tox-lower-constraints/0504ffd/ : SUCCESS in 4m 54s\n- openstackclient-check-plugins http://logs.openstack.org/25/663425/7/check/openstackclient-check-plugins/3d9f912/ : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/7/check/openstack-tox-pep8/569792a/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/7/check/openstack-tox-py27/fba4921/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/7/check/openstack-tox-py35/c9a0ebf/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/7/check/openstack-tox-py36/10b6b0c/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/7/check/openstack-tox-py37/4c75d08/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/25/663425/7/check/openstack-tox-docs/9b7d9af/html/ : SUCCESS in 3m 25s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-buildimage-overcloud-full-centos-7/2c936c6/ : SUCCESS in 26m 12s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7/8384de4/ : SUCCESS in 44m 17s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-stein/e3dd77e/ : SUCCESS in 45m 08s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-rocky/faadb68/ : SUCCESS in 42m 59s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah/9ba25f7/ : SUCCESS in 41m 54s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/7/check/tripleo-build-containers-centos-7-buildah-stein/7354ca5/ : SUCCESS in 1h 06m 35s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containers-multinode/8597282/ : SUCCESS in 1h 55m 56s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-undercloud-containers/bcc94d2/ : SUCCESS in 1h 26m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b60e8a1/ : SUCCESS in 2h 16m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a963228/ : SUCCESS in 2h 23m 46s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone/c3f3932/ : SUCCESS in 1h 57m 42s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/7/check/tripleo-ci-fedora-28-standalone/88ef369/ : SUCCESS in 1h 48m 35s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade/ddc5f19/ : SUCCESS in 2h 21m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/d9c0b4b/ : SUCCESS in 2h 06m 25s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c015b72/ : FAILURE in 2h 12m 39s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8cb1fb6/ : SUCCESS in 2h 20m 42s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/25/663425/7/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/aa4de67/ : SUCCESS in 2h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"52d776b04db0377dae7d5c4f054fd672814e4c4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 17:36:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"06f5aab4f4cdb3857a03af4cb747a0883f596e94","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-15 18:29:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/aac16c49e66f4f3d93e29d43cad8ba87 : POST_FAILURE in 3h 34m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze07.softwarefactory-project.io:7979/f2369a612c4f4fa5b1285895865dea9e : POST_FAILURE in 3h 48m 12s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/25/663425/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/8ea0d47/ : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":7},{"id":"7e9cd9c24c01c1baec0fb2ff7a20fb636f7bac0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 22:06:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/25/663425/7/gate/requirements-check/85ef2d3/ : SUCCESS in 2m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/663425/7/gate/openstack-tox-lower-constraints/011aa39/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/25/663425/7/gate/openstack-tox-pep8/2bcd2e0/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py27/f6e92a4/ : SUCCESS in 2m 50s\n- openstack-tox-py35 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py35/995a1f5/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py36/ba7775d/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/25/663425/7/gate/openstack-tox-py37/1702fac/ : SUCCESS in 3m 38s\n- openstack-tox-docs http://logs.openstack.org/25/663425/7/gate/openstack-tox-docs/9cc7a19/html/ : SUCCESS in 2m 47s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/25/663425/7/gate/tripleo-buildimage-overcloud-full-centos-7/c66c790/ : SUCCESS in 23m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7/068b944/ : SUCCESS in 45m 27s\n- tripleo-build-containers-centos-7-stein http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-stein/7bcb98c/ : SUCCESS in 44m 51s\n- tripleo-build-containers-centos-7-rocky http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-rocky/99b40ec/ : SUCCESS in 37m 04s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-buildah/b5f3b3b/ : SUCCESS in 50m 34s\n- tripleo-build-containers-centos-7-buildah-stein http://logs.openstack.org/25/663425/7/gate/tripleo-build-containers-centos-7-buildah-stein/165dffa/ : SUCCESS in 56m 32s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-containers-multinode/48a5b50/ : SUCCESS in 2h 09m 16s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-standalone/fafe25e/ : SUCCESS in 1h 57m 50s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/25/663425/7/gate/tripleo-ci-fedora-28-standalone/655ab25/ : SUCCESS in 1h 57m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/507cee4/ : SUCCESS in 2h 04m 20s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-standalone-upgrade-stein/a97a006/ : SUCCESS in 2h 01m 58s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/25/663425/7/gate/tripleo-ci-centos-7-undercloud-containers/4ec1802/ : SUCCESS in 1h 33m 26s","accounts_in_message":[],"_revision_number":7},{"id":"5ecc7e7ecf6169050f08c5e58b01df63ccd0b69f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 22:06:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6aa1902322e2256b55a40af46224aabfbfd11be7","revisions":{"42ca45b58e0bde95b87d1dcd03d9aae8b4f4d942":{"kind":"REWORK","_number":1,"created":"2019-06-05 20:41:42.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/25/663425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/25/663425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/1"}}},"commit":{"parents":[{"commit":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","subject":"Change source for upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"subject":"doc/requirements: require sphinx\u003e\u003d1.6.2","message":"doc/requirements: require sphinx\u003e\u003d1.6.2\n\nThis change is useful to match OpenStack general requirements.\n\nChange-Id: I1e39fa5c220390fcd658147d9f47ae2196eefe56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/42ca45b58e0bde95b87d1dcd03d9aae8b4f4d942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/42ca45b58e0bde95b87d1dcd03d9aae8b4f4d942"}]},"branch":"refs/heads/master"},"5c4907a49d158608a6528726ff992557119fbf3a":{"kind":"REWORK","_number":2,"created":"2019-06-06 13:24:36.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/25/663425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/25/663425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/2"}}},"commit":{"parents":[{"commit":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","subject":"Change source for upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-06 13:24:30.000000000","tz":-360},"subject":"doc/requirements: require sphinx\u003e\u003d1.6.2","message":"doc/requirements: require sphinx\u003e\u003d1.6.2\n\nThis change is useful to match OpenStack general requirements.\n\nChange-Id: I1e39fa5c220390fcd658147d9f47ae2196eefe56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5c4907a49d158608a6528726ff992557119fbf3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5c4907a49d158608a6528726ff992557119fbf3a"}]},"branch":"refs/heads/master"},"cb4eedc597004e9f2caa97c4cc287e9ea2e03a4f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-06 13:25:21.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/25/663425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/25/663425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/3"}}},"commit":{"parents":[{"commit":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","subject":"Change source for upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-06 13:25:05.000000000","tz":-360},"subject":"doc/requirements: fix sphinx requirement lines","message":"doc/requirements: fix sphinx requirement lines\n\nThis change is useful to match OpenStack general requirements.\n\nChange-Id: I1e39fa5c220390fcd658147d9f47ae2196eefe56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cb4eedc597004e9f2caa97c4cc287e9ea2e03a4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cb4eedc597004e9f2caa97c4cc287e9ea2e03a4f"}]},"branch":"refs/heads/master"},"d21648a359ddb0f99cf3606a32930e4e61f3104e":{"kind":"REWORK","_number":4,"created":"2019-06-06 18:51:25.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/25/663425/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/25/663425/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/4"}}},"commit":{"parents":[{"commit":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","subject":"Change source for upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-06 18:51:20.000000000","tz":-360},"subject":"doc/requirements: fix sphinx requirement lines","message":"doc/requirements: fix sphinx requirement lines\n\nThis change is useful to match OpenStack general requirements.\n\nChange-Id: I1e39fa5c220390fcd658147d9f47ae2196eefe56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d21648a359ddb0f99cf3606a32930e4e61f3104e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d21648a359ddb0f99cf3606a32930e4e61f3104e"}]},"branch":"refs/heads/master"},"53b55236c372ce0ed8195c96bd0f9b6ea48bf0cb":{"kind":"REWORK","_number":5,"created":"2019-06-10 13:25:47.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/25/663425/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/25/663425/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/5"}}},"commit":{"parents":[{"commit":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","subject":"Change source for upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-10 13:25:43.000000000","tz":-360},"subject":"doc/requirements: fix sphinx requirement lines","message":"doc/requirements: fix sphinx requirement lines\n\nThis change is useful to match OpenStack general requirements.\n\nChange-Id: I1e39fa5c220390fcd658147d9f47ae2196eefe56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/53b55236c372ce0ed8195c96bd0f9b6ea48bf0cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/53b55236c372ce0ed8195c96bd0f9b6ea48bf0cb"}]},"branch":"refs/heads/master"},"2bf370cfd00692e28767fa2c383b5110bf280802":{"kind":"REWORK","_number":6,"created":"2019-06-10 16:45:25.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/25/663425/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/25/663425/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/6"}}},"commit":{"parents":[{"commit":"8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5","subject":"Change source for upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a3dc5121c8f09cda8d42cc78741d2d90dc9fba5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-06-10 16:45:22.000000000","tz":-360},"subject":"doc/requirements: fix sphinx requirement lines","message":"doc/requirements: fix sphinx requirement lines\n\nThis change is useful to match OpenStack general requirements.\n\nChange-Id: I1e39fa5c220390fcd658147d9f47ae2196eefe56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2bf370cfd00692e28767fa2c383b5110bf280802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2bf370cfd00692e28767fa2c383b5110bf280802"}]},"branch":"refs/heads/master"},"6aa1902322e2256b55a40af46224aabfbfd11be7":{"kind":"REWORK","_number":7,"created":"2019-06-12 14:03:13.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/25/663425/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/25/663425/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/25/663425/7"}}},"commit":{"parents":[{"commit":"53d0237c8cee3275e77021a74f661b3df01380a9","subject":"Merge \"Update sphinx requirements for python 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/53d0237c8cee3275e77021a74f661b3df01380a9"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-05 20:40:37.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-06-12 14:02:57.000000000","tz":-240},"subject":"doc/requirements: fix sphinx requirement lines","message":"doc/requirements: fix sphinx requirement lines\n\nThis change is useful to match OpenStack general requirements.\n\nChange-Id: I1e39fa5c220390fcd658147d9f47ae2196eefe56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6aa1902322e2256b55a40af46224aabfbfd11be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6aa1902322e2256b55a40af46224aabfbfd11be7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
