)]}'
{"id":"openstack%2Fpython-tripleoclient~663686","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fqueens~Idcdbd38129f5694c5452f3f8aca0388df80476b2","project":"openstack/python-tripleoclient","branch":"stable/queens","hashtags":[],"change_id":"Idcdbd38129f5694c5452f3f8aca0388df80476b2","subject":"Raise proper exception at webscocket close","status":"MERGED","created":"2019-06-06 14:36:25.000000000","updated":"2019-06-12 05:44:25.000000000","submitted":"2019-06-12 05:44:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663686-1560318265360-8840775f","meta_rev_id":"3ea054186873814e5dcf1f9d21a9d589810b2253","_number":663686,"virtual_id_number":663686,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-07 09:52:59.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"value":2,"date":"2019-06-12 05:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-07 11:58:54.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-06-12 00:53:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-06-06 21:22:36.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-06-12 00:53:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-06-06 14:36:25.000000000","_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 14:36:25.000000000","updated_by":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"reviewer":{"_account_id":27898,"name":"Quique Llorente","email":"ellorent@redhat.com","username":"quiquell"},"state":"REVIEWER"},{"updated":"2019-06-06 21:22:36.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-07 11:58:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-12 00:53:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-12 05:44:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"485f54327ce02fcbfd1441c52702de6fecd032aa","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-06 14:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d68ac6e691cacbe2a4d6d307b4f0b4ba0d855db","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 16:45:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/86/663686/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/329115e/ : SUCCESS in 2h 08m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/86/663686/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/513f94a/ : FAILURE in 1h 47m 05s","accounts_in_message":[],"_revision_number":1},{"id":"604036b0934d10850ec52809cc7021ca42cf51ab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 21:22:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c709fa4d2a1fa748eba04f4d15e68bc62d79aaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:34:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-undercloud-oooq/2e42b64/ : SUCCESS in 1h 19m 15s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-undercloud-upgrades/c252edf/ : SUCCESS in 1h 16m 02s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-containers-multinode/4451baf/ : SUCCESS in 2h 08m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9663727/ : FAILURE in 2h 00m 03s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/86/663686/1/check/tripleo-buildimage-overcloud-full-centos-7/58e8f73/ : SUCCESS in 21m 58s\n- openstack-tox-pep8 http://logs.openstack.org/86/663686/1/check/openstack-tox-pep8/76f7c4c/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/86/663686/1/check/openstack-tox-py27/9cbf35a/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/86/663686/1/check/openstack-tox-py35/3bae455/ : SUCCESS in 3m 57s\n- openstackclient-check-plugins http://logs.openstack.org/86/663686/1/check/openstackclient-check-plugins/f13e9f0/ : SUCCESS in 4m 02s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/86/663686/1/check/build-openstack-sphinx-docs/4412b89/html/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"381fe814d56a59add1ab86f51499d5452796d476","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-07 09:52:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f67169addddeba4095bf4fc7e151934884427f9d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 11:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/86/663686/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/379524c/ : SUCCESS in 2h 05m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/86/663686/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d8afb58/ : SUCCESS in 2h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f25a8e411d3d5d4a28b35aed6ec182e153a3a7f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:31:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-undercloud-oooq/7732fb4/ : SUCCESS in 1h 18m 46s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-undercloud-upgrades/7335d63/ : SUCCESS in 1h 24m 12s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-containers-multinode/8d01e77/ : SUCCESS in 2h 05m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/86/663686/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b8bc17f/ : FAILURE in 1h 57m 52s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/86/663686/1/check/tripleo-buildimage-overcloud-full-centos-7/277a886/ : SUCCESS in 24m 14s\n- openstack-tox-pep8 http://logs.openstack.org/86/663686/1/check/openstack-tox-pep8/5a2b5b1/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/86/663686/1/check/openstack-tox-py27/516150a/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/86/663686/1/check/openstack-tox-py35/b06b23b/ : SUCCESS in 4m 09s\n- openstackclient-check-plugins http://logs.openstack.org/86/663686/1/check/openstackclient-check-plugins/9f98931/ : SUCCESS in 5m 21s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/86/663686/1/check/build-openstack-sphinx-docs/f0cc200/html/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f7e4699d254b11d53883e0f9473fd83e0452db9b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 00:53:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc787cde71cc45789d45813b5583507cf4e9a14f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 00:53:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20fb546286e752b4b4aaeb2669ad6888f9dfe2bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 05:44:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/86/663686/1/gate/tripleo-ci-centos-7-undercloud-oooq/1cc4e1e/ : SUCCESS in 1h 19m 51s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/86/663686/1/gate/tripleo-ci-centos-7-containers-multinode/53a2757/ : SUCCESS in 2h 25m 40s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/86/663686/1/gate/tripleo-buildimage-overcloud-full-centos-7/2df99ee/ : SUCCESS in 27m 05s\n- openstack-tox-pep8 http://logs.openstack.org/86/663686/1/gate/openstack-tox-pep8/361bf16/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/86/663686/1/gate/openstack-tox-py27/90f2456/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/86/663686/1/gate/openstack-tox-py35/0b7e6be/ : SUCCESS in 4m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/663686/1/gate/build-openstack-sphinx-docs/fa54af3/html/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3ea054186873814e5dcf1f9d21a9d589810b2253","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 05:44:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"531c1504ab6d43066d1c17a1fb87b5a060d7d932","revisions":{"531c1504ab6d43066d1c17a1fb87b5a060d7d932":{"kind":"REWORK","_number":1,"created":"2019-06-06 14:36:25.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/86/663686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/86/663686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/86/663686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/86/663686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/86/663686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/86/663686/1"}}},"commit":{"parents":[{"commit":"7f8d8999d3b0cdc182d52ae83d7539679e71c1ea","subject":"Handle tls endpoint for zaqar websocket client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7f8d8999d3b0cdc182d52ae83d7539679e71c1ea"}]}],"author":{"name":"Quique Llorente","email":"ellorent@redhat.com","date":"2018-09-26 10:09:41.000000000","tz":120},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-06-06 13:29:23.000000000","tz":60},"subject":"Raise proper exception at webscocket close","message":"Raise proper exception at webscocket close\n\nAfter fixing a bug with https://review.openstack.org/#/c/603802/ the\nreturn introduced there is bogus, we have to raise a proper exception\nand handle it like timeouts, to get all the mistral allright.\n\nChange-Id: Idcdbd38129f5694c5452f3f8aca0388df80476b2\n(cherry picked from commit fee9f8cf15d708e4b3b8bdf0f0ddd6df8ce5c223)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/531c1504ab6d43066d1c17a1fb87b5a060d7d932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/531c1504ab6d43066d1c17a1fb87b5a060d7d932"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
