)]}'
{"id":"openstack%2Fpython-tripleoclient~663687","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fqueens~Ie6473d6a1044cdf76552d62645b4d63da2df9398","project":"openstack/python-tripleoclient","branch":"stable/queens","hashtags":[],"change_id":"Ie6473d6a1044cdf76552d62645b4d63da2df9398","subject":"Filter messages not from waiting execution","status":"MERGED","created":"2019-06-06 14:36:25.000000000","updated":"2019-08-22 16:32:39.000000000","submitted":"2019-08-22 16:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663687-1566491559598-07838efc","meta_rev_id":"1b91f03f5d0011c45f7a76bd40a2f44e201930f7","_number":663687,"virtual_id_number":663687,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-22 16:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-22 11:05:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-08-22 13:45:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-08-22 13:45:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:32:17.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-22 11:05:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 13:45:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-22 16:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddec9218d0abcfb1414a152de632f3846a0336bc","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-06 14:36:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"284d69a518cb1ac3be74eae463e931ff17d01262","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-06 18:21:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/87/663687/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0f43d16/ : TIMED_OUT in 3h 35m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/87/663687/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/81de5da/ : FAILURE in 1h 42m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c967a1660f07d055f8f0eed4e26ffb0df94a6110","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 21:23:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"900d9537d2feac03a8d3ea07de4175ccd9e72e39","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 21:24:26.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56b19c3d48c7cc922f2d37a142bb1e82fb830b54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:32:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/87/663687/1/check/tripleo-ci-centos-7-undercloud-oooq/69409ab/ : SUCCESS in 1h 11m 57s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/87/663687/1/check/tripleo-ci-centos-7-undercloud-upgrades/ea3f52a/ : SUCCESS in 1h 20m 19s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/87/663687/1/check/tripleo-ci-centos-7-containers-multinode/492289b/ : SUCCESS in 2h 04m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/87/663687/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7950667/ : FAILURE in 2h 06m 03s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/87/663687/1/check/tripleo-buildimage-overcloud-full-centos-7/279c61d/ : SUCCESS in 24m 13s\n- openstack-tox-pep8 http://logs.openstack.org/87/663687/1/check/openstack-tox-pep8/3cd4df4/ : FAILURE in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/87/663687/1/check/openstack-tox-py27/e239482/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/87/663687/1/check/openstack-tox-py35/4555dab/ : SUCCESS in 4m 09s\n- openstackclient-check-plugins http://logs.openstack.org/87/663687/1/check/openstackclient-check-plugins/f64736b/ : SUCCESS in 4m 39s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/87/663687/1/check/build-openstack-sphinx-docs/754c261/html/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"52c944ea65bbf5c82343283500f10eafc6d0ef71","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-07 09:52:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fe31047f4827fad85140dfd558071329e8a6e42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 09:54:20.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f44455b994d269b298c2b3dd4e01d00f33bc5f4c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-07 09:54:21.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"fb8e6a10b7f934b5cac549a305a4cce1c5daface","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-10 10:12:27.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4ddff0d78c7f5b5f367577ba1bd8cd9435312b2d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-06-11 20:32:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d410a7b639213dab15fadfed3512ef917c84e9de","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-22 08:46:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0cf0a7361818e1e0cec67ff4f9ab84d7c9f421b1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 11:05:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/87/663687/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7ab52f6/ : FAILURE in 2h 03m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/87/663687/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a168b51/ : SUCCESS in 2h 16m 17s","accounts_in_message":[],"_revision_number":3},{"id":"4e873cbfe5bfdcc3312f7a09313fb133c83a1aea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 12:13:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/dd4625cbabde41dcabfd23910817847d : SUCCESS in 1h 24m 34s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1869689dcd5440e2b672a19ab4b45785 : SUCCESS in 1h 24m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ae2f6b39befe41aeb672adb325c6be66 : SUCCESS in 2h 46m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f8cbf8fa42dc42e6852039df262d0367 : FAILURE in 2h 33m 29s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/c96ed255246c48b99eb48caa7ef5eba5 : SUCCESS in 35m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8772b9d10a7242af8c510898a0fbc1aa : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b683d4669724436b52bf950067d77ea : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e8a51c61fff943e782b8687dd1d8a6f1 : SUCCESS in 7m 43s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/03de1eb30aa94b62a6c2fb7e7c3455fa : SUCCESS in 4m 19s (non-voting)\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c7bb6c99df304e3dbce1fae5baa163ea : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"9f28b43e50b4cd0b2771f33b85c347ab9e26c240","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 13:45:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"19c0aeccf52afce367a31345ad50c83473fa4f91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 13:45:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"96a7bfecaa8843fc9f113a5a6cd10bfa046ea316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 16:32:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/d5dc864c96634c9480f058c70c59f6b4 : SUCCESS in 1h 32m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3a5326cd578b4be09087d8a57bf7aa2f : SUCCESS in 2h 39m 36s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/f562f10b39f04ecc9ee80702c174a5e4 : SUCCESS in 33m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2094ff1e52a744c3b3769adbaf25e636 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13a1e703c0364878abef03dc0d5314cc : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/545160c14a7e486dbc34b479c28a777c : SUCCESS in 4m 41s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a46a955030814b23b3abe7429bfeacde : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"1b91f03f5d0011c45f7a76bd40a2f44e201930f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 16:32:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"074bb6e3a06f1d49a21506ffbd16b521db462c9c","revisions":{"d6bb90771581689826ac9d123d827362cbf67c24":{"kind":"REWORK","_number":1,"created":"2019-06-06 14:36:25.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/87/663687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/87/663687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/1"}}},"commit":{"parents":[{"commit":"531c1504ab6d43066d1c17a1fb87b5a060d7d932","subject":"Raise proper exception at webscocket close","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/531c1504ab6d43066d1c17a1fb87b5a060d7d932"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2018-09-26 20:08:10.000000000","tz":-240},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-06-06 13:59:19.000000000","tz":60},"subject":"Filter messages not from waiting execution","message":"Filter messages not from waiting execution\n\nThe convention is to use the same queue name (\"tripleo\") for all\nworkflows. This can lead to messages showing from other tripleoclient\ntriggered workflows showing up during message polling if multiple\nworkflows are running at the same time.\n\nThis patch adds a check that will filter out any messages that do not\nbelong to the execution that is being waited on by comparing the\nexecution id with the root_execution_id returned in the execution\npayload.\n\nDepends-On: Icbe80c338d69efc6ce8fceb0f73f833bec588536\nChange-Id: Ie6473d6a1044cdf76552d62645b4d63da2df9398\nRelated-Bug: #1794277\n(cherry picked from commit 339c1f334c5b4dba7a95740e35b63aab931601a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d6bb90771581689826ac9d123d827362cbf67c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d6bb90771581689826ac9d123d827362cbf67c24"}]},"branch":"refs/heads/stable/queens"},"02699015d83092497dfcf6495b47c418e66bf83e":{"kind":"REWORK","_number":2,"created":"2019-06-07 09:52:13.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/87/663687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/87/663687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/2"}}},"commit":{"parents":[{"commit":"531c1504ab6d43066d1c17a1fb87b5a060d7d932","subject":"Raise proper exception at webscocket close","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/531c1504ab6d43066d1c17a1fb87b5a060d7d932"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2018-09-26 20:08:10.000000000","tz":-240},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-06-07 09:49:34.000000000","tz":60},"subject":"Filter messages not from waiting execution","message":"Filter messages not from waiting execution\n\nThe convention is to use the same queue name (\"tripleo\") for all\nworkflows. This can lead to messages showing from other tripleoclient\ntriggered workflows showing up during message polling if multiple\nworkflows are running at the same time.\n\nThis patch adds a check that will filter out any messages that do not\nbelong to the execution that is being waited on by comparing the\nexecution id with the root_execution_id returned in the execution\npayload.\n\nDepends-On: Icbe80c338d69efc6ce8fceb0f73f833bec588536\nChange-Id: Ie6473d6a1044cdf76552d62645b4d63da2df9398\nRelated-Bug: #1794277\n(cherry picked from commit 339c1f334c5b4dba7a95740e35b63aab931601a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/02699015d83092497dfcf6495b47c418e66bf83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/02699015d83092497dfcf6495b47c418e66bf83e"}]},"branch":"refs/heads/stable/queens"},"074bb6e3a06f1d49a21506ffbd16b521db462c9c":{"kind":"REWORK","_number":3,"created":"2019-08-22 08:46:02.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/87/663687/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/87/663687/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/87/663687/3"}}},"commit":{"parents":[{"commit":"f7badfea523af31e7c46b5e978c4b9caeae15896","subject":"Merge \"Make ipmi lanplus the default in auto fencing\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f7badfea523af31e7c46b5e978c4b9caeae15896"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2018-09-26 20:08:10.000000000","tz":-240},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-22 08:45:05.000000000","tz":60},"subject":"Filter messages not from waiting execution","message":"Filter messages not from waiting execution\n\nThe convention is to use the same queue name (\"tripleo\") for all\nworkflows. This can lead to messages showing from other tripleoclient\ntriggered workflows showing up during message polling if multiple\nworkflows are running at the same time.\n\nThis patch adds a check that will filter out any messages that do not\nbelong to the execution that is being waited on by comparing the\nexecution id with the root_execution_id returned in the execution\npayload.\n\nDepends-On: Icbe80c338d69efc6ce8fceb0f73f833bec588536\nChange-Id: Ie6473d6a1044cdf76552d62645b4d63da2df9398\nRelated-Bug: #1794277\n(cherry picked from commit 339c1f334c5b4dba7a95740e35b63aab931601a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/074bb6e3a06f1d49a21506ffbd16b521db462c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/074bb6e3a06f1d49a21506ffbd16b521db462c9c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
