)]}'
{"id":"openstack%2Fpython-tripleoclient~665690","triplet_id":"openstack%2Fpython-tripleoclient~master~I054a805e820087a30bd8d27a0b460b6e519849ac","project":"openstack/python-tripleoclient","branch":"master","topic":"mistral_to_ansible","hashtags":[],"change_id":"I054a805e820087a30bd8d27a0b460b6e519849ac","subject":"Move undercloud backup from mistral to ansible","status":"MERGED","created":"2019-06-17 14:31:23.000000000","updated":"2020-01-29 01:55:25.000000000","submitted":"2020-01-29 01:54:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665690-1580262843900-a46f390a","meta_rev_id":"0a2bc0faa65398955e0363d99369bbba0ee30ad3","_number":665690,"virtual_id_number":665690,"owner":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 01:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-27 20:21:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2020-01-27 17:12:33.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2020-01-27 17:12:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2020-01-28 21:51:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2020-01-28 21:51:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 15:50:19.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-09-27 08:30:42.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2020-01-22 00:47:54.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2020-01-24 10:08:52.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-01-27 14:35:46.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-01-27 20:21:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-28 21:51:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-01-29 01:54:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db4dc94bd01dff27a78916bc6a9c946ee5416ae8","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-06-17 14:31:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03f8d60e34b864f2c59d6f429cb33b65bf0c9ace","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-17 14:55:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c682658/ : FAILURE in 23m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/90/665690/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3fc5614/ : FAILURE in 20m 52s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/90/665690/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/1d4d5d8/ : FAILURE in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"549264dd87891ca6cbf3a5c7805dc2552422d937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 21:39:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/665690/1/check/openstack-tox-lower-constraints/c90b74d/ : FAILURE in 4m 30s\n- openstackclient-check-plugins http://logs.openstack.org/90/665690/1/check/openstackclient-check-plugins/d9ad188/ : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/90/665690/1/check/openstack-tox-pep8/247d06a/ : FAILURE in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/90/665690/1/check/openstack-tox-py27/20f166a/ : FAILURE in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/90/665690/1/check/openstack-tox-py35/3354338/ : FAILURE in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/90/665690/1/check/openstack-tox-py36/ac9254f/ : FAILURE in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/90/665690/1/check/openstack-tox-py37/e6be5e6/ : FAILURE in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/90/665690/1/check/openstack-tox-docs/23760c8/ : FAILURE in 2m 58s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/90/665690/1/check/tripleo-buildimage-overcloud-full-centos-7/3a875f2/ : SUCCESS in 27m 01s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/90/665690/1/check/tripleo-ci-centos-7-undercloud-containers/1da2cbe/ : FAILURE in 24m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/90/665690/1/check/tripleo-ci-centos-7-standalone/0d0598b/ : FAILURE in 28m 04s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/90/665690/1/check/tripleo-ci-fedora-28-standalone/28af88f/ : FAILURE in 24m 23s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/90/665690/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/194fa2a/ : SUCCESS in 1h 52m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2eb87c77f3676db52a539d7bfee9a910a344c271","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-06-20 09:23:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ec70ccccae1ad6eaf1aca3c511d3fe16300b479","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-20 09:49:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cc574dc/ : FAILURE in 24m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/90/665690/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b5ea0a2/ : FAILURE in 21m 06s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/90/665690/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/43a2e5a/ : FAILURE in 21m 49s","accounts_in_message":[],"_revision_number":2},{"id":"375c8dc46c6cc1abfc803bca6b79871f7b86465a","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-06-20 10:13:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"202554e02a4ac24086b850062917bc24809d8abe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 12:21:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/665690/2/check/openstack-tox-lower-constraints/abb23dd/ : FAILURE in 5m 48s\n- openstackclient-check-plugins http://logs.openstack.org/90/665690/2/check/openstackclient-check-plugins/01b7ff4/ : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/90/665690/2/check/openstack-tox-pep8/584c621/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/90/665690/2/check/openstack-tox-py27/de6483a/ : FAILURE in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/90/665690/2/check/openstack-tox-py35/6b74d5c/ : FAILURE in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/90/665690/2/check/openstack-tox-py36/c43c3ae/ : FAILURE in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/90/665690/2/check/openstack-tox-py37/76dda6f/ : FAILURE in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/90/665690/2/check/openstack-tox-docs/8c04dca/html/ : SUCCESS in 3m 32s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/90/665690/2/check/tripleo-buildimage-overcloud-full-centos-7/e7d5156/ : SUCCESS in 20m 00s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/90/665690/2/check/tripleo-ci-centos-7-undercloud-containers/54c4cf4/ : FAILURE in 21m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/90/665690/2/check/tripleo-ci-centos-7-standalone/7d00dc3/ : FAILURE in 21m 40s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/90/665690/2/check/tripleo-ci-fedora-28-standalone/af0aa22/ : FAILURE in 23m 30s\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/90/665690/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/eeb1d32/ : SUCCESS in 2h 06m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d1641f1e9a9d771649819519158f64af828359a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-11 20:27:07.000000000","message":"Patch Set 2:\n\nAdriano,\n\nMaybe you could be interested by this patch[1] which is adding the extra_vars support.\nThis new parameter accepts either a Dict or the absolute path of a JSON|YAML file.\n\n[1] - https://review.opendev.org/#/c/669697/","accounts_in_message":[],"_revision_number":2},{"id":"6fa150c184937917fbc7724e701004b688b27d1f","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-02 17:18:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aeace9ef7e7ada5c4a025ff9d18a528495bb6497","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 17:39:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d28f9a3/ : FAILURE in 18m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/90/665690/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6df46a2/ : FAILURE in 18m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ada294d6769c049fed8546c47154a1fbd543a54d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 17:54:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/90/665690/3/check/openstack-tox-lower-constraints/2d9b10d/ : FAILURE in 5m 02s\n- openstackclient-check-plugins https://logs.opendev.org/90/665690/3/check/openstackclient-check-plugins/75d23cc/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/90/665690/3/check/openstack-tox-pep8/226413d/ : SUCCESS in 3m 38s\n- openstack-tox-py27 https://logs.opendev.org/90/665690/3/check/openstack-tox-py27/ac9d6cc/ : FAILURE in 4m 47s\n- openstack-tox-py35 https://logs.opendev.org/90/665690/3/check/openstack-tox-py35/b7cf173/ : FAILURE in 4m 41s\n- openstack-tox-py36 https://logs.opendev.org/90/665690/3/check/openstack-tox-py36/f4b99ca/ : FAILURE in 3m 37s\n- openstack-tox-py37 https://logs.opendev.org/90/665690/3/check/openstack-tox-py37/d88a69f/ : FAILURE in 4m 15s\n- openstack-tox-docs https://logs.opendev.org/90/665690/3/check/openstack-tox-docs/5e041ae/html/ : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://logs.opendev.org/90/665690/3/check/build-openstack-releasenotes/fbcc85d/html/ : SUCCESS in 3m 06s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/90/665690/3/check/tripleo-buildimage-overcloud-full-centos-7/9c58506/ : SUCCESS in 24m 40s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/90/665690/3/check/tripleo-ci-centos-7-undercloud-containers/8d0c9ea/ : FAILURE in 23m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/90/665690/3/check/tripleo-ci-centos-7-standalone/21bb63b/ : FAILURE in 23m 49s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/90/665690/3/check/tripleo-ci-fedora-28-standalone/969a8e6/ : FAILURE in 28m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ba0ab04f993c72734e3e5f03f9ff1017cca59c9","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-05 08:05:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b7c28c789d153a503cf911075ac84fbce2f77a8c","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-05 10:12:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e7ce8631d8bab903ea55eca5c062b5033dbb39b","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-05 10:15:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"db3d4ee3727ad79e3d07b32593fbf94552d26cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:48:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/90/665690/4/check/openstack-tox-lower-constraints/3ba2cae/ : SUCCESS in 7m 43s\n- openstackclient-check-plugins https://logs.opendev.org/90/665690/4/check/openstackclient-check-plugins/0be94ba/ : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/90/665690/4/check/openstack-tox-pep8/cd9608c/ : SUCCESS in 5m 36s\n- openstack-tox-py27 https://logs.opendev.org/90/665690/4/check/openstack-tox-py27/a6d6f07/ : SUCCESS in 5m 20s\n- openstack-tox-py35 https://logs.opendev.org/90/665690/4/check/openstack-tox-py35/eca9250/ : SUCCESS in 5m 34s\n- openstack-tox-py36 https://logs.opendev.org/90/665690/4/check/openstack-tox-py36/5135ec7/ : SUCCESS in 6m 05s\n- openstack-tox-py37 https://logs.opendev.org/90/665690/4/check/openstack-tox-py37/5acc68f/ : SUCCESS in 8m 24s\n- openstack-tox-docs https://logs.opendev.org/90/665690/4/check/openstack-tox-docs/b4438ec/html/ : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://logs.opendev.org/90/665690/4/check/build-openstack-releasenotes/085d03b/html/ : SUCCESS in 5m 04s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/90/665690/4/check/tripleo-buildimage-overcloud-full-centos-7/2197d7e/ : SUCCESS in 28m 47s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-containers-multinode/23d06f8/ : FAILURE in 1h 05m 49s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-undercloud-containers/fa982d0/ : SUCCESS in 1h 42m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cf31c47/ : SUCCESS in 2h 23m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/85719f4/ : FAILURE in 1h 13m 19s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-standalone/d5945ec/ : FAILURE in 55m 49s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/90/665690/4/check/tripleo-ci-fedora-28-standalone/ffe9828/ : FAILURE in 1h 08m 45s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-standalone-upgrade/1f41f31/ : FAILURE in 39m 12s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/19d7749/ : FAILURE in 55m 45s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/b8cd11b/ : FAILURE in 1h 08m 23s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/9ea0229/ : FAILURE in 1h 11m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5c46779e05d9c5e4025e6445da08a817cbd2c72e","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-05 11:01:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4b49d64edd6c5a5abc20e2f4c605049d8b729bdf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 11:13:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/63f2715/ : SUCCESS in 3h 06m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/90/665690/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/36970c3/ : FAILURE in 2h 41m 57s","accounts_in_message":[],"_revision_number":4},{"id":"ff086dde725e93dc7e95b9d65da388f49e5c6078","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-06 07:24:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4b07812922b1fae393f8f36c74bc911e7b4d8f8a","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-08-06 07:25:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"984b6fe2eda2ba30d24289271446d0f51d9158cd","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-06 08:00:33.000000000","message":"Patch Set 4:\n\nAdriano Petrich wrote:\n\n\u003e (1 comment)\nThat\u0027s good then, no need!","accounts_in_message":[],"_revision_number":4},{"id":"515b63d611385364853e3e07f82fba88415c104e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 10:47:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4fb0782/ : SUCCESS in 3h 20m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/90/665690/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bb94734/ : SUCCESS in 2h 55m 26s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/90/665690/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/c38d2bf/ : FAILURE in 8m 31s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/90/665690/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/a743c8c/ : FAILURE in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f0c6ae2b7bdc0797179c3200f3bfd51dbedda8b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 11:31:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/90/665690/4/check/openstack-tox-lower-constraints/c7c1202/ : SUCCESS in 6m 44s\n- openstackclient-check-plugins https://logs.opendev.org/90/665690/4/check/openstackclient-check-plugins/cb3bb17/ : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/90/665690/4/check/openstack-tox-pep8/9215890/ : SUCCESS in 3m 50s\n- openstack-tox-py27 https://logs.opendev.org/90/665690/4/check/openstack-tox-py27/411114d/ : SUCCESS in 4m 17s\n- openstack-tox-py35 https://logs.opendev.org/90/665690/4/check/openstack-tox-py35/9a8310c/ : SUCCESS in 4m 29s\n- openstack-tox-py36 https://logs.opendev.org/90/665690/4/check/openstack-tox-py36/a06d830/ : SUCCESS in 5m 15s\n- openstack-tox-py37 https://logs.opendev.org/90/665690/4/check/openstack-tox-py37/771312c/ : SUCCESS in 4m 15s\n- openstack-tox-docs https://logs.opendev.org/90/665690/4/check/openstack-tox-docs/e172340/html/ : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://logs.opendev.org/90/665690/4/check/build-openstack-releasenotes/56d7bd8/html/ : SUCCESS in 3m 51s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/90/665690/4/check/tripleo-buildimage-overcloud-full-centos-7/4d325f0/ : SUCCESS in 23m 56s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-containers-multinode/a33f5f1/ : SUCCESS in 2h 30m 24s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-undercloud-containers/c7228e1/ : SUCCESS in 1h 37m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f608f5f/ : SUCCESS in 2h 29m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/31ffb95/ : FAILURE in 2h 44m 27s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-standalone/43c9989/ : SUCCESS in 1h 59m 39s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/90/665690/4/check/tripleo-ci-fedora-28-standalone/f270dc4/ : SUCCESS in 3h 13m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-standalone-upgrade/7c9dcc0/ : SUCCESS in 2h 20m 53s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1a45f54/ : FAILURE in 2h 17m 51s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/cc18f10/ : SUCCESS in 2h 24m 57s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/90/665690/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/e9a3611/ : FAILURE in 2h 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10e28d29cb0da5596c3da0575716a2be7005233d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-22 15:50:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9b8aeda7bd4057649c48e57eb2f4053f7e7c9174","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-29 14:49:12.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bbf0986a58fb2dc3fcd48bbb56f8954ac141cd34","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-09-27 08:30:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d9a43104acc6896b745919dd204b6b0bf15859f2","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2019-11-04 12:12:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"576260a5363814ac7ccc8f3d103c260edec252c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 15:11:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/606bba94b52e48199b8695f0937472ec : SUCCESS in 5m 25s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/dcaee93934ce4415904bbb85b2f69f4e : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d82daf6965464d308f922113512ab448 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdb2496e47d44c2589dc8554a06c93da : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae6473bd12748c98e3400213b9aaf1a : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb0fb1c533cf4403ab2dcd0e01c4c685 : SUCCESS in 4m 03s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/166e6c9b6f5248ec8eeecc5cb3d2ff7d : SUCCESS in 26m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9d0a5419113448628cb760aca2d5fe18 : SUCCESS in 2h 16m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2dd7fab1b1b54db3932042f402c8b0bd : SUCCESS in 1h 27m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5ac8d5bd80424d3da3793dd93120827a : SUCCESS in 2h 00m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7f568b6793d14bc583d44c0856a376ee : SUCCESS in 2h 14m 16s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/98c19a61b42d4ad3b4c7e28a30ed5632 : SUCCESS in 1h 42m 12s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/4404fdf9a82341c2940f02a08744af1d : FAILURE in 1h 32m 32s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b4cbee9610434b5bbe17745e1c8a54b3 : FAILURE in 2h 01m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83214632d95043f7a6d43e147dac41e9 : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/433a1d07ea5c4fa2aef642d2f1685fc2 : FAILURE in 3m 14s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9dcbf36e52794fce939723a97eb7c8c0 : SUCCESS in 2h 48m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b84e85a0a587663299b616e5623e5717e6655be6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-04 15:17:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8d9e50a/ : SUCCESS in 3h 02m 16s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/90/665690/5/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/4b12017/ : FAILURE in 17m 42s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/90/665690/5/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/e7beb00/ : SUCCESS in 18m 23s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/90/665690/5/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/75c24e9/ : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":5},{"id":"aee0dda864b49f5caf15a4ea34a62b4ffd00eb19","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-21 21:34:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"99d84d9430a01b8e86d927052fed3730c4edf4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 22:06:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f118622183ed49c0a6ec23d44153ca7e : FAILURE in 5m 15s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e40faf8198e04948a66b660aca898233 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ffca71d786c4961bfcf18f9e8004816 : FAILURE in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ce5c27916144658a3e5ace82e604ed8 : FAILURE in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c10fccb8fd044b88776a7a8072018ba : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71b881ca5d634844b7c6d92e926517d6 : SUCCESS in 3m 52s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/d2414936d1a1473c97d6021faed81fe4 : SUCCESS in 26m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/43877b6709b747a78a746784fcd6e40b : FAILURE in 25m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cb7f92fca2b74ebea740362815f778d8 : FAILURE in 25m 29s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5865fd7f9522401c9a1d9eebed77137b : FAILURE in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bb36b2eb8314376b3ac79fbd98e4c53 : FAILURE in 2m 56s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3aafce3dd679f98c9f2f2d262e9ce28e4f6c9d63","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-21 22:18:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e2724b9/ : FAILURE in 23m 11s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/90/665690/6/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/f94768d/ : FAILURE in 20m 42s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/90/665690/6/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/471390b/ : SUCCESS in 18m 56s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/90/665690/6/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/df73741/ : FAILURE in 20m 08s","accounts_in_message":[],"_revision_number":6},{"id":"0193a31f8a6f1970ea76e453dd1d869a17b71aa1","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-21 23:45:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"5b8215a618fb30986f2725a305d3e1594d98f080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 00:09:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad02a8f23fad49dc870d83e804ac325a : FAILURE in 4m 29s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ebfe5ca0267c4c47a628de611f8d3766 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4531b328a0f4354ad4174cad721fcc7 : FAILURE in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a35e6ac7be534b39b58ceb2373f80a10 : FAILURE in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61840f7f5f7c4ac3a412d648e6ae8108 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4e3f6944bdc46ea8b5ce8dae04c943b : SUCCESS in 3m 32s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/8241bc1c4e7e4f918474293a4c9faa60 : SUCCESS in 22m 38s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cf79cabcf74c43589d977529ed7e30b9 : FAILURE in 22m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d328d1b6d6ea40028a76b3cebfdbc2de : FAILURE in 22m 34s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82367673326643f49cec6315ce4b579b : FAILURE in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42dd785cc987457d8ee4b484d51c4632 : FAILURE in 3m 26s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2dd188221f4cb38f785bfbe30b1821e2dffae29d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 00:11:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a1afa92/ : FAILURE in 23m 09s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/90/665690/7/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/8482a7a/ : FAILURE in 16m 49s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/90/665690/7/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/fd57a07/ : SUCCESS in 18m 47s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/90/665690/7/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/ef307a8/ : FAILURE in 22m 27s","accounts_in_message":[],"_revision_number":7},{"id":"5ff0e1404ffb2c0ae5c948a0c5d366d44880dc94","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-22 02:48:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d399e6b9a49df106de2ca495aa9f92c8c4ef352c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 04:35:32.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ed863056f20414eab483b924bd2d015 : SUCCESS in 4m 48s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/92fa2acfb62d48e8afc2a29502f4b3d5 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd4654fb75174637a44af620a0508062 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cec79b3846e242419c8574e3816d4cc3 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a070371a414b19b387c8fe10e70cc8 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a94dce7b5bd4425b768496bb64be55e : SUCCESS in 4m 07s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/277807897c294ecb85c8b08765223cac : SUCCESS in 20m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/eff89d97b8b64c919b9162e34f43dd68 : SUCCESS in 1h 44m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a7e69998130c4e85aa057cac2a353d22 : SUCCESS in 1h 45m 58s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf279e5b5c1427eb5a3291ebb8c4300 : FAILURE in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c26376b26e624f8fad06afa1dc074df5 : FAILURE in 4m 17s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f008b61fa3e807b8edc451800838edd38402d61f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 05:09:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/25afbee/ : FAILURE in 2h 19m 53s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/90/665690/8/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/fb532e1/ : SUCCESS in 22m 34s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/90/665690/8/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/2eeb727/ : SUCCESS in 18m 34s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/90/665690/8/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/6eaa0a3/ : SUCCESS in 1h 21m 21s","accounts_in_message":[],"_revision_number":8},{"id":"1e88ecc1b65a24c8d38b76da9ad4a8bbaa17027d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-22 22:42:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"adb210ea1cbc17f9d6f84411ddcee97dc145c1cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-23 01:37:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cab4f77/ : SUCCESS in 2h 54m 12s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/90/665690/9/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/a675812/ : FAILURE in 17m 20s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/90/665690/9/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/f11496d/ : SUCCESS in 18m 39s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/90/665690/9/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/c0970f3/ : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":9},{"id":"0e784acca8166af3f54d177b7f691d8820023413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-23 01:58:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc89040c442e4b2dbfd345c78ded85fb : SUCCESS in 6m 01s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/a83d755495d04e3c81306019187dc229 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b65e4c41e684ec3826a460afb5265fb : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4636471c3334698b5125ad7b0315271 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97e5d46e7684ce48c12cb04dc2106e8 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/106a9b517ed1420f8f06135aff2e6891 : SUCCESS in 4m 06s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/bdbcaa73e4334fd5b78d98286fe2d35b : SUCCESS in 22m 39s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2300521cb7ec470386f3385f9d19b52c : SUCCESS in 2h 20m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/235de982760f4858a551806b506112c1 : SUCCESS in 1h 56m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/97e9cd8efd0c48faa3f0626a38663129 : SUCCESS in 2h 23m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ec9694087e64493791737676565e3133 : SUCCESS in 3h 00m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1e51c42d424a419e98c0b53655542bac : SUCCESS in 2h 10m 40s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7abe7ac037114fc583d747c9416e884e : FAILURE in 1h 36m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/099bd5031df84925af9896e458081a70 : FAILURE in 1h 58m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b440fa6f05344dee950f5f182c4ff98a : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7378823dd1804572a2ce03124ce81b1b : FAILURE in 3m 26s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3dd87e7dac8e428e8f9c6a6a4e0f4232 : SUCCESS in 2h 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"046986534f699e9bd301f9d7cfad026aed496028","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-25 15:01:17.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"cd3fd6173da1e8749645a2bce0eb149aae7a3a9a","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-25 15:13:23.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"382f861f3d8ccc9b2d8ab9fff2e1a6fc634a100d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-11-25 15:16:19.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"5d984a6454057356de525c85925393762f640595","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-25 18:13:30.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/90/665690/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/923488e/ : FAILURE in 2h 24m 41s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/90/665690/12/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/b651df9/ : FAILURE in 18m 04s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/90/665690/12/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/78deb98/ : SUCCESS in 18m 30s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/90/665690/12/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/55ae66f/ : FAILURE in 26m 18s","accounts_in_message":[],"_revision_number":12},{"id":"0827e64a464c752c139ea6d48969446c35ea96c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:22:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e15749bd5a34a6f919c5c7f6c691eb2 : SUCCESS in 5m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/90c4dd905ff44a7f921451b7f85286c9 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b24f40de95c41a6907a538637361564 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a7480513d1c401a8384691bce411276 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5539ae50c02a4ccfa8d0875d14e46dae : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1847aec2ee83439e875dceb7e50995a7 : SUCCESS in 4m 12s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/f4c2b98be68247ba8d7b73f0151ef68a : SUCCESS in 32m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a8df37f6688946d8a6486c3bb1b2d546 : SUCCESS in 2h 36m 25s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/09e08803cc7f4d76baf99a400cac2a63 : SUCCESS in 1h 42m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/391770fa3f5c429889c168c2af9fc69c : SUCCESS in 2h 58m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ad758e9a816148b6860fb19a9cd6e44d : SUCCESS in 2h 30m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c32621d6038d46869922743f6ecd2fad : SUCCESS in 1h 42m 45s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/565c14730fb84d6cb459132f4ae870ef : SUCCESS in 2h 32m 23s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/489bf4cf72b14c939655aeede01b5dea : SUCCESS in 2h 21m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c21cface74b40d5b497d8c4ac1d0aa9 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b57faeb4d824dbca9271fba91ff3c6a : FAILURE in 3m 26s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/fee449cabd334fd29f64d7d53d92f70e : SUCCESS in 2h 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e869c44440ff357a5b310142537950cf096e6b9f","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-28 12:41:52.000000000","message":"Patch Set 12: Code-Review-1\n\nI think we want to move this to a new v2 file. Something we had discussed a while back but I think we all forgot about. I am trying to follow that approach here; https://review.opendev.org/695893","accounts_in_message":[],"_revision_number":12},{"id":"e703b50c353eae4703f853309fe828342e7e04c3","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-21 13:54:19.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"63b7294df91048125930e56196f624a9cc8746cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 14:17:39.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/2ba37a1b074e44b58b32f45a4d016961 : SUCCESS in 17m 34s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/ed053e26c7b3425a848310edda6d960a : SUCCESS in 16m 39s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9ea57f8535ba4c9fb79c78b45ec656cf : FAILURE in 22m 24s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/852c4c16fd6c41a48db74dbe234dba4e : FAILURE in 21m 45s","accounts_in_message":[],"_revision_number":13},{"id":"6d5889ae873cf8d5f09cd62daacf0ecfa649a02f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-21 15:26:44.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"da28a48322fb1191489602529cfdf61f431e517c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-21 15:28:57.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"9a4f6c9b71b16a8c8fda983c53d374e2b1dc389b","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-21 15:30:08.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"484b023ae1fa32aa5855ef5e9e79723c3f19adcf","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-21 15:31:05.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7bcc3d26822855d66f6b357fa1860b02ed957947","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-21 15:50:37.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"3dae2cb535452c503b10b40a4eb45f7d2fdd4602","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-01-21 15:51:17.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"b7f647e5e0632fb00b27d42fd2f99df5ce73a12d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 18:35:58.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/3c652b7cd904434ba1eeffae46ebc329 : SUCCESS in 17m 53s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/82ada22bbf1c49abad0ae9ba4b41a65c : SUCCESS in 16m 00s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2f5b36e9581e410d8f222559c6b89617 : SUCCESS in 2h 44m 27s\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/7fa08b83926b490d9db1930ab6c3357c : SUCCESS in 26m 07s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/2edb980372c04c149b8b49cd189a5592 : SUCCESS in 1h 12m 35s","accounts_in_message":[],"_revision_number":18},{"id":"4bd9e4144735e7ff71449d99051f7995202e0418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 19:20:42.000000000","message":"Patch Set 18: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":18},{"id":"9b2c4a7d3ad6db44eb8f09b72b7750013df63f48","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-21 19:22:20.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"e7382b66f43d83ebbd59f8d9d53285e5793b573f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 21:58:01.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c966c3efd0684ce8be185448c91362e5 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60ff6ed6436d43bbbd8f6a7e3f5466f8 : SUCCESS in 7m 26s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5ef39d8641b14caf93082662d4925c50 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43cf10db489e4283801e5ab8894ca7ca : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bea6965099f403fb9064944a3246867 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f563b591d3348c49fa51e33ad7efecf : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bb4f4136240450cbd7d6cd41b763403 : SUCCESS in 4m 49s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/9355d2a7107542f98e33cb512104d8e7 : SUCCESS in 29m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3f6ae7089d9a491bbfbb7413de1c65d6 : SUCCESS in 1h 00m 56s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/1c58b6b54cfe4f2598849ba0f60247c6 : SUCCESS in 1h 14m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d88069c7def140fdb36d0a6b307363b9 : SUCCESS in 2h 00m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/55164a1ac5fa49fbb7d4774486fc3918 : SUCCESS in 1h 30m 26s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526662b69d5e4da5af5e65df16424a30 : FAILURE in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/add23e33ac22413a8a4eb7e59d8e6f3a : FAILURE in 6m 51s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"5d5947d332ed2abcc5b1232ce8295db3ab73de73","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 22:12:55.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/b7ab21395620436ab1f57655fdf9a14c : SUCCESS in 17m 34s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/9c25a421f4a64c5990fd3285e9f52b32 : SUCCESS in 14m 40s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/caf493bb0fa9487fb98dfb079ef864e5 : SUCCESS in 2h 42m 52s\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a80ec2226a26455c9ef588b33f794526 : SUCCESS in 26m 57s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/6cdb002e28c14af78d0f7ddb3f158551 : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":18},{"id":"cc85c8739b59aaf45e90d0b93d5bdc8f6a6ff0db","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-01-22 00:47:54.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"6370fe8daa72aa8a8e739652ac5decac23a06032","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-22 13:57:24.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"b5c05becdd2367ad1ec2621c933870ff61991afb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-22 16:53:47.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/f13ffb7d707a474da4036e729d363985 : SUCCESS in 18m 33s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/7aae1d4acc704bab8acedfa1309b3fce : SUCCESS in 17m 33s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1814c77ab6ab472fb8060ef18cd2a3fb : SUCCESS in 2h 54m 33s\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/5449f8ba36694f8db21a5c6c67954202 : SUCCESS in 30m 52s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/89dcf266247943a5b5799c1890aa5159 : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":19},{"id":"9b5b341e19d5db859d505de540479c613a914907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:37:36.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af9be42e94ed4fcca42b9c92705f4be4 : SUCCESS in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4b058cb84804091b11518f7afd219c2 : SUCCESS in 6m 39s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/1150e6eae89540bbbe890abc9968306f : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f98505e17724466869b4e839dfd439a : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a588a1e68194640b5cd3e85c9b87879 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f84b70a8225438aad16b90b86f67638 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f939a3dab9214f269c54e622cff3712f : SUCCESS in 5m 23s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/fba5e91069d34ee3936d3c08ec2b7f71 : SUCCESS in 34m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cb5aa4ffb0684f85bdfb503acd0f016d : SUCCESS in 54m 53s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/02798b17ff7c4949a306475badb76a06 : SUCCESS in 1h 02m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f95cd045249d4f6796146a032a7be908 : SUCCESS in 2h 56m 35s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/388ce70dba41469cab243d480ae54cd6 : SUCCESS in 2h 33m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/650f940d13f14ce59f837fe606c886a0 : SUCCESS in 2h 28m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4ee3d53cbc91464193fcf2b4c47fb764 : SUCCESS in 2h 48m 24s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/78eb5d59d50146b590dd5cb1a80a612d : SUCCESS in 1h 42m 42s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/eaaa57bdd02c42f1930936ce6fd183ff : SUCCESS in 2h 51m 54s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/0470195b0f80408eafae7815a922706c : FAILURE in 2h 25m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7a84a9e949b42aa9becaec6866e8209 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb1fcc0f639847a999d741cb347184c7 : FAILURE in 5m 12s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/862045fb76c441e9915fcf45b9c559f8 : SUCCESS in 3h 01m 44s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"68e800fc0ba7f89528710b0d02bf1b56a91f4b1e","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-01-23 11:06:57.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"151d7272d0e6eb0391ac878ec013c6840b877f51","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-01-23 13:45:24.000000000","message":"Patch Set 19: Code-Review-1\n\nI think we should go ahead and delete the v1 version of this command.","accounts_in_message":[],"_revision_number":19},{"id":"1f6711b39b33155f949949c55d272ccc0b743a69","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-23 13:55:30.000000000","message":"Patch Set 19:\n\n\u003e I think we should go ahead and delete the v1 version of this\n \u003e command.\n\ndone.","accounts_in_message":[],"_revision_number":19},{"id":"afc8d6f67d027d910dedfb11c44a207b3e07ff30","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-23 13:55:42.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"d432b3cacee026ef7f5cbeb8129ca88813f7fbd8","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-01-23 13:57:28.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"593b0c3e7a33774540ddd63afbc5d52b4a17a519","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-23 15:21:03.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/1a189a9d4cf44c559987fb524e40d686 : SUCCESS in 17m 36s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/dd34152a9fab4781b40bf59a0deece48 : SUCCESS in 14m 22s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2b9865db6b5045b4a7e4a8f625b030a5 : FAILURE in 1h 24m 30s\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/4fefad81291a4e3c81a6adf680e7cf45 : SUCCESS in 25m 01s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/368d8d96a49d4f3ba1bd7d579d6580ee : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":20},{"id":"6b425c56ca31a1f442c3c7a9806ffa63e2b8856d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 17:24:34.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b3cc569399d4394b5606c8f847f3d8f : SUCCESS in 2m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a4eab5196ef406aa177b1de7ca27595 : SUCCESS in 5m 55s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/bcce21666ff94b5fb3d0e1e92a497a0c : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860166d3d454481f9e3725a9d6acc266 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78f1f72349d34c0b800ccf3381ba6d6e : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d486967f01f04768abffdcbec363688a : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4875f6269c394099bdc4f33a5949892a : SUCCESS in 4m 16s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/53beb898658449bbab3e9e58f11c212d : SUCCESS in 28m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/901fada3dd624edfbdd6de5f92215c6b : SUCCESS in 52m 11s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/d589ca03755a4beb9cce3443d4bbc03a : SUCCESS in 1h 30m 15s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d075652849894db482f21713ce3487cc : SUCCESS in 2h 46m 43s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b208a8f072d441d2870c0a8caffeea70 : SUCCESS in 2h 22m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/acef2dc2a23f4d6280324e6a53e0f370 : SUCCESS in 3h 08m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5441090163044085847c58fe0adc66c8 : SUCCESS in 2h 20m 20s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9a5e599bf96b42838268347b52c52485 : SUCCESS in 1h 38m 15s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/69cb5964caaa45e98c6d9fb5cad4c9ff : SUCCESS in 2h 19m 13s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6463a1d47db240389e1584664c26717a : SUCCESS in 2h 58m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92f5bf83d5a14e31baa06d5847d9c2f2 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc5c9ac101d54f5aacdf72d82d99fdb5 : FAILURE in 6m 50s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/789ef692ebd24de3a121038eb334cb8a : SUCCESS in 2h 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"1c0daf28d2c04169cc2329e258709678f4503819","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-01-24 09:59:42.000000000","message":"Patch Set 21: Commit message was updated.","accounts_in_message":[],"_revision_number":21},{"id":"43dc37708f735b21b4ae2dab310226b994896789","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-01-24 10:08:52.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"c0afb86f7b3d8177bc32f1a538356724a7dc7189","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-24 12:54:34.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/37ed9c91767a48238b5568faac8dd311 : SUCCESS in 17m 54s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/820bec883d484c4188a054414d3e8257 : SUCCESS in 14m 54s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1387af5476bb4bf8b498bdd8fb671d26 : SUCCESS in 2h 54m 07s\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/fd185a18cfac46b1b2f7774c1c3affc9 : SUCCESS in 24m 49s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/da2318eedaca4352b0e931ceec1494d1 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":21},{"id":"0a4378df096eefd67600d8b4272695c776a8f83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 13:31:03.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d0bcedfad98941e8b6edc511260f51be : SUCCESS in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe56c22b6b72455ab6198949a961289c : SUCCESS in 5m 57s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5a847cb7bfd9425b9f35ed66e2386661 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc1ee32f20fe4c4db58241f47497f865 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bab8af25381499983a57bd5c66b48d9 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1f45a3806d4a2fa32be96bbb00a3e1 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad2612c02b374405b3322edfa8ad7653 : SUCCESS in 4m 36s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/4327ea3696b54b8a89d34da4bec0af40 : SUCCESS in 27m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/20d5cc13988b4304bfa40c9f1b1effd7 : SUCCESS in 1h 07m 53s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/9591e0f7c85840e5bb3b1f64b2c2667c : SUCCESS in 1h 04m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/eba585da1e3b46e48263e0a6ec38cd85 : SUCCESS in 3h 03m 05s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2e4f00e6807243faafa8deb3bb6a3910 : SUCCESS in 1h 58m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6b5e284518f34f488a6b739f470d5900 : SUCCESS in 2h 15m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/41efa73e55e9440c863dda3fd6ce5d03 : SUCCESS in 2h 47m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/902f2b61c25a4f52ad1b8c6f8526e668 : SUCCESS in 1h 46m 29s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/8ebb585857cf40aabc4865aed1d2ab35 : SUCCESS in 2h 29m 33s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/80e67bb9a7f8498fb01d05cf7a0389f3 : FAILURE in 1h 34m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fae4642c5ab48ee850d40bf3011d14d : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c392479d422c49a19c620e00cc2d9a2c : FAILURE in 7m 07s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7c0a96fcb5e84fd58eb3c2cf10902cae : SUCCESS in 3h 06m 34s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"66de3cafd7c63c74b905950cb2a22566496a56f4","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-27 14:35:46.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"9a7d697fdcbb917d759ef5ce662c2540cdbd7e9c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-27 14:48:49.000000000","message":"Patch Set 21:\n\nthis will need a rebase in order to be landed","accounts_in_message":[],"_revision_number":21},{"id":"b2a14c41d5c4f9f007ec4c26c28cc4b8a02cc6e1","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-01-27 17:12:33.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"e1261eee5d3f1fcf0087c39d0a9634599793c85f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-27 20:21:38.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-rhel-8-buildimage-overcloud-full https://review.rdoproject.org/zuul/build/f58f96e7e1ae41258f02c55b36ea4166 : SUCCESS in 17m 30s\n- tripleo-rhel-8-buildimage-ironic-python-agent https://review.rdoproject.org/zuul/build/b591e70422584be1aee6b95b0a5ee788 : SUCCESS in 15m 03s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/08d1970e2ad24078a4ecde910f0d624e : SUCCESS in 2h 35m 53s\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/2a810815df834214a6c5a9c6c998095c : SUCCESS in 26m 10s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo https://review.rdoproject.org/zuul/build/b3915a5e069e40d09ec3075db3e004a7 : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":22},{"id":"eb746a2461d4575cf45f39152e4b83a5c59dd3dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 20:37:05.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/072d3597fa624bbca0d38b9b43cf785f : SUCCESS in 3m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db3b3c97895b4b92b4b43466e1bdfcb6 : SUCCESS in 5m 32s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/661991dfefe34e86baa3400cc4c5e039 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ac8b45883204dc4b362ebdbed441d5f : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceb80cfaef0842c58e26911ddbcae328 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89a31e063f94758be0ceb81b0dd8697 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/594d46110142408999085de4beb7bf69 : SUCCESS in 3m 41s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/2701248a5ba74193890ed4f3ecbf0f75 : SUCCESS in 31m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8d2a42e575bb445881e7c24dffef0f9f : SUCCESS in 1h 00m 41s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/d13b952506e14fb88f41cfdc660c615e : SUCCESS in 1h 14m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2e20ea20149440eaa287af51077fb613 : SUCCESS in 3h 08m 31s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/db15501484e54cf9bef27a01cdabed80 : SUCCESS in 1h 56m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bbf5546dd3a04b53b0c434c9a09e7bd8 : SUCCESS in 2h 27m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8bcbe6c0fdd64ff9b386f9d483ee1509 : SUCCESS in 2h 21m 54s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4c46d11fea7c44049beef5954e8bc61a : SUCCESS in 1h 32m 47s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fa3793db417b4b5180d2a49f83a67770 : FAILURE in 2h 10m 08s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/fcbf0d0144d5427d8e551e0de1420307 : FAILURE in 1h 53m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa8988a193af4bc8a0d9ee988a85700d : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d4e745b42724f14b57360dfd6e34ebc : FAILURE in 4m 38s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3f652bcc027b48e0831904df5493d7ae : SUCCESS in 2h 53m 29s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"e77784646030a394d6bb7545692c6aa40a591b3c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-27 22:50:08.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"2a5a7db4ca2026c45a31159a00624ababf00586e","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-01-28 21:43:53.000000000","message":"Patch Set 22:\n\n\u003e (1 comment)\n\nUnfortunately we\u0027ve not figured out how to keep both v1 and v2 active at the same time; it seems osc wants only one active api and our client plugin doesn\u0027t seem to provide for version switching. We could update the v1 backup command and tests in-place, and then push the v2 migration to a follow-on review?","accounts_in_message":[],"_revision_number":22},{"id":"993e78af84305108d1e8ce1722a777b8f004bfe3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-28 21:51:35.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\nI guess we can go with this. I think you\u0027d have to add some python code to go back to execute the previous version. For example a new --legacy option that would just invoke the v1 code in take_action.  I was just thinking it would be nice to allow a fallback if we ultimately find issues with this code but still want to offer a legacy way of invoking for this cycle.  \n\nI\u0027m of the opinion that we need to cut these over similar to how we did the undercloud switch where we use a cli command argument as a feature flag that allowed for additional testing without breaking the world. I\u0027m not aware of CI jobs for this that could exercise that functionality but it might be beneficial for some of the other functions.","accounts_in_message":[],"_revision_number":22},{"id":"6e12c70853c6de5310623df3559e7bf7a5eb3da1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 21:51:57.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"90f5d66545a89954777a16e05d3b00f630873c46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 01:54:03.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f96ef23dd3de46e2af696417976854a7 : SUCCESS in 3m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ff4611d7f634143a2192195b4f440ea : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e06e2b509e84c83bc9c31d063eeaadf : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f8fb9d00a7d443f8a944178896e7c3c : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c7fb26ce76474c9453d03e07d21395 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2264a474636f42878a93117dde8cc5fe : SUCCESS in 5m 27s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/66f3e8197b6740da8816d9cbf6a7cb9a : SUCCESS in 24m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/01943cfed6ab4ab6bd82a348bde10837 : SUCCESS in 49m 37s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/b1adf23376b04f1791189cc922132640 : SUCCESS in 58m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4f4ec0eb098e4be7a48e98c1e3ebe39e : SUCCESS in 2h 47m 05s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1bd4a93aae5e450e8b73c2095f02b90c : SUCCESS in 1h 30m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2804b4ab51ab449ab2e5f08cc3c74483 : SUCCESS in 2h 08m 50s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/64c4f13f202143bb91430938cd59e78e : SUCCESS in 1h 56m 38s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/d6dd6041562a4d03b216b283ab5e4a59 : SUCCESS in 1h 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950098abcba3406eababafb8aacd9f1a : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":22},{"id":"dd3c558b545132262be4a4d72ca21ff138102a8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 01:54:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"358c460562bb217946cd0923fe04416e443cc129","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 01:55:25.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ad8f10e0f249e78d9736941cd36ca4 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a67bf7eb2c34e9bb89dff201a087823 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"8bc80382d3a940cc34f4f4377cf53a80364959f4","revisions":{"9ad7202d5a44345fd86a62e5742c682db86602ba":{"kind":"REWORK","_number":1,"created":"2019-06-17 14:31:23.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/90/665690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/1"}}},"commit":{"parents":[{"commit":"05f0218a0c78e66fb0527266e6cedae14b16fe2c","subject":"Merge \"Improve yaml output of \"openstack overcloud node provision\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/05f0218a0c78e66fb0527266e6cedae14b16fe2c"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nDepends-on: Ie09fda68b631725c3667961d271ae627d1f89038\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9ad7202d5a44345fd86a62e5742c682db86602ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9ad7202d5a44345fd86a62e5742c682db86602ba"}]},"branch":"refs/heads/master"},"7e75856d73993da427f5d6955b6b491459c9cee1":{"kind":"REWORK","_number":2,"created":"2019-06-20 09:23:53.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/90/665690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/2"}}},"commit":{"parents":[{"commit":"4bfd472b38919a24be15dd49f03d5b6e13bdd45b","subject":"Merge \"Allow run_update_ansible_action run with Mistral or Ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4bfd472b38919a24be15dd49f03d5b6e13bdd45b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-20 09:23:23.000000000","tz":120},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nDepends-on: Ie09fda68b631725c3667961d271ae627d1f89038\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7e75856d73993da427f5d6955b6b491459c9cee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7e75856d73993da427f5d6955b6b491459c9cee1"}]},"branch":"refs/heads/master"},"8646958322905f9281e0d6f69694182d9d460084":{"kind":"REWORK","_number":3,"created":"2019-08-02 17:18:29.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/90/665690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/3"}}},"commit":{"parents":[{"commit":"7b365a586ea301ab83c9441889ab1cc4c882e48a","subject":"Merge \"Write minion preflight checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7b365a586ea301ab83c9441889ab1cc4c882e48a"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-08-02 17:18:11.000000000","tz":120},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name.\nIt adds the functionality discussed on this external bug\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem\n\nDepends-on: Icf964c23745941dd596f65e2b0bfa81e515508d0\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8646958322905f9281e0d6f69694182d9d460084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8646958322905f9281e0d6f69694182d9d460084"}]},"branch":"refs/heads/master"},"b5672d714457b7ee743dcebab2fc61c50e1976be":{"kind":"REWORK","_number":4,"created":"2019-08-05 08:05:49.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/90/665690/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/4"}}},"commit":{"parents":[{"commit":"7b365a586ea301ab83c9441889ab1cc4c882e48a","subject":"Merge \"Write minion preflight checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7b365a586ea301ab83c9441889ab1cc4c882e48a"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-08-05 08:05:38.000000000","tz":120},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name.\nIt adds the functionality discussed on this external bug\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem\n\nDepends-on: Icf964c23745941dd596f65e2b0bfa81e515508d0\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b5672d714457b7ee743dcebab2fc61c50e1976be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b5672d714457b7ee743dcebab2fc61c50e1976be"}]},"branch":"refs/heads/master"},"705755b7550bf6a8cdbcca8e843c034c4c7421a9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-04 12:12:42.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/90/665690/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/5"}}},"commit":{"parents":[{"commit":"e794286d47818f787eff50f8b2a16eb7093d4537","subject":"undercloud: introduce undercloud_enable_paunch option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e794286d47818f787eff50f8b2a16eb7093d4537"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Adriano Petrich","email":"apetrich@redhat.com","date":"2019-11-04 12:12:42.000000000","tz":0},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name.\nIt adds the functionality discussed on this external bug\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem\n\nDepends-on: Icf964c23745941dd596f65e2b0bfa81e515508d0\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/705755b7550bf6a8cdbcca8e843c034c4c7421a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/705755b7550bf6a8cdbcca8e843c034c4c7421a9"}]},"branch":"refs/heads/master"},"50e4df7146e06ad4007fa91398aa609c1f109611":{"kind":"REWORK","_number":6,"created":"2019-11-21 21:34:59.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/6"}}},"commit":{"parents":[{"commit":"f34481f35199e6c63df1eac4c27b77b16233ab47","subject":"Migrate the openstack overcloud delete command to use Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f34481f35199e6c63df1eac4c27b77b16233ab47"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-11-21 21:33:56.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: RHBZ:1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/50e4df7146e06ad4007fa91398aa609c1f109611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/50e4df7146e06ad4007fa91398aa609c1f109611"}]},"branch":"refs/heads/master"},"ab15456f582641fef582f95dec52e7492cb376b3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-11-21 23:45:45.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/7"}}},"commit":{"parents":[{"commit":"28a1cac6a1caa289d8264eb8fce47f65db22b40b","subject":"Migrate the openstack overcloud delete command to use Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/28a1cac6a1caa289d8264eb8fce47f65db22b40b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-11-21 23:45:45.000000000","tz":0},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: RHBZ:1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ab15456f582641fef582f95dec52e7492cb376b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ab15456f582641fef582f95dec52e7492cb376b3"}]},"branch":"refs/heads/master"},"e3749bac733d2590aad9dfd474f49752c267a1b0":{"kind":"REWORK","_number":8,"created":"2019-11-22 02:48:32.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/8"}}},"commit":{"parents":[{"commit":"28a1cac6a1caa289d8264eb8fce47f65db22b40b","subject":"Migrate the openstack overcloud delete command to use Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/28a1cac6a1caa289d8264eb8fce47f65db22b40b"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-11-22 02:48:27.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: RHBZ:1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e3749bac733d2590aad9dfd474f49752c267a1b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e3749bac733d2590aad9dfd474f49752c267a1b0"}]},"branch":"refs/heads/master"},"f41523e820d51f622f9fc1d32fa00d4a7283d277":{"kind":"REWORK","_number":9,"created":"2019-11-22 22:42:00.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/9"}}},"commit":{"parents":[{"commit":"2224dc9a257ef1f61761fe46500914352873066c","subject":"Migrate the openstack overcloud delete command to use Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2224dc9a257ef1f61761fe46500914352873066c"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-11-22 22:41:31.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f41523e820d51f622f9fc1d32fa00d4a7283d277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f41523e820d51f622f9fc1d32fa00d4a7283d277"}]},"branch":"refs/heads/master"},"b68e5a5e8b9b2ca6eaa98057927dd0395d4dc997":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-11-25 15:01:17.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/10"}}},"commit":{"parents":[{"commit":"8f7339a9ff0db5bd51bbe815ed36dffb1fc6a963","subject":"Migrate the openstack overcloud delete command to use Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8f7339a9ff0db5bd51bbe815ed36dffb1fc6a963"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-11-25 15:01:17.000000000","tz":0},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b68e5a5e8b9b2ca6eaa98057927dd0395d4dc997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b68e5a5e8b9b2ca6eaa98057927dd0395d4dc997"}]},"branch":"refs/heads/master"},"98c6e7fa8a93981917256bfcf6bb05af6f2eacb1":{"kind":"REWORK","_number":11,"created":"2019-11-25 15:13:23.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/11"}}},"commit":{"parents":[{"commit":"8f7339a9ff0db5bd51bbe815ed36dffb1fc6a963","subject":"Migrate the openstack overcloud delete command to use Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8f7339a9ff0db5bd51bbe815ed36dffb1fc6a963"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-11-25 15:13:18.000000000","tz":0},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/98c6e7fa8a93981917256bfcf6bb05af6f2eacb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/98c6e7fa8a93981917256bfcf6bb05af6f2eacb1"}]},"branch":"refs/heads/master"},"a29ad5c551bb342da8961725fa98393d4da9b42d":{"kind":"REWORK","_number":12,"created":"2019-11-25 15:16:19.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/12"}}},"commit":{"parents":[{"commit":"8f7339a9ff0db5bd51bbe815ed36dffb1fc6a963","subject":"Migrate the openstack overcloud delete command to use Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8f7339a9ff0db5bd51bbe815ed36dffb1fc6a963"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2019-11-25 15:16:16.000000000","tz":0},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a29ad5c551bb342da8961725fa98393d4da9b42d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a29ad5c551bb342da8961725fa98393d4da9b42d"}]},"branch":"refs/heads/master"},"773de79c5e65682698223d7768de089fd7eb55a8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-01-21 13:54:19.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/13"}}},"commit":{"parents":[{"commit":"6417b0cfd993b20deb21279630b8a29a0a5e2195","subject":"Merge \"Add tripleo launch heat command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6417b0cfd993b20deb21279630b8a29a0a5e2195"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2020-01-21 13:54:19.000000000","tz":0},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nDepends-on: Ia82ddeeae53b2c61a914e14ac8bae4adcee90c76\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/773de79c5e65682698223d7768de089fd7eb55a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/773de79c5e65682698223d7768de089fd7eb55a8"}]},"branch":"refs/heads/master"},"92a14ab6bdcc12811d73ecbfa419f7b223a3e0b1":{"kind":"REWORK","_number":14,"created":"2020-01-21 15:26:44.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/14"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-01-21 15:25:34.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/92a14ab6bdcc12811d73ecbfa419f7b223a3e0b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/92a14ab6bdcc12811d73ecbfa419f7b223a3e0b1"}]},"branch":"refs/heads/master"},"967be318441bb6b86200d6e24792b5d4f366cd91":{"kind":"NO_CHANGE","_number":15,"created":"2020-01-21 15:28:57.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/15"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-01-21 15:28:50.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/967be318441bb6b86200d6e24792b5d4f366cd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/967be318441bb6b86200d6e24792b5d4f366cd91"}]},"branch":"refs/heads/master"},"aac6f4e59ef39c0e62092ec7d231efd626794827":{"kind":"REWORK","_number":16,"created":"2020-01-21 15:30:08.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/16"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-01-21 15:30:04.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/aac6f4e59ef39c0e62092ec7d231efd626794827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/aac6f4e59ef39c0e62092ec7d231efd626794827"}]},"branch":"refs/heads/master"},"25e535e86391c8e3585b7de2b83bca24baab4699":{"kind":"REWORK","_number":17,"created":"2020-01-21 15:31:05.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/17"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-01-21 15:31:00.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nRelated-Bug: rhbz#1659888\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/25e535e86391c8e3585b7de2b83bca24baab4699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/25e535e86391c8e3585b7de2b83bca24baab4699"}]},"branch":"refs/heads/master"},"755d80fb0eeb6d4689f8bf8ce0286f7d67413974":{"kind":"REWORK","_number":18,"created":"2020-01-21 15:50:37.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/18"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-01-21 15:50:23.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nDepends-On: I309d52eb424a73bd04af167a2a7411ba3bf35674\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/755d80fb0eeb6d4689f8bf8ce0286f7d67413974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/755d80fb0eeb6d4689f8bf8ce0286f7d67413974"}]},"branch":"refs/heads/master"},"0a37e4be2b94a7318823d5002c3636bb45a464db":{"kind":"REWORK","_number":19,"created":"2020-01-22 13:57:24.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/19"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-01-22 13:57:14.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nDepends-On: I309d52eb424a73bd04af167a2a7411ba3bf35674\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0a37e4be2b94a7318823d5002c3636bb45a464db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0a37e4be2b94a7318823d5002c3636bb45a464db"}]},"branch":"refs/heads/master"},"9c5bc75df75b7688343684e345bc09c225f31698":{"kind":"REWORK","_number":20,"created":"2020-01-23 13:55:42.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/90/665690/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/20"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2020-01-23 13:55:33.000000000","tz":-360},"subject":"Move undercloud backup from ansible to mistral","message":"Move undercloud backup from ansible to mistral\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nDepends-On: I309d52eb424a73bd04af167a2a7411ba3bf35674\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c5bc75df75b7688343684e345bc09c225f31698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c5bc75df75b7688343684e345bc09c225f31698"}]},"branch":"refs/heads/master"},"4b3245c23fab6646ced155888edf29455f9e50a2":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2020-01-24 09:59:42.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/90/665690/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/21"}}},"commit":{"parents":[{"commit":"690a0a4d6da17302d32b7665cc5125f65f8ecf88","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/690a0a4d6da17302d32b7665cc5125f65f8ecf88"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Adriano Petrich","email":"apetrich@redhat.com","date":"2020-01-24 09:59:37.000000000","tz":0},"subject":"Move undercloud backup from mistral to ansible","message":"Move undercloud backup from mistral to ansible\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nDepends-On: I309d52eb424a73bd04af167a2a7411ba3bf35674\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4b3245c23fab6646ced155888edf29455f9e50a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4b3245c23fab6646ced155888edf29455f9e50a2"}]},"branch":"refs/heads/master"},"8bc80382d3a940cc34f4f4377cf53a80364959f4":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2020-01-27 17:12:33.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/90/665690/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/90/665690/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/90/665690/22"}}},"commit":{"parents":[{"commit":"f456ac9312c879c0489d1b3c09cc67696ff3d7c2","subject":"Prepare tripleoclient for v2 implementations of the commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f456ac9312c879c0489d1b3c09cc67696ff3d7c2"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2019-06-17 14:30:24.000000000","tz":120},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-01-27 17:12:33.000000000","tz":0},"subject":"Move undercloud backup from mistral to ansible","message":"Move undercloud backup from mistral to ansible\n\nThis playbook reflects the same behaviour as the mistral workbook\nof the same name and adds functionality discussed on this external\nbug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1659888\n\nThis also adds a --save-swift parameter (default false)\nthat makes the backup be saved on swift otherwise it will\nbe saved on the filesystem.\n\nDepends-On: I309d52eb424a73bd04af167a2a7411ba3bf35674\nChange-Id: I054a805e820087a30bd8d27a0b460b6e519849ac\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8bc80382d3a940cc34f4f4377cf53a80364959f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8bc80382d3a940cc34f4f4377cf53a80364959f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
