)]}'
{"id":"openstack%2Fpython-tripleoclient~668685","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fstein~I15511b4f36260292e0ea4100b15b8e65a701b38b","project":"openstack/python-tripleoclient","branch":"stable/stein","topic":"bp/validation-framework-stable/stein","hashtags":[],"change_id":"I15511b4f36260292e0ea4100b15b8e65a701b38b","subject":"Allow run_update_ansible_action run with Mistral or Ansible","status":"MERGED","created":"2019-07-02 14:57:18.000000000","updated":"2019-07-04 15:41:35.000000000","submitted":"2019-07-04 15:41:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668685-1562254895618-6efaa91b","meta_rev_id":"ecc7fa6de8031e5e26d297923176c3342e29b769","_number":668685,"virtual_id_number":668685,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-04 15:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-04 09:47:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"all":[{"value":2,"date":"2019-07-04 12:30:44.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-07-04 07:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"all":[{"value":1,"date":"2019-07-04 12:30:44.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 14:57:31.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2019-07-02 14:57:34.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-02 14:57:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-02 17:19:12.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-04 09:47:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 12:30:44.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2019-07-04 15:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fdd9cbae3d09dbb83359700a9ab8f7feccb60d4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-02 14:57:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"317aedd7b7f4b73b45087507f5b99202095936e6","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-02 17:19:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"62297eb052c2c36f28833fd58c2bec774a4562e1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 18:07:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/85/668685/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ae2bc66/ : FAILURE in 2h 53m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/85/668685/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/13670ce/ : SUCCESS in 3h 08m 21s","accounts_in_message":[],"_revision_number":1},{"id":"671380c9dcdc8106f9783a2d19b1aab3ec861922","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 20:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/668685/1/check/openstack-tox-lower-constraints/da5c271/ : SUCCESS in 6m 40s\n- openstackclient-check-plugins http://logs.openstack.org/85/668685/1/check/openstackclient-check-plugins/dfc6d79/ : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/668685/1/check/openstack-tox-pep8/4a9d8d3/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/85/668685/1/check/openstack-tox-py27/e66d1ee/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/85/668685/1/check/openstack-tox-py35/be3abc7/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/85/668685/1/check/openstack-tox-py36/e57a512/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/85/668685/1/check/openstack-tox-py37/e68007e/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/85/668685/1/check/openstack-tox-docs/26d1bcf/html/ : SUCCESS in 4m 29s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/85/668685/1/check/tripleo-buildimage-overcloud-full-centos-7/407dabd/ : SUCCESS in 24m 14s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-containers-multinode/eb43646/ : SUCCESS in 2h 04m 08s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-undercloud-containers/11fc7bf/ : SUCCESS in 1h 37m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/de81181/ : SUCCESS in 2h 21m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/81bce86/ : SUCCESS in 2h 38m 45s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-standalone/91888eb/ : SUCCESS in 1h 40m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/668685/1/check/tripleo-ci-fedora-28-standalone/9e498fc/ : SUCCESS in 1h 55m 13s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1c31b91/ : FAILURE in 1h 25m 53s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/604e97c/ : SUCCESS in 3h 11m 27s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/110c59d/ : FAILURE in 1h 31m 28s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/85/668685/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4ff26d4/ : SUCCESS in 2h 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4691c47668a866053457b32ec9b2a8651546b72d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-04 07:07:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d543e80161ea42eb183fd841d5ec25e7286f1eec","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-04 07:09:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3f8aca55be9ba89077a9a43f3c3ae7c151e956f4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 09:47:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/85/668685/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/59d65f4/ : SUCCESS in 2h 37m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/85/668685/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a597fb9/ : SUCCESS in 2h 33m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4007208361e95957b3230fb98cdc05e34f38620d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:34:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/668685/2/check/openstack-tox-lower-constraints/9712d4d/ : SUCCESS in 5m 04s\n- openstackclient-check-plugins http://logs.openstack.org/85/668685/2/check/openstackclient-check-plugins/aedde9f/ : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/668685/2/check/openstack-tox-pep8/6b382b5/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/85/668685/2/check/openstack-tox-py27/27e2dc9/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/85/668685/2/check/openstack-tox-py35/5d2ff23/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/85/668685/2/check/openstack-tox-py36/4f6b5f2/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/85/668685/2/check/openstack-tox-py37/99a189f/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/85/668685/2/check/openstack-tox-docs/70b1914/html/ : SUCCESS in 3m 59s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/85/668685/2/check/tripleo-buildimage-overcloud-full-centos-7/5bff2e4/ : SUCCESS in 25m 06s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-containers-multinode/51a4a39/ : SUCCESS in 2h 20m 18s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-undercloud-containers/5951e88/ : SUCCESS in 1h 46m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/dd7fab0/ : SUCCESS in 2h 27m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/757916e/ : SUCCESS in 3h 02m 52s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-standalone/69f6504/ : SUCCESS in 1h 58m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/668685/2/check/tripleo-ci-fedora-28-standalone/749ac1e/ : SUCCESS in 1h 58m 32s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6a5cfe2/ : FAILURE in 1h 33m 46s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/ca32737/ : SUCCESS in 2h 42m 16s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/c8b9f23/ : FAILURE in 1h 21m 09s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/85/668685/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/97c26ae/ : SUCCESS in 2h 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf53e0cb82bca9606b47e071ff1d46bb26b7b711","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-07-04 12:30:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a12f1484271499db77cd4f269fd9e2478c0ef5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:30:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d541b31e1335d2a5830c5c7abe75dda9c44f63ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 15:41:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"846bb2c1eeb889aa994a36094c9f69337bc0c653","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 15:41:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/85/668685/2/gate/openstack-tox-lower-constraints/2105918/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/85/668685/2/gate/openstack-tox-pep8/c614157/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/85/668685/2/gate/openstack-tox-py27/84bf183/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/85/668685/2/gate/openstack-tox-py35/d002f7b/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/85/668685/2/gate/openstack-tox-py36/2de5467/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/85/668685/2/gate/openstack-tox-py37/7241050/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/85/668685/2/gate/openstack-tox-docs/a26c849/html/ : SUCCESS in 3m 21s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/85/668685/2/gate/tripleo-buildimage-overcloud-full-centos-7/c356e19/ : SUCCESS in 23m 48s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/668685/2/gate/tripleo-ci-centos-7-containers-multinode/8cb7357/ : SUCCESS in 2h 11m 40s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/668685/2/gate/tripleo-ci-centos-7-standalone/3641565/ : SUCCESS in 1h 58m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/668685/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6287125/ : SUCCESS in 2h 33m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/668685/2/gate/tripleo-ci-centos-7-undercloud-containers/5c9e37a/ : SUCCESS in 2h 00m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09d0a03d3dc6ff0b8108d1e415beeb508490f3b1","revisions":{"b802e028ed701c3d2524eb75a70f0869735d2342":{"kind":"REWORK","_number":1,"created":"2019-07-02 14:57:18.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/85/668685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/85/668685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/1"}}},"commit":{"parents":[{"commit":"f78809a3e3122f268bcc8fa1a9c12974735654b9","subject":"Ensure files exist before building/uploading plan","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f78809a3e3122f268bcc8fa1a9c12974735654b9"}]}],"author":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-02-12 08:48:44.000000000","tz":60},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-02 14:57:18.000000000","tz":0},"subject":"Allow run_update_ansible_action run with Mistral or Ansible","message":"Allow run_update_ansible_action run with Mistral or Ansible\n\nWe extend run_update_ansible_action\nto allow running the Ansible playbooks with Mistral\nor executing directly thru Ansible.\n\nThis is needed  in case we need to run\nexceptionally a task depending on Mistral\nbut Mistral is broken. For example, retry\nan upgrade operation after having Mistral broken.\n\nChange-Id: I15511b4f36260292e0ea4100b15b8e65a701b38b\n(cherry picked from commit 5249fbb2623165ef40098631fad18104579450c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b802e028ed701c3d2524eb75a70f0869735d2342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b802e028ed701c3d2524eb75a70f0869735d2342"}]},"branch":"refs/heads/stable/stein"},"09d0a03d3dc6ff0b8108d1e415beeb508490f3b1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-04 07:09:30.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/85/668685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/85/668685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/85/668685/2"}}},"commit":{"parents":[{"commit":"7d41f81ecad3d4692a5ca97d3ad7760d6287ca82","subject":"Send terminal erase sequence on tripleo deploy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7d41f81ecad3d4692a5ca97d3ad7760d6287ca82"}]}],"author":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-02-12 08:48:44.000000000","tz":60},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-04 07:08:20.000000000","tz":120},"subject":"Allow run_update_ansible_action run with Mistral or Ansible","message":"Allow run_update_ansible_action run with Mistral or Ansible\n\nWe extend run_update_ansible_action\nto allow running the Ansible playbooks with Mistral\nor executing directly thru Ansible.\n\nThis is needed  in case we need to run\nexceptionally a task depending on Mistral\nbut Mistral is broken. For example, retry\nan upgrade operation after having Mistral broken.\n\nChange-Id: I15511b4f36260292e0ea4100b15b8e65a701b38b\n(cherry picked from commit 5249fbb2623165ef40098631fad18104579450c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/09d0a03d3dc6ff0b8108d1e415beeb508490f3b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/09d0a03d3dc6ff0b8108d1e415beeb508490f3b1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
