)]}'
{"id":"openstack%2Fpython-tripleoclient~668940","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fqueens~Ifd4229b68571fd6b91574675c1f04d7b5c755eab","project":"openstack/python-tripleoclient","branch":"stable/queens","topic":"bug/1834307-stable/stein-stable/rocky-stable/queens","hashtags":[],"change_id":"Ifd4229b68571fd6b91574675c1f04d7b5c755eab","subject":"Ensure files exist before building/uploading plan","status":"MERGED","created":"2019-07-03 15:38:42.000000000","updated":"2019-07-04 16:09:36.000000000","submitted":"2019-07-04 16:09:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668940-1562256576570-6efaa91b","meta_rev_id":"11e5be306f2db10e8683f48528b4a77c9dcb6e5c","_number":668940,"virtual_id_number":668940,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-04 16:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-03 17:58:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-04 14:06:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-04 14:06:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 17:58:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 14:06:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-04 16:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ab631794028b022cfc07a44b142aa36b212fdd2","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-03 15:38:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"e7b1bb1ac893eb034208e97092d6afa60e354526","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-03 17:58:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/40/668940/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/facbf18/ : SUCCESS in 2h 12m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/40/668940/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4af2920/ : SUCCESS in 2h 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"48344934901470e28c7101f97b016a105d2e9abe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:34:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/40/668940/1/check/tripleo-ci-centos-7-undercloud-oooq/4291b13/ : SUCCESS in 1h 16m 11s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/40/668940/1/check/tripleo-ci-centos-7-undercloud-upgrades/78d1eaf/ : SUCCESS in 1h 22m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/40/668940/1/check/tripleo-ci-centos-7-containers-multinode/ef32bc1/ : SUCCESS in 2h 18m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/40/668940/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/bc0d372/ : FAILURE in 1h 59m 05s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/40/668940/1/check/tripleo-buildimage-overcloud-full-centos-7/41e677b/ : SUCCESS in 27m 29s\n- openstack-tox-pep8 http://logs.openstack.org/40/668940/1/check/openstack-tox-pep8/a179fb5/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/40/668940/1/check/openstack-tox-py27/0cb7d84/ : SUCCESS in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/40/668940/1/check/openstack-tox-py35/58f9cdc/ : SUCCESS in 3m 16s\n- openstackclient-check-plugins http://logs.openstack.org/40/668940/1/check/openstackclient-check-plugins/358736c/ : SUCCESS in 2m 58s (non-voting)\n- build-openstack-sphinx-docs http://logs.openstack.org/40/668940/1/check/build-openstack-sphinx-docs/595a9e4/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ca6a3ccf1205b6d4372ffcec2f58f7afe05d2325","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 14:06:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1b5744e4127e11577cd8f394ce32624e88eaf89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 14:06:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"680ae5f9013087b260b73d6d4b3255400aade649","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 16:09:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"05903c1153f4ea9a49fd84cfc4c1a70c69880da1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 16:09:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/40/668940/1/gate/tripleo-ci-centos-7-undercloud-oooq/5dc14a3/ : SUCCESS in 1h 07m 49s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/40/668940/1/gate/tripleo-ci-centos-7-containers-multinode/9e0695e/ : SUCCESS in 2h 01m 35s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/40/668940/1/gate/tripleo-buildimage-overcloud-full-centos-7/6f66f3e/ : SUCCESS in 26m 10s\n- openstack-tox-pep8 http://logs.openstack.org/40/668940/1/gate/openstack-tox-pep8/b7f8440/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/40/668940/1/gate/openstack-tox-py27/dbe426b/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/40/668940/1/gate/openstack-tox-py35/8f5b7fd/ : SUCCESS in 5m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/40/668940/1/gate/build-openstack-sphinx-docs/b05f100/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e5ec97754eeb152fb92e5a27904bb8eb1378db4","revisions":{"9e5ec97754eeb152fb92e5a27904bb8eb1378db4":{"kind":"REWORK","_number":1,"created":"2019-07-03 15:38:42.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/40/668940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/40/668940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/40/668940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/40/668940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/40/668940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/40/668940/1"}}},"commit":{"parents":[{"commit":"e717083e2af9d73badef959977ea7608467458eb","subject":"Retry heat stack polling on gateway problems","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e717083e2af9d73badef959977ea7608467458eb"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-06-26 09:24:14.000000000","tz":120},"committer":{"name":"Cédric Jeanneret (Tengu)","email":"cjeanner@redhat.com","date":"2019-07-03 15:38:42.000000000","tz":0},"subject":"Ensure files exist before building/uploading plan","message":"Ensure files exist before building/uploading plan\n\nUntil now, only environment files were tested for their eXistenZ. This\npatch corrects this issue, allowing to get a graceful failure before\nplan is compiled and uploaded.\n\nThe following files are now checked, and deploy will gracefully fail in\ncase of non-existing file:\n- plan-environment-file\n- answers-file\n- networks-file\n\nNo Bioport nor Pod was harmed for this patch.\n\nChange-Id: Ifd4229b68571fd6b91574675c1f04d7b5c755eab\nCloses-Bug: #1834307\n(cherry picked from commit bd354f1fc8ae7a36898a22e01e52815d85d8378d)\n(cherry picked from commit f78809a3e3122f268bcc8fa1a9c12974735654b9)\n(cherry picked from commit cb976f672959f3a14360390d6bc8b6ba8e873fa2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9e5ec97754eeb152fb92e5a27904bb8eb1378db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9e5ec97754eeb152fb92e5a27904bb8eb1378db4"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
