)]}'
{"id":"openstack%2Fpython-tripleoclient~670295","triplet_id":"openstack%2Fpython-tripleoclient~master~I77f1a4efb5c60ffb5d7363d0343101fb6e702292","project":"openstack/python-tripleoclient","branch":"master","topic":"ansible/switch/option","hashtags":[],"change_id":"I77f1a4efb5c60ffb5d7363d0343101fb6e702292","subject":"Make mistral optional for Ansible Update and Upgrade","status":"MERGED","created":"2019-07-11 12:53:45.000000000","updated":"2019-08-24 02:42:27.000000000","submitted":"2019-08-24 02:40:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670295-1566614453138-1b5df26a","meta_rev_id":"cab462f5141c12eff9b6be08b8876124100d2628","_number":670295,"virtual_id_number":670295,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-08-24 02:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-19 14:16:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2019-08-16 17:47:01.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-08-23 21:51:10.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-07-24 06:25:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-08-23 21:51:10.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 12:54:26.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-07-16 13:52:38.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-07-19 14:16:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-24 06:25:37.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-08-16 17:47:01.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-23 21:51:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-24 02:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3409b06a3f5d5f44b47e7c80a6152ca4573ea83","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-07-11 12:53:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ad3ec921937f33c5319b948f090d49e4ef2bfbd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 13:14:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/95/670295/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9c5c818/ : FAILURE in 18m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/95/670295/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/28a1c35/ : FAILURE in 18m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7f5929b750c2e86003df0e02b6b66935f028730c","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-07-11 13:22:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97026ab989a9cd4e885f79d9f1d662d183570491","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:14:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/95/670295/2/check/openstack-tox-lower-constraints/f5a3188/ : SUCCESS in 4m 56s\n- openstackclient-check-plugins http://logs.openstack.org/95/670295/2/check/openstackclient-check-plugins/6fdfbd5/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/670295/2/check/openstack-tox-pep8/782bcff/ : FAILURE in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/95/670295/2/check/openstack-tox-py27/2183e10/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/95/670295/2/check/openstack-tox-py35/2e59d54/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/95/670295/2/check/openstack-tox-py36/cd14e89/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/95/670295/2/check/openstack-tox-py37/15112ef/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/95/670295/2/check/openstack-tox-docs/b0d6a7e/html/ : SUCCESS in 3m 30s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/95/670295/2/check/tripleo-buildimage-overcloud-full-centos-7/92c9574/ : SUCCESS in 20m 06s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/670295/2/check/tripleo-ci-centos-7-undercloud-containers/8dd9dec/ : SUCCESS in 1h 26m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/670295/2/check/tripleo-ci-centos-7-standalone/cb4e799/ : SUCCESS in 1h 56m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/670295/2/check/tripleo-ci-fedora-28-standalone/f732e18/ : SUCCESS in 1h 53m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18efc3104bc4ee6c29d9de50cc609ff0cf0521c4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 16:25:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/95/670295/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/355ffb3/ : SUCCESS in 3h 00m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/95/670295/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5682f9b/ : FAILURE in 2h 37m 49s","accounts_in_message":[],"_revision_number":2},{"id":"04c9b512a380116cd13f08b57e338f1c3af6e140","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-07-12 07:18:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bcc600423f8a8acd9a378a35e048d60a7a5c83c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 10:20:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/95/670295/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3153830/ : SUCCESS in 2h 35m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/95/670295/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/550e5c1/ : SUCCESS in 2h 59m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3108b53ad30fb689c3bd242656535660cbdf05aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 10:36:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/95/670295/3/check/openstack-tox-lower-constraints/80e6706/ : SUCCESS in 5m 42s\n- openstackclient-check-plugins http://logs.openstack.org/95/670295/3/check/openstackclient-check-plugins/7dd3d15/ : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/670295/3/check/openstack-tox-pep8/aad68f7/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/95/670295/3/check/openstack-tox-py27/d9959ed/ : SUCCESS in 6m 53s\n- openstack-tox-py35 http://logs.openstack.org/95/670295/3/check/openstack-tox-py35/4d12369/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/95/670295/3/check/openstack-tox-py36/c4a1621/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/95/670295/3/check/openstack-tox-py37/f1530b8/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/95/670295/3/check/openstack-tox-docs/52e44e1/html/ : SUCCESS in 3m 10s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/95/670295/3/check/tripleo-buildimage-overcloud-full-centos-7/bc1b57f/ : SUCCESS in 23m 58s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-containers-multinode/3c5b4e2/ : SUCCESS in 2h 01m 24s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-undercloud-containers/0f7997f/ : SUCCESS in 1h 35m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2e21941/ : SUCCESS in 2h 11m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6877837/ : FAILURE in 2h 53m 32s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-standalone/185bc86/ : SUCCESS in 1h 51m 19s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/670295/3/check/tripleo-ci-fedora-28-standalone/2e229b7/ : SUCCESS in 1h 39m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-standalone-upgrade/a444d8a/ : FAILURE in 1h 47m 48s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/981c646/ : FAILURE in 2h 08m 33s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/df91e5c/ : SUCCESS in 2h 11m 57s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/95/670295/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/9d4c1ed/ : SUCCESS in 2h 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21bc268f58e3484b8ed289e080793bb5b44721e5","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-12 12:54:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"052a231e2e551b3620c85da703039bc21862f8b9","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-15 08:31:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ad8c6e59836c401bce3ebfabddae8eb439801759","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-07-16 14:31:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b63ac09782ffcdd9a1bd4bfa53bad50088061a87","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 15:57:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/95/670295/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b6795c1/ : FAILURE in 1h 22m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/95/670295/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/43c70ef/ : FAILURE in 1h 14m 02s","accounts_in_message":[],"_revision_number":4},{"id":"28fe76318964b4e910a7c7e0424594f54e7e0aa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:17:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/95/670295/4/check/openstack-tox-lower-constraints/71a33a3/ : SUCCESS in 5m 34s\n- openstackclient-check-plugins http://logs.openstack.org/95/670295/4/check/openstackclient-check-plugins/16c8dc6/ : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/670295/4/check/openstack-tox-pep8/90f696f/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/95/670295/4/check/openstack-tox-py27/dd2e8fc/ : SUCCESS in 4m 56s\n- openstack-tox-py35 http://logs.openstack.org/95/670295/4/check/openstack-tox-py35/6a86a4f/ : SUCCESS in 17m 59s\n- openstack-tox-py36 http://logs.openstack.org/95/670295/4/check/openstack-tox-py36/3d9767e/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/95/670295/4/check/openstack-tox-py37/3a71a4a/ : SUCCESS in 7m 12s\n- openstack-tox-docs http://logs.openstack.org/95/670295/4/check/openstack-tox-docs/3aa5b7b/html/ : SUCCESS in 3m 23s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/95/670295/4/check/tripleo-buildimage-overcloud-full-centos-7/8665011/ : SUCCESS in 26m 55s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-containers-multinode/04a3934/ : SUCCESS in 2h 36m 59s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-undercloud-containers/175fbc0/ : SUCCESS in 1h 32m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4d5d3af/ : SUCCESS in 2h 46m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ed3bb83/ : FAILURE in 2h 50m 32s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-standalone/b0ebbdf/ : SUCCESS in 1h 45m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/670295/4/check/tripleo-ci-fedora-28-standalone/b7ef93d/ : SUCCESS in 1h 41m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-standalone-upgrade/eb65b26/ : FAILURE in 2h 02m 12s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ae043e3/ : FAILURE in 1h 55m 44s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/a8eda70/ : SUCCESS in 2h 18m 32s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/95/670295/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/7bb987d/ : SUCCESS in 2h 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"baa192a41331408573d1750c9792e711905af8fe","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2019-07-17 09:30:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"800511c3aa547f013039e62bd80c15f58ac2c763","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:09:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/95/670295/5/check/openstack-tox-lower-constraints/8ef5e72/ : SUCCESS in 5m 06s\n- openstackclient-check-plugins http://logs.openstack.org/95/670295/5/check/openstackclient-check-plugins/4d80d2c/ : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/670295/5/check/openstack-tox-pep8/44ce935/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/95/670295/5/check/openstack-tox-py27/f9ac3c4/ : SUCCESS in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/95/670295/5/check/openstack-tox-py35/9b1ad54/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/95/670295/5/check/openstack-tox-py36/4abca0d/ : SUCCESS in 4m 45s\n- openstack-tox-py37 http://logs.openstack.org/95/670295/5/check/openstack-tox-py37/be738f0/ : SUCCESS in 4m 06s\n- openstack-tox-docs http://logs.openstack.org/95/670295/5/check/openstack-tox-docs/d1629d8/html/ : SUCCESS in 3m 22s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/95/670295/5/check/tripleo-buildimage-overcloud-full-centos-7/8047a28/ : SUCCESS in 22m 11s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-containers-multinode/49e8b5c/ : SUCCESS in 2h 40m 11s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-undercloud-containers/8cfa6e3/ : SUCCESS in 1h 49m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/be1344b/ : SUCCESS in 2h 17m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4f37487/ : FAILURE in 2h 29m 53s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-standalone/746f8d9/ : SUCCESS in 2h 01m 18s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/670295/5/check/tripleo-ci-fedora-28-standalone/9a0a15c/ : SUCCESS in 1h 41m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-standalone-upgrade/419b5e8/ : FAILURE in 2h 10m 32s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d609c92/ : FAILURE in 2h 22m 42s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/97b0627/ : SUCCESS in 2h 22m 18s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/95/670295/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/ba7cb7e/ : SUCCESS in 2h 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a83cf929ecc17668de49812a3c72f458a330b571","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 14:16:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/a204abf9ff4e44508386b03f28cf9cc4 : POST_FAILURE in 3h 47m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/95/670295/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2eff166/ : RETRY_LIMIT in 3m 59s","accounts_in_message":[],"_revision_number":5},{"id":"1bb7e365aa73f35e15c05c3c4506a01364405caa","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-24 06:25:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"691b33c5df02dd031599656ea15747a39e0691ef","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-16 17:47:01.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b02ea647aaa76a2781a1c539d594de9375a2b068","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-23 21:51:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c7e86c6ffb7b0fdf7f3d30a677bb75bf220fd959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:51:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9b286f9774718cba066c0b701ff80f4b7e7e504d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 02:40:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7978a8a14a5849c184c483c50511fe6f : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835af6d21fd846338f12c456adad4b13 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff2e5c01555549cc8d2dbf40082cf95d : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71cabc9a656344768e6eb6e0c6f58fdd : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/523958e90a5146ddb820b146b6768539 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998e35fd82e44abebfbf6bc32bc05be0 : SUCCESS in 5m 04s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/82ab9ecddec04148b068fee5fee3112f : SUCCESS in 20m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fa390903b38d4128bcbffb22117edfc6 : SUCCESS in 2h 42m 32s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e7381015c35c4a2e8c12201d3900b6b4 : SUCCESS in 1h 46m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b1a60f17d8584b858d9d4298bb3501c4 : SUCCESS in 2h 27m 19s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b802e1bcc584662921f5c5d2046fc61 : SUCCESS in 1h 35m 30s","accounts_in_message":[],"_revision_number":5},{"id":"8004d0a1dc80ebb78280fc48a50842b4105d9408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 02:40:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"6dc2c3f09895fcf5f8bf1c83d781004b361daa26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 02:42:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7147fea9982a417d97bcf94cf01ec262 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"92a2fda8e697b8c754bc13d1558058fa61fe3211","revisions":{"7607be023a9194438c9c84ccc7b2d8a181a19db6":{"kind":"REWORK","_number":1,"created":"2019-07-11 12:53:45.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/95/670295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/95/670295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/1"}}},"commit":{"parents":[{"commit":"cd42f4476983c4eab4e9822b671902a7c261ce30","subject":"Merge \"Allow running validations by group using Ansible by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cd42f4476983c4eab4e9822b671902a7c261ce30"}]}],"author":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-11 12:53:14.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-11 12:53:14.000000000","tz":120},"subject":"Make mistral optional for Ansible Update and Upgrade","message":"Make mistral optional for Ansible Update and Upgrade\n\nChange-Id: I77f1a4efb5c60ffb5d7363d0343101fb6e702292\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7607be023a9194438c9c84ccc7b2d8a181a19db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7607be023a9194438c9c84ccc7b2d8a181a19db6"}]},"branch":"refs/heads/master"},"08964e5789e70d03a7abcd182c427184dd980c84":{"kind":"REWORK","_number":2,"created":"2019-07-11 13:22:44.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/95/670295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/95/670295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/2"}}},"commit":{"parents":[{"commit":"cd42f4476983c4eab4e9822b671902a7c261ce30","subject":"Merge \"Allow running validations by group using Ansible by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cd42f4476983c4eab4e9822b671902a7c261ce30"}]}],"author":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-11 12:53:14.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-11 13:22:32.000000000","tz":120},"subject":"Make mistral optional for Ansible Update and Upgrade","message":"Make mistral optional for Ansible Update and Upgrade\n\nChange-Id: I77f1a4efb5c60ffb5d7363d0343101fb6e702292\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/08964e5789e70d03a7abcd182c427184dd980c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/08964e5789e70d03a7abcd182c427184dd980c84"}]},"branch":"refs/heads/master"},"f1c1a850fb78e5f729b3efb822bf4eeb559a0a44":{"kind":"REWORK","_number":3,"created":"2019-07-12 07:18:42.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/95/670295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/95/670295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/3"}}},"commit":{"parents":[{"commit":"cd42f4476983c4eab4e9822b671902a7c261ce30","subject":"Merge \"Allow running validations by group using Ansible by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cd42f4476983c4eab4e9822b671902a7c261ce30"}]}],"author":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-11 12:53:14.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-12 07:18:30.000000000","tz":120},"subject":"Make mistral optional for Ansible Update and Upgrade","message":"Make mistral optional for Ansible Update and Upgrade\n\nChange-Id: I77f1a4efb5c60ffb5d7363d0343101fb6e702292\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f1c1a850fb78e5f729b3efb822bf4eeb559a0a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f1c1a850fb78e5f729b3efb822bf4eeb559a0a44"}]},"branch":"refs/heads/master"},"706632ff8b02de64fd5b17c24acb36f27ceae191":{"kind":"REWORK","_number":4,"created":"2019-07-16 14:31:22.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/95/670295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/95/670295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/4"}}},"commit":{"parents":[{"commit":"cd42f4476983c4eab4e9822b671902a7c261ce30","subject":"Merge \"Allow running validations by group using Ansible by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cd42f4476983c4eab4e9822b671902a7c261ce30"}]}],"author":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-11 12:53:14.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-16 14:29:54.000000000","tz":120},"subject":"Make mistral optional for Ansible Update and Upgrade","message":"Make mistral optional for Ansible Update and Upgrade\n\nAdd option: --no-workflow to enable running ansible\nupdate and upgrade outside of mistral.\n\nDepends-On: I0c106cf795a6830a87790f1075889a6a4db65d22\nChange-Id: I77f1a4efb5c60ffb5d7363d0343101fb6e702292\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/706632ff8b02de64fd5b17c24acb36f27ceae191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/706632ff8b02de64fd5b17c24acb36f27ceae191"}]},"branch":"refs/heads/master"},"92a2fda8e697b8c754bc13d1558058fa61fe3211":{"kind":"REWORK","_number":5,"created":"2019-07-17 09:30:22.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/95/670295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/95/670295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/95/670295/5"}}},"commit":{"parents":[{"commit":"cd42f4476983c4eab4e9822b671902a7c261ce30","subject":"Merge \"Allow running validations by group using Ansible by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cd42f4476983c4eab4e9822b671902a7c261ce30"}]}],"author":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-11 12:53:14.000000000","tz":120},"committer":{"name":"Mathieu Bultel","email":"mbultel@redhat.com","date":"2019-07-17 09:30:08.000000000","tz":120},"subject":"Make mistral optional for Ansible Update and Upgrade","message":"Make mistral optional for Ansible Update and Upgrade\n\nAdd option: --no-workflow to enable running ansible\nupdate and upgrade outside of mistral.\n\nDepends-On: I0c106cf795a6830a87790f1075889a6a4db65d22\nChange-Id: I77f1a4efb5c60ffb5d7363d0343101fb6e702292\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/92a2fda8e697b8c754bc13d1558058fa61fe3211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/92a2fda8e697b8c754bc13d1558058fa61fe3211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
