)]}'
{"id":"openstack%2Fpython-tripleoclient~670821","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fstein~I4a80a339b2cc0c3fca992cfc456581a91dbf2eef","project":"openstack/python-tripleoclient","branch":"stable/stein","topic":"bug/1836379-stable/stein","hashtags":[],"change_id":"I4a80a339b2cc0c3fca992cfc456581a91dbf2eef","subject":"Ensure files are closed, ensure no ResourceWarning on console","status":"MERGED","created":"2019-07-15 12:05:43.000000000","updated":"2019-07-19 01:06:47.000000000","submitted":"2019-07-19 01:06:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670821-1563498407634-22ab38c7","meta_rev_id":"c8e09e0bc3db3ba0c47b75aa90d55135cd54e83a","_number":670821,"virtual_id_number":670821,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"date":"2019-07-18 13:45:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-19 01:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-18 22:21:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-07-16 11:43:39.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-16 12:01:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 11:43:39.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2019-07-18 13:45:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-18 22:21:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-19 01:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52b40f4860a5a853d034a14d4dbeab302861fcd1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-15 12:05:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"406ac7e44ad71d9d1b958fc2745672b4807be300","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 12:25:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/78414c7/ : FAILURE in 18m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/31729d3/ : FAILURE in 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9e1c7706574e09e7a41ba4e5eafae6641926fb04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 13:04:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/1/check/openstack-tox-lower-constraints/5c9f184/ : FAILURE in 4m 56s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/1/check/openstackclient-check-plugins/b709fde/ : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/1/check/openstack-tox-pep8/6f6e504/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/1/check/openstack-tox-py27/a816bb6/ : FAILURE in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/1/check/openstack-tox-py35/f2f0d54/ : FAILURE in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/1/check/openstack-tox-py36/db7e783/ : FAILURE in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/1/check/openstack-tox-py37/2d8bf46/ : FAILURE in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/21/670821/1/check/openstack-tox-docs/2c57513/html/ : SUCCESS in 4m 14s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/1/check/tripleo-buildimage-overcloud-full-centos-7/5e4bb50/ : SUCCESS in 21m 57s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/1/check/tripleo-ci-centos-7-undercloud-containers/cc206c0/ : FAILURE in 24m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/1/check/tripleo-ci-centos-7-standalone/aeb2980/ : FAILURE in 24m 26s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/1/check/tripleo-ci-fedora-28-standalone/3504e04/ : FAILURE in 31m 05s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"400a1307a46afa3cbb0178d25c58eb1e8a99b09c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-15 13:13:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d862d8b9167d30932404151e211c0c74ef7c5088","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 13:34:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/380e710/ : FAILURE in 18m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bd48cdb/ : FAILURE in 18m 24s","accounts_in_message":[],"_revision_number":2},{"id":"50f59658d380b810429f631e928ff542f6704456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:56:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/2/check/openstack-tox-lower-constraints/64e2661/ : FAILURE in 5m 18s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/2/check/openstackclient-check-plugins/0418337/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/2/check/openstack-tox-pep8/c9d94c2/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/2/check/openstack-tox-py27/526f85d/ : FAILURE in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/2/check/openstack-tox-py35/440cd66/ : FAILURE in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/2/check/openstack-tox-py36/6f9eab9/ : FAILURE in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/2/check/openstack-tox-py37/17acbec/ : FAILURE in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/21/670821/2/check/openstack-tox-docs/02a4934/html/ : SUCCESS in 3m 20s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/2/check/tripleo-buildimage-overcloud-full-centos-7/122a30b/ : SUCCESS in 24m 01s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/2/check/tripleo-ci-centos-7-undercloud-containers/98bbee3/ : FAILURE in 25m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/2/check/tripleo-ci-centos-7-standalone/f6563dd/ : FAILURE in 24m 57s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/2/check/tripleo-ci-fedora-28-standalone/44d6e26/ : FAILURE in 34m 07s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container tripleo-ci-centos-7-scenario008-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a46b0f8cc1de61b91beb9474338d22a7aae7eb83","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-15 14:57:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae47dce8adcc3fce68102e3a1c6c209d3ac3ee11","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 17:44:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/041f331/ : FAILURE in 1h 11m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ea03bcc/ : SUCCESS in 2h 45m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b785e20437574472ac0eade1255894b1227267a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:20:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/check/openstack-tox-lower-constraints/186bcf8/ : SUCCESS in 4m 40s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/3/check/openstackclient-check-plugins/bb5c9e3/ : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/check/openstack-tox-pep8/a779807/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/check/openstack-tox-py27/2bccce9/ : SUCCESS in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/check/openstack-tox-py35/7cf6185/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/check/openstack-tox-py36/084ab32/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/check/openstack-tox-py37/d19a3d2/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/check/openstack-tox-docs/69d090c/html/ : SUCCESS in 3m 57s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/check/tripleo-buildimage-overcloud-full-centos-7/4c64ef7/ : SUCCESS in 26m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containers-multinode/c89d6a9/ : SUCCESS in 2h 19m 24s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-undercloud-containers/b4a8614/ : SUCCESS in 1h 46m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/075dcfc/ : SUCCESS in 2h 39m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/daa1235/ : SUCCESS in 3h 01m 53s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-standalone/8a5b361/ : SUCCESS in 2h 09m 58s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-fedora-28-standalone/646ffb1/ : SUCCESS in 2h 02m 49s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/410c810/ : FAILURE in 1h 55m 42s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/59a280f/ : SUCCESS in 2h 29m 01s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/2e1844d/ : FAILURE in 1h 32m 29s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/fbb7f3c/ : SUCCESS in 2h 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af54efe7425a7680a637d2454103680bc3a20f40","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-07-16 11:43:39.000000000","message":"Patch Set 3: Code-Review+2\n\ncheck-rdo","accounts_in_message":[],"_revision_number":3},{"id":"8dce4a566f9876b9d07dac90d852e115763fa647","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-16 12:01:40.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ddededca9995224d1789c7d9720bd68733e1e74d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 12:01:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8316b5cf5c284fbe19c2c4e3dd3c9b6bcffeb69a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 14:27:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/41c8975/ : SUCCESS in 2h 30m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/872f51e/ : SUCCESS in 2h 42m 17s","accounts_in_message":[],"_revision_number":3},{"id":"9d42e6560c3b1b87e86628a5f4caf3d86cc086f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 14:53:51.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/gate/openstack-tox-lower-constraints/4f0e298/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/gate/openstack-tox-pep8/2ea6a41/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py27/9a06947/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py35/ec09db8/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py36/d6573c1/ : SUCCESS in 3m 01s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py37/5e046eb/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/gate/openstack-tox-docs/8158dc3/html/ : SUCCESS in 3m 21s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/gate/tripleo-buildimage-overcloud-full-centos-7/645e1bc/ : SUCCESS in 25m 18s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-containers-multinode/0c3a96f/ : FAILURE in 2h 38m 39s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-standalone/0fb0035/ : SUCCESS in 2h 00m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ec84719/ : FAILURE in 1h 31m 00s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-undercloud-containers/1596de8/ : SUCCESS in 1h 51m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4f6d5ca86a524e605bca8c3f6b76f259d996a452","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-16 17:47:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"727a3920678a23ed3e3b71245b0fa09742168e21","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 20:24:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/97b76a2/ : SUCCESS in 2h 30m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5a19961/ : SUCCESS in 2h 34m 50s","accounts_in_message":[],"_revision_number":3},{"id":"7baf03639a3db52baf06dc293644c4f5d611a4c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 00:58:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/check/openstack-tox-lower-constraints/1c20ef3/ : SUCCESS in 5m 06s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/3/check/openstackclient-check-plugins/a226e62/ : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/check/openstack-tox-pep8/6784b93/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/check/openstack-tox-py27/ad7be2a/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/check/openstack-tox-py35/9820178/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/check/openstack-tox-py36/d99fd4c/ : SUCCESS in 3m 19s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/check/openstack-tox-py37/4abae60/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/check/openstack-tox-docs/8176cce/html/ : SUCCESS in 3m 14s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/check/tripleo-buildimage-overcloud-full-centos-7/1c77a69/ : SUCCESS in 25m 29s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containers-multinode/b63a84a/ : FAILURE in 2h 15m 43s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-undercloud-containers/8ff2b0b/ : SUCCESS in 1h 37m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2b3a634/ : SUCCESS in 2h 23m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d6706f0/ : SUCCESS in 2h 36m 25s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-standalone/c4767f5/ : TIMED_OUT in 3h 16m 33s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-fedora-28-standalone/2938ad6/ : SUCCESS in 1h 52m 24s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/05d1076/ : FAILURE in 1h 29m 46s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/2a611be/ : SUCCESS in 2h 40m 07s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/fe5b11e/ : FAILURE in 1h 20m 28s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/cbe9e05/ : FAILURE in 2h 01m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0eb1122478d57e7fe2c6f3ba054d5657043528d7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-17 13:37:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"be301df03fe1108c85f9a8d5e62480917cc1243a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 16:19:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f802d37/ : SUCCESS in 2h 37m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/48228c7/ : SUCCESS in 2h 40m 06s","accounts_in_message":[],"_revision_number":3},{"id":"db1f0bcc292d9b0a298e24b624e0e9c7fb007e53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 19:11:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/check/openstack-tox-lower-constraints/f9230bb/ : SUCCESS in 5m 06s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/3/check/openstackclient-check-plugins/19b117f/ : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/check/openstack-tox-pep8/75513fe/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/check/openstack-tox-py27/271af8b/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/check/openstack-tox-py35/2db4db5/ : SUCCESS in 6m 46s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/check/openstack-tox-py36/6a6a0a5/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/check/openstack-tox-py37/bd54285/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/check/openstack-tox-docs/382ad06/html/ : SUCCESS in 3m 07s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/check/tripleo-buildimage-overcloud-full-centos-7/9c3cb45/ : SUCCESS in 23m 19s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containers-multinode/7a4f191/ : FAILURE in 2h 26m 23s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-undercloud-containers/6005309/ : SUCCESS in 1h 44m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d467e2c/ : SUCCESS in 2h 34m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d6c6dac/ : TIMED_OUT in 3h 23m 04s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-standalone/36c00be/ : FAILURE in 1h 09m 29s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-fedora-28-standalone/4ffb03e/ : SUCCESS in 1h 51m 49s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/af2fbc2/ : FAILURE in 1h 36m 39s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8463414/ : SUCCESS in 2h 37m 24s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/cc64f4e/ : FAILURE in 1h 33m 31s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/5b1c918/ : FAILURE in 2h 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e308a6ba7533fb16d666ade6dd9e200c3485f2a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-18 01:59:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3df535fe4998c579cce73745ba3d8493fb8953b8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 04:30:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0454a1e/ : RETRY_LIMIT in 6m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3ade6c8/ : RETRY_LIMIT in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"454709f32d23ab61ed209025829b3b476fa0cde7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 08:11:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/check/openstack-tox-lower-constraints/41941c6/ : SUCCESS in 4m 50s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/3/check/openstackclient-check-plugins/6fd0cf7/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/check/openstack-tox-pep8/3764c93/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/check/openstack-tox-py27/2f1e67e/ : SUCCESS in 5m 58s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/check/openstack-tox-py35/18c535a/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/check/openstack-tox-py36/9856fdb/ : SUCCESS in 3m 18s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/check/openstack-tox-py37/53217d2/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/check/openstack-tox-docs/d861eb5/html/ : SUCCESS in 2m 54s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/check/tripleo-buildimage-overcloud-full-centos-7/00cf44d/ : SUCCESS in 21m 34s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containers-multinode/881964e/ : FAILURE in 2h 16m 18s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-undercloud-containers/69788f4/ : SUCCESS in 1h 57m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b98ad7b/ : SUCCESS in 2h 26m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0e608fa/ : SUCCESS in 2h 50m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-standalone/31fa98c/ : SUCCESS in 2h 22m 22s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-fedora-28-standalone/cf1e66d/ : FAILURE in 2h 08m 27s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/622513b/ : FAILURE in 1h 35m 08s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8f41924/ : SUCCESS in 3h 18m 15s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/2c2f13f/ : FAILURE in 1h 32m 50s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6f2dd96/ : FAILURE in 2h 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3093a81ce4e01ece6eabc825a4bca10098c13af8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-18 13:45:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2dfa81d519fd04df0d42ca27cea784aefc16a8c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 17:27:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d302448/ : TIMED_OUT in 3h 40m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d3591c7/ : SUCCESS in 3h 15m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f7a45163d7193e011861cfca7575fcdeea0734bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 18:43:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/check/openstack-tox-lower-constraints/c486291/ : SUCCESS in 4m 50s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/3/check/openstackclient-check-plugins/736eb38/ : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/check/openstack-tox-pep8/2bf3e89/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/check/openstack-tox-py27/b84700b/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/check/openstack-tox-py35/a7b4f17/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/check/openstack-tox-py36/6a7bd33/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/check/openstack-tox-py37/3c4b744/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/check/openstack-tox-docs/3ab0891/html/ : SUCCESS in 3m 34s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/check/tripleo-buildimage-overcloud-full-centos-7/a0ead85/ : SUCCESS in 27m 59s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containers-multinode/f3a5871/ : FAILURE in 42m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-undercloud-containers/ae41126/ : FAILURE in 53m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0f43394/ : FAILURE in 41m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8985ce1/ : FAILURE in 47m 16s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-standalone/8ba8882/ : SUCCESS in 2h 08m 20s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-fedora-28-standalone/44b11cb/ : FAILURE in 52m 08s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c027287/ : FAILURE in 45m 39s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/6e0f772/ : FAILURE in 46m 15s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/772e67c/ : FAILURE in 40m 15s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/6399f58/ : FAILURE in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"99f89153bc6f210ad5131b1fb1f59e5147ebe490","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-18 19:12:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a2ea89f1c3d0c81a95789dab6e8b0e19c51adee1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 22:21:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/183c131/ : SUCCESS in 2h 58m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/21/670821/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d097c47/ : SUCCESS in 3h 07m 13s","accounts_in_message":[],"_revision_number":3},{"id":"66d9c56df48b35229a9abde5fb9e79497e95669d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:33:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/check/openstack-tox-lower-constraints/511454f/ : SUCCESS in 5m 26s\n- openstackclient-check-plugins http://logs.openstack.org/21/670821/3/check/openstackclient-check-plugins/c1a1baa/ : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/check/openstack-tox-pep8/c3bf1e3/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/check/openstack-tox-py27/3ffb873/ : SUCCESS in 6m 12s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/check/openstack-tox-py35/58401b5/ : SUCCESS in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/check/openstack-tox-py36/c3370f2/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/check/openstack-tox-py37/d287a62/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/check/openstack-tox-docs/67dab42/html/ : SUCCESS in 3m 37s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/check/tripleo-buildimage-overcloud-full-centos-7/029257e/ : SUCCESS in 22m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containers-multinode/1aea2ee/ : SUCCESS in 2h 33m 30s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-undercloud-containers/80aff59/ : SUCCESS in 1h 46m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0875a92/ : SUCCESS in 2h 40m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3ee12e5/ : SUCCESS in 2h 59m 23s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-standalone/acd602b/ : SUCCESS in 2h 06m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/21/670821/3/check/tripleo-ci-fedora-28-standalone/ab23b39/ : SUCCESS in 2h 06m 37s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/60c2522/ : FAILURE in 1h 30m 22s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/97de65b/ : SUCCESS in 2h 30m 03s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/ad190ac/ : FAILURE in 1h 37m 01s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/21/670821/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/32bbeae/ : FAILURE in 2h 06m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"10c1113b61a0677849f7e0c469dbe24783163ba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 22:33:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6c9b6e36cf7891df8cd6e5b4591129d80761fff4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:06:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/21/670821/3/gate/openstack-tox-lower-constraints/0e08075/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/21/670821/3/gate/openstack-tox-pep8/fd1fd1d/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py27/3afcbd5/ : SUCCESS in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py35/8785b06/ : SUCCESS in 7m 37s\n- openstack-tox-py36 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py36/8c50375/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/21/670821/3/gate/openstack-tox-py37/d9be0b2/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/21/670821/3/gate/openstack-tox-docs/3c01325/html/ : SUCCESS in 3m 31s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/21/670821/3/gate/tripleo-buildimage-overcloud-full-centos-7/61b6517/ : SUCCESS in 25m 35s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-containers-multinode/7792a81/ : SUCCESS in 2h 21m 22s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-standalone/c7be060/ : SUCCESS in 1h 45m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b9fe905/ : SUCCESS in 2h 27m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/21/670821/3/gate/tripleo-ci-centos-7-undercloud-containers/149f12f/ : SUCCESS in 1h 49m 46s","accounts_in_message":[],"_revision_number":3},{"id":"78f2258e246a3db3c49d55ab1a39698c267760ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 01:06:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bffda0183cd27e8cc859e9e0142534689047ddf7","revisions":{"928f76f5daaed4e7af655c49b50580f9293a28a2":{"kind":"REWORK","_number":1,"created":"2019-07-15 12:05:43.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/21/670821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/21/670821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/1"}}},"commit":{"parents":[{"commit":"3f572d6f8c73fb75021325f60721a6e5d2a15356","subject":"Allow running validations by group using Ansible by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3f572d6f8c73fb75021325f60721a6e5d2a15356"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-12 14:50:01.000000000","tz":120},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-07-15 12:05:43.000000000","tz":0},"subject":"Ensure files are closed, ensure no ResourceWarning on console","message":"Ensure files are closed, ensure no ResourceWarning on console\n\nFixes an issue wich cause warnings such as the one below on the\nconsole when using python3.\n  ResourceWarning : unclosed file \u003c_io.TextIOWrapper name\u003d\u0027FILENAME\u0027\n\nCloses-Bug: #1836379\nChange-Id: I4a80a339b2cc0c3fca992cfc456581a91dbf2eef\n(cherry picked from commit 317f3cff7d00d9031a482e26d9149bca6010677c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/928f76f5daaed4e7af655c49b50580f9293a28a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/928f76f5daaed4e7af655c49b50580f9293a28a2"}]},"branch":"refs/heads/stable/stein"},"6549c13f63ff050548735d8694e536427ff22f0c":{"kind":"NO_CHANGE","_number":2,"created":"2019-07-15 13:13:29.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/21/670821/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/21/670821/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/2"}}},"commit":{"parents":[{"commit":"3f572d6f8c73fb75021325f60721a6e5d2a15356","subject":"Allow running validations by group using Ansible by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3f572d6f8c73fb75021325f60721a6e5d2a15356"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-12 14:50:01.000000000","tz":120},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-15 13:13:29.000000000","tz":0},"subject":"Ensure files are closed, ensure no ResourceWarning on console","message":"Ensure files are closed, ensure no ResourceWarning on console\n\nFixes an issue wich cause warnings such as the one below on the\nconsole when using python3.\n  ResourceWarning : unclosed file \u003c_io.TextIOWrapper name\u003d\u0027FILENAME\u0027\n\nCloses-Bug: #1836379\nChange-Id: I4a80a339b2cc0c3fca992cfc456581a91dbf2eef\n(cherry picked from commit 317f3cff7d00d9031a482e26d9149bca6010677c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6549c13f63ff050548735d8694e536427ff22f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6549c13f63ff050548735d8694e536427ff22f0c"}]},"branch":"refs/heads/stable/stein"},"bffda0183cd27e8cc859e9e0142534689047ddf7":{"kind":"REWORK","_number":3,"created":"2019-07-15 14:57:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/21/670821/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/21/670821/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/21/670821/3"}}},"commit":{"parents":[{"commit":"3f572d6f8c73fb75021325f60721a6e5d2a15356","subject":"Allow running validations by group using Ansible by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3f572d6f8c73fb75021325f60721a6e5d2a15356"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-12 14:50:01.000000000","tz":120},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-15 14:57:44.000000000","tz":-360},"subject":"Ensure files are closed, ensure no ResourceWarning on console","message":"Ensure files are closed, ensure no ResourceWarning on console\n\nFixes an issue wich cause warnings such as the one below on the\nconsole when using python3.\n  ResourceWarning : unclosed file \u003c_io.TextIOWrapper name\u003d\u0027FILENAME\u0027\n\nCloses-Bug: #1836379\nChange-Id: I4a80a339b2cc0c3fca992cfc456581a91dbf2eef\n(cherry picked from commit 317f3cff7d00d9031a482e26d9149bca6010677c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bffda0183cd27e8cc859e9e0142534689047ddf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bffda0183cd27e8cc859e9e0142534689047ddf7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
