)]}'
{"id":"openstack%2Fpython-tripleoclient~671667","triplet_id":"openstack%2Fpython-tripleoclient~master~Ibd6b3a191c6bee5b094de94680bab6df07616deb","project":"openstack/python-tripleoclient","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"Ibd6b3a191c6bee5b094de94680bab6df07616deb","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-07-19 07:39:54.000000000","updated":"2019-08-13 00:41:49.000000000","submitted":"2019-08-13 00:41:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671667-1565656909135-e4a4d76b","meta_rev_id":"4a1393ab44a14c9ca2e988d13b3c838ed1603a7a","_number":671667,"virtual_id_number":671667,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-13 00:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-08-01 13:25:55.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-08-12 22:19:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 22:19:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-13 00:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0348beabe05979e7d3ce2ae9b055b5d6faccc341","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-19 07:39:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1135f45472ac85df6d9cd96b956c2e72bae2b647","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 08:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/67/671667/1/check/requirements-check/e0d0049/ : SUCCESS in 3m 05s\n- openstackclient-check-plugins http://logs.openstack.org/67/671667/1/check/openstackclient-check-plugins/8ef6b09/ : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/671667/1/check/openstack-tox-pep8/75059eb/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/67/671667/1/check/openstack-tox-docs/b9ba18e/html/ : SUCCESS in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/67/671667/1/check/build-openstack-releasenotes/8c09562/html/ : SUCCESS in 3m 30s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/67/671667/1/check/tripleo-buildimage-overcloud-full-centos-7/ca8f479/ : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6567acba0166a679ec4131c2a34cf3bd7b87dd96","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 22:49:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04fd96780a4b7c3cbb168e45962ae09ff502c698","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-01 01:49:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d92ed20b758000f227af6aff0b9f12979b70fca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 02:17:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://logs.opendev.org/67/671667/2/check/requirements-check/02cf23a/ : SUCCESS in 2m 18s\n- openstackclient-check-plugins https://logs.opendev.org/67/671667/2/check/openstackclient-check-plugins/0e38422/ : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/67/671667/2/check/openstack-tox-pep8/208cfe6/ : SUCCESS in 2m 57s\n- openstack-tox-docs https://logs.opendev.org/67/671667/2/check/openstack-tox-docs/3fc7951/html/ : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://logs.opendev.org/67/671667/2/check/build-openstack-releasenotes/4c97ff5/html/ : SUCCESS in 4m 43s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/67/671667/2/check/tripleo-buildimage-overcloud-full-centos-7/bf0adc0/ : SUCCESS in 22m 29s","accounts_in_message":[],"_revision_number":2},{"id":"5806c2b3615ec79d88f5a9c96d86541b4b1c0010","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 13:25:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0739d25a0faadc459737fa80e778b308be0f0e5a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 22:19:07.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f69f1cd3ad01ead21b9b98e97a9749a8b27c9e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:19:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d77f60de7c70cc8eb697de3ce06e04d582fa5748","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 00:41:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/65cc7cdb9c9d486090c283e744f8e34c : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80a0b2064da54d7e9b8ec005779320ef : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e9d0532af2442b9cea52d79ca32b5b : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7714dd9c3614feb8c55e81db06531a1 : SUCCESS in 3m 02s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/fcd713bc93744c6d853d9cb4c0b3a31e : SUCCESS in 24m 19s","accounts_in_message":[],"_revision_number":2},{"id":"643963c9380b7c3f47e1f19663fbc728215c5008","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 00:41:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7de33c13404a0a7ac3e388aa06184f08f7f76413","revisions":{"68c8559e7146a00da30bfc1270556e0258e76bf7":{"kind":"REWORK","_number":1,"created":"2019-07-19 07:39:54.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/67/671667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/671667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/1"}}},"commit":{"parents":[{"commit":"bc56ab81a8ee69b207d1000e38e540f323ae8ade","subject":"Merge \"Implement tripleo container image show command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bc56ab81a8ee69b207d1000e38e540f323ae8ade"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-19 07:39:04.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-19 07:40:03.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ibd6b3a191c6bee5b094de94680bab6df07616deb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/68c8559e7146a00da30bfc1270556e0258e76bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/68c8559e7146a00da30bfc1270556e0258e76bf7"}]},"branch":"refs/heads/master"},"7de33c13404a0a7ac3e388aa06184f08f7f76413":{"kind":"REWORK","_number":2,"created":"2019-08-01 01:49:33.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/67/671667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/67/671667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/67/671667/2"}}},"commit":{"parents":[{"commit":"bc56ab81a8ee69b207d1000e38e540f323ae8ade","subject":"Merge \"Implement tripleo container image show command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bc56ab81a8ee69b207d1000e38e540f323ae8ade"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-19 07:39:04.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-01 01:49:37.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Ibd6b3a191c6bee5b094de94680bab6df07616deb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7de33c13404a0a7ac3e388aa06184f08f7f76413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7de33c13404a0a7ac3e388aa06184f08f7f76413"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
