)]}'
{"id":"openstack%2Fpython-tripleoclient~671983","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fstein~I1151923f3f0de7acf505a0beda3b7488c8ed8dec","project":"openstack/python-tripleoclient","branch":"stable/stein","topic":"bp/validation-framework-stable/stein","hashtags":[],"change_id":"I1151923f3f0de7acf505a0beda3b7488c8ed8dec","subject":"Renaming \u0027Metadata\u0027 column by \u0027Parameters\u0027 when listing validations","status":"MERGED","created":"2019-07-22 08:33:55.000000000","updated":"2019-07-23 17:42:49.000000000","submitted":"2019-07-23 17:42:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671983-1563903769364-a431606b","meta_rev_id":"39bc63c2f5b713502df6fe0d22eb9a3c186ce4f2","_number":671983,"virtual_id_number":671983,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2019-07-23 10:59:46.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-07-23 17:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 13:38:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":2,"date":"2019-07-23 14:20:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-07-22 08:34:20.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-07-22 08:34:17.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-07-22 08:34:08.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":1,"date":"2019-07-23 14:20:25.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-07-22 08:34:14.000000000","_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-22 08:34:03.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 08:34:03.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-22 08:34:08.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2019-07-22 08:34:14.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-07-22 08:34:17.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-07-22 08:34:20.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-23 13:38:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-23 14:20:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-23 17:42:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92db6f683b8776e9cc38bf296f9ccc348b4d5614","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-22 08:33:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dd026e82e7be202ea2faa6c3945030e6f0707692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 11:49:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/671983/1/check/openstack-tox-lower-constraints/c97c5ea/ : SUCCESS in 5m 08s\n- openstackclient-check-plugins http://logs.openstack.org/83/671983/1/check/openstackclient-check-plugins/ffce1c3/ : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/671983/1/check/openstack-tox-pep8/8cbbd68/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/83/671983/1/check/openstack-tox-py27/b31e719/ : SUCCESS in 4m 39s\n- openstack-tox-py35 http://logs.openstack.org/83/671983/1/check/openstack-tox-py35/38697a8/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/83/671983/1/check/openstack-tox-py36/411eb98/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/83/671983/1/check/openstack-tox-py37/4e77d07/ : SUCCESS in 6m 21s\n- openstack-tox-docs http://logs.openstack.org/83/671983/1/check/openstack-tox-docs/c7206cb/html/ : SUCCESS in 4m 19s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/83/671983/1/check/tripleo-buildimage-overcloud-full-centos-7/61d2264/ : SUCCESS in 26m 19s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-containers-multinode/aa47211/ : FAILURE in 2h 37m 49s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-undercloud-containers/0c68525/ : SUCCESS in 1h 41m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ea8faee/ : SUCCESS in 2h 56m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b99f98a/ : SUCCESS in 3h 00m 45s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-standalone/5ecd7f3/ : SUCCESS in 2h 08m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/83/671983/1/check/tripleo-ci-fedora-28-standalone/b865460/ : SUCCESS in 1h 54m 36s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/130f8ea/ : FAILURE in 1h 50m 44s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/1200dec/ : SUCCESS in 2h 56m 40s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/dd81f7a/ : FAILURE in 1h 33m 34s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/8a314b1/ : SUCCESS in 2h 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60e1038c64281e216dd608161182f6bae091f372","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 13:50:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/83/671983/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f134be3/ : RETRY_LIMIT in 11m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/83/671983/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fbbd340/ : FAILURE in 2h 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8bfb100e0c8e673b79e997545e2a0b9b92202a2a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-23 10:59:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdfd5b1de2068a3314b876e00546a61c0c28eb2b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 13:38:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/83/671983/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/de2b4d0/ : SUCCESS in 2h 35m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/83/671983/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a6ee72b/ : SUCCESS in 2h 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"dd70cb68219e6c055cacd2f1724f1e19545a3de2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:10:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/671983/1/check/openstack-tox-lower-constraints/51783d3/ : SUCCESS in 4m 39s\n- openstackclient-check-plugins http://logs.openstack.org/83/671983/1/check/openstackclient-check-plugins/49d7278/ : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/83/671983/1/check/openstack-tox-pep8/2502739/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/83/671983/1/check/openstack-tox-py27/f5f32c8/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/83/671983/1/check/openstack-tox-py35/14d8783/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/83/671983/1/check/openstack-tox-py36/78ce9ba/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/83/671983/1/check/openstack-tox-py37/e2b4163/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/83/671983/1/check/openstack-tox-docs/8aea40e/html/ : SUCCESS in 3m 24s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/83/671983/1/check/tripleo-buildimage-overcloud-full-centos-7/0be1cf4/ : SUCCESS in 28m 40s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-containers-multinode/1a5e772/ : SUCCESS in 2h 09m 13s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-undercloud-containers/491ece2/ : SUCCESS in 1h 34m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/719f575/ : SUCCESS in 2h 39m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/eded365/ : SUCCESS in 2h 29m 18s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-standalone/7631ee4/ : SUCCESS in 2h 00m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/83/671983/1/check/tripleo-ci-fedora-28-standalone/1c21d0c/ : SUCCESS in 1h 43m 50s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/976acef/ : FAILURE in 1h 33m 17s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/52c500e/ : SUCCESS in 2h 22m 59s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/5f83399/ : FAILURE in 1h 27m 46s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/83/671983/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/fcce349/ : FAILURE in 1h 53m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1646240349ce6fdd39530d4ab9894a7c2af354d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-23 14:20:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0df77a082a9ef0bde0e606b2aa9304f4aef60e18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:20:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4eb3a59b35ccc3f2f5d2585da054ee80ce78f81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:42:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/83/671983/1/gate/openstack-tox-lower-constraints/0423ea0/ : SUCCESS in 4m 56s\n- openstack-tox-pep8 http://logs.openstack.org/83/671983/1/gate/openstack-tox-pep8/c7d6a32/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/83/671983/1/gate/openstack-tox-py27/f17374a/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/83/671983/1/gate/openstack-tox-py35/b52625a/ : SUCCESS in 5m 53s\n- openstack-tox-py36 http://logs.openstack.org/83/671983/1/gate/openstack-tox-py36/b3381cf/ : SUCCESS in 5m 18s\n- openstack-tox-py37 http://logs.openstack.org/83/671983/1/gate/openstack-tox-py37/c14cfc6/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/83/671983/1/gate/openstack-tox-docs/99ebc9b/html/ : SUCCESS in 4m 20s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/83/671983/1/gate/tripleo-buildimage-overcloud-full-centos-7/2b209cd/ : SUCCESS in 25m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/83/671983/1/gate/tripleo-ci-centos-7-containers-multinode/dd2334d/ : SUCCESS in 2h 09m 45s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/83/671983/1/gate/tripleo-ci-centos-7-standalone/a953174/ : SUCCESS in 1h 58m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/83/671983/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6d267d7/ : SUCCESS in 3h 17m 25s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/83/671983/1/gate/tripleo-ci-centos-7-undercloud-containers/8a1a29b/ : SUCCESS in 2h 04m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1b5526e283ae86c179f4f8994a95c8f3c41a58db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:42:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f54216dfd59183dc460c2c54b90ea7891ada19a5","revisions":{"f54216dfd59183dc460c2c54b90ea7891ada19a5":{"kind":"REWORK","_number":1,"created":"2019-07-22 08:33:55.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/83/671983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/83/671983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/83/671983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/83/671983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/83/671983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/83/671983/1"}}},"commit":{"parents":[{"commit":"bffda0183cd27e8cc859e9e0142534689047ddf7","subject":"Ensure files are closed, ensure no ResourceWarning on console","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bffda0183cd27e8cc859e9e0142534689047ddf7"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-08 11:04:25.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-22 08:33:55.000000000","tz":0},"subject":"Renaming \u0027Metadata\u0027 column by \u0027Parameters\u0027 when listing validations","message":"Renaming \u0027Metadata\u0027 column by \u0027Parameters\u0027 when listing validations\n\nAll the available Ansible variables for each Validations are now\nprocessed as parameters and not as metadata anymore.\n\nChange-Id: I1151923f3f0de7acf505a0beda3b7488c8ed8dec\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit e0d35b30de42455b098174ce754da323849958cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f54216dfd59183dc460c2c54b90ea7891ada19a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f54216dfd59183dc460c2c54b90ea7891ada19a5"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
