)]}'
{"id":"openstack%2Fpython-tripleoclient~672070","triplet_id":"openstack%2Fpython-tripleoclient~master~Id7fdbf029a6dd1b45e9801c9cf8814a15a157ee0","project":"openstack/python-tripleoclient","branch":"master","topic":"cell_export","hashtags":[],"change_id":"Id7fdbf029a6dd1b45e9801c9cf8814a15a157ee0","subject":"Add overcloud cell export command","status":"MERGED","created":"2019-07-22 15:22:33.000000000","updated":"2019-07-29 10:08:05.000000000","submitted":"2019-07-29 10:08:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672070-1564394885338-a91269ad","meta_rev_id":"a92ec0818ef72e040e0b3814355788a2030feeac","_number":672070,"virtual_id_number":672070,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-27 05:54:19.000000000","_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":2,"date":"2019-07-29 10:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":1,"date":"2019-07-27 08:51:39.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":2,"date":"2019-07-25 19:13:31.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-07-25 21:30:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-26 06:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"date":"2019-07-25 21:21:10.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-07-24 14:36:32.000000000","_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"value":0,"date":"2019-07-25 21:21:22.000000000","_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-07-29 06:54:14.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 14:36:32.000000000","updated_by":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"reviewer":{"_account_id":23811,"name":"Oliver Walsh","email":"owalsh@redhat.com","username":"owalsh"},"state":"REVIEWER"},{"updated":"2019-07-25 19:13:31.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-07-25 21:21:10.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-25 21:21:22.000000000","updated_by":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"reviewer":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"state":"REVIEWER"},{"updated":"2019-07-25 21:30:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-26 06:21:49.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2019-07-27 08:51:39.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-29 06:54:14.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-07-29 10:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f4826a27b7b38197733e706191632c9c1546793","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-22 15:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f95636e5ca83a68875b57e20b94d59feb77659b","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-22 15:28:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb69a84e5cb840045ef82bfe0ab34e67fea35076","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 18:15:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/672070/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6b8e19a/ : SUCCESS in 2h 41m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/672070/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/474d4f5/ : SUCCESS in 2h 44m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ba9efa6436d3846232df9da1fca2d96c6f7bf866","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:25:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/70/672070/2/check/test-release-openstack/a78045a/ : SUCCESS in 2m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/2/check/openstack-tox-lower-constraints/09f2318/ : SUCCESS in 5m 23s\n- openstackclient-check-plugins http://logs.openstack.org/70/672070/2/check/openstackclient-check-plugins/375e694/ : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/2/check/openstack-tox-pep8/b48a9f4/ : FAILURE in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/2/check/openstack-tox-py27/216239f/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/2/check/openstack-tox-py35/53dfa26/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/2/check/openstack-tox-py36/bbd32a4/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/2/check/openstack-tox-py37/f4a0fe6/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/70/672070/2/check/openstack-tox-docs/9020d2e/html/ : SUCCESS in 4m 34s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/2/check/build-openstack-releasenotes/37d90bd/html/ : SUCCESS in 3m 14s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/2/check/tripleo-buildimage-overcloud-full-centos-7/e6f31da/ : SUCCESS in 25m 38s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/2/check/tripleo-build-containers-centos-7/7705b96/ : SUCCESS in 46m 22s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/2/check/tripleo-build-containers-centos-7-buildah/4fb421e/ : SUCCESS in 1h 04m 22s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/2/check/tripleo-ci-centos-7-undercloud-containers/4bfeb2a/ : SUCCESS in 1h 37m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/2/check/tripleo-ci-centos-7-standalone/637c864/ : SUCCESS in 2h 01m 19s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/672070/2/check/tripleo-ci-fedora-28-standalone/498de14/ : FAILURE in 1h 56m 28s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a5c91482b36435279d13c8497eb767eebe98508","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-23 06:55:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3a43b288505cd5b5a8d0b5370f1dd75eee1b220e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 08:47:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/70/672070/3/check/test-release-openstack/00d3277/ : SUCCESS in 2m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/3/check/openstack-tox-lower-constraints/7d9212d/ : SUCCESS in 5m 25s\n- openstackclient-check-plugins http://logs.openstack.org/70/672070/3/check/openstackclient-check-plugins/ec53a6f/ : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/3/check/openstack-tox-pep8/90d1e4f/ : FAILURE in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/3/check/openstack-tox-py27/e0380cf/ : SUCCESS in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/3/check/openstack-tox-py35/a346b45/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/3/check/openstack-tox-py36/39ca68f/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/3/check/openstack-tox-py37/1e754a4/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/70/672070/3/check/openstack-tox-docs/a76ea4d/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/3/check/build-openstack-releasenotes/b936065/html/ : SUCCESS in 3m 22s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/3/check/tripleo-buildimage-overcloud-full-centos-7/742005c/ : SUCCESS in 26m 51s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/3/check/tripleo-build-containers-centos-7/62ed901/ : SUCCESS in 52m 17s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/3/check/tripleo-build-containers-centos-7-buildah/09c809b/ : SUCCESS in 48m 20s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/3/check/tripleo-ci-centos-7-undercloud-containers/f0303f0/ : SUCCESS in 1h 29m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/3/check/tripleo-ci-centos-7-standalone/acafde4/ : SUCCESS in 1h 47m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/672070/3/check/tripleo-ci-fedora-28-standalone/9d5a5f2/ : SUCCESS in 1h 43m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ef6f75c6c7618dcd6afe7984658ff078f0b81d7","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-23 09:16:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a0aea999d059bf83868826f97c199a2c708a5832","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:17:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/70/672070/4/check/test-release-openstack/02f34cd/ : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/4/check/openstack-tox-lower-constraints/0cb704e/ : SUCCESS in 6m 56s\n- openstackclient-check-plugins http://logs.openstack.org/70/672070/4/check/openstackclient-check-plugins/55e1137/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/4/check/openstack-tox-pep8/5ee8988/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/4/check/openstack-tox-py27/13f7c58/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/4/check/openstack-tox-py35/dc08060/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/4/check/openstack-tox-py36/d3cdd20/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/4/check/openstack-tox-py37/33783f5/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/70/672070/4/check/openstack-tox-docs/5b3476b/html/ : SUCCESS in 5m 26s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/4/check/build-openstack-releasenotes/94387ef/html/ : SUCCESS in 4m 27s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/4/check/tripleo-buildimage-overcloud-full-centos-7/bbf16d3/ : SUCCESS in 22m 06s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/4/check/tripleo-build-containers-centos-7/599a3a6/ : SUCCESS in 41m 19s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/4/check/tripleo-build-containers-centos-7-buildah/9649e1b/ : SUCCESS in 49m 33s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-containers-multinode/7cbd305/ : SUCCESS in 2h 28m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-undercloud-containers/d1cb6ce/ : SUCCESS in 1h 48m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3ea8ae4/ : SUCCESS in 2h 16m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/150254e/ : SUCCESS in 2h 27m 50s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-standalone/7f063c9/ : SUCCESS in 2h 06m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/672070/4/check/tripleo-ci-fedora-28-standalone/10d7ea6/ : SUCCESS in 1h 50m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-standalone-upgrade/ab380ab/ : FAILURE in 2h 02m 13s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9ab7099/ : FAILURE in 2h 14m 40s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/966e3ee/ : SUCCESS in 2h 21m 20s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/70/672070/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/1566fc6/ : FAILURE in 1h 25m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"183717129f284913479fb607fb90d5ef3c0f98d9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 12:21:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/672070/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8d8ff0c/ : SUCCESS in 2h 45m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/672070/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/04fb2ac/ : SUCCESS in 3h 03m 17s","accounts_in_message":[],"_revision_number":4},{"id":"5588e50d8479bdc6cba8e8dfc601e16e51849605","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-24 08:01:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"83d9e88f31d323942f6a96ab06494686df8c9531","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:44:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/70/672070/5/check/test-release-openstack/1906023/ : SUCCESS in 2m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/5/check/openstack-tox-lower-constraints/d52e427/ : SUCCESS in 4m 47s\n- openstackclient-check-plugins http://logs.openstack.org/70/672070/5/check/openstackclient-check-plugins/65e5dd2/ : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/5/check/openstack-tox-pep8/2683817/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/5/check/openstack-tox-py27/21fab5a/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/5/check/openstack-tox-py35/1137c29/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/5/check/openstack-tox-py36/7c8d961/ : SUCCESS in 3m 21s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/5/check/openstack-tox-py37/d5b93d3/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/70/672070/5/check/openstack-tox-docs/8ed1ecc/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/5/check/build-openstack-releasenotes/66e159a/html/ : SUCCESS in 3m 59s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-buildimage-overcloud-full-centos-7/51a9133/ : SUCCESS in 26m 39s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7/1f66efc/ : SUCCESS in 42m 02s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7-buildah/895ddf3/ : SUCCESS in 54m 32s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containers-multinode/5ce3a94/ : FAILURE in 2h 16m 02s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-undercloud-containers/110d794/ : SUCCESS in 1h 34m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/deac0b7/ : FAILURE in 1h 50m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/36f6e17/ : FAILURE in 1h 49m 10s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone/702d123/ : FAILURE in 1h 04m 53s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-fedora-28-standalone/97cfd4b/ : SUCCESS in 1h 58m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone-upgrade/5541b6a/ : FAILURE in 2h 08m 49s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/debf706/ : FAILURE in 2h 01m 15s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/95d68df/ : FAILURE in 1h 49m 23s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/d4cb0c8/ : FAILURE in 1h 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff3dc06516200bea2caecad77ccea19dbfb4a3df","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 10:59:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/383238d/ : SUCCESS in 2h 39m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/96cc871/ : SUCCESS in 2h 56m 31s","accounts_in_message":[],"_revision_number":5},{"id":"b75a0b737d20d73be1616e8e4f01ef7f1874fb0d","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-25 11:51:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"01531b67fa3652082ad2adbc9c7e7a49cce6795a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 15:34:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/01edcaf/ : TIMED_OUT in 3h 40m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7236ceb/ : SUCCESS in 2h 50m 30s","accounts_in_message":[],"_revision_number":5},{"id":"82ed53ce49cc67b8e323939eec1847f9b3730ee1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 15:42:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/70/672070/5/check/test-release-openstack/73ec12a/ : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/5/check/openstack-tox-lower-constraints/8cae732/ : SUCCESS in 5m 57s\n- openstackclient-check-plugins http://logs.openstack.org/70/672070/5/check/openstackclient-check-plugins/2232445/ : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/5/check/openstack-tox-pep8/bdd5193/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/5/check/openstack-tox-py27/9081752/ : SUCCESS in 4m 56s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/5/check/openstack-tox-py35/4ce368b/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/5/check/openstack-tox-py36/4e92549/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/5/check/openstack-tox-py37/19f8411/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/70/672070/5/check/openstack-tox-docs/92fd318/html/ : SUCCESS in 4m 06s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/5/check/build-openstack-releasenotes/a64a373/html/ : SUCCESS in 3m 52s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-buildimage-overcloud-full-centos-7/f487bd5/ : SUCCESS in 25m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7/af5c41c/ : SUCCESS in 45m 28s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7-buildah/41819a7/ : SUCCESS in 1h 01m 13s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containers-multinode/1385004/ : SUCCESS in 2h 40m 13s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-undercloud-containers/d02729d/ : SUCCESS in 1h 42m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e4c1017/ : FAILURE in 1h 44m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/619e377/ : FAILURE in 1h 51m 28s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone/9853c00/ : SUCCESS in 1h 51m 38s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-fedora-28-standalone/325d4bd/ : SUCCESS in 1h 53m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone-upgrade/e75c514/ : FAILURE in 1h 51m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/cd64372/ : FAILURE in 2h 09m 03s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/f39ecac/ : FAILURE in 1h 48m 50s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/ea5d4a9/ : FAILURE in 1h 40m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"efcaaf9f5f314cd117c70df25e7c39493b689e3e","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-25 19:13:31.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"03382aa9985ec0d09e7b24667774a31b37b9125a","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-25 21:22:29.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4ea7778dbb03357b13e8b54eb505785c6aab709c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-25 21:30:07.000000000","message":"Patch Set 5: Code-Review+2\n\ntests would be nice","accounts_in_message":[],"_revision_number":5},{"id":"5683a215b12eb6d2337581d2cba1cabd15b099a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-26 00:13:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2f9a6d9/ : SUCCESS in 2h 28m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/537e6de/ : SUCCESS in 2h 49m 26s","accounts_in_message":[],"_revision_number":5},{"id":"2133f37f3b9c1db1e1498036c494cf84c241de9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 00:38:04.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/70/672070/5/check/test-release-openstack/b73f7c6/ : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/5/check/openstack-tox-lower-constraints/22f484a/ : SUCCESS in 6m 22s\n- openstackclient-check-plugins http://logs.openstack.org/70/672070/5/check/openstackclient-check-plugins/0f4fa83/ : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/5/check/openstack-tox-pep8/c77c945/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/5/check/openstack-tox-py27/368b221/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/5/check/openstack-tox-py35/1a47ee3/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/5/check/openstack-tox-py36/83575a4/ : SUCCESS in 3m 48s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/5/check/openstack-tox-py37/ba4921d/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/70/672070/5/check/openstack-tox-docs/defbcc4/html/ : SUCCESS in 5m 29s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/5/check/build-openstack-releasenotes/7df4523/html/ : SUCCESS in 5m 26s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-buildimage-overcloud-full-centos-7/a38c250/ : SUCCESS in 34m 03s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7/ec4fe0f/ : SUCCESS in 51m 30s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7-buildah/9fc07dd/ : SUCCESS in 1h 03m 24s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containers-multinode/e3d688a/ : SUCCESS in 2h 15m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-undercloud-containers/1a26490/ : SUCCESS in 1h 29m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c910ffe/ : FAILURE in 1h 32m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/264e365/ : FAILURE in 1h 34m 01s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone/d4a4c22/ : SUCCESS in 1h 48m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-fedora-28-standalone/2e30772/ : SUCCESS in 2h 07m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone-upgrade/1bb317a/ : FAILURE in 1h 45m 19s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0bd9c5f/ : FAILURE in 2h 59m 56s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/ff29c0e/ : FAILURE in 1h 56m 07s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/1cc0ff4/ : FAILURE in 1h 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"14946f83cc814183b05299d81025a88c53c1954b","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2019-07-26 06:21:49.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"aa66708fec65d0f8e07d87b7a3874861abc6f851","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-26 13:35:41.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e5200914f2bfbcf55ac9ba479bc2e282607ae936","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 13:44:01.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"d6ee7ec5b5a292ea593c8210fa11ab1fc34ba254","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-26 13:44:01.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"b33047f802c01140e69aa98ec0a1d9badb8d2e94","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-07-27 05:54:19.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"78caa03a71396e3e94e5b72838482a5ffd272fd0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-27 08:51:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ac30b67/ : SUCCESS in 2h 42m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/70/672070/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/015e353/ : SUCCESS in 2h 56m 05s","accounts_in_message":[],"_revision_number":5},{"id":"55e9c00d53d1b5c95a78c96e669638f18f434dec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 08:59:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/70/672070/5/check/test-release-openstack/9b43f3f/ : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/5/check/openstack-tox-lower-constraints/4896f5f/ : SUCCESS in 5m 09s\n- openstackclient-check-plugins http://logs.openstack.org/70/672070/5/check/openstackclient-check-plugins/cd03e10/ : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/5/check/openstack-tox-pep8/46471b3/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/5/check/openstack-tox-py27/d706590/ : SUCCESS in 3m 08s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/5/check/openstack-tox-py35/767d2ac/ : SUCCESS in 5m 46s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/5/check/openstack-tox-py36/20594b8/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/5/check/openstack-tox-py37/8f19ee4/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/70/672070/5/check/openstack-tox-docs/a7ebc52/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/5/check/build-openstack-releasenotes/39e0420/html/ : SUCCESS in 3m 03s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-buildimage-overcloud-full-centos-7/1e14942/ : SUCCESS in 26m 22s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7/e245b5e/ : SUCCESS in 50m 56s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/5/check/tripleo-build-containers-centos-7-buildah/a407df1/ : SUCCESS in 57m 23s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containers-multinode/91c55fd/ : SUCCESS in 2h 32m 11s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-undercloud-containers/8d6e0c1/ : SUCCESS in 1h 25m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bbf1a17/ : SUCCESS in 2h 15m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/015fd97/ : SUCCESS in 2h 33m 33s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone/f636e60/ : SUCCESS in 1h 35m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/672070/5/check/tripleo-ci-fedora-28-standalone/db958fc/ : SUCCESS in 1h 38m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-standalone-upgrade/a6b2b2b/ : SUCCESS in 2h 04m 53s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c97c8d8/ : FAILURE in 2h 22m 30s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8c34ac0/ : SUCCESS in 2h 21m 01s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/70/672070/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/3e67b63/ : FAILURE in 1h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3b48074f0e296f80890b1c129b31d5cc2b5a8e66","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-29 06:54:14.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"270440114aca2eb8fe9d57c4e34939128dfedef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 06:54:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"345c6de94f6b92d38e0848323c5f20d33a8782a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 10:08:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ea0e937bb7a25ebd958d7f2ff4a8b74b2d2fc61d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 10:08:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/70/672070/5/gate/test-release-openstack/1c9d648/ : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/672070/5/gate/openstack-tox-lower-constraints/fcd276d/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/70/672070/5/gate/openstack-tox-pep8/1e5e8b4/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/70/672070/5/gate/openstack-tox-py27/1fde4a7/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/70/672070/5/gate/openstack-tox-py35/cc0c3ef/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/70/672070/5/gate/openstack-tox-py36/5755d1e/ : SUCCESS in 5m 56s\n- openstack-tox-py37 http://logs.openstack.org/70/672070/5/gate/openstack-tox-py37/d6b64bd/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/70/672070/5/gate/openstack-tox-docs/8c6be4e/html/ : SUCCESS in 4m 22s\n- build-openstack-releasenotes http://logs.openstack.org/70/672070/5/gate/build-openstack-releasenotes/6c22c8a/html/ : SUCCESS in 3m 38s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/70/672070/5/gate/tripleo-buildimage-overcloud-full-centos-7/bb8ae8a/ : SUCCESS in 23m 18s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/70/672070/5/gate/tripleo-build-containers-centos-7/79783ab/ : SUCCESS in 50m 57s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/70/672070/5/gate/tripleo-build-containers-centos-7-buildah/b06051a/ : SUCCESS in 54m 01s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/672070/5/gate/tripleo-ci-centos-7-containers-multinode/fe9f50f/ : SUCCESS in 2h 35m 23s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/672070/5/gate/tripleo-ci-centos-7-standalone/1f66604/ : SUCCESS in 2h 10m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/672070/5/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/201bc4b/ : SUCCESS in 2h 48m 16s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/672070/5/gate/tripleo-ci-centos-7-undercloud-containers/ef27eb3/ : SUCCESS in 1h 32m 21s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aa0d29fb4413c150af873c7d93b249ea2fc09573","revisions":{"96bd3f26814b7bc527911bad79f00fbfef75c1a1":{"kind":"REWORK","_number":1,"created":"2019-07-22 15:22:33.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/70/672070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/70/672070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/1"}}},"commit":{"parents":[{"commit":"c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0","subject":"Merge \"Implement tripleo container image delete command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-09 11:52:13.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-22 15:22:22.000000000","tz":120},"subject":"Add overcloud cell export command","message":"Add overcloud cell export command\n\n[1] provides the steps on how to setup multiple cells using tripleo.\nThis requires to extract deployment information from the overcloud\n/control plane stack which then is used as input for the cell\ndeployment.\n\nWith this patch we provide a new tripleoclient functionality which\nhelps\nto automate the export steps from [1]:\n* Export the default cell EndpointMap\n* Export the default cell HostsEntry\n* Export AllNodesConfig and GlobalConfig information\n* Export passwords\n\n[1] https://docs.openstack.org/tripleo-docs/latest/install/advanced_deployment/deploy_cellv2.html#deploy-an-additional-nova-cell-v2\n\nChange-Id: Id7fdbf029a6dd1b45e9801c9cf8814a15a157ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/96bd3f26814b7bc527911bad79f00fbfef75c1a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/96bd3f26814b7bc527911bad79f00fbfef75c1a1"}]},"branch":"refs/heads/master"},"39d45feaac4c5622b34cb75cda4ff73b4c99e735":{"kind":"REWORK","_number":2,"created":"2019-07-22 15:28:57.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/70/672070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/70/672070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/2"}}},"commit":{"parents":[{"commit":"c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0","subject":"Merge \"Implement tripleo container image delete command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-09 11:52:13.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-22 15:28:47.000000000","tz":120},"subject":"Add overcloud cell export command","message":"Add overcloud cell export command\n\n[1] provides the steps on how to setup multiple cells using tripleo.\nThis requires to extract deployment information from the overcloud\n/control plane stack which then is used as input for the cell\ndeployment.\n\nWith this patch we provide a new tripleoclient functionality which\nhelps\nto automate the export steps from [1]:\n* Export the default cell EndpointMap\n* Export the default cell HostsEntry\n* Export AllNodesConfig and GlobalConfig information\n* Export passwords\n\n[1] https://docs.openstack.org/tripleo-docs/latest/install/advanced_deployment/deploy_cellv2.html#deploy-an-additional-nova-cell-v2\n\nChange-Id: Id7fdbf029a6dd1b45e9801c9cf8814a15a157ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/39d45feaac4c5622b34cb75cda4ff73b4c99e735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/39d45feaac4c5622b34cb75cda4ff73b4c99e735"}]},"branch":"refs/heads/master"},"4989f22de218022998c35797c046f678eada2e89":{"kind":"REWORK","_number":3,"created":"2019-07-23 06:55:49.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/70/672070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/70/672070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/3"}}},"commit":{"parents":[{"commit":"c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0","subject":"Merge \"Implement tripleo container image delete command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-09 11:52:13.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-23 06:55:36.000000000","tz":120},"subject":"Add overcloud cell export command","message":"Add overcloud cell export command\n\n[1] provides the steps on how to setup multiple cells using tripleo.\nThis requires to extract deployment information from the overcloud\n/control plane stack which then is used as input for the cell\ndeployment.\n\nWith this patch we provide a new tripleoclient functionality which\nhelps\nto automate the export steps from [1]:\n* Export the default cell EndpointMap\n* Export the default cell HostsEntry\n* Export AllNodesConfig and GlobalConfig information\n* Export passwords\n\n[1] https://docs.openstack.org/tripleo-docs/latest/install/advanced_deployment/deploy_cellv2.html#deploy-an-additional-nova-cell-v2\n\nChange-Id: Id7fdbf029a6dd1b45e9801c9cf8814a15a157ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4989f22de218022998c35797c046f678eada2e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4989f22de218022998c35797c046f678eada2e89"}]},"branch":"refs/heads/master"},"3987dbc6776cf5c272de70fec21e259e4be3a97e":{"kind":"REWORK","_number":4,"created":"2019-07-23 09:16:46.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/70/672070/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/70/672070/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/4"}}},"commit":{"parents":[{"commit":"c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0","subject":"Merge \"Implement tripleo container image delete command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-09 11:52:13.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-23 09:16:24.000000000","tz":120},"subject":"Add overcloud cell export command","message":"Add overcloud cell export command\n\n[1] provides the steps on how to setup multiple cells using tripleo.\nThis requires to extract deployment information from the overcloud\n/control plane stack which then is used as input for the cell\ndeployment.\n\nWith this patch we provide a new tripleoclient functionality which\nhelps\nto automate the export steps from [1]:\n* Export the default cell EndpointMap\n* Export the default cell HostsEntry\n* Export AllNodesConfig and GlobalConfig information\n* Export passwords\n\n[1] https://docs.openstack.org/tripleo-docs/latest/install/advanced_deployment/deploy_cellv2.html#deploy-an-additional-nova-cell-v2\n\nChange-Id: Id7fdbf029a6dd1b45e9801c9cf8814a15a157ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3987dbc6776cf5c272de70fec21e259e4be3a97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3987dbc6776cf5c272de70fec21e259e4be3a97e"}]},"branch":"refs/heads/master"},"aa0d29fb4413c150af873c7d93b249ea2fc09573":{"kind":"REWORK","_number":5,"created":"2019-07-24 08:01:05.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/70/672070/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/70/672070/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/70/672070/5"}}},"commit":{"parents":[{"commit":"c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0","subject":"Merge \"Implement tripleo container image delete command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c6d441b24c2f367a6d95772a7d83e4f6fcd8caa0"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-09 11:52:13.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2019-07-24 07:57:40.000000000","tz":120},"subject":"Add overcloud cell export command","message":"Add overcloud cell export command\n\n[1] provides the steps on how to setup multiple cells using tripleo.\nThis requires to extract deployment information from the overcloud\n/control plane stack which then is used as input for the cell\ndeployment.\n\nWith this patch we provide a new tripleoclient functionality which\nhelps\nto automate the export steps from [1]:\n* Export the default cell EndpointMap\n* Export the default cell HostsEntry\n* Export AllNodesConfig and GlobalConfig information\n* Export passwords\n\n[1] https://docs.openstack.org/tripleo-docs/latest/install/advanced_deployment/deploy_cellv2.html#deploy-an-additional-nova-cell-v2\n\nDepends-On: https://review.opendev.org/672415\n\nChange-Id: Id7fdbf029a6dd1b45e9801c9cf8814a15a157ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/aa0d29fb4413c150af873c7d93b249ea2fc09573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/aa0d29fb4413c150af873c7d93b249ea2fc09573"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
