)]}'
{"id":"openstack%2Fpython-tripleoclient~672849","triplet_id":"openstack%2Fpython-tripleoclient~master~Ic7705c290220df7c55656fa080ff74315fb2030f","project":"openstack/python-tripleoclient","branch":"master","topic":"bp/nova-less-deploy","hashtags":[],"change_id":"Ic7705c290220df7c55656fa080ff74315fb2030f","subject":"New command \"openstack overcloud node unprovision\"","status":"MERGED","created":"2019-07-26 03:05:43.000000000","updated":"2019-09-06 06:33:00.000000000","submitted":"2019-09-06 06:31:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672849-1567751501045-7e60562c","meta_rev_id":"bc9d6fa66b71d681e5f8b6c6b0ab5b617c624a11","_number":672849,"virtual_id_number":672849,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":2,"date":"2019-09-06 06:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-27 07:24:00.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":2,"date":"2019-08-29 19:06:19.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-09-04 15:23:43.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-30 08:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-09-04 15:23:43.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 07:24:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-29 19:06:19.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-30 08:35:43.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-09-04 15:23:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-06 06:31:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"956a541e669051558dcdb06209c6cbb498806eac","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-07-26 03:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48b1fbc6d90c7518870011c087cc76256b86cb3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-26 03:25:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/672849/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/05d90d2/ : FAILURE in 17m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/672849/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8044207/ : FAILURE in 17m 51s","accounts_in_message":[],"_revision_number":1},{"id":"513aa1b062af13ef7df7412b67eb8d4a0367188f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 04:04:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/49/672849/1/check/test-release-openstack/2781b5f/ : SUCCESS in 2m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/672849/1/check/openstack-tox-lower-constraints/04645c7/ : FAILURE in 4m 27s\n- openstackclient-check-plugins http://logs.openstack.org/49/672849/1/check/openstackclient-check-plugins/65ee14f/ : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/672849/1/check/openstack-tox-pep8/f66c71a/ : FAILURE in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/49/672849/1/check/openstack-tox-py27/25dc447/ : FAILURE in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/49/672849/1/check/openstack-tox-py35/4f89a24/ : FAILURE in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/49/672849/1/check/openstack-tox-py36/33ec481/ : FAILURE in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/49/672849/1/check/openstack-tox-py37/41a5869/ : FAILURE in 4m 01s\n- openstack-tox-docs http://logs.openstack.org/49/672849/1/check/openstack-tox-docs/63a4500/ : FAILURE in 3m 37s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/672849/1/check/tripleo-buildimage-overcloud-full-centos-7/9c8d64b/ : SUCCESS in 19m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/672849/1/check/tripleo-build-containers-centos-7/a6e7340/ : SUCCESS in 41m 23s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/672849/1/check/tripleo-build-containers-centos-7-buildah/20c7e2d/ : SUCCESS in 56m 38s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/49/672849/1/check/tripleo-ci-centos-7-undercloud-containers/85d5d63/ : FAILURE in 18m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/672849/1/check/tripleo-ci-centos-7-standalone/a8b227b/ : FAILURE in 19m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/672849/1/check/tripleo-ci-fedora-28-standalone/3050510/ : FAILURE in 21m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a77196434070762d3f401fdf84094b4f323b58f0","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-07-29 23:36:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42efa8641c2b1b31815f0aadb759708b526a07b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 02:30:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/672849/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1110c6d/ : SUCCESS in 2h 52m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/672849/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/911012e/ : SUCCESS in 2h 49m 36s","accounts_in_message":[],"_revision_number":2},{"id":"2b2aa241ed0c8934a085b683875b399870690f54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 02:33:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/672849/2/check/test-release-openstack/d824721/ : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/672849/2/check/openstack-tox-lower-constraints/4ddf2a3/ : SUCCESS in 4m 56s\n- openstackclient-check-plugins http://logs.openstack.org/49/672849/2/check/openstackclient-check-plugins/6165cf0/ : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/672849/2/check/openstack-tox-pep8/8094d91/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/49/672849/2/check/openstack-tox-py27/0494181/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/49/672849/2/check/openstack-tox-py35/cbcc4e8/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/49/672849/2/check/openstack-tox-py36/7145a4f/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/49/672849/2/check/openstack-tox-py37/4f4c919/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/49/672849/2/check/openstack-tox-docs/8985a55/html/ : SUCCESS in 3m 22s\n- build-openstack-releasenotes http://logs.openstack.org/49/672849/2/check/build-openstack-releasenotes/4df9eb3/html/ : SUCCESS in 3m 50s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/49/672849/2/check/tripleo-buildimage-overcloud-full-centos-7/2a65162/ : SUCCESS in 23m 54s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/49/672849/2/check/tripleo-build-containers-centos-7/624342f/ : SUCCESS in 37m 29s\n- tripleo-build-containers-centos-7-buildah http://logs.openstack.org/49/672849/2/check/tripleo-build-containers-centos-7-buildah/3d38a87/ : SUCCESS in 1h 08m 59s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-containers-multinode/0406f54/ : SUCCESS in 2h 01m 13s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-undercloud-containers/51afac4/ : SUCCESS in 1h 51m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/21d8a2e/ : SUCCESS in 2h 20m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/96f48d9/ : SUCCESS in 2h 41m 10s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-standalone/93b30e8/ : SUCCESS in 1h 41m 52s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/49/672849/2/check/tripleo-ci-fedora-28-standalone/78ffee1/ : SUCCESS in 1h 51m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-standalone-upgrade/ce2e4b7/ : SUCCESS in 2h 09m 48s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/31ead00/ : FAILURE in 1h 51m 09s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/5d861c8/ : SUCCESS in 2h 18m 29s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/49/672849/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/0346134/ : FAILURE in 1h 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e3d4baae2da4c4b062c6b3cd930d0d9ae75c488","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-21 18:11:51.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"749871f8abfa275a8a5171120792d96bac17888b","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-27 04:07:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"089b41db3c8b63e43594d8b77017edabf2c84049","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-08-27 04:28:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a40e7ae0d15901a0b687f819a70867ce20e6e00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 07:20:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83f241b7ba5c45fe9c664cd940f95006 : SUCCESS in 3m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9218913b0f84747ba39b02357870958 : SUCCESS in 5m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/240e23790e0e47879c988eb6a3618a2d : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c4ab25881354db5a5bb937c07b6a2dd : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59dcf77407a64ba593de085a47eb0ef5 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac249c0866f845c68fa13fcca5329ff5 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ace028ebc3a47a4b5264349c4adf7da : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba5ee20969647bab97c9293dc7f397e : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a99fbda6dc84845901cf98b2dab9e75 : SUCCESS in 3m 51s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/8e1a1b5208594a82ba08ae936361c89f : SUCCESS in 24m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5a97c16cdaf34151a61381940c11bda0 : SUCCESS in 52m 18s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2e7bd4c42fd6491f813b01debee20047 : SUCCESS in 1h 14m 41s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/aa56ee3a957147928ea7c80cc40cfc43 : SUCCESS in 2h 58m 02s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3f4f5007b2024448945bb1d01e552acc : SUCCESS in 2h 00m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c7ec9b26c3ed48c396787759a3082be0 : SUCCESS in 2h 36m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d3a430b8e1834616a3e660ea9aae5108 : SUCCESS in 2h 44m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e7fbfd9729e548f6804e7ca511ef1e42 : SUCCESS in 2h 09m 01s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/00ea864ea5ab421f9339aa7c27f580d6 : SUCCESS in 2h 19m 50s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/137037e7f95f4ccdadf2f6fcb639d4fc : FAILURE in 2h 17m 03s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b70a2e5ee82443a4aa6c2675724fc91f : SUCCESS in 2h 45m 36s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/182d8cb8439f4b488c339e5e4aa59e22 : FAILURE in 1h 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ebe26ab4619e0de70b4b48c197e09a124fb1ba2e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 07:24:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/672849/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2a61688/ : FAILURE in 3h 16m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/49/672849/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/250bff9/ : SUCCESS in 3h 08m 22s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/49/672849/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/a779eee/ : SUCCESS in 20m 30s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/49/672849/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/ecab5c2/ : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ccddb0a944b22657c48a2c882d51ce350248bfa1","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-29 19:06:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"12fcf953c62903a31f6a796396a394ffa1cafb5c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-30 08:35:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7bed50b93cdbc7244ff38bcfc5d4665d5353e175","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-04 15:23:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"23b74be1be22b243d4992d68969e7998a12cd121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 15:24:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"40a205c805ad95ef9352e41884b09d84209e7821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 23:36:46.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58b01c7c752d4f3ebeb33fae802f6418 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b05b5ececac24a05b98c6000e23328de : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eaff71d90df4991869e11815a824ee9 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c121fbbea144242a5cbb7ad7820276a : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e52e7af3b9c44399d72287242e4f415 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e27ad4d6ca4d4ecaa1de696d0cbdc663 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc43d20103943ab8d3298d0c5a9345d : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35f19a463ea149b0b66e41ea58615d5d : SUCCESS in 4m 10s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/84f27d48449441d78578eef032e2a403 : SUCCESS in 24m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a745e1834e71464f93e2bc545898ab27 : SUCCESS in 46m 52s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/90f441de3e1244119d8507f5e1f1e674 : SUCCESS in 1h 16m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d3117bc4a19c488aa6fc689328abc38e : SUCCESS in 2h 38m 10s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d42ff02596414df688e649b24e07e502 : SUCCESS in 2h 15m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ddc16e7723c54e99bfdcb52226760d79 : RETRY_LIMIT in 12m 59s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d73b7bae11c34940a657ff3ad2d38af4 : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":3},{"id":"1c90124b3334cb3bdb3b800995dc430ce6c8bf59","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-05 13:33:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d4dc0dee178df97b491684efeb14002c1284db88","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-05 15:26:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/49/672849/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/acbb637/ : FAILURE in 1h 51m 45s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/49/672849/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/90d845c/ : SUCCESS in 20m 29s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/49/672849/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/5e910d6/ : SUCCESS in 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3bc936ba43e0aec5d82fee6abcfe942cd4b47a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:27:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2290adeb2b94c21b08a567c9ddfd9fc : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aea813b2cf314599912168be532eb08b : SUCCESS in 10m 03s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0170cb57be3a4517889bed082f407654 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d23d9bd675942a697ca433e31d37dba : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7146173e569f4829bf96ca51f186103a : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff147b64b3aa44bebcf14d1c195a6408 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2754d852543546c0bf234848afba5a00 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07501df13714b82954da24d54651aba : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7cf5eecdbdc4a5cbeaeaa5007a2bd58 : SUCCESS in 5m 13s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/44f87673c07245bfbcbd102a0d876ec3 : SUCCESS in 24m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/028bccdbd793429093214c0210994337 : SUCCESS in 51m 02s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/4c70a357757040a1bde18444ac1931c9 : SUCCESS in 1h 08m 02s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9a3a72cd09584749aa9ddb6396e5a2b5 : SUCCESS in 3h 19m 29s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2f6484f0ceee4a64aa34745f6330289f : SUCCESS in 1h 47m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f7c6695aab2b4557827e15e3b2aea86b : SUCCESS in 2h 11m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4f6eb231b7d24685b92c734ea653220f : SUCCESS in 3h 03m 12s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c1e1c46b81d049b092e5bb9233f88346 : SUCCESS in 2h 05m 36s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/8e2bbccbe2564cb3b7d3c014238124d2 : SUCCESS in 2h 23m 17s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/07de6f633d034c58aec13186c7bd2fd3 : FAILURE in 2h 31m 31s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/77fcf9f5086248769d3902b33c045877 : SUCCESS in 3h 15m 30s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/062822267eb342bd8dcc4d103d47937f : FAILURE in 1h 52m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb139a7d24df0065a957289f0483d9b8c0c2f108","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:28:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2696b56aa146cc95df01058200a6714013b9df2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 06:31:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7dbdf7d66e5c44a0a91b68df0959e877 : SUCCESS in 2m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84138e1061b94a5291dfe740b2b5a5d7 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8fe93c050b4a9d9b3644c7d8eef231 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b290c5be0ebb410c8216111a400503f6 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a44ee4dc7e64d1d9bd21750485864b8 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67e2cb05191549359128379a3081c267 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78774ab320046f3856583308c555dfe : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1929e1798f864144815c3de5faea96d6 : SUCCESS in 3m 32s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/5cd072d0c62144faa3decabda4f8c45b : SUCCESS in 20m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e6e38632fba54b12a3ce9899565fce6d : SUCCESS in 55m 33s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/3e6bb009903648699b562ad1a7b11d3a : SUCCESS in 1h 06m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/01f4844f236a4782b9f4ce0c12f8fe99 : SUCCESS in 2h 57m 08s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/758722647c454dd986ab60056b056aca : SUCCESS in 2h 02m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/76949d81592642368747678e2adbeefb : SUCCESS in 2h 02m 42s","accounts_in_message":[],"_revision_number":3},{"id":"33238c12d1f432e18d9c089cc83a6d7f23a40489","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 06:31:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8e5149e7b73f43e41fddb18872b780f702a033c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 06:33:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ded2589c6b41ed89b3609e690da1bf : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2a0266dae18db129498ae99b4b2a5e152a6ad448","revisions":{"fc2d3bdb32bfa4f1746083933063bb6b6f06f89a":{"kind":"REWORK","_number":1,"created":"2019-07-26 03:05:43.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/49/672849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/672849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/1"}}},"commit":{"parents":[{"commit":"7022d93e81b1c573a03f53b13dfa18bf8fae37f2","subject":"Merge \"Renaming \u0027Metadata\u0027 column by \u0027Parameters\u0027 when listing validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7022d93e81b1c573a03f53b13dfa18bf8fae37f2"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-07-26 03:03:34.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-07-26 03:03:34.000000000","tz":0},"subject":"WIP new command \"openstack overcloud node unprovision\"","message":"WIP new command \"openstack overcloud node unprovision\"\n\nChange-Id: Ic7705c290220df7c55656fa080ff74315fb2030f\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/#/c/672848/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/fc2d3bdb32bfa4f1746083933063bb6b6f06f89a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/fc2d3bdb32bfa4f1746083933063bb6b6f06f89a"}]},"branch":"refs/heads/master"},"4d36a230d2f8c04c4838b237f58bde6eb756732f":{"kind":"REWORK","_number":2,"created":"2019-07-29 23:36:13.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/49/672849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/672849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/2"}}},"commit":{"parents":[{"commit":"7022d93e81b1c573a03f53b13dfa18bf8fae37f2","subject":"Merge \"Renaming \u0027Metadata\u0027 column by \u0027Parameters\u0027 when listing validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7022d93e81b1c573a03f53b13dfa18bf8fae37f2"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-07-26 03:03:34.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-07-29 23:36:12.000000000","tz":0},"subject":"New command \"openstack overcloud node unprovision\"","message":"New command \"openstack overcloud node unprovision\"\n\nThis is a companion command to \"openstack overcloud node provision\"\nwhich is required for scale-down and should be run after \"openstack\novercloud deploy\".\n\nIt will undeploy any instance which has provisioned\u003dFalse in the\nsupplied roles yaml, see this change[1] for the provisioned format.\n\n[1] https://review.opendev.org/#/c/672605/\n\nChange-Id: Ic7705c290220df7c55656fa080ff74315fb2030f\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/#/c/672848/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4d36a230d2f8c04c4838b237f58bde6eb756732f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4d36a230d2f8c04c4838b237f58bde6eb756732f"}]},"branch":"refs/heads/master"},"2a0266dae18db129498ae99b4b2a5e152a6ad448":{"kind":"REWORK","_number":3,"created":"2019-08-27 04:07:05.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/49/672849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/49/672849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/49/672849/3"}}},"commit":{"parents":[{"commit":"c513e4975fc4e4164c9a6c0f39ee59f2adbb72d8","subject":"Merge \"Remove converge_nodes functions, unused package_update code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c513e4975fc4e4164c9a6c0f39ee59f2adbb72d8"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-07-26 03:03:34.000000000","tz":0},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2019-08-27 04:06:35.000000000","tz":0},"subject":"New command \"openstack overcloud node unprovision\"","message":"New command \"openstack overcloud node unprovision\"\n\nThis is a companion command to \"openstack overcloud node provision\"\nwhich is required for scale-down and should be run after \"openstack\novercloud deploy\".\n\nIt will undeploy any instance which has provisioned\u003dFalse in the\nsupplied roles yaml, see this change[1] for the provisioned format.\n\n[1] https://review.opendev.org/#/c/672605/\n\nChange-Id: Ic7705c290220df7c55656fa080ff74315fb2030f\nBlueprint: nova-less-deploy\nDepends-On: https://review.opendev.org/#/c/672848/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2a0266dae18db129498ae99b4b2a5e152a6ad448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2a0266dae18db129498ae99b4b2a5e152a6ad448"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
