)]}'
{"id":"openstack%2Fpython-tripleoclient~673264","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fstein~Ib25ee9593528ad680b14ca09c62addbbd0b773a3","project":"openstack/python-tripleoclient","branch":"stable/stein","topic":"bp/validation-framework-stable/stein","hashtags":[],"change_id":"Ib25ee9593528ad680b14ca09c62addbbd0b773a3","subject":"Add extra_vars parameter to run_ansible_playbook function","status":"MERGED","created":"2019-07-29 10:33:45.000000000","updated":"2019-07-29 16:45:42.000000000","submitted":"2019-07-29 16:45:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673264-1564418742084-622e802a","meta_rev_id":"f09acb1d7262fd1bafa74065849ca0f288342b51","_number":673264,"virtual_id_number":673264,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":2,"date":"2019-07-29 16:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-29 14:04:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-29 12:32:30.000000000","_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-07-29 10:35:31.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-07-29 10:35:28.000000000","_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"date":"2019-07-29 10:35:34.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":1,"date":"2019-07-29 13:48:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-29 10:35:18.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 10:35:18.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-29 10:35:28.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2019-07-29 10:35:31.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-07-29 10:35:34.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2019-07-29 12:32:30.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-07-29 13:48:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-29 14:04:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-29 16:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9b585cb8bf74037285fd1b1088d329236685419","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-29 10:33:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"521e24e65ca2f57601496e909f8018c3fe70bba1","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-07-29 12:32:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"07cb4f190394aacc588ace3b01a0b95dce55a769","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 13:33:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/64/673264/1/check/openstack-tox-lower-constraints/a0462dc/ : SUCCESS in 7m 23s\n- openstackclient-check-plugins http://logs.openstack.org/64/673264/1/check/openstackclient-check-plugins/c26079c/ : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/64/673264/1/check/openstack-tox-pep8/2bed225/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/64/673264/1/check/openstack-tox-py27/0fd099c/ : SUCCESS in 7m 26s\n- openstack-tox-py35 http://logs.openstack.org/64/673264/1/check/openstack-tox-py35/eebdd3a/ : SUCCESS in 6m 14s\n- openstack-tox-py36 http://logs.openstack.org/64/673264/1/check/openstack-tox-py36/509f7fb/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/64/673264/1/check/openstack-tox-py37/44962f2/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/64/673264/1/check/openstack-tox-docs/fac8464/html/ : SUCCESS in 4m 30s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/64/673264/1/check/tripleo-buildimage-overcloud-full-centos-7/db95b9b/ : SUCCESS in 25m 24s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-containers-multinode/563663a/ : SUCCESS in 2h 10m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-undercloud-containers/d81c2c9/ : SUCCESS in 1h 43m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/87c79b8/ : SUCCESS in 2h 22m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fe1f72b/ : SUCCESS in 2h 42m 08s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-standalone/9b3f498/ : SUCCESS in 1h 59m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/64/673264/1/check/tripleo-ci-fedora-28-standalone/1bbccbc/ : SUCCESS in 1h 41m 50s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d463776/ : FAILURE in 1h 23m 48s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/d59e4f1/ : SUCCESS in 2h 24m 44s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/eafc70a/ : FAILURE in 1h 17m 25s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/64/673264/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/2ba3d33/ : SUCCESS in 2h 09m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f493fb3a90118ba244789046766e08035a14fbdf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-29 13:48:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0bd48e49c4fb40dd5fdcc6a85d65ea2f9371acb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 13:48:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c05bf3d0d91c0487ac52909c672943ee24bf876f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 14:04:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/64/673264/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7b466fd/ : SUCCESS in 3h 27m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/64/673264/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/89a1ef7/ : SUCCESS in 3h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"20573a1583959f27bf967d66903b25f03de3d020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 16:45:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/64/673264/1/gate/openstack-tox-lower-constraints/efbcfcf/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/64/673264/1/gate/openstack-tox-pep8/88a919f/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/64/673264/1/gate/openstack-tox-py27/d81f7db/ : SUCCESS in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/64/673264/1/gate/openstack-tox-py35/146e439/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/64/673264/1/gate/openstack-tox-py36/1268050/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/64/673264/1/gate/openstack-tox-py37/9f3ee67/ : SUCCESS in 5m 59s\n- openstack-tox-docs http://logs.openstack.org/64/673264/1/gate/openstack-tox-docs/b7263f5/html/ : SUCCESS in 5m 13s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/64/673264/1/gate/tripleo-buildimage-overcloud-full-centos-7/54c95d1/ : SUCCESS in 22m 38s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/64/673264/1/gate/tripleo-ci-centos-7-containers-multinode/5139f34/ : SUCCESS in 2h 08m 33s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/64/673264/1/gate/tripleo-ci-centos-7-standalone/e8fe910/ : SUCCESS in 1h 54m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/64/673264/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7f12737/ : SUCCESS in 2h 49m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/64/673264/1/gate/tripleo-ci-centos-7-undercloud-containers/66c8278/ : SUCCESS in 1h 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"16fc8784d893d50531e7c0bacfdecf27a85fd960","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 16:45:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9b7f02095bdd7089099bfad06f4f62f46b30655","revisions":{"e9b7f02095bdd7089099bfad06f4f62f46b30655":{"kind":"REWORK","_number":1,"created":"2019-07-29 10:33:45.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/64/673264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/64/673264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/64/673264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/64/673264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/64/673264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/64/673264/1"}}},"commit":{"parents":[{"commit":"f54216dfd59183dc460c2c54b90ea7891ada19a5","subject":"Renaming \u0027Metadata\u0027 column by \u0027Parameters\u0027 when listing validations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f54216dfd59183dc460c2c54b90ea7891ada19a5"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-08 15:09:44.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-07-29 10:33:45.000000000","tz":0},"subject":"Add extra_vars parameter to run_ansible_playbook function","message":"Add extra_vars parameter to run_ansible_playbook function\n\nThis patch adds a new parameter called \u0027extra_vars (Defaults to None) to\nthe \u0027run_ansible_playbook\u0027 function. This parameter will set additional\nvariables to the \u0027ansible-playbook\u0027 command. It will accept either a\ndict or the absolute path of a file (YAML or JSON format).\n\nChange-Id: Ib25ee9593528ad680b14ca09c62addbbd0b773a3\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 35952145a1d5c5880ee3f66059fa00ec2156c8c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e9b7f02095bdd7089099bfad06f4f62f46b30655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e9b7f02095bdd7089099bfad06f4f62f46b30655"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
