)]}'
{"id":"openstack%2Fpython-tripleoclient~673868","triplet_id":"openstack%2Fpython-tripleoclient~master~I993fc09fd571fb93d78ffdbecf1823e806f744f4","project":"openstack/python-tripleoclient","branch":"master","topic":"bug/1838540","hashtags":[],"change_id":"I993fc09fd571fb93d78ffdbecf1823e806f744f4","subject":"Pass deployment_options to config download","status":"MERGED","created":"2019-07-31 16:23:34.000000000","updated":"2019-08-01 18:01:25.000000000","submitted":"2019-08-01 18:01:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673868-1564682485912-3342d8ab","meta_rev_id":"14be00c5d270f9e0811bc4f2d70c0c16c8907da7","_number":673868,"virtual_id_number":673868,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2019-08-01 03:51:19.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-01 18:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2019-08-01 07:03:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-01 07:58:09.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-08-01 10:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-01 07:08:48.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-08-01 03:51:34.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-08-01 13:33:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-01 03:51:53.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"date":"2019-08-01 03:52:06.000000000","_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 03:51:34.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-01 03:51:53.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-08-01 03:52:06.000000000","updated_by":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"reviewer":{"_account_id":17823,"name":"Michele Baldessari","email":"michele@redhat.com","username":"mbaldessari"},"state":"REVIEWER"},{"updated":"2019-08-01 07:03:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 07:08:48.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-08-01 07:58:09.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-01 10:28:57.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-08-01 13:33:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-01 18:01:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"437c6aa290d0d8049c52fcb436886acea62578bc","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-07-31 16:23:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49c63e5a88e3d7fb41ad4c8d314d068b642222e3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 19:31:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/68/673868/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cf62b31/ : SUCCESS in 2h 51m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/68/673868/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/05a5248/ : FAILURE in 3h 04m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3e207664fa016feb1a95828fd64ea972f1c1ee96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 22:44:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/68/673868/1/check/openstack-tox-lower-constraints/6fbc54d/ : SUCCESS in 6m 10s\n- openstackclient-check-plugins http://logs.openstack.org/68/673868/1/check/openstackclient-check-plugins/6b042f9/ : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/68/673868/1/check/openstack-tox-pep8/e708763/ : FAILURE in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/68/673868/1/check/openstack-tox-py27/2c78354/ : SUCCESS in 4m 35s\n- openstack-tox-py35 http://logs.openstack.org/68/673868/1/check/openstack-tox-py35/52ed5c6/ : SUCCESS in 6m 06s\n- openstack-tox-py36 http://logs.openstack.org/68/673868/1/check/openstack-tox-py36/80bad9d/ : SUCCESS in 3m 52s\n- openstack-tox-py37 http://logs.openstack.org/68/673868/1/check/openstack-tox-py37/810db71/ : SUCCESS in 5m 41s\n- openstack-tox-docs http://logs.openstack.org/68/673868/1/check/openstack-tox-docs/5d0c699/html/ : SUCCESS in 4m 39s\n- tripleo-buildimage-overcloud-full-centos-7 http://logs.openstack.org/68/673868/1/check/tripleo-buildimage-overcloud-full-centos-7/8a6887b/ : SUCCESS in 23m 51s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/68/673868/1/check/tripleo-ci-centos-7-undercloud-containers/2c8138f/ : SUCCESS in 1h 35m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/68/673868/1/check/tripleo-ci-centos-7-standalone/f87a403/ : SUCCESS in 2h 09m 10s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/68/673868/1/check/tripleo-ci-fedora-28-standalone/cbcfcbb/ : FAILURE in 56m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd84672ef88a2afe4cb87e627b36c18510904d4e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-01 03:50:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c3c53b5709e7c12f27a5a3b5766ccf199f1f038","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-01 03:51:19.000000000","message":"Patch Set 2:\n\nTest patch result:- https://logs.rdoproject.org/74/673874/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9286448/logs/undercloud/var/lib/mistral/overcloud/tripleo-ansible-inventory.yaml.txt.gz","accounts_in_message":[],"_revision_number":2},{"id":"7cfcbee8dfd6f4e70a5df11bbbd1f2da356f1e25","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 07:03:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/68/673868/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fcbb8b8/ : SUCCESS in 2h 46m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/68/673868/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f2095a7/ : SUCCESS in 3h 11m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ce4b5ff6583c5aab024618c62ebb01dfa60dd992","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-01 07:08:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"72840ba9467b6f5f4ffcc4ee7e535f10bb73642c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-01 07:58:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8311b60b2ff4861c6944cfc044b729919495859","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 10:00:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/68/673868/2/check/openstack-tox-lower-constraints/162c4f8/ : SUCCESS in 5m 34s\n- openstackclient-check-plugins https://logs.opendev.org/68/673868/2/check/openstackclient-check-plugins/a69221d/ : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/68/673868/2/check/openstack-tox-pep8/2f6fa03/ : SUCCESS in 4m 03s\n- openstack-tox-py27 https://logs.opendev.org/68/673868/2/check/openstack-tox-py27/f43d3fd/ : SUCCESS in 4m 19s\n- openstack-tox-py35 https://logs.opendev.org/68/673868/2/check/openstack-tox-py35/1466a13/ : SUCCESS in 4m 18s\n- openstack-tox-py36 https://logs.opendev.org/68/673868/2/check/openstack-tox-py36/f699b4c/ : SUCCESS in 4m 00s\n- openstack-tox-py37 https://logs.opendev.org/68/673868/2/check/openstack-tox-py37/a51b10e/ : SUCCESS in 4m 10s\n- openstack-tox-docs https://logs.opendev.org/68/673868/2/check/openstack-tox-docs/873c51b/html/ : SUCCESS in 3m 07s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/68/673868/2/check/tripleo-buildimage-overcloud-full-centos-7/cd837aa/ : SUCCESS in 25m 24s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-containers-multinode/ce94010/ : SUCCESS in 2h 13m 58s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-undercloud-containers/c01f02b/ : SUCCESS in 1h 38m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2845931/ : SUCCESS in 2h 28m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/139465b/ : SUCCESS in 2h 34m 25s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-standalone/240eb71/ : SUCCESS in 1h 54m 30s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/68/673868/2/check/tripleo-ci-fedora-28-standalone/c3a6a16/ : FAILURE in 1h 18m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-standalone-upgrade/26181d3/ : FAILURE in 1h 52m 43s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/dfb65b5/ : TIMED_OUT in 3h 11m 57s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/353b9fb/ : SUCCESS in 2h 25m 01s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/68/673868/2/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/4060101/ : FAILURE in 1h 48m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f0ab633506f65db579969624f388fa9f74baae2d","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-08-01 10:28:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9a69fc2d074c9f996964afa77118dfe0db78a644","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-01 13:33:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5d3c39bb96529da8bf97e73c69d176942f2eafc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 13:33:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7f1fff0792c0f2e4173e027158b59ed8e389d43c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 15:02:10.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2647a4081ba53095ce61ab475851609a54527197","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 18:01:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/68/673868/2/gate/openstack-tox-lower-constraints/969b614/ : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://logs.opendev.org/68/673868/2/gate/openstack-tox-pep8/f16f9fa/ : SUCCESS in 6m 59s\n- openstack-tox-py27 https://logs.opendev.org/68/673868/2/gate/openstack-tox-py27/055cf8a/ : SUCCESS in 6m 32s\n- openstack-tox-py35 https://logs.opendev.org/68/673868/2/gate/openstack-tox-py35/cc16576/ : SUCCESS in 4m 50s\n- openstack-tox-py36 https://logs.opendev.org/68/673868/2/gate/openstack-tox-py36/3a920c2/ : SUCCESS in 3m 50s\n- openstack-tox-py37 https://logs.opendev.org/68/673868/2/gate/openstack-tox-py37/9fe5db7/ : SUCCESS in 5m 23s\n- openstack-tox-docs https://logs.opendev.org/68/673868/2/gate/openstack-tox-docs/5cd4bbf/html/ : SUCCESS in 5m 20s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/68/673868/2/gate/tripleo-buildimage-overcloud-full-centos-7/66a92cc/ : SUCCESS in 26m 35s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/68/673868/2/gate/tripleo-ci-centos-7-containers-multinode/06104a3/ : SUCCESS in 2h 34m 28s\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/68/673868/2/gate/tripleo-ci-centos-7-standalone/068b93a/ : SUCCESS in 1h 55m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/68/673868/2/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c8f8ce4/ : SUCCESS in 2h 29m 16s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/68/673868/2/gate/tripleo-ci-centos-7-undercloud-containers/0788630/ : SUCCESS in 1h 46m 48s","accounts_in_message":[],"_revision_number":2},{"id":"dc75c999a4ef449e9264fb6309c08214e4a93359","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 18:01:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac7ae9e8d638676196c4897a30c978d51a091900","revisions":{"8f9c4f391392abc892d654d5102c8e20056767f0":{"kind":"REWORK","_number":1,"created":"2019-07-31 16:23:34.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/68/673868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/68/673868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/1"}}},"commit":{"parents":[{"commit":"c3fb3cb6622dbf7dd577578a55999954d544265a","subject":"Merge \"Add --parameters and --create-vars-file arguments to the list subcommand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c3fb3cb6622dbf7dd577578a55999954d544265a"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 16:09:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 16:10:05.000000000","tz":330},"subject":"Pass deployment_options to config download","message":"Pass deployment_options to config download\n\nhttps://review.opendev.org/673862 allows config_download\nworkflow to take deployment_options as a input, this patch\npasses it to the workflow so ansible_python_interpreter\nis set correctly in generated inventory and thus used\nduring ansible deployment.\n\nDepends-On: https://review.opendev.org/#/c/673862/\nCloses-Bug: #1838540\nChange-Id: I993fc09fd571fb93d78ffdbecf1823e806f744f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8f9c4f391392abc892d654d5102c8e20056767f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8f9c4f391392abc892d654d5102c8e20056767f0"}]},"branch":"refs/heads/master"},"ac7ae9e8d638676196c4897a30c978d51a091900":{"kind":"REWORK","_number":2,"created":"2019-08-01 03:50:36.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/68/673868/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/68/673868/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/68/673868/2"}}},"commit":{"parents":[{"commit":"75beaf85ab4efd9b7c8dbf9ca6a9094c3fdceff0","subject":"Merge \"Mock deployment action in upgrade tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/75beaf85ab4efd9b7c8dbf9ca6a9094c3fdceff0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 16:09:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-08-01 03:49:01.000000000","tz":330},"subject":"Pass deployment_options to config download","message":"Pass deployment_options to config download\n\nhttps://review.opendev.org/673862 allows config_download\nworkflow to take deployment_options as a input, this patch\npasses it to the workflow so ansible_python_interpreter\nis set correctly in generated inventory and thus used\nduring ansible deployment.\n\nDepends-On: https://review.opendev.org/#/c/673862/\nCloses-Bug: #1838540\nChange-Id: I993fc09fd571fb93d78ffdbecf1823e806f744f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ac7ae9e8d638676196c4897a30c978d51a091900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ac7ae9e8d638676196c4897a30c978d51a091900"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
