)]}'
{"id":"openstack%2Fpython-tripleoclient~674006","triplet_id":"openstack%2Fpython-tripleoclient~master~I7487d32305be36971e5f79e587f8c5c212b0feb8","project":"openstack/python-tripleoclient","branch":"master","topic":"mistral-removal","hashtags":[],"change_id":"I7487d32305be36971e5f79e587f8c5c212b0feb8","subject":"Remove wait_for_provision_state, a unused utility function","status":"MERGED","created":"2019-08-01 11:17:10.000000000","updated":"2019-08-13 21:34:53.000000000","submitted":"2019-08-13 21:34:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674006-1565732093888-0fc9d827","meta_rev_id":"8c9837dfef9b9df0ee63be680a325fad33ce009d","_number":674006,"virtual_id_number":674006,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-13 21:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-13 17:05:18.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-10 12:21:49.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-10 12:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":2,"date":"2019-08-10 12:21:49.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":1,"date":"2019-08-13 13:37:09.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 16:54:09.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2019-08-05 19:20:04.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-13 13:37:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-13 17:05:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-13 21:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"689f9c88359622a8fb7354e95382aaf391e88a71","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 11:17:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f00cd71e80cca25bf43757f027c548c5bdeb6111","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 11:36:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"df05a1bc13337208f44969e4dacb6bd9234cab7d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 12:02:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/06/674006/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e28eacb/ : FAILURE in 21m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/06/674006/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bac3352/ : FAILURE in 24m 00s","accounts_in_message":[],"_revision_number":2},{"id":"6fd3899b680dfd2ee4fbc0c258711caa4940153b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 12:20:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/06/674006/2/check/openstack-tox-lower-constraints/8200743/ : FAILURE in 5m 25s\n- openstackclient-check-plugins https://logs.opendev.org/06/674006/2/check/openstackclient-check-plugins/294088f/ : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/06/674006/2/check/openstack-tox-pep8/37992e5/ : FAILURE in 3m 19s\n- openstack-tox-py27 https://logs.opendev.org/06/674006/2/check/openstack-tox-py27/ffda46e/ : FAILURE in 3m 24s\n- openstack-tox-py35 https://logs.opendev.org/06/674006/2/check/openstack-tox-py35/25570d0/ : FAILURE in 4m 35s\n- openstack-tox-py36 https://logs.opendev.org/06/674006/2/check/openstack-tox-py36/7a1243b/ : FAILURE in 4m 48s\n- openstack-tox-py37 https://logs.opendev.org/06/674006/2/check/openstack-tox-py37/31594ff/ : FAILURE in 4m 30s\n- openstack-tox-docs https://logs.opendev.org/06/674006/2/check/openstack-tox-docs/4751dcf/html/ : SUCCESS in 3m 47s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/06/674006/2/check/tripleo-buildimage-overcloud-full-centos-7/3f4ecf0/ : SUCCESS in 24m 19s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/06/674006/2/check/tripleo-ci-centos-7-undercloud-containers/2d75b4b/ : FAILURE in 25m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/06/674006/2/check/tripleo-ci-centos-7-standalone/6d01c51/ : FAILURE in 24m 33s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/06/674006/2/check/tripleo-ci-fedora-28-standalone/c2023d6/ : FAILURE in 40m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bb02466420e041ba33aeece91856862cba8ddbe","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 13:34:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"154bf0ab60770886ef443487b3b1bc6a4bdcd81c","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 13:52:25.000000000","message":"Topic set to mistral-removal","accounts_in_message":[],"_revision_number":3},{"id":"be49cc41614e0b73743cf9dd874ec085a5bc47c7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 16:41:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a31b3cc/ : SUCCESS in 2h 43m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6b313f7/ : SUCCESS in 3h 05m 38s","accounts_in_message":[],"_revision_number":3},{"id":"6ec530e8f40fea91aaded03bd39af3e999e46efd","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-01 16:54:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d32e2599d5937d43ce54ea18831400584540fc83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 00:26:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/06/674006/3/check/openstack-tox-lower-constraints/5b8675e/ : SUCCESS in 6m 39s\n- openstackclient-check-plugins https://logs.opendev.org/06/674006/3/check/openstackclient-check-plugins/7c7be3a/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/06/674006/3/check/openstack-tox-pep8/75124a5/ : SUCCESS in 4m 48s\n- openstack-tox-py27 https://logs.opendev.org/06/674006/3/check/openstack-tox-py27/81d504a/ : SUCCESS in 4m 50s\n- openstack-tox-py35 https://logs.opendev.org/06/674006/3/check/openstack-tox-py35/d510da0/ : SUCCESS in 4m 48s\n- openstack-tox-py36 https://logs.opendev.org/06/674006/3/check/openstack-tox-py36/eeab245/ : SUCCESS in 5m 22s\n- openstack-tox-py37 https://logs.opendev.org/06/674006/3/check/openstack-tox-py37/188ae48/ : SUCCESS in 4m 11s\n- openstack-tox-docs https://logs.opendev.org/06/674006/3/check/openstack-tox-docs/6b08ad3/html/ : SUCCESS in 4m 29s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/06/674006/3/check/tripleo-buildimage-overcloud-full-centos-7/8e67a98/ : SUCCESS in 24m 51s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-containers-multinode/c3030ff/ : SUCCESS in 2h 24m 51s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-undercloud-containers/152eb8f/ : SUCCESS in 1h 32m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/07dd5ea/ : SUCCESS in 2h 25m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/03bfcb1/ : SUCCESS in 2h 26m 14s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-standalone/bd92364/ : SUCCESS in 1h 53m 11s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/06/674006/3/check/tripleo-ci-fedora-28-standalone/9bfa71e/ : FAILURE in 57m 31s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-standalone-upgrade/db435e8/ : FAILURE in 1h 47m 24s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5702a05/ : TIMED_OUT in 3h 18m 52s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/0cd0649/ : SUCCESS in 2h 31m 42s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/44065c9/ : FAILURE in 2h 07m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"90f5cd2a815b9441ec12d2daf1363cf9602c48d5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 19:20:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fac2da2cb443d2d2e9fc683ab9bfca79f56c9c50","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 23:42:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0c90f8fb0743c66867fb3dbf0ff06b0bc11f96f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 23:42:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"579bf0f4c5c521927bad67195edf9f248a28892e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 04:53:28.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"dba5ca757d4c9e56192dcb683482eafff9012933","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-06 05:50:20.000000000","message":"Patch Set 3:\n\nrecheck - OpenStack CI log server was done","accounts_in_message":[],"_revision_number":3},{"id":"992faf75c5fe6cedabcd0079495d55b2d715f59b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 10:29:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a2e30ce/ : FAILURE in 21m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e3373b0/ : TIMED_OUT in 4h 32m 17s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/4640c78/ : FAILURE in 9m 00s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/9787ab0/ : FAILURE in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f2bcb181bde1495352ec0c9023035a93b041469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 10:32:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/06/674006/3/check/openstack-tox-lower-constraints/dcb2404/ : SUCCESS in 6m 08s\n- openstackclient-check-plugins https://logs.opendev.org/06/674006/3/check/openstackclient-check-plugins/b253c1f/ : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/06/674006/3/check/openstack-tox-pep8/6d571da/ : SUCCESS in 4m 57s\n- openstack-tox-py27 https://logs.opendev.org/06/674006/3/check/openstack-tox-py27/179ad57/ : SUCCESS in 5m 00s\n- openstack-tox-py35 https://logs.opendev.org/06/674006/3/check/openstack-tox-py35/19a0015/ : SUCCESS in 5m 42s\n- openstack-tox-py36 https://logs.opendev.org/06/674006/3/check/openstack-tox-py36/fcc7794/ : SUCCESS in 4m 42s\n- openstack-tox-py37 https://logs.opendev.org/06/674006/3/check/openstack-tox-py37/44f477f/ : SUCCESS in 5m 23s\n- openstack-tox-docs https://logs.opendev.org/06/674006/3/check/openstack-tox-docs/a0c0b07/html/ : SUCCESS in 4m 36s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/06/674006/3/check/tripleo-buildimage-overcloud-full-centos-7/782305b/ : SUCCESS in 26m 19s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-containers-multinode/0ce6003/ : SUCCESS in 2h 18m 01s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-undercloud-containers/d888a85/ : SUCCESS in 1h 36m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5e975dd/ : SUCCESS in 2h 23m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/79bda71/ : SUCCESS in 2h 52m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-standalone/7cede69/ : SUCCESS in 1h 56m 24s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/06/674006/3/check/tripleo-ci-fedora-28-standalone/ffb6fa0/ : SUCCESS in 1h 48m 51s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-standalone-upgrade/a30fc7c/ : SUCCESS in 2h 12m 30s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1ccf5ee/ : FAILURE in 2h 09m 14s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/75c2d6c/ : SUCCESS in 2h 08m 46s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/2488bd1/ : SUCCESS in 1h 56m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"279a5eb2293b4c76641660a55ed24d15d25a3b75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 10:37:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"98e21d50328c15892fffc78becc87c3a19ae4ff9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 13:15:11.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"0af5e159903cb8c6c2a5bfad310529ef7427f001","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-06 13:28:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8c06940319f4b4e25c2bea36307f73f24358daac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 16:25:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9d91ff7/ : FAILURE in 2h 55m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/04d3d9a/ : FAILURE in 2h 54m 54s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/43cfcab/ : FAILURE in 9m 01s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/06/674006/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/5b398bf/ : FAILURE in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6aa1fdd1c3cc2c37671012346b7e282597aed2f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:30:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/06/674006/3/check/openstack-tox-lower-constraints/f6502e6/ : SUCCESS in 7m 49s\n- openstackclient-check-plugins https://logs.opendev.org/06/674006/3/check/openstackclient-check-plugins/e0bd806/ : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/06/674006/3/check/openstack-tox-pep8/30df8ad/ : SUCCESS in 5m 26s\n- openstack-tox-py27 https://logs.opendev.org/06/674006/3/check/openstack-tox-py27/7b0288d/ : SUCCESS in 3m 31s\n- openstack-tox-py35 https://logs.opendev.org/06/674006/3/check/openstack-tox-py35/2b5bc05/ : SUCCESS in 4m 59s\n- openstack-tox-py36 https://logs.opendev.org/06/674006/3/check/openstack-tox-py36/7031426/ : SUCCESS in 4m 07s\n- openstack-tox-py37 https://logs.opendev.org/06/674006/3/check/openstack-tox-py37/d230cdc/ : SUCCESS in 4m 29s\n- openstack-tox-docs https://logs.opendev.org/06/674006/3/check/openstack-tox-docs/996b87a/html/ : SUCCESS in 4m 11s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/06/674006/3/check/tripleo-buildimage-overcloud-full-centos-7/7e44760/ : SUCCESS in 25m 37s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-containers-multinode/3363cca/ : SUCCESS in 2h 33m 45s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-undercloud-containers/f6fce96/ : SUCCESS in 1h 42m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b7f5d4e/ : SUCCESS in 2h 13m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a58bd43/ : SUCCESS in 2h 43m 18s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-standalone/ef3deae/ : SUCCESS in 3h 04m 23s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/06/674006/3/check/tripleo-ci-fedora-28-standalone/ede6b91/ : FAILURE in 2h 04m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-standalone-upgrade/9c9fbac/ : SUCCESS in 2h 07m 28s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/46372bb/ : FAILURE in 1h 58m 08s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/fbe4774/ : SUCCESS in 2h 21m 05s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/06/674006/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/092eee8/ : FAILURE in 1h 56m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"89f904e483008e88c01dd63e76cb48eb16b44b2b","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-10 12:21:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"27790614b4ab207f8a4a555e032dfeaf3d542169","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 15:08:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/06/674006/4/check/openstack-tox-lower-constraints/44973ad/ : SUCCESS in 3m 55s\n- openstackclient-check-plugins https://logs.opendev.org/06/674006/4/check/openstackclient-check-plugins/418f260/ : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/06/674006/4/check/openstack-tox-pep8/170ca25/ : SUCCESS in 3m 38s\n- openstack-tox-py27 https://logs.opendev.org/06/674006/4/check/openstack-tox-py27/d1e0fb1/ : SUCCESS in 3m 34s\n- openstack-tox-py35 https://logs.opendev.org/06/674006/4/check/openstack-tox-py35/2706fe7/ : SUCCESS in 3m 00s\n- openstack-tox-py36 https://logs.opendev.org/06/674006/4/check/openstack-tox-py36/adb7a32/ : SUCCESS in 4m 23s\n- openstack-tox-py37 https://logs.opendev.org/06/674006/4/check/openstack-tox-py37/cc0f7a5/ : SUCCESS in 2m 57s\n- openstack-tox-docs https://logs.opendev.org/06/674006/4/check/openstack-tox-docs/66a7c11/html/ : SUCCESS in 3m 12s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/06/674006/4/check/tripleo-buildimage-overcloud-full-centos-7/b2d874a/ : SUCCESS in 16m 31s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-containers-multinode/3f8350c/ : FAILURE in 58m 19s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-undercloud-containers/56ccddb/ : SUCCESS in 1h 26m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/535bcee/ : SUCCESS in 2h 30m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/18cc9ab/ : FAILURE in 59m 34s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-standalone/ea1336f/ : SUCCESS in 2h 08m 47s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/06/674006/4/check/tripleo-ci-fedora-28-standalone/c8e1b53/ : FAILURE in 55m 39s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-standalone-upgrade/123be1d/ : SUCCESS in 2h 05m 53s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9d24da0/ : FAILURE in 1h 46m 50s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/06dd23a/ : SUCCESS in 2h 32m 55s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/06/674006/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/121dd64/ : FAILURE in 1h 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4ad86138eaefd318d9a3899b26c113d21f8dd05b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-10 15:43:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ce577f4/ : SUCCESS in 3h 17m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1545f27/ : FAILURE in 3h 10m 25s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/002a971/ : FAILURE in 14m 27s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/0cde79a/ : SUCCESS in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ccad8616cd0edf11340d0c62831751a7225112be","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-13 13:37:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"68ba1015f86b645a317648c1c684ce42046bca92","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-13 13:37:09.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"14a65bd237e258c45e05c933a8716e6a58ca2eea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 17:05:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/687216f/ : SUCCESS in 2h 43m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4b87b21/ : SUCCESS in 3h 17m 47s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/821e7c3/ : SUCCESS in 21m 29s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/06/674006/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/95b7e2b/ : SUCCESS in 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b87b7ffdd6961574b5d294da2dffbf37db74291d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 17:47:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62ded458be0b416f9d8394b873799a67 : SUCCESS in 4m 58s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/fc7d5f53c83b4776b39283d69d324797 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bdc3fd7ea9b44d39589a62d87b95d66 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08b82fe1d18d461e8caf17009856b2b8 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8bc1fe563ce4d1fbbb9f6aa59ea928c : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa7b6aaec29646c1b031d0c5327a4ebd : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebe99b31b3e48ca94267234c3ad3fff : SUCCESS in 2m 55s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/36e9bb6226fc45c38cd45f8fa95a57f1 : SUCCESS in 24m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/adff756bfd4c47e18330274b2d242dfc : SUCCESS in 2h 33m 16s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0a61a850b3254836952bfd5060de52b5 : SUCCESS in 1h 47m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f91f6f64f3ee43b8a1db0764987c653a : SUCCESS in 2h 36m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/158be9dbed1449d4a5f0aab9457f24f0 : SUCCESS in 2h 23m 45s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/eb7ca9da911e40709ae6563fbb790c0b : SUCCESS in 1h 57m 26s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/a8b26698133f426eb5886edb4d39f67c : FAILURE in 1h 06m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5d66a22f49dd48c4894f6a3f94115ce3 : SUCCESS in 3h 05m 18s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/06cf0027da54493db79cbc8f1689cb9a : FAILURE in 2h 09m 37s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/834b97820511478e8243f46561c5a282 : SUCCESS in 2h 08m 54s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7883756d47014a2e86e34733834407dd : FAILURE in 1h 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d7ae041eb1732a6353f0ef2b76f968b2117ef1ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 17:50:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9dc5dd3e9bde66210060b22b370804dea7e759be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 21:34:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8c9837dfef9b9df0ee63be680a325fad33ce009d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 21:34:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42c02db55d3a4a3ea26606a3b8d9b457 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a091fec2d124b9698398cc6835782ed : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/677d4bc44b8d499b87b411887b7d1918 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5e7aaddbb464644b2e00b97281c2eb0 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b54cbce81a94405bb0b13bebdb3f200 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96c85a6265e4b18b55fe17e9ca2313a : SUCCESS in 3m 14s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/e9099454083c47d5a2398ee3f6d05233 : SUCCESS in 23m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1ba05c8df6524b839ccc36d59a567c0c : SUCCESS in 2h 41m 26s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/2866a3129bc649a4b2413e6ccbe06ac9 : SUCCESS in 2h 12m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c352e66b79224a5bab26e0f7c7a5c4af : SUCCESS in 3h 08m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c9b8dc019f4e4e6fb776bfa4d19ea201 : SUCCESS in 2h 00m 19s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"90be7d7a9b2e1e94209ff6ea07beeb3f794b608d","revisions":{"e0ca7728f8976f8b1cc1297216e2a2b7e150bdcb":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:17:10.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/06/674006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/06/674006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/1"}}},"commit":{"parents":[{"commit":"eb4863d527aa95f5b321a3eae695f63b15d1638f","subject":"Removed set_nodes_state, a unused utility function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/eb4863d527aa95f5b321a3eae695f63b15d1638f"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:16:20.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:16:20.000000000","tz":60},"subject":"Remove wait_for_provision_state, a unused utility function","message":"Remove wait_for_provision_state, a unused utility function\n\nChange-Id: I7487d32305be36971e5f79e587f8c5c212b0feb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e0ca7728f8976f8b1cc1297216e2a2b7e150bdcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e0ca7728f8976f8b1cc1297216e2a2b7e150bdcb"}]},"branch":"refs/heads/master"},"f482766ff423d18db82006b879a95778044c6438":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-01 11:36:04.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/06/674006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/06/674006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/2"}}},"commit":{"parents":[{"commit":"3f183987a4bf0df20298bd130174a898dcb0514d","subject":"Removed set_nodes_state, a unused utility function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3f183987a4bf0df20298bd130174a898dcb0514d"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:16:20.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:34:11.000000000","tz":60},"subject":"Remove wait_for_provision_state, a unused utility function","message":"Remove wait_for_provision_state, a unused utility function\n\nChange-Id: I7487d32305be36971e5f79e587f8c5c212b0feb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f482766ff423d18db82006b879a95778044c6438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f482766ff423d18db82006b879a95778044c6438"}]},"branch":"refs/heads/master"},"b7aa90738d69104a351b6e0ed78157485012f6be":{"kind":"REWORK","_number":3,"created":"2019-08-01 13:34:34.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/06/674006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/06/674006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/3"}}},"commit":{"parents":[{"commit":"3f183987a4bf0df20298bd130174a898dcb0514d","subject":"Removed set_nodes_state, a unused utility function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3f183987a4bf0df20298bd130174a898dcb0514d"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:16:20.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 13:30:06.000000000","tz":60},"subject":"Remove wait_for_provision_state, a unused utility function","message":"Remove wait_for_provision_state, a unused utility function\n\nChange-Id: I7487d32305be36971e5f79e587f8c5c212b0feb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b7aa90738d69104a351b6e0ed78157485012f6be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b7aa90738d69104a351b6e0ed78157485012f6be"}]},"branch":"refs/heads/master"},"90be7d7a9b2e1e94209ff6ea07beeb3f794b608d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-10 12:21:49.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/06/674006/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/06/674006/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/06/674006/4"}}},"commit":{"parents":[{"commit":"48ed65c2050e508a7fe444ce0b765c3613c38c3c","subject":"Removed set_nodes_state, a unused utility function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/48ed65c2050e508a7fe444ce0b765c3613c38c3c"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:16:20.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-10 12:21:49.000000000","tz":0},"subject":"Remove wait_for_provision_state, a unused utility function","message":"Remove wait_for_provision_state, a unused utility function\n\nChange-Id: I7487d32305be36971e5f79e587f8c5c212b0feb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/90be7d7a9b2e1e94209ff6ea07beeb3f794b608d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/90be7d7a9b2e1e94209ff6ea07beeb3f794b608d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
