)]}'
{"id":"openstack%2Fpython-tripleoclient~674015","triplet_id":"openstack%2Fpython-tripleoclient~master~Id510b52e8391f65e93cb9dac171c978a5ee882e5","project":"openstack/python-tripleoclient","branch":"master","topic":"mistral-removal","hashtags":[],"change_id":"Id510b52e8391f65e93cb9dac171c978a5ee882e5","subject":"Remove converge_nodes functions, unused package_update code","status":"MERGED","created":"2019-08-01 11:36:04.000000000","updated":"2019-08-26 16:28:08.000000000","submitted":"2019-08-26 16:25:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674015-1566836700857-934252a3","meta_rev_id":"0d0a4772ee48a34f69d0d7a07576d27546e4cf84","_number":674015,"virtual_id_number":674015,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-26 16:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-22 12:53:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":2,"date":"2019-08-22 08:23:11.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-26 13:31:05.000000000","_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"all":[{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-26 13:31:05.000000000","_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 23:40:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-05 23:40:38.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"REVIEWER"},{"updated":"2019-08-22 12:53:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 13:31:05.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-08-26 16:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60f1c6d7bc55e400ce4bc03c9f089ade267ee813","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 11:36:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b8a559e6afe97a3b739b2297dd74915a60d7db0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 12:03:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/15/674015/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/eb292df/ : FAILURE in 25m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/15/674015/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/902df36/ : FAILURE in 24m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1e16f4029af9aed7aeaeec0e9c5050e769fff070","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 12:29:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/15/674015/1/check/openstack-tox-lower-constraints/e0c7491/ : FAILURE in 4m 49s\n- openstackclient-check-plugins https://logs.opendev.org/15/674015/1/check/openstackclient-check-plugins/21d6f6f/ : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/15/674015/1/check/openstack-tox-pep8/bdea12a/ : FAILURE in 3m 21s\n- openstack-tox-py27 https://logs.opendev.org/15/674015/1/check/openstack-tox-py27/17e4055/ : FAILURE in 3m 07s\n- openstack-tox-py35 https://logs.opendev.org/15/674015/1/check/openstack-tox-py35/d2f74c6/ : FAILURE in 3m 17s\n- openstack-tox-py36 https://logs.opendev.org/15/674015/1/check/openstack-tox-py36/e37d74f/ : FAILURE in 3m 08s\n- openstack-tox-py37 https://logs.opendev.org/15/674015/1/check/openstack-tox-py37/4210427/ : FAILURE in 3m 25s\n- openstack-tox-docs https://logs.opendev.org/15/674015/1/check/openstack-tox-docs/f7e7c8e/html/ : SUCCESS in 3m 59s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/15/674015/1/check/tripleo-buildimage-overcloud-full-centos-7/c33e75d/ : SUCCESS in 25m 09s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/15/674015/1/check/tripleo-ci-centos-7-undercloud-containers/b642972/ : FAILURE in 21m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/15/674015/1/check/tripleo-ci-centos-7-standalone/a665829/ : FAILURE in 22m 14s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/15/674015/1/check/tripleo-ci-fedora-28-standalone/9997ee0/ : FAILURE in 29m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a866ab317f02a50a64cc5edb1cd0761b1ab424eb","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 13:34:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"236f14f57d50b299691bb97326b566dfdf02e843","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 13:35:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e40080bb5d9697eb2108481832e31db1f31a2751","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-01 13:52:52.000000000","message":"Topic set to mistral-removal","accounts_in_message":[],"_revision_number":3},{"id":"cff75300f1955f414033d5ff69c78a0526daeffe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 16:37:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/15/674015/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6ff98e9/ : FAILURE in 2h 52m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/15/674015/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2bd2fab/ : SUCCESS in 3h 00m 41s","accounts_in_message":[],"_revision_number":3},{"id":"32019c86a79da5328594f41e58014f5cf19ed3fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 01:32:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/15/674015/3/check/openstack-tox-lower-constraints/2ce83be/ : SUCCESS in 6m 05s\n- openstackclient-check-plugins https://logs.opendev.org/15/674015/3/check/openstackclient-check-plugins/7c0d9ea/ : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/15/674015/3/check/openstack-tox-pep8/bd72c1b/ : SUCCESS in 3m 58s\n- openstack-tox-py27 https://logs.opendev.org/15/674015/3/check/openstack-tox-py27/fbf37da/ : SUCCESS in 3m 15s\n- openstack-tox-py35 https://logs.opendev.org/15/674015/3/check/openstack-tox-py35/9a4d369/ : SUCCESS in 4m 56s\n- openstack-tox-py36 https://logs.opendev.org/15/674015/3/check/openstack-tox-py36/8bdb46e/ : SUCCESS in 4m 08s\n- openstack-tox-py37 https://logs.opendev.org/15/674015/3/check/openstack-tox-py37/c5e5eba/ : SUCCESS in 4m 36s\n- openstack-tox-docs https://logs.opendev.org/15/674015/3/check/openstack-tox-docs/12342bb/html/ : SUCCESS in 3m 17s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/15/674015/3/check/tripleo-buildimage-overcloud-full-centos-7/d4661ba/ : SUCCESS in 27m 04s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-containers-multinode/59b89fd/ : SUCCESS in 2h 18m 34s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-undercloud-containers/53baca4/ : SUCCESS in 1h 33m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9502baf/ : SUCCESS in 2h 13m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9daa5e0/ : SUCCESS in 2h 41m 01s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-standalone/177e1d7/ : SUCCESS in 1h 50m 23s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/15/674015/3/check/tripleo-ci-fedora-28-standalone/bda81eb/ : FAILURE in 55m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-standalone-upgrade/849ef2f/ : FAILURE in 1h 46m 20s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/61adcc0/ : TIMED_OUT in 3h 20m 25s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/e6b313d/ : SUCCESS in 2h 26m 51s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/15/674015/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/116a1de/ : SUCCESS in 2h 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2908fd60e96aa8d4b3ee3c3ee842dbd43a5dce7b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 23:40:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"16e7cccbe87d2960ad159ae4129a953e8a60caa9","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-10 12:22:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b8dd1a846346cee91d941fcfc45a600638e18212","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 15:15:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/15/674015/4/check/openstack-tox-lower-constraints/e21744b/ : SUCCESS in 4m 04s\n- openstackclient-check-plugins https://logs.opendev.org/15/674015/4/check/openstackclient-check-plugins/657b76a/ : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/15/674015/4/check/openstack-tox-pep8/159e7c6/ : SUCCESS in 3m 21s\n- openstack-tox-py27 https://logs.opendev.org/15/674015/4/check/openstack-tox-py27/167a8ee/ : SUCCESS in 3m 27s\n- openstack-tox-py35 https://logs.opendev.org/15/674015/4/check/openstack-tox-py35/7866062/ : SUCCESS in 5m 19s\n- openstack-tox-py36 https://logs.opendev.org/15/674015/4/check/openstack-tox-py36/f1e7ed6/ : SUCCESS in 3m 50s\n- openstack-tox-py37 https://logs.opendev.org/15/674015/4/check/openstack-tox-py37/03d7e20/ : SUCCESS in 3m 19s\n- openstack-tox-docs https://logs.opendev.org/15/674015/4/check/openstack-tox-docs/3ef9881/html/ : SUCCESS in 3m 32s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/15/674015/4/check/tripleo-buildimage-overcloud-full-centos-7/53e5248/ : SUCCESS in 22m 21s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-containers-multinode/6d1d89a/ : SUCCESS in 2h 12m 05s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-undercloud-containers/48c4fc1/ : SUCCESS in 1h 35m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cf078ab/ : SUCCESS in 2h 04m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c61d6f3/ : SUCCESS in 2h 36m 06s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-standalone/05df0a6/ : SUCCESS in 1h 48m 40s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/15/674015/4/check/tripleo-ci-fedora-28-standalone/a854ea5/ : SUCCESS in 1h 52m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-standalone-upgrade/60637c7/ : SUCCESS in 1h 58m 12s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/347824b/ : FAILURE in 1h 58m 15s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/8199394/ : SUCCESS in 2h 26m 47s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/15/674015/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/27f7d9a/ : FAILURE in 1h 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d0fbc52a9a1382d32592ea3e73bb99f7d853a4b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-10 16:07:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/15/674015/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f1d1f79/ : FAILURE in 3h 41m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/15/674015/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1255e49/ : SUCCESS in 3h 38m 51s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/15/674015/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/765e924/ : SUCCESS in 20m 21s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/15/674015/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/41cfb8e/ : SUCCESS in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b056d30ebc1da0b06302b69821ef3652328b623d","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-19 15:56:20.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"155f2e6a53160c5d30b633c89d6251a35537087f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 19:14:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/15/674015/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d1269fb/ : SUCCESS in 3h 07m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/15/674015/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/34cc33f/ : SUCCESS in 3h 16m 18s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/15/674015/5/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/2041088/ : RETRY_LIMIT in 7m 09s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/15/674015/5/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/dec2cff/ : RETRY_LIMIT in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc94399a6d1341cf3fdbe00e5ed338c64650954a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 19:41:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9c754f5844b417c919780216c69e5a9 : SUCCESS in 6m 46s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0bd960e02e6a493896b886a73d371c0b : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8dccbd054ce46eba951142eb2ae2e2d : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d120624346443d59d7eebf65b840de6 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a1bbec1a7f64969bee5530a5825fab3 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea050370849d46b3baadd7da55273bff : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb3ab73f33743d4b9b04ba0a964600a : SUCCESS in 5m 40s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/f935dd7ccb444ba5b7e2ab3b8e251de2 : RETRY_LIMIT in 15m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/da1481e2bf914cf2a86667345888965f : SUCCESS in 2h 42m 32s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5682f4f7e98443c282bec8e8037ad89f : SUCCESS in 1h 30m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/66f69660f1c3472483e398549664cbe0 : SUCCESS in 2h 36m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e3401bb437ab401ca52179019d8f3f7a : SUCCESS in 2h 54m 41s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4d8d226f6877428caa14e4b78c27f1cc : SUCCESS in 2h 04m 04s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f0c45b1f36dd428ca18725552a40ce03 : SUCCESS in 2h 06m 20s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f4325cc2e031485e8936e737da6262f5 : FAILURE in 2h 10m 49s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9acea3c165a245619797f1f97e2482c8 : SUCCESS in 2h 30m 10s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1b9bf209892c4ebba86357988166557a : FAILURE in 1h 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8ce05503f4794d0e1117c0fddd1b41f00e73d4ef","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-08-22 08:23:11.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7bfdf3c74ccf7751ba11e816e62886460c51b7d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 12:20:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6f93322838f4cff941a4b6efa2b5950 : SUCCESS in 4m 59s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0678078de6da47ad83e133bd07808bb7 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8feb9f70dcc04a18abebf8c0fb23d78f : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57553791eb7e40cab9b3f548f75759a0 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ef83b65b2ac4cfea8ffbd1ee505826c : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cee7fc308fdf41f19949881ce4a0c6c4 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168aa711692b4202a541cf958c78cb54 : SUCCESS in 4m 37s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/c9a4e50903ca4378a938455f48707968 : SUCCESS in 24m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/063d8bcbff434c7e9136dab16e5a9429 : SUCCESS in 2h 42m 54s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a643d83b274f4f779934c91e8e8560e8 : SUCCESS in 1h 46m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d22dfbb463664f678f83d7a96c85f015 : SUCCESS in 2h 25m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f1dcbad50b3b417096bed27d5490b0ec : SUCCESS in 2h 47m 49s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/9ffea593372b4f91aa70cf39bd329d73 : SUCCESS in 2h 05m 33s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d95f4f3bc16c4f2ba2642b2a0bb9b032 : SUCCESS in 2h 22m 25s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/612b37310cb940f1b55a306daa34c47b : FAILURE in 2h 03m 44s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/40ee0b22c4324065bb8eed2e53aaa205 : SUCCESS in 2h 44m 04s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/73743850be584fc2aea45ef91908eba9 : FAILURE in 1h 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ed3c999eeec0d7b1bbc702f185bb185a7209ecb0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 12:53:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/15/674015/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f01c651/ : TIMED_OUT in 4h 05m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/3c12e0d0d47b4ab0aacd383a5edc5b2b : POST_FAILURE in 3h 52m 39s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/15/674015/6/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/e3bfe38/ : SUCCESS in 22m 07s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/15/674015/6/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/30899d3/ : SUCCESS in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"827d5ad941a0d9fc222062e557cb4145b80fb38b","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2019-08-26 13:31:05.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b71ddb53875270c5230952083e60b5ad1a9fb67a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 13:31:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"302690323ab2fcc49e4930df98107c15f6f69014","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:24:56.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0f18b567f3148d681fd1a120529eecc : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/536567e24dff473a97ca93d15e9c2d18 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20da930c85d940e3b743e8a95141e85e : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c82d181fef24e7bb10d1cbc08ec2d04 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f2b27f7434f4e2cad3d38ac8d32b12d : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd76f6c9ea444c1acd44f7c7d0a0d30 : SUCCESS in 4m 31s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/47d09f1b26114380beb01e45fa16b981 : SUCCESS in 23m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f0eeb2faaa5f487eab205f7d280b3de6 : SUCCESS in 2h 48m 45s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e1937d44f60641cd92cd08d1c43ffb1c : SUCCESS in 1h 54m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fc21a9b64de04545a40f8a4c69c18ee4 : SUCCESS in 2h 45m 09s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dcf89f38195f4ab49ee17ae936f00ec0 : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":6},{"id":"8e1dee8a99325023830248b10dcae24a6dca99a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:25:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"0d0a4772ee48a34f69d0d7a07576d27546e4cf84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:28:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70bab9eaf96444a2814486e8812f8d20 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"671313de681ff0f08db90be11d14445590bb2a9d","revisions":{"4c02d556d90dbf9e4a19916c53f189364b592669":{"kind":"REWORK","_number":1,"created":"2019-08-01 11:36:04.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/15/674015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/15/674015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/1"}}},"commit":{"parents":[{"commit":"7c94750df713b049767940541dc2f8328508d2f0","subject":"Remove all_unique, a unused utility function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7c94750df713b049767940541dc2f8328508d2f0"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:32:36.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:35:42.000000000","tz":60},"subject":"Remove converge_nodes functions, unused package_update code","message":"Remove converge_nodes functions, unused package_update code\n\nIt feels like this code should be used, but no sign of usage;\nhttp://codesearch.openstack.org/?q\u003dconverge_nodes\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nChange-Id: Id510b52e8391f65e93cb9dac171c978a5ee882e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4c02d556d90dbf9e4a19916c53f189364b592669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4c02d556d90dbf9e4a19916c53f189364b592669"}]},"branch":"refs/heads/master"},"b748c6ef05146205a5663383dd54c9c15079e780":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-01 13:34:34.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/15/674015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/15/674015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/2"}}},"commit":{"parents":[{"commit":"3266e2087dcd21e0c25cee1d722d4add0ea26470","subject":"Remove all_unique, a unused utility function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3266e2087dcd21e0c25cee1d722d4add0ea26470"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:32:36.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 13:33:51.000000000","tz":60},"subject":"Remove converge_nodes functions, unused package_update code","message":"Remove converge_nodes functions, unused package_update code\n\nIt feels like this code should be used, but no sign of usage;\nhttp://codesearch.openstack.org/?q\u003dconverge_nodes\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nChange-Id: Id510b52e8391f65e93cb9dac171c978a5ee882e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b748c6ef05146205a5663383dd54c9c15079e780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b748c6ef05146205a5663383dd54c9c15079e780"}]},"branch":"refs/heads/master"},"d40598b0b154067aeb763d82b063e294b38ca48a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-01 13:35:30.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/15/674015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/15/674015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/3"}}},"commit":{"parents":[{"commit":"7b365a586ea301ab83c9441889ab1cc4c882e48a","subject":"Merge \"Write minion preflight checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7b365a586ea301ab83c9441889ab1cc4c882e48a"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:32:36.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 13:35:30.000000000","tz":0},"subject":"Remove converge_nodes functions, unused package_update code","message":"Remove converge_nodes functions, unused package_update code\n\nIt feels like this code should be used, but no sign of usage;\nhttp://codesearch.openstack.org/?q\u003dconverge_nodes\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nChange-Id: Id510b52e8391f65e93cb9dac171c978a5ee882e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d40598b0b154067aeb763d82b063e294b38ca48a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d40598b0b154067aeb763d82b063e294b38ca48a"}]},"branch":"refs/heads/master"},"0fea6d880951adefc5525f4e28c7b9e64e9dcf4b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-10 12:22:40.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/15/674015/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/15/674015/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/4"}}},"commit":{"parents":[{"commit":"a8e81658532dcf06ab29ff35554c1f76407abdc3","subject":"Merge \"Allows to disable in-flight validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a8e81658532dcf06ab29ff35554c1f76407abdc3"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:32:36.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-10 12:22:40.000000000","tz":0},"subject":"Remove converge_nodes functions, unused package_update code","message":"Remove converge_nodes functions, unused package_update code\n\nIt feels like this code should be used, but no sign of usage;\nhttp://codesearch.openstack.org/?q\u003dconverge_nodes\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nChange-Id: Id510b52e8391f65e93cb9dac171c978a5ee882e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0fea6d880951adefc5525f4e28c7b9e64e9dcf4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0fea6d880951adefc5525f4e28c7b9e64e9dcf4b"}]},"branch":"refs/heads/master"},"6eeaff34326040fa989b4ab9f70056bece9a2609":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-19 15:56:20.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/15/674015/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/15/674015/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/5"}}},"commit":{"parents":[{"commit":"5a66fe5446809a62efb9f1bf5d9a60903a10fd33","subject":"Merge \"update constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5a66fe5446809a62efb9f1bf5d9a60903a10fd33"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:32:36.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-19 15:56:20.000000000","tz":0},"subject":"Remove converge_nodes functions, unused package_update code","message":"Remove converge_nodes functions, unused package_update code\n\nIt feels like this code should be used, but no sign of usage;\nhttp://codesearch.openstack.org/?q\u003dconverge_nodes\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nChange-Id: Id510b52e8391f65e93cb9dac171c978a5ee882e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6eeaff34326040fa989b4ab9f70056bece9a2609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6eeaff34326040fa989b4ab9f70056bece9a2609"}]},"branch":"refs/heads/master"},"671313de681ff0f08db90be11d14445590bb2a9d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-22 08:23:11.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/15/674015/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/15/674015/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/674015/6"}}},"commit":{"parents":[{"commit":"097210977f9da89adc7b8744ec814427813f6390","subject":"Merge \"Move enable_nova after enable_ironic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/097210977f9da89adc7b8744ec814427813f6390"}]}],"author":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-01 11:32:36.000000000","tz":60},"committer":{"name":"Dougal Matthews","email":"dougal@redhat.com","date":"2019-08-22 08:23:11.000000000","tz":0},"subject":"Remove converge_nodes functions, unused package_update code","message":"Remove converge_nodes functions, unused package_update code\n\nIt feels like this code should be used, but no sign of usage;\nhttp://codesearch.openstack.org/?q\u003dconverge_nodes\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nChange-Id: Id510b52e8391f65e93cb9dac171c978a5ee882e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/671313de681ff0f08db90be11d14445590bb2a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/671313de681ff0f08db90be11d14445590bb2a9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
