)]}'
{"id":"openstack%2Fpython-tripleoclient~674631","triplet_id":"openstack%2Fpython-tripleoclient~master~I5a286cca6e48b0abc9cec3f15b2648977ae838c2","project":"openstack/python-tripleoclient","branch":"master","topic":"fix_ssh_output","hashtags":[],"change_id":"I5a286cca6e48b0abc9cec3f15b2648977ae838c2","subject":"Suppress output for ssh-keygen","status":"MERGED","created":"2019-08-05 15:15:04.000000000","updated":"2019-08-20 22:19:06.000000000","submitted":"2019-08-20 21:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674631-1566337756457-8b5338a5","meta_rev_id":"25eeb4b2fd156fee3ef77cfdc8ce5733311aa6bf","_number":674631,"virtual_id_number":674631,"owner":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-20 21:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 16:19:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":1,"date":"2019-08-16 16:11:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-08-14 17:43:56.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":1,"date":"2019-08-19 07:11:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":1,"date":"2019-08-15 05:51:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"date":"2019-08-14 17:49:10.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-08-20 18:59:27.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-08-14 17:48:32.000000000","_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 15:18:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-06 09:49:03.000000000","updated_by":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"reviewer":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"state":"REVIEWER"},{"updated":"2019-08-06 14:07:07.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-08-14 16:19:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 17:48:32.000000000","updated_by":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"reviewer":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"state":"REVIEWER"},{"updated":"2019-08-14 17:49:10.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-15 05:51:22.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-16 16:11:54.000000000","updated_by":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2019-08-19 07:11:06.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-20 18:59:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-20 21:49:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bab041a7c7a443ed4f63db079cc2ff95606fdeb","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-05 15:15:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"742bf81967f006ea9c7fc801a205f040e96bf181","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-05 15:17:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d3ed5ee1a42086a02cd4feff173ec5ef61469f7","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-05 15:21:34.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c54990242e1987d8566a1b47aa1da38631b78ddb","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-05 15:27:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5a05d6c69353d050631bec6d6e797946f1f5a974","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-05 15:58:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"46333a2b3632202da94782c69ba8440d04afcb5f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 16:20:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674631/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d849c6c/ : FAILURE in 19m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674631/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c6de868/ : FAILURE in 20m 19s","accounts_in_message":[],"_revision_number":5},{"id":"eddf0513e61cf6feb9b25129326362a06c8f8964","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-05 16:42:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2285316b24ebeed62a635be7d697e18e6f9e1cb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:50:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674631/5/check/openstack-tox-lower-constraints/cb96728/ : SUCCESS in 6m 00s\n- openstackclient-check-plugins https://logs.opendev.org/31/674631/5/check/openstackclient-check-plugins/574056f/ : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/31/674631/5/check/openstack-tox-pep8/68b4163/ : SUCCESS in 3m 54s\n- openstack-tox-py27 https://logs.opendev.org/31/674631/5/check/openstack-tox-py27/b77caa5/ : FAILURE in 3m 52s\n- openstack-tox-py35 https://logs.opendev.org/31/674631/5/check/openstack-tox-py35/ca4e515/ : SUCCESS in 3m 52s\n- openstack-tox-py36 https://logs.opendev.org/31/674631/5/check/openstack-tox-py36/4165c09/ : SUCCESS in 4m 28s\n- openstack-tox-py37 https://logs.opendev.org/31/674631/5/check/openstack-tox-py37/5460b24/ : SUCCESS in 4m 27s\n- openstack-tox-docs https://logs.opendev.org/31/674631/5/check/openstack-tox-docs/655593e/html/ : SUCCESS in 4m 53s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/31/674631/5/check/tripleo-buildimage-overcloud-full-centos-7/9e0fefa/ : SUCCESS in 23m 44s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674631/5/check/tripleo-ci-centos-7-undercloud-containers/a690e7a/ : FAILURE in 27m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674631/5/check/tripleo-ci-centos-7-standalone/87e527f/ : FAILURE in 24m 42s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674631/5/check/tripleo-ci-fedora-28-standalone/57bca5f/ : SUCCESS in 1h 42m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a688685418d9e657e9ccd7476f7ae2568c41e913","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-08-06 09:49:03.000000000","message":"Patch Set 5:\n\nunit tests seem to be failing with:\nAttributeError: \u003cmodule \u0027subprocess\u0027 from \u0027/usr/lib/python2.7/subprocess.pyc\u0027\u003e does not have the attribute \u0027run\u0027","accounts_in_message":[],"_revision_number":5},{"id":"dcff9733aa89943f387d42bfe7311225b6fb640a","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-06 11:17:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fe47c77f342f362723b58e763e516c7ea060f19a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-08-06 14:07:07.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"10c1638e39fbffc990061d4227bf714d3a76c57c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 14:18:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674631/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/73e115c/ : FAILURE in 1h 53m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674631/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/70272cb/ : FAILURE in 2h 57m 45s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/31/674631/6/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/97569c0/ : FAILURE in 8m 54s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/31/674631/6/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/9b046e3/ : FAILURE in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4c9aa0c9330f7517cc5635509a9dcc39c841410f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 17:19:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674631/6/check/openstack-tox-lower-constraints/4850533/ : SUCCESS in 6m 58s\n- openstackclient-check-plugins https://logs.opendev.org/31/674631/6/check/openstackclient-check-plugins/41dfd77/ : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/31/674631/6/check/openstack-tox-pep8/f8c0359/ : SUCCESS in 4m 58s\n- openstack-tox-py27 https://logs.opendev.org/31/674631/6/check/openstack-tox-py27/6613b8a/ : SUCCESS in 5m 12s\n- openstack-tox-py35 https://logs.opendev.org/31/674631/6/check/openstack-tox-py35/db3b4e4/ : SUCCESS in 6m 11s\n- openstack-tox-py36 https://logs.opendev.org/31/674631/6/check/openstack-tox-py36/5ee40fd/ : SUCCESS in 4m 41s\n- openstack-tox-py37 https://logs.opendev.org/31/674631/6/check/openstack-tox-py37/241f7d9/ : SUCCESS in 3m 36s\n- openstack-tox-docs https://logs.opendev.org/31/674631/6/check/openstack-tox-docs/0d8543d/html/ : SUCCESS in 3m 18s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/31/674631/6/check/tripleo-buildimage-overcloud-full-centos-7/d2262a3/ : SUCCESS in 24m 39s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-containers-multinode/3ffe6f3/ : FAILURE in 1h 35m 36s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-undercloud-containers/3588f16/ : SUCCESS in 1h 44m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/de09a8c/ : FAILURE in 1h 25m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6237184/ : FAILURE in 1h 36m 02s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-standalone/751a945/ : FAILURE in 50m 25s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674631/6/check/tripleo-ci-fedora-28-standalone/d4dd68c/ : SUCCESS in 1h 56m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-standalone-upgrade/a9497f4/ : SUCCESS in 2h 14m 11s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a02f69c/ : TIMED_OUT in 3h 21m 04s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/3918e5d/ : FAILURE in 1h 03m 40s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/31/674631/6/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/2bbce8b/ : FAILURE in 1h 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"85f425fa64e968586abbbded29137e26565d16a7","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-06 23:25:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e12a5433b22647cef99d11ef16c7ee5f3271343e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 23:49:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674631/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1cec38d/ : FAILURE in 23m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674631/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/28b24da/ : FAILURE in 20m 44s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/31/674631/7/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/03da157/ : FAILURE in 8m 44s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/31/674631/7/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/a5b3d8c/ : FAILURE in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"aa0eb9d3068738eea5edb501f05ae87a1cd3b0b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:13:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://logs.opendev.org/31/674631/7/check/openstack-tox-lower-constraints/5abdc6c/ : FAILURE in 5m 54s\n- openstackclient-check-plugins https://logs.opendev.org/31/674631/7/check/openstackclient-check-plugins/c83915c/ : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/31/674631/7/check/openstack-tox-pep8/fcc30c3/ : SUCCESS in 5m 34s\n- openstack-tox-py27 https://logs.opendev.org/31/674631/7/check/openstack-tox-py27/5c9844a/ : FAILURE in 4m 04s\n- openstack-tox-py35 https://logs.opendev.org/31/674631/7/check/openstack-tox-py35/d65d4b4/ : FAILURE in 6m 02s\n- openstack-tox-py36 https://logs.opendev.org/31/674631/7/check/openstack-tox-py36/ba3cf82/ : FAILURE in 4m 19s\n- openstack-tox-py37 https://logs.opendev.org/31/674631/7/check/openstack-tox-py37/e4e4430/ : FAILURE in 5m 11s\n- openstack-tox-docs https://logs.opendev.org/31/674631/7/check/openstack-tox-docs/4642c8d/html/ : SUCCESS in 5m 00s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/31/674631/7/check/tripleo-buildimage-overcloud-full-centos-7/557dc2a/ : SUCCESS in 27m 23s\n- tripleo-ci-centos-7-containers-multinode tripleo-ci-centos-7-containers-multinode : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/31/674631/7/check/tripleo-ci-centos-7-undercloud-containers/7b6d3fa/ : FAILURE in 21m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/31/674631/7/check/tripleo-ci-centos-7-standalone/79cedbb/ : FAILURE in 26m 56s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/31/674631/7/check/tripleo-ci-fedora-28-standalone/7f8f857/ : FAILURE in 24m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade tripleo-ci-centos-7-standalone-upgrade : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades tripleo-ci-centos-7-containerized-undercloud-upgrades : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container tripleo-ci-centos-7-scenario007-multinode-oooq-container : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container tripleo-ci-centos-7-scenario009-multinode-oooq-container : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e48dbab263044d7a6f50d73e5668f6f1fba7b426","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-08-14 13:11:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8167ef037f7c912c3bfb35063e8e7aa2710e7f18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 16:18:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/860494f27e764b7e9addb5c2a3b9cc51 : SUCCESS in 5m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5866fce528324806a3bf88aa5c310be2 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d492ec8ca584847ba58762c7feeb88b : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/600df563c8234383873b362c47054305 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06e83f6148b046d0a9641886a8d2c0ad : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4108340b9fd499684c196eaedeb0aaa : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ad527b1f8149c194af19077571e39b : SUCCESS in 4m 36s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/44549d89a6394cb8a08eec49596ddcd7 : SUCCESS in 21m 23s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1ce0cfa94ee841179d04a95e5160a402 : SUCCESS in 2h 33m 39s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d236683b85db4c88916602c0cd03b3b5 : SUCCESS in 1h 57m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e3748111fda84220aac1ccd7566318a7 : SUCCESS in 2h 24m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4ea61a59ee96403aa383479bab828e44 : SUCCESS in 2h 42m 19s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ce59b465ea1142059d7e31a4913dd15c : SUCCESS in 1h 40m 22s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/a0c0f15560164c2c93a065be82dc6ab5 : SUCCESS in 2h 07m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/73667f0f61a9447f8930ef7d4aef7701 : SUCCESS in 2h 04m 37s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d8e9543015df49a79d62050306b07c7f : FAILURE in 2h 11m 25s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/25e4b53a6d4345b6800e042dade54a41 : SUCCESS in 2h 38m 08s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2e2bf9a8e62e4bc48ea1ea75254dd6fd : FAILURE in 1h 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8d99ef6d3be758a74649985032e771ecb3c2c283","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 16:19:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/feceeb4/ : SUCCESS in 2h 45m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/10ae107/ : SUCCESS in 3h 07m 20s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/f1d1b7b/ : SUCCESS in 19m 39s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/f4bebdf/ : SUCCESS in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"85935c5cdfd6db36fd97915466f59b3a2ebb1815","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-14 17:43:56.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"eda4198a3ded3e956428e8a94f96ac6f5ac2126c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-15 05:51:22.000000000","message":"Patch Set 8: Code-Review+1\n\nAlthough I don\u0027t see where you change anything related to open/close files?\nAnyway, LGTM, better output/logging is always important!","accounts_in_message":[],"_revision_number":8},{"id":"0d5135f535dcd502448be96bc94dc6fe4417b205","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2019-08-16 16:11:54.000000000","message":"Patch Set 8: Code-Review+1\n\nNeat approach. Nice work.","accounts_in_message":[],"_revision_number":8},{"id":"5a85a4904463a729a3aba2ed826615b2ccb2f6bf","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-19 07:11:06.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e62c613e77fe2d129a9ed2c0dc02992d3fc77893","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-19 13:55:29.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"df4daa7b8b7d1200a1711d88bc406a897f7f8811","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:55:43.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3d30f081f7aea9f24e3004fb66952ee39a5e428e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:42:02.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b41a5f79731f4b2a8df2dce6b2a812bd : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1771716ee3d1478191f64d5525f54a65 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9548438d423b43bd92714f7ba0087430 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8394718ebc94adaa96577d435836b40 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca48783d788042fba0633fa84714a475 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8e6095ec524041832271ef8a793d26 : SUCCESS in 4m 29s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/7927e2b493e84fbab0d043846b736645 : RETRY_LIMIT in 10m 39s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/49859e10e8ac4fa2b48fcd0fd37806f9 : SUCCESS in 2h 40m 58s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f75886a15073407785405f091c1afb0d : SUCCESS in 1h 59m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/afad00014eb945788c320dc115ef8f2a : SUCCESS in 2h 32m 30s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/10f71c9f52f146dcbbcef04c4a1f2027 : SUCCESS in 1h 45m 13s","accounts_in_message":[],"_revision_number":8},{"id":"ce4d87478dd5005bcdb85eb0e1b151e52a42e198","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-19 18:07:20.000000000","message":"Patch Set 8: -Workflow\n\nmeh forgot we needed https://review.opendev.org/#/c/677063","accounts_in_message":[],"_revision_number":8},{"id":"bf628b31b0137c7a46acb60784d0dad8b49f49ef","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-20 15:50:09.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"2e8d25b4bcc609fbb896beb53810079ed04df81a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 18:51:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4c1fee1/ : SUCCESS in 2h 42m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/61e551f/ : SUCCESS in 3h 00m 05s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/fc0c8c5/ : SUCCESS in 22m 52s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/31/674631/8/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/2eab556/ : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"373e701186e4c563b941af5721edeea5e0a3c702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:55:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb67a8f6fa4c4d82ba77c5c6a1daa0bc : SUCCESS in 5m 02s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ff4d5c0c824b4bfcbe1473d929a1adc0 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b97df31a59a847d0a3f2a4b6073c344d : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e1884ce1ffd5429a8080630fd76c0cd6 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/144f783f51e14c56aadba70bba557320 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21c27890fd794e0299c6dfaac549142d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fef712c55ca4b5cb4fb60a64f1a25a9 : SUCCESS in 4m 56s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/5484655f40f0420ea26a7e47c82681e5 : SUCCESS in 24m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/43fc0ca859a5435094ba3d3a0ce575ba : SUCCESS in 2h 40m 14s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/692c96acd6844bef8372ee0e1e9d91fa : SUCCESS in 1h 34m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/863df9793f8140358c4074dc414bf464 : SUCCESS in 2h 29m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8fa0d6c8686941a295fddc548d62aaea : SUCCESS in 2h 48m 53s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6cbae4f556bb452481734f5cb835ea0a : SUCCESS in 2h 06m 29s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7a0a54fd5d754e21ae09f8c80db7d477 : SUCCESS in 2h 30m 53s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c34538d80b664c1eb8b9f7910abf5b02 : FAILURE in 2h 15m 55s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/18a3617bc8f24247862ce567fd7d69cc : SUCCESS in 2h 38m 45s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e677db7a87ec43879754878659f7930f : FAILURE in 1h 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"33c31dddae7138f00001698443c1e788144ae0fd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-20 18:59:27.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"0b8d78d27a4e692d8622253d86c05c7f2a87c272","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 18:59:51.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fb0e9cf7896f3397fe53398c976ef051d37c9e94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 21:49:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"c1a8445fb578045829d6dc05247b0304ddbced3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 21:49:16.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a76c44b3cb234b799f7efeb0e78fdf83 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54dff8ff6c4b419fa9cb76e4b2498f56 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2e59b3ed9e84361b6d938e072377c21 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e9fdda87615453495a3ccf659598a00 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83e8f785932c4906aa36ac0084261095 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8c9d1011f74fcba424b1c082e05464 : SUCCESS in 4m 22s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/bc80528ce0ba4e05b542a656c354dcf2 : SUCCESS in 22m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4bfff54acee34037a744ee3942521870 : SUCCESS in 2h 39m 07s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/48b993f32ec846819e50ab67637288f4 : SUCCESS in 2h 01m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e9cb2f3974604ae794b634d0bc6c1dc5 : SUCCESS in 2h 28m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0e84a53600e8450db4d2177deb6a01a5 : SUCCESS in 1h 45m 20s","accounts_in_message":[],"_revision_number":8},{"id":"c520c05fbae181915e088ee7b70e176df9b887cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 21:50:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b2a46b0d3254187b036b74577120315 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":8},{"id":"09cfa4d696f96172d0db8c86116e5b1038a9624c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-20 22:19:06.000000000","message":"Patch Set 8: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 7a18efe47f0f3c739cd7c7beb381d94c09fb4b5a","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b40c9e3b1ca9ed74dd86d002451005faabda89b9","revisions":{"c56d5ba1098c28bf5e2e1e90cf3391ed96f47e6e":{"kind":"REWORK","_number":1,"created":"2019-08-05 15:15:04.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/1"}}},"commit":{"parents":[{"commit":"cbaff2dd996c60db1b654750f02856e00041f82f","subject":"Merge \"Replace warn with warning method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cbaff2dd996c60db1b654750f02856e00041f82f"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 15:14:00.000000000","tz":120},"subject":"Suppress output for ssh-keygen","message":"Suppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure and noth\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1619704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c56d5ba1098c28bf5e2e1e90cf3391ed96f47e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c56d5ba1098c28bf5e2e1e90cf3391ed96f47e6e"}]},"branch":"refs/heads/master"},"f2d453ba24f23decfb0a8e56da04a362d91a0f86":{"kind":"REWORK","_number":2,"created":"2019-08-05 15:17:49.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/2"}}},"commit":{"parents":[{"commit":"cbaff2dd996c60db1b654750f02856e00041f82f","subject":"Merge \"Replace warn with warning method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cbaff2dd996c60db1b654750f02856e00041f82f"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 15:17:27.000000000","tz":120},"subject":"uppress output for ssh-keygen","message":"uppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure and noth\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1619704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f2d453ba24f23decfb0a8e56da04a362d91a0f86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f2d453ba24f23decfb0a8e56da04a362d91a0f86"}]},"branch":"refs/heads/master"},"74715c186e684e8897c4c5628566660d25a3303d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-05 15:21:34.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/3"}}},"commit":{"parents":[{"commit":"cbaff2dd996c60db1b654750f02856e00041f82f","subject":"Merge \"Replace warn with warning method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cbaff2dd996c60db1b654750f02856e00041f82f"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 15:21:03.000000000","tz":120},"subject":"Suppress output for ssh-keygen","message":"Suppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure and noth\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1734356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/74715c186e684e8897c4c5628566660d25a3303d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/74715c186e684e8897c4c5628566660d25a3303d"}]},"branch":"refs/heads/master"},"e7f75afc1a09b4822bb31e061c598205084cf272":{"kind":"REWORK","_number":4,"created":"2019-08-05 15:27:44.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/4"}}},"commit":{"parents":[{"commit":"cbaff2dd996c60db1b654750f02856e00041f82f","subject":"Merge \"Replace warn with warning method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cbaff2dd996c60db1b654750f02856e00041f82f"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 15:26:55.000000000","tz":120},"subject":"Suppress output for ssh-keygen","message":"Suppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure. Also this\npatch closes file descriptor once key is read.\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1734356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e7f75afc1a09b4822bb31e061c598205084cf272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e7f75afc1a09b4822bb31e061c598205084cf272"}]},"branch":"refs/heads/master"},"0ff681bdea530882501ee961186613cfa6206ab7":{"kind":"REWORK","_number":5,"created":"2019-08-05 15:58:39.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/5"}}},"commit":{"parents":[{"commit":"cbaff2dd996c60db1b654750f02856e00041f82f","subject":"Merge \"Replace warn with warning method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cbaff2dd996c60db1b654750f02856e00041f82f"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 15:58:15.000000000","tz":120},"subject":"Suppress output for ssh-keygen","message":"Suppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure. Also this\npatch closes file descriptor once key is read.\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1734356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0ff681bdea530882501ee961186613cfa6206ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0ff681bdea530882501ee961186613cfa6206ab7"}]},"branch":"refs/heads/master"},"9c4f0421d440682c2ffed5cae9474859127d4353":{"kind":"REWORK","_number":6,"created":"2019-08-06 11:17:31.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/6"}}},"commit":{"parents":[{"commit":"cbaff2dd996c60db1b654750f02856e00041f82f","subject":"Merge \"Replace warn with warning method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cbaff2dd996c60db1b654750f02856e00041f82f"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-06 11:17:07.000000000","tz":120},"subject":"Suppress output for ssh-keygen","message":"Suppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure. Also this\npatch closes file descriptor once key is read.\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1734356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c4f0421d440682c2ffed5cae9474859127d4353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c4f0421d440682c2ffed5cae9474859127d4353"}]},"branch":"refs/heads/master"},"ab22a5357859fb7e10bf32020e8d5a1b0c9b341d":{"kind":"REWORK","_number":7,"created":"2019-08-06 23:25:28.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/7"}}},"commit":{"parents":[{"commit":"f64bf489ce29efb993a70edfafa6dcfdb60e5f4e","subject":"Merge \"Additional resource cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f64bf489ce29efb993a70edfafa6dcfdb60e5f4e"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-06 23:14:32.000000000","tz":120},"subject":"Suppress output for ssh-keygen","message":"Suppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure. Also this\npatch closes file descriptor once key is read.\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1734356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ab22a5357859fb7e10bf32020e8d5a1b0c9b341d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ab22a5357859fb7e10bf32020e8d5a1b0c9b341d"}]},"branch":"refs/heads/master"},"b40c9e3b1ca9ed74dd86d002451005faabda89b9":{"kind":"REWORK","_number":8,"created":"2019-08-14 13:11:51.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/31/674631/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/31/674631/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/31/674631/8"}}},"commit":{"parents":[{"commit":"055443d4e224bd141ec217c4dd0af5fbc9317f3b","subject":"Merge \"Remove download_ansible_playbooks, a unused util function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/055443d4e224bd141ec217c4dd0af5fbc9317f3b"}]}],"author":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-05 13:21:16.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2019-08-14 13:11:37.000000000","tz":120},"subject":"Suppress output for ssh-keygen","message":"Suppress output for ssh-keygen\n\nssh-keygen has private key which shouldn\u0027t be visible in output. This\npatch change the behaviour to print only exit code on failure. Also this\npatch closes file descriptor once key is read.\n\nChange-Id: I5a286cca6e48b0abc9cec3f15b2648977ae838c2\nRelated-Bug: rhbz#1734356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b40c9e3b1ca9ed74dd86d002451005faabda89b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b40c9e3b1ca9ed74dd86d002451005faabda89b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
