)]}'
{"id":"openstack%2Fpython-tripleoclient~675291","triplet_id":"openstack%2Fpython-tripleoclient~master~I3171adabc639789b0ea805a140e4b1220ed4c93a","project":"openstack/python-tripleoclient","branch":"master","topic":"bp/validation-framework","hashtags":[],"change_id":"I3171adabc639789b0ea805a140e4b1220ed4c93a","subject":"Exit listing validations with a proper return code","status":"MERGED","created":"2019-08-08 10:46:36.000000000","updated":"2019-09-03 08:26:39.000000000","submitted":"2019-08-22 13:10:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675291-1566479402601-66c16398","meta_rev_id":"f409d11828deb9e3da8a69832ce8d3a92f78d9bc","_number":675291,"virtual_id_number":675291,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"date":"2019-08-19 08:44:01.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-22 13:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-19 11:33:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":2,"date":"2019-08-15 13:50:45.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":1,"date":"2019-08-14 12:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 07:34:36.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":1,"date":"2019-08-15 13:50:45.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 10:47:06.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-13 16:00:28.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-14 06:07:48.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-14 06:36:36.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2019-08-14 12:39:26.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-15 07:34:36.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-08-15 13:50:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-19 11:33:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-22 13:10:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7194b8f340bb84b8e9b557564800aedabc24c4a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-08 10:46:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aff70af3662a86716b9559a8bd622c99afed4597","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 13:51:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/91/675291/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7b2f0aa/ : SUCCESS in 3h 02m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/91/675291/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/61695df/ : SUCCESS in 2h 57m 20s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/e95fe5f/ : FAILURE in 9m 32s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/00c702f/ : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e7513f789206241ebb33f633191cfae7ab75155","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-08 13:57:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6967a27fd2822087557b900eeec888792905fb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:31:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/91/675291/1/check/openstack-tox-lower-constraints/dfb0f35/ : SUCCESS in 7m 05s\n- openstackclient-check-plugins https://logs.opendev.org/91/675291/1/check/openstackclient-check-plugins/1ac54d4/ : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/91/675291/1/check/openstack-tox-pep8/a36e432/ : SUCCESS in 4m 06s\n- openstack-tox-py27 https://logs.opendev.org/91/675291/1/check/openstack-tox-py27/2be9394/ : SUCCESS in 3m 12s\n- openstack-tox-py35 https://logs.opendev.org/91/675291/1/check/openstack-tox-py35/e870d92/ : SUCCESS in 3m 54s\n- openstack-tox-py36 https://logs.opendev.org/91/675291/1/check/openstack-tox-py36/a5e370f/ : SUCCESS in 4m 00s\n- openstack-tox-py37 https://logs.opendev.org/91/675291/1/check/openstack-tox-py37/cbca46f/ : SUCCESS in 6m 57s\n- openstack-tox-docs https://logs.opendev.org/91/675291/1/check/openstack-tox-docs/809c1ed/html/ : SUCCESS in 5m 43s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/91/675291/1/check/tripleo-buildimage-overcloud-full-centos-7/1267a9d/ : SUCCESS in 23m 44s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-containers-multinode/83a115d/ : SUCCESS in 2h 40m 36s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-undercloud-containers/58b021d/ : SUCCESS in 1h 36m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bbfe370/ : SUCCESS in 2h 11m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b4a5bb8/ : SUCCESS in 2h 32m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-standalone/a7c4cf5/ : SUCCESS in 2h 36m 00s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/91/675291/1/check/tripleo-ci-fedora-28-standalone/8895195/ : FAILURE in 1h 37m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-standalone-upgrade/ad67958/ : FAILURE in 1h 37m 21s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/26f8e6a/ : FAILURE in 2h 03m 10s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/086f1ae/ : SUCCESS in 2h 27m 06s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/91/675291/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/cc1aac5/ : FAILURE in 2h 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08537f3bdc40cb7d371ad82c091e8b77ac302536","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-09 07:59:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nseems reasonable - +1 to encourage consideration of my comment but i may be wrong/missed something and in general won\u0027t block on that. let me know if you disagree and i will revote thanks.","accounts_in_message":[],"_revision_number":1},{"id":"ea657aa35d2b986a51af6ffc8d8a9bd79f4fab9b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-09 08:24:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bdef754db94f461518df82888ee0f5a206f84bc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 15:03:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f028a346fe0cdf41e0c60a283d826b87caf72643","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-13 08:41:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a61a71d6225cf70c91fdcb4ddc9e6364f353a54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 12:30:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/e0beb095b3cd45538476e44898f182c6 : POST_FAILURE in 3h 47m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze02.softwarefactory-project.io:7979/51ac376229624ed3828f8036b2bf3cfa : POST_FAILURE in 3h 47m 23s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/36a71ab/ : SUCCESS in 23m 11s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/6f269f5/ : SUCCESS in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"131a906659aeafc9bad8d50f8837b5a9e9c0a16a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 13:07:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f45c8aebfdac4a3295e0f98d3d67ccdf : SUCCESS in 5m 31s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/212aee9cd8dc450396e712a45bf61fc5 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94c35d6cff18424284e8914f49d5d5c4 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d97c7f0ff37a4052ae8f2b80721aa992 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f260059eef4802bc6a00020ae0fd9a : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4274a584ac3f4b0dbb37800110984efe : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004014a2033a43009f40e17759c1600d : SUCCESS in 3m 46s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/1a75eb86209b49baabd7154fcb3e8ca0 : SUCCESS in 23m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/41272df8adb344789c209cdac9903eb4 : SUCCESS in 2h 23m 52s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1809029a3567459ba30cd8d478855439 : SUCCESS in 1h 33m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e69a8905d01b4b7f9821a2c8b2c9955b : SUCCESS in 2h 18m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5ea9ae3232ac49b783219394e5b4bb7c : TIMED_OUT in 3h 15m 42s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/451bef314aa444f1bfef8a541280f178 : SUCCESS in 2h 09m 23s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/5aa4676de3424ff6a05e23226da47116 : SUCCESS in 1h 47m 02s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/58fea6b9428249229cd04015f6a9887a : SUCCESS in 2h 13m 35s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/facaff32a0ea4c35a1a74b62070c780e : FAILURE in 2h 04m 02s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/fc417a60d2b74789b230eef5a42f7cf0 : SUCCESS in 3h 12m 22s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/746c4adf6c3f469baae6173fcde829b1 : FAILURE in 1h 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae9a06ee0502f46bd8beffdc401431b5c2bc1397","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-13 13:45:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ad72fd6028456dc3a440a5c15d767b138c22c533","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-08-13 16:00:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c6a5a0f0eca9df7da58cfadcf894bccd37e8fafe","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-14 06:07:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nOK - holding on approval in case you were going to update tests for the new exception raising? (ping me if you want me to +A)","accounts_in_message":[],"_revision_number":2},{"id":"591e4f72e26533ae1a7dbb60062cf7439b2a6668","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-14 06:22:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"458a8398b74fb78483b5c04f92c9c2d847f44026","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2019-08-14 06:36:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4b66249c4c575f0a92b78a87f37cbad6013d35bc","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-14 10:59:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"27bf6d721b1ba063500160d47791a2d2e8cfbf14","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-14 12:39:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b74bc1aed5342961743c4e6f18fdf73962dc51c1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 13:49:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5f4d14e/ : SUCCESS in 2h 49m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2ea50c6/ : SUCCESS in 2h 44m 07s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/5280fb7/ : SUCCESS in 19m 51s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/df9f444/ : SUCCESS in 17m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d87ab764eab83d9d0957a0d35ffbca836bcf10f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 14:08:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ec2b014e5f14e49ab991375193af391 : SUCCESS in 6m 22s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/1df8ac21b5f84264bbe39e9faf919f2e : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8af1a9257344228db286810e75ae93 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08f6fb7cbd0e4619ae8c6a70cbfe0301 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6206ac4d82549829b35397a8526f522 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6e3a8baaebb4026af09d59fa0442c6e : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4084243f49ca491b9e6e99fea132978e : SUCCESS in 4m 03s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/7650e352256843a2b1d0d986e0ee028f : SUCCESS in 19m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cf6e6f61cec741a0b3326f941e2b2444 : SUCCESS in 2h 09m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/eab2364d95104cd9b767de8d41ca5862 : SUCCESS in 1h 56m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/bc6ec8ae39c6420381e3464643456ff9 : SUCCESS in 2h 44m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/575222d50fa542c197eff4e98bc9e0ba : SUCCESS in 2h 30m 05s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/6987dac12bb7455ab0606c3973803490 : FAILURE in 54m 51s\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/b6ae33fdfeb84e68bd682df7919b9531 : FAILURE in 57m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c473b91346e24b1ea5c156d54448d4af : SUCCESS in 2h 16m 36s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ed63a81e9275406d8e986ac97dceaf02 : FAILURE in 1h 57m 55s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/475a12b85a7445b8b9813b47bb06e17f : SUCCESS in 2h 18m 16s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6262d70673044ccb9b34822fb48eb12f : FAILURE in 1h 39m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a0198ddb829fb410c7b4da45e1c733ed4fc3fc24","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-15 07:34:36.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0df2fd04790049a9e0fd925f92d1d1504d884eef","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 10:51:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5eeccaf/ : SUCCESS in 2h 53m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3718308/ : SUCCESS in 3h 15m 39s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/a7fccc9/ : SUCCESS in 21m 00s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/5f4808e/ : SUCCESS in 17m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c9d7d0de2ae21405558c96eb8500754c0618b23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 11:25:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0525dc329e046d0a6dc0c8345dd5eae : SUCCESS in 5m 00s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/530ecff7b91d4b5184f1a910eb9b6c01 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbeebb95409544e2a92550aed02e8bf3 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39f7029c180d41f4bf288da81287e8d9 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a413cd3ff25d4406aab139fa7f364c35 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2199feb7ef6d49b6a5c356b2a3d55ca4 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f921d1807b41ea9d4a4c338211c86d : SUCCESS in 3m 11s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/ea14781ed3544aaeaa5503b736cfce18 : SUCCESS in 24m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e019e9a7dbde48e7adf2e4b837868a93 : SUCCESS in 2h 17m 11s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/81d59cff041f462a851cae8530189001 : SUCCESS in 1h 37m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0d6425018749404fbe67f4a87fbc827a : SUCCESS in 2h 15m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f335fd0b9bf14581a7846cb9c09734ae : SUCCESS in 2h 19m 54s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/cb4cff3fce9c46f8b8244236f5ac8b98 : SUCCESS in 2h 13m 41s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c288332a599141759081fdc92d638c70 : SUCCESS in 2h 05m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/691ee9b9ba9c4c55a2505816e1488485 : FAILURE in 2h 00m 41s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8579a572e1d34ecba34e89d23aaf1a6a : SUCCESS in 2h 37m 51s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b8e6bca740fc4b09b027e02f9ebe10e9 : FAILURE in 1h 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"18d5df2333a967a53e4fe91cf92c38fb169b45ae","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 13:50:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c612129303bd1aece9ba5384dd6d9f3424b264e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 13:50:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e8e5fa5352eb58ec4d915ee0a0a6c207bcb1ceb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 17:11:12.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a69c66d1ccd340d78db1d1363a916d35 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1961887e94443b4ab828b23abe3d837 : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/928a4b3fe4fd4c39986576893f874717 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f6219040e884c178b8bd9b53d02ef85 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b20ba87655454ce1b9d2c5f2f3754e55 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a945d39ae4840d18383b9d91db6209d : SUCCESS in 4m 46s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/2a541db2919b46d5a2abfd09708e48cd : SUCCESS in 23m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/eb705ad2933d4d83a1834fdf2a1ee3b0 : SUCCESS in 2h 15m 05s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/90dea4e1fe2240639c3bda13a25c9157 : SUCCESS in 1h 58m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4b77a4b475d44918b0c23248bd1d64af : FAILURE in 1h 58m 55s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/74838e3b833d4f5b85d837688f8c262c : SUCCESS in 1h 35m 32s","accounts_in_message":[],"_revision_number":3},{"id":"8c3919c51e751d45a6075881e5fd1f1ce9781cd2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 19:00:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0d41d7d3b8a51cb47b967f42a5c30e7f99f6a243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 22:12:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e23e4d1c47c4d799293e49f72f775d2 : SUCCESS in 6m 40s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d6d84ad5dfaa402dadd4cf91410029d1 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0161a05827148b98381d5e768aab478 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d653d2311151481593b95aa36fd05e86 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6ca2bbc073a477a9a7f63e2c98916fc : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/914ef4f1cfc24f18906d526139d6111f : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb0bba862a84c809b41ca314c50cd3a : SUCCESS in 3m 50s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/9396d7aafb334f3587fd4a93a954d76c : SUCCESS in 21m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1ed1d26d54c143abbe332f13f56d4f94 : SUCCESS in 2h 45m 42s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/38c1d35f4fb24a1fb1da1d2dd17843c1 : SUCCESS in 1h 43m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c05aa8e39ed5425e917d965d37a9fec9 : SUCCESS in 2h 13m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/178ebd1995054ee991b0bdb16ead8d07 : SUCCESS in 2h 32m 27s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/3122d3ad58c64dceb751d15a9d154a78 : SUCCESS in 2h 06m 12s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/47c80c9f180b4ee197b7034b60d38605 : SUCCESS in 2h 24m 18s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4fa4bf27cd24494c8fb2c2427b124d92 : FAILURE in 2h 04m 29s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/45c60eea12474dbfa7d896dfcdc8575b : SUCCESS in 2h 22m 40s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9b7915ab450849e0b5af7d9148cce12b : FAILURE in 1h 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4cfbd8c5d1791f60767b5083b1e4d491a6772879","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 22:12:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8fecc8b211cafadc651c8258028b7dfd9836946d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 00:22:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/2122f3bcdc124e60812895ebebe267fc : POST_FAILURE in 3h 50m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze07.softwarefactory-project.io:7979/48046038ffa84f8cab4441d969399d32 : POST_FAILURE in 4h 05m 00s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/72fa44b/ : FAILURE in 16m 23s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/29ac113/ : SUCCESS in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ffc43345fb1e47faafe2c1eb6f090d6e6b8b9191","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 02:59:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31d743b37f8845e19c1901af3fd442a8 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6808d58d7a9c4e3faaed9ef4fb4137df : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c18ef00c18a54019b9945010b78649e5 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b439b5e3dbc34396b17b5ca746edd727 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/850d4ff53b38420fb3bcec86e311b610 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210fa52266b0431ba1c62be40586744e : SUCCESS in 3m 55s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/6ef75df7270c4e98bb0998b9fa0ab065 : RETRY_LIMIT in 7m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6209ee119b8648989dc99886c8d91f0d : FAILURE in 1h 32m 59s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/47e3265deeb94e50b2abae032c51bdeb : SUCCESS in 1h 51m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/27c8263979a5414c89d530833be72ddf : SUCCESS in 2h 30m 32s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fc7ec3a14ee349e495b8a61378f13f80 : SUCCESS in 1h 49m 31s","accounts_in_message":[],"_revision_number":3},{"id":"afbaa4e5aafcd1624846f98f7983091935870db8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-16 13:56:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a8d60318b2bed2988b3c75b555559e855eaea671","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 15:20:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/aa020c8/ : FAILURE in 1h 17m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/177fc4d/ : FAILURE in 1h 21m 15s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/6810dd5/ : RETRY_LIMIT in 7m 56s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/d80cb3d/ : RETRY_LIMIT in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41aaf00c3f5fd114a3dd3d4e549dae70678df604","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 17:01:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36fd6926c7374048beed1efe62129491 : SUCCESS in 6m 23s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/449af60d01814441a788c9865986e94f : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963ef80bb5b94fd7a4df8d4e2567181a : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/460d61d668d54d61b493288e6c0c7e9f : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0cc5dafe3024fcaaee6b6ebedc47694 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/08b99a61a8094684b7509b01856658fc : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ce18bb502a47b8a8a2df12d56fb89a : SUCCESS in 3m 46s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/d69db11950cb40b995f1100853d4ca6c : RETRY_LIMIT in 7m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6e08e27f18c64e5d82148d75340090ad : SUCCESS in 2h 29m 33s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/64a31455921f47deb042ae98b496264c : SUCCESS in 1h 47m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/70728ef8f2d4414eaf289e5db28b5f4f : SUCCESS in 2h 13m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/0d4a3ce0c82e4667817a7a064f3a7011 : SUCCESS in 2h 43m 09s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/50f18cd0f32e4fd08e733821b1f40cd3 : SUCCESS in 1h 52m 46s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5ba69e8b71e343ad9ef789b73adcd4ba : SUCCESS in 2h 14m 47s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/9e6fab17036b4e33908981637c878818 : FAILURE in 2h 06m 05s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ba682d9fe8a3408c88a9e1e618f38f94 : SUCCESS in 2h 38m 32s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9c9f2f31bb61400794ac4e4744a79df9 : FAILURE in 1h 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3121afc5bb8ec2a51e18bc5a317134436a74acbf","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-19 08:44:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ecddba3f9e29fc688722c30e0405ca448a162c7d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 11:33:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b69187c/ : SUCCESS in 2h 48m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4c9776c/ : SUCCESS in 2h 41m 26s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/b5e6a33/ : RETRY_LIMIT in 7m 04s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/c3c4995/ : RETRY_LIMIT in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"79630136f8e6aaab218adcd2439c22e0450fb743","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:51:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e065c3080e4404d8f2aa7cbe7558666 : SUCCESS in 5m 54s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/93b76a7eac7242eb8a9e968cf5566db0 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b979b437350d4dbca746d7223336dc9a : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e07834dff0f6445e9cbfcf283b083f1a : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4af189d75b6d47749b2439e6b89b077e : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87ea08624cbb4dfcaefad59ec0b05467 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8c2bf0c685408dab96cb7a6fdbfc5d : SUCCESS in 4m 55s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/cda432de30034cb58c9f4062e173c876 : RETRY_LIMIT in 9m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/88345523ad704aca82cd028425e95edc : SUCCESS in 2h 45m 25s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ba1d69e6af8d43e3907d18a2e5fddb4e : SUCCESS in 1h 31m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a26bd3ee8a344ac2a6f3c0a5d39f8459 : SUCCESS in 2h 32m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/37b8498433a44ed5b4c4f0df04e9698e : SUCCESS in 2h 56m 17s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/be884fff9b484b47a965069602bba56a : SUCCESS in 2h 06m 22s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d35d2c7f27784a949c8e2159cb404fe5 : SUCCESS in 1h 57m 55s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4423b4c52f8a4e79bd3a46b891d3aa42 : FAILURE in 2h 14m 54s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/65c3208b96a94810ba7c5db518d74c7f : SUCCESS in 2h 50m 01s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/190b719d4cb549b6bb3c2cee6dddeaa6 : FAILURE in 1h 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d161e6b4a805fc6f527bef9ba66dd63ddde545e","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-08-22 06:59:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a22a5b6f1a0d2d2531da0d535d8e682fcd76dc9f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 10:04:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f3aa4e5/ : SUCCESS in 2h 41m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/db33359/ : SUCCESS in 3h 04m 19s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/fa5ad79/ : SUCCESS in 19m 55s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/91/675291/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/7f1fbbc/ : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e1f441424c87df5b551b3f279d4b967dcbc7f839","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:23:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4af91587fb60441bba71fc5c391026be : SUCCESS in 5m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c4c02d3ab80e4fdd886a8d8e8a7b34a7 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e72503789124588acdf7b8ecad3ce92 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/83554d299aec4d7eae84c7580022d8fb : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66c35f15799747e486837ca0895e6b22 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5dc399e930b24ae3b80eb16967cd3a8b : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00996c640c484965a3d2a37cac587227 : SUCCESS in 5m 16s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/af7ff8bd841b4d9faf0282eacd036210 : SUCCESS in 23m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/db1d5e7de8fc40a88b11809a2a7a5bbc : SUCCESS in 2h 41m 36s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/97860a3d39954a69b134f6766364ba50 : SUCCESS in 1h 56m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/776d9b9a95944d20a711e7f104fa1eef : SUCCESS in 2h 39m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/faf0681f8dfb4684b90737698183bdc7 : SUCCESS in 2h 38m 30s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/de705cddd93540bcbdc097cbfc3eed59 : SUCCESS in 1h 56m 34s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9e5b07e034c447bf89b6956434125c48 : SUCCESS in 2h 31m 52s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/6ad7bfe9eef242d69bb3ff32dc183d0b : FAILURE in 2h 24m 19s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d0c1b1e34f01406893e646ce8a6edb39 : SUCCESS in 3h 03m 16s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f3a0bf5409d4414386174d8b663501f5 : FAILURE in 1h 47m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"07f6ba29ffb7764a901fabe5d0f3cf5ed43c7c0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:23:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"64fbe80244f6863a787536899db643378786b74d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 13:10:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"28b5965472eb36268687eaceb4e56f6614bf7019","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 13:10:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4563166e49dc4950a9170f6e65d95472 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fddf1914471497f84684a8fb1b77e24 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d553d394748347bebe84f1403645bf98 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/244ba7835dbe4f29bb9becf1c0a88818 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38e9765f45f54f188a3be8f781caa871 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b6febfa4b740998f2cfe98bea9992f : SUCCESS in 5m 14s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/3dc971f357884586915e4148ca9ca5a6 : SUCCESS in 23m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f351da85281c49d79644f0c16189f42f : SUCCESS in 2h 40m 08s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ef8af28659814ddd980589c299ce19b2 : SUCCESS in 2h 31m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4a8256b899214337be28ec2c177efe9b : SUCCESS in 2h 32m 46s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/898531b19d4f463fa788c61344c71556 : SUCCESS in 1h 45m 54s","accounts_in_message":[],"_revision_number":3},{"id":"277fac6da15e1553b006f9d7a61db55609e38b67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 13:11:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/283815aa206940b2abee1e89ff82f7e0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3},{"id":"da3e214cf9158d9e0752845553774ec776c0ccd7","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-09-03 08:26:39.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 781d1051b4abc2cb3dffd5361c0c2c2358ba97eb","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"768d20cf2a3cad03da7333871e71b2fbc6d93eb9","revisions":{"1ed5cb117bba404944dace1edf9a2a0327e291f4":{"kind":"REWORK","_number":1,"created":"2019-08-08 10:46:36.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/91/675291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/91/675291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/1"}}},"commit":{"parents":[{"commit":"44c76f8c386e70eba85ea17ea665141db081024b","subject":"Merge \"Disable telemetry if it\u0027s not enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/44c76f8c386e70eba85ea17ea665141db081024b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-08 10:39:03.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-08 10:46:10.000000000","tz":120},"subject":"Exit listing validations with a proper return code","message":"Exit listing validations with a proper return code\n\nWhen listing validations through the CLI, the return code was always 0.\n\nThis patch adds a proper return code management to validations listing\nCLI.\n\nChange-Id: I3171adabc639789b0ea805a140e4b1220ed4c93a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1ed5cb117bba404944dace1edf9a2a0327e291f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1ed5cb117bba404944dace1edf9a2a0327e291f4"}]},"branch":"refs/heads/master"},"cd91e99df18c1a1500bc30508e33cd3f94079689":{"kind":"REWORK","_number":2,"created":"2019-08-13 08:41:29.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/91/675291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/91/675291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/2"}}},"commit":{"parents":[{"commit":"a8e81658532dcf06ab29ff35554c1f76407abdc3","subject":"Merge \"Allows to disable in-flight validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a8e81658532dcf06ab29ff35554c1f76407abdc3"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-08 10:39:03.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-13 08:41:16.000000000","tz":120},"subject":"Exit listing validations with a proper return code","message":"Exit listing validations with a proper return code\n\nWhen listing validations through the CLI, the return code was always 0.\n\nThis patch adds a proper return code management to validations listing\nCLI.\n\nChange-Id: I3171adabc639789b0ea805a140e4b1220ed4c93a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cd91e99df18c1a1500bc30508e33cd3f94079689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/cd91e99df18c1a1500bc30508e33cd3f94079689"}]},"branch":"refs/heads/master"},"768d20cf2a3cad03da7333871e71b2fbc6d93eb9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2019-08-14 10:59:31.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/91/675291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/91/675291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/91/675291/3"}}},"commit":{"parents":[{"commit":"055443d4e224bd141ec217c4dd0af5fbc9317f3b","subject":"Merge \"Remove download_ansible_playbooks, a unused util function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/055443d4e224bd141ec217c4dd0af5fbc9317f3b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-08 10:39:03.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-08-14 10:57:19.000000000","tz":120},"subject":"Exit listing validations with a proper return code","message":"Exit listing validations with a proper return code\n\nWhen listing validations through the CLI, the return code was always 0.\n\nThis patch adds a proper return code management to validations listing\nCLI by raising an exception.\n\nChange-Id: I3171adabc639789b0ea805a140e4b1220ed4c93a\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/768d20cf2a3cad03da7333871e71b2fbc6d93eb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/768d20cf2a3cad03da7333871e71b2fbc6d93eb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
