)]}'
{"id":"openstack%2Fpython-tripleoclient~675817","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Fstein~I993fc09fd571fb93d78ffdbecf1823e806f744f4","project":"openstack/python-tripleoclient","branch":"stable/stein","topic":"bug/1838540","hashtags":[],"change_id":"I993fc09fd571fb93d78ffdbecf1823e806f744f4","subject":"Pass deployment_options to config download","status":"MERGED","created":"2019-08-12 06:11:15.000000000","updated":"2019-08-23 05:21:00.000000000","submitted":"2019-08-23 05:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675817-1566537569051-aa19dd0d","meta_rev_id":"fde5b5b264bb423fe29b17738813de6d03f0347c","_number":675817,"virtual_id_number":675817,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-23 05:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-22 22:44:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 22:12:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-22 22:12:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-12 06:22:24.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-12 15:41:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-22 17:38:53.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-22 22:12:38.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 22:44:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 05:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69076a03452516a6dbe3930831f0b83af3465ad6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-12 06:11:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e2f229c84b6b836a0b7d3735babbd65c21a76a5","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-12 06:22:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e6834a25ac688093029a3e4f09a5e655d1b5c9a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 09:43:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/17/675817/1/check/openstack-tox-lower-constraints/02abf06/ : SUCCESS in 5m 48s\n- openstackclient-check-plugins https://logs.opendev.org/17/675817/1/check/openstackclient-check-plugins/b63594a/ : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/17/675817/1/check/openstack-tox-pep8/075ef8c/ : SUCCESS in 3m 36s\n- openstack-tox-py27 https://logs.opendev.org/17/675817/1/check/openstack-tox-py27/16ca15d/ : SUCCESS in 5m 17s\n- openstack-tox-py35 https://logs.opendev.org/17/675817/1/check/openstack-tox-py35/7ba91ad/ : SUCCESS in 2m 55s\n- openstack-tox-py36 https://logs.opendev.org/17/675817/1/check/openstack-tox-py36/b00e117/ : SUCCESS in 5m 00s\n- openstack-tox-py37 https://logs.opendev.org/17/675817/1/check/openstack-tox-py37/3554142/ : SUCCESS in 5m 34s\n- openstack-tox-docs https://logs.opendev.org/17/675817/1/check/openstack-tox-docs/29f1d61/html/ : SUCCESS in 3m 45s\n- tripleo-buildimage-overcloud-full-centos-7 https://logs.opendev.org/17/675817/1/check/tripleo-buildimage-overcloud-full-centos-7/f2ab4c3/ : SUCCESS in 20m 13s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-containers-multinode/36244d6/ : SUCCESS in 2h 10m 16s\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-undercloud-containers/7d8d4cf/ : SUCCESS in 1h 40m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/859bb77/ : SUCCESS in 2h 32m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f6707ad/ : SUCCESS in 2h 45m 56s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-standalone/e10f334/ : SUCCESS in 1h 55m 37s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/17/675817/1/check/tripleo-ci-fedora-28-standalone/0896e6e/ : SUCCESS in 1h 48m 06s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c4c9cee/ : FAILURE in 1h 32m 55s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-scenario007-multinode-oooq-container/46e871e/ : SUCCESS in 2h 31m 23s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-scenario008-multinode-oooq-container/06edec7/ : FAILURE in 1h 34m 33s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/17/675817/1/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/aa564b0/ : SUCCESS in 2h 23m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1eb73856dd618fb6e3d7ffee59e284bd3463c794","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 09:47:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/66356fb47aca48b8b33b7c8591bd044c : POST_FAILURE in 3h 34m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze02.softwarefactory-project.io:7979/d90d8c9bb0304a03b504912b39445993 : POST_FAILURE in 3h 34m 39s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/17/675817/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/2f4275a/ : RETRY_LIMIT in 6m 19s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/17/675817/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/97a389a/ : RETRY_LIMIT in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab8ae4a5bce67aa8994bfa43a08fc3536d700de9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 15:41:07.000000000","message":"Patch Set 1: Code-Review+2\n\nWe\u0027ll want to wait until tripleo-common has promoted in containers before landing this","accounts_in_message":[],"_revision_number":1},{"id":"f40db1a587a0cafa9b85b3a9cd717369219e5dd7","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-22 17:38:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"adb339a0f99c54db84b3d17444876d2ff9a62e0a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 18:06:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad12e777728af4c9c6c72f6ab1a07b250c28694e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 18:31:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/675817/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/79c4ba1/ : FAILURE in 24m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/675817/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f06491a/ : FAILURE in 23m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0d7253ed64b3c09fa964d3163b747c22757a7c9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 18:51:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49bda3a8f07b4546848880c2f260f9ac : FAILURE in 5m 05s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f5db6df3320f40ff9189d6fb27c1fb63 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5286fd8aed4bdba768db1794a9df11 : FAILURE in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a995a00cb374f13a802b8fe213e4cfd : FAILURE in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/26b78922316d4c259e340c5ca9f22ef9 : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2720e52d404a86ad06d5c1231060bd : FAILURE in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7725c18a48e4adaa1cc711608e628c5 : FAILURE in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aaa9ecb598f43a2953a66b98a97095d : FAILURE in 3m 59s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/5f303e0624914178b7e8d503ff532af4 : SUCCESS in 23m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4d34b8583dcf4297b8c72444896b3034 : FAILURE in 32m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d9508144b8b4446ebaceb8acc17b1476 : FAILURE in 28m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8ab86e99503db4d266071cbbd3cb36f92e56f12","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 19:16:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"668673ae5255a4b25d09b7a86b98253d762d5701","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 22:04:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6212c1a0acc452d80043777a4fe474d : SUCCESS in 6m 01s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/dd421fd5c9e34f44be49d4ef075088c5 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690b8e3c714b46889b87d08bd2654e22 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b29542c698804376a0d7119e0b4049ec : SUCCESS in 3m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/688800059704425cba3cba801fa710b7 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b4318286f8f40f4a31d3ef1ea59796a : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8660ba26f774d62881009de9c200533 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb38239f32040e8bf566bf2322acf28 : SUCCESS in 4m 53s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/f227a59754ac4374bc4786a7b7068399 : SUCCESS in 22m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/db5cbe60d01b4aafad05da7bb8bb596e : SUCCESS in 2h 20m 56s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b83d810e2354e23baa87a3eef584e2e : SUCCESS in 1h 48m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ce424aebad944303a615bba7aafe06cf : SUCCESS in 2h 19m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/868bd97837424cc5b1a40bef7d180a98 : SUCCESS in 2h 25m 31s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0d97220d2a9142998e34578a26393ca9 : SUCCESS in 1h 51m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7f2dda431ce84a2d95eff703ae6a12aa : FAILURE in 1h 38m 42s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f6c8e8a6ed5d4ab4bd93b5f49c8419f4 : SUCCESS in 2h 27m 21s (non-voting)\n- tripleo-ci-centos-7-scenario008-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c2b904f597bf4558a8e6d29cb0e75334 : FAILURE in 1h 39m 08s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/fddbd2b837404e1c99460c710c581f52 : SUCCESS in 2h 07m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"872b6323646c84926ad36215dd7945bee44a7d06","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 22:12:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8a1c87d28b7a51fab0682d4c0f893a83c9a314aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 22:13:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bd17d4057dace3fb1c908c52e7a79dd2299e6c40","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 22:44:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/675817/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6cd4ffe/ : FAILURE in 3h 04m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/675817/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/172a482/ : SUCCESS in 3h 26m 55s","accounts_in_message":[],"_revision_number":3},{"id":"4800483182fdf90b6ba60450d4bbb1ba829536a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:19:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1183e2a7cadf40dca65995ab3bc22531 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f8ab4993cc4d0ebc06775f7e43c8f1 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9496a8412234492b3714b0cde1f8ea6 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7e6f722bf7cf466387e0b2c52aa65eba : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b8f6286b5cd48c7bca81c10bda97346 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7019a0bcfcc040dbb6e20fbb9266c338 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/361ffbd706d949ab97ef27d5a0a867f0 : SUCCESS in 4m 04s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/3aa70778a61f4323bc4fecb245d22df1 : SUCCESS in 24m 35s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/81b96a031b8347ee82a75deb4ab2e9dc : SUCCESS in 2h 13m 59s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/4903bf7a736745958b4a6e4257dcc71b : SUCCESS in 1h 45m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9ae574c070e846a39eb1c0fbde902301 : SUCCESS in 2h 28m 16s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8449860c88a341c186681695561bc081 : SUCCESS in 1h 47m 24s","accounts_in_message":[],"_revision_number":3},{"id":"2e52fbc7804aa4bb48c7ab6bbcd172e77a77c210","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:19:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bc90e0a76c9da5e832003fe70c7b0b15b84af8e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:21:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782643ca278f4d52bd66b4d8d6c9a08f : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"17079abbbbad82322f5326d4ea50c474349eefae","revisions":{"9c421958ea5409b38e507010e47b0c778ea27c4e":{"kind":"REWORK","_number":1,"created":"2019-08-12 06:11:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/17/675817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/17/675817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/1"}}},"commit":{"parents":[{"commit":"f83a1ed916372d519a82184d5e0809c5e65e421c","subject":"Merge \"Add gathering_policy parameter to run_ansible_playbook function\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f83a1ed916372d519a82184d5e0809c5e65e421c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 16:09:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-08-12 06:07:46.000000000","tz":330},"subject":"Pass deployment_options to config download","message":"Pass deployment_options to config download\n\nhttps://review.opendev.org/673862 allows config_download\nworkflow to take deployment_options as a input, this patch\npasses it to the workflow so ansible_python_interpreter\nis set correctly in generated inventory and thus used\nduring ansible deployment.\n\nDepends-On: https://review.opendev.org/#/c/675804/\nCloses-Bug: #1838540\nChange-Id: I993fc09fd571fb93d78ffdbecf1823e806f744f4\n(cherry picked from commit ac7ae9e8d638676196c4897a30c978d51a091900)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c421958ea5409b38e507010e47b0c778ea27c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c421958ea5409b38e507010e47b0c778ea27c4e"}]},"branch":"refs/heads/stable/stein"},"85e3852c7421a062751f335cc0edb72a4d795208":{"kind":"REWORK","_number":2,"created":"2019-08-22 18:06:13.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/17/675817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/17/675817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/2"}}},"commit":{"parents":[{"commit":"4d0c4ab51cea49eb8e620bd2f4439535e263c5d9","subject":"Merge \"Run Validations with ThreadPoolExecutor\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4d0c4ab51cea49eb8e620bd2f4439535e263c5d9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 16:09:38.000000000","tz":330},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 18:06:06.000000000","tz":-360},"subject":"Pass deployment_options to config download","message":"Pass deployment_options to config download\n\nhttps://review.opendev.org/673862 allows config_download\nworkflow to take deployment_options as a input, this patch\npasses it to the workflow so ansible_python_interpreter\nis set correctly in generated inventory and thus used\nduring ansible deployment.\n\nDepends-On: https://review.opendev.org/#/c/675804/\nCloses-Bug: #1838540\nChange-Id: I993fc09fd571fb93d78ffdbecf1823e806f744f4\n(cherry picked from commit ac7ae9e8d638676196c4897a30c978d51a091900)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/85e3852c7421a062751f335cc0edb72a4d795208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/85e3852c7421a062751f335cc0edb72a4d795208"}]},"branch":"refs/heads/stable/stein"},"17079abbbbad82322f5326d4ea50c474349eefae":{"kind":"REWORK","_number":3,"created":"2019-08-22 19:16:33.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/17/675817/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/17/675817/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/17/675817/3"}}},"commit":{"parents":[{"commit":"4d0c4ab51cea49eb8e620bd2f4439535e263c5d9","subject":"Merge \"Run Validations with ThreadPoolExecutor\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4d0c4ab51cea49eb8e620bd2f4439535e263c5d9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2019-07-31 16:09:38.000000000","tz":330},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 19:16:32.000000000","tz":0},"subject":"Pass deployment_options to config download","message":"Pass deployment_options to config download\n\nhttps://review.opendev.org/673862 allows config_download\nworkflow to take deployment_options as a input, this patch\npasses it to the workflow so ansible_python_interpreter\nis set correctly in generated inventory and thus used\nduring ansible deployment.\n\nDepends-On: https://review.opendev.org/#/c/675804/\nCloses-Bug: #1838540\nChange-Id: I993fc09fd571fb93d78ffdbecf1823e806f744f4\n(cherry picked from commit ac7ae9e8d638676196c4897a30c978d51a091900)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/17079abbbbad82322f5326d4ea50c474349eefae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/17079abbbbad82322f5326d4ea50c474349eefae"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
