)]}'
{"id":"openstack%2Fpython-tripleoclient~682763","triplet_id":"openstack%2Fpython-tripleoclient~master~Ibf7d52aa1482936dc5f2340b115b0730544b7b3c","project":"openstack/python-tripleoclient","branch":"master","topic":"space","hashtags":[],"change_id":"Ibf7d52aa1482936dc5f2340b115b0730544b7b3c","subject":"container_image: avoid extra space in logs","status":"MERGED","created":"2019-09-17 22:57:10.000000000","updated":"2019-10-01 11:22:40.000000000","submitted":"2019-10-01 11:21:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682763-1569928888798-a241494d","meta_rev_id":"13fb0fa1d0ff3d4df1da79d20a7f901bb4eccb3f","_number":682763,"virtual_id_number":682763,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 11:21:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-01 06:03:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-01 01:28:15.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2019-10-01 01:28:15.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-10-01 01:28:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-01 01:28:17.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 13:56:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-25 17:17:25.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-09-25 21:45:06.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-10-01 06:03:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 11:21:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ed360df73e1898e8b899e6575a7ade645111662","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-17 22:57:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2717e2fe0b15cb75e113fda982a91e9f85491c5e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 02:36:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/682763/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6594521/ : FAILURE in 3h 38m 34s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/682763/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/d7db5d7/ : SUCCESS in 19m 49s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/682763/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/c724c9b/ : SUCCESS in 17m 57s (non-voting)\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/63/682763/1/openstack-check/tripleo-build-containers-rhel-8-master/7507bcb/ : SUCCESS in 37m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"036031c1311f1c4ddc501ba15339a4bc18fa4069","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 02:41:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d5a7b32a2b840a0807f87f696be85e3 : SUCCESS in 4m 59s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9032d0fdc16141dda7cbe4a62e40de9b : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a18b865a19c47fcad61fb1c3d6170ed : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c7fdddb8d17473fbe4e2f0e6786349a : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e20831dee2c40458125b701d3b0cff1 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2af01ad01b8b42a8a2a0914abaf52e10 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab27aadb125f4070bef9bdfb559e2ada : SUCCESS in 5m 03s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/415d5ee82d794bc8835a00168ca9da6e : SUCCESS in 23m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/12b6423c049b4f728ebe7e17e83a49c6 : SUCCESS in 52m 41s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/171663193270435097de25f7438e352e : SUCCESS in 1h 14m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2ec14ba280214a6aa70ccf11a440b520 : SUCCESS in 2h 57m 06s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fcb256ef2b5143e6bb1d5d67cecd185b : SUCCESS in 2h 03m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0be536f48f62425da116fb830fc13686 : SUCCESS in 2h 27m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/471ee030e156488fb55032b8486e5e9b : SUCCESS in 2h 46m 38s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0ec0483f75dd4f9391a8cbc1d764afc9 : SUCCESS in 2h 33m 44s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/84d12ce3ed4344c493abb87393ab381f : SUCCESS in 2h 40m 35s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cee8cbdbf8434fd3b2606cd394b6fb52 : FAILURE in 2h 10m 13s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d713fe29b0174ec8a433cbafe3797e02 : TIMED_OUT in 3h 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ef163e889ed738a4e29bd54169c6e2715319e97","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-23 13:56:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bde39d7bd922a1ffda1ab7f0bea2b399c4bef38e","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-09-25 17:17:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bfcfbf156cfc7a045030beda258e7a58a3104cc8","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-25 21:45:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1fc0cbc9990498687040ac01dd8868be2ce58df7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-30 14:29:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ce6b9f398acbea729503a2c6ca2d09b244fcd34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 14:29:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40fe0db07e82b968247740f061db9980cc42d45f","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-30 16:42:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"5d751d7de4b222b7508c97d73accb7ee2bb5afce","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-30 16:42:48.000000000","message":"Restored\n\npy27 failed in the gate https://0af6baf55c6d82500c87-3a8bdf82174e44fa5c863ba9b47d4ff0.ssl.cf5.rackcdn.com/682763/1/gate/openstack-tox-py27/b0fe4c6/","accounts_in_message":[],"_revision_number":1},{"id":"2c020480c4df2e31a7796a8927581915cabe1542","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-30 16:42:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"94cfcf1df6d630e50858ace67c3d4c2cd8b98002","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-09-30 16:42:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac4eb04fcdd2e5042e675d4264a5b4f026341d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 19:39:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5e567c39c134065be638c729b8a2f91 : SUCCESS in 6m 04s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7d05eca71f324a8cb58436822425b7f1 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41d8822ddff847fcbeac2881691daee7 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c13e833197e4e5eaa808f7a18e6b748 : FAILURE in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7abd9384a40941928616805449dcc6b6 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b05f5eee09a34378a7618124ee71d3d4 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131fe2bd44144c5ca751e8a3616a5888 : SUCCESS in 5m 14s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/9f1f99b716624114b8a581d80353696f : SUCCESS in 26m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/07c80468f9d34ad7a1b8499e033ec3a5 : SUCCESS in 44m 52s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/a67a66cd6a9e49be885d863362ca8182 : SUCCESS in 1h 17m 43s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d893769404bb41bc920718794c45b568 : SUCCESS in 2h 07m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fcf3078c7afd46c5bac8f15fabd22ef3 : SUCCESS in 2h 42m 22s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a31516476f169862fb7ba7306bc087494645c411","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-30 20:25:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/682763/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6be53d6/ : TIMED_OUT in 3h 41m 39s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/682763/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/f250682/ : SUCCESS in 21m 35s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/682763/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/b693848/ : SUCCESS in 18m 01s (non-voting)\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/63/682763/2/openstack-check/tripleo-build-containers-rhel-8-master/878b1b8/ : SUCCESS in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"554fefd0e47db9f3730fb07328838665b17f484b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-01 01:28:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"39666bf2eed2ba3b95ccb8f94ac2cf3547461b4c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-01 01:28:17.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dbf8d7b22d453d00bdc2b29998737fa6afac2b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 04:53:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dceb572270b44c7f896b420b40b64f64 : SUCCESS in 4m 31s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7c53c3c8e26f490881467129b0e60a9b : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce56d404711e4542a3ae90f5b63e907a : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2185f19fe1f7448e973c9d3ba38772a6 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490e521f78424bdbb7045aa4ce9dd6d4 : SUCCESS in 4m 41s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/582a3f3214914adc895e6a73604ec018 : SUCCESS in 19m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/355ce68559d641bfb157b75bfbf58fb8 : SUCCESS in 45m 24s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/26a630d170fd4b298539324ed8747c29 : SUCCESS in 58m 43s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/965d9a6505fa44a2b36465c00828969d : SUCCESS in 3h 11m 44s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d1a58def669e4692b89db55a134dad8f : SUCCESS in 1h 46m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/34b2f68a416d41f1941963b09e3605d4 : SUCCESS in 2h 54m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8ca2049cc8b24eefbb0a5e9892acf3d7 : SUCCESS in 3h 10m 31s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/a4d7bb65cd96407db4d1c641e78fbf24 : SUCCESS in 2h 09m 46s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b1e4b66b38654315bbd983dea49736eb : FAILURE in 2h 06m 55s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2c40c525f193497ba28b3fc68b5ac515 : FAILURE in 2h 13m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a85d045f564ad7b5665c8546af5deb : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bd66fd3e7274bc49ed390581e5199ee : FAILURE in 3m 21s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/52ba486ffcc1480fb13c7df0089928ca : SUCCESS in 3h 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"756b6969f92289ac07d3e512ec8250de7391b741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 04:54:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"900b258ab406640e1375520165778e5cf799064b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 06:03:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/682763/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fda0e5a/ : TIMED_OUT in 4h 32m 12s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/682763/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/3eddf5c/ : SUCCESS in 20m 48s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/682763/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/d1a7c9f/ : SUCCESS in 18m 24s (non-voting)\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/63/682763/3/openstack-check/tripleo-build-containers-rhel-8-master/9c049ee/ : SUCCESS in 39m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37d698f0fc00b5099e0de92e4863df26ffecc936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:21:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a0fb88d3c134f33841a18beba2c3fe8 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f21a04905fc04df3bea2ae4b8eaa238f : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/200a13a964e04e3883f21bd008c77899 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e67ab7cbec43b385b4c18141851005 : SUCCESS in 5m 43s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/b85ec10913634cc6985458feee91e94c : SUCCESS in 25m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fe28add2704d43ae905c543221894f08 : SUCCESS in 1h 04m 50s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/7e9391dab680494685d8109a26f8551a : SUCCESS in 1h 20m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f066fce1c65c4a1088e924d267fe3584 : SUCCESS in 3h 18m 09s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0a1ad1c7f9654998b66386f82afd0809 : SUCCESS in 2h 34m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0d6499fc61fc493393c4b08708d8a708 : SUCCESS in 3h 14m 16s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e53a3c1506e44346bf2b6484633a6cab : SUCCESS in 2h 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ef134612374d6aba892f4e10895f03 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"c44bae9fd78ad69f793a05bf0c0b995c4e275e4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:21:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"58f6894f8234ab5dccad02a792551ea17b22125e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 11:22:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af85d5109c52425095f698956d771cc4 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5f22dfdbf846b58b81f12fce3b09556d32b8315a","revisions":{"358a9725d19905aa4be5f5f3f16bca7ed9136ff8":{"kind":"REWORK","_number":1,"created":"2019-09-17 22:57:10.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/63/682763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/63/682763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/1"}}},"commit":{"parents":[{"commit":"c28df729e7ec08091f0de5374214616b5d93a207","subject":"Merge \"Rename \u0027undercloud-\u0027 to \u0027\u003cstack_name\u003e-\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c28df729e7ec08091f0de5374214616b5d93a207"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-17 22:55:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-17 22:55:42.000000000","tz":-240},"subject":"container_image: avoid extra space in logs","message":"container_image: avoid extra space in logs\n\nIt\u0027ll avoid this logging:\n\nOutput env file exists,                                  moving it to backup.\n\nChange-Id: Ibf7d52aa1482936dc5f2340b115b0730544b7b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/358a9725d19905aa4be5f5f3f16bca7ed9136ff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/358a9725d19905aa4be5f5f3f16bca7ed9136ff8"}]},"branch":"refs/heads/master"},"2c41be18c3d8b49f2b3aeee39b894b14396a2274":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-30 16:42:54.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/63/682763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/63/682763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/2"}}},"commit":{"parents":[{"commit":"782cd28ca98adf0733fde190930532a146a879db","subject":"Merge \"Add ovn_dbs_vip to export_data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/782cd28ca98adf0733fde190930532a146a879db"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-17 22:55:42.000000000","tz":-240},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-09-30 16:42:54.000000000","tz":0},"subject":"container_image: avoid extra space in logs","message":"container_image: avoid extra space in logs\n\nIt\u0027ll avoid this logging:\n\nOutput env file exists,                                  moving it to backup.\n\nChange-Id: Ibf7d52aa1482936dc5f2340b115b0730544b7b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2c41be18c3d8b49f2b3aeee39b894b14396a2274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2c41be18c3d8b49f2b3aeee39b894b14396a2274"}]},"branch":"refs/heads/master"},"5f22dfdbf846b58b81f12fce3b09556d32b8315a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-01 01:28:15.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/63/682763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/63/682763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/682763/3"}}},"commit":{"parents":[{"commit":"d31fbf2e4cac5f30443723667b22974cf098d0c5","subject":"Disable voting on openstack-tox-py27","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d31fbf2e4cac5f30443723667b22974cf098d0c5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-17 22:55:42.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-10-01 01:28:15.000000000","tz":0},"subject":"container_image: avoid extra space in logs","message":"container_image: avoid extra space in logs\n\nIt\u0027ll avoid this logging:\n\nOutput env file exists,                                  moving it to backup.\n\nChange-Id: Ibf7d52aa1482936dc5f2340b115b0730544b7b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5f22dfdbf846b58b81f12fce3b09556d32b8315a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5f22dfdbf846b58b81f12fce3b09556d32b8315a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
