)]}'
{"id":"openstack%2Fpython-tripleoclient~685063","triplet_id":"openstack%2Fpython-tripleoclient~master~I95b8d7abc632b190ac6731393bd490bfa3aedcca","project":"openstack/python-tripleoclient","branch":"master","topic":"healthcheck","hashtags":[],"change_id":"I95b8d7abc632b190ac6731393bd490bfa3aedcca","subject":"(actually) disable the inflight validations by default","status":"MERGED","created":"2019-09-26 12:55:28.000000000","updated":"2019-10-01 15:57:13.000000000","submitted":"2019-10-01 15:55:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685063-1569945347815-01385176","meta_rev_id":"f88afc9b28aac299150adc022f5b114511525744","_number":685063,"virtual_id_number":685063,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 15:55:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":1,"date":"2019-10-01 08:32:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-01 01:27:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-10-01 01:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-10-01 01:27:52.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-01 01:27:52.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":1,"date":"2019-10-01 08:46:29.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 14:40:34.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-09-26 19:04:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-09-27 08:18:19.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-09-28 10:45:40.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-09-30 21:48:06.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-10-01 08:32:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-01 08:46:29.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-10-01 15:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68ae5b386c6ad80a9062e8789de40d18d832b9e7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-26 12:55:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c278cd57a0f39e11d89312e386faede440b364b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-09-26 12:57:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e0c924a71b9c25ce2574cc877b404bc7620093bb","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-09-26 12:58:55.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks!","accounts_in_message":[],"_revision_number":1},{"id":"2dbec9db7c13881bf16666c9818b213467a335f3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-26 13:15:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"26c2a5325455526668883d1c56581994e08af382","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-09-26 13:57:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"478c53f6f722d2d72883fd140c7dd2ff665c9eb2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-09-26 14:40:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e47c9265b3566ed3a74b67a8108039707bc90fec","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-26 14:42:27.000000000","message":"Patch Set 3: Code-Review+2\n\nRunning Ansible deploy tasks: ansible-playbook-2 -i /home/vagrant/undercloud-ansible-1ixbE1/inventory.yaml deploy_steps_playbook.yaml --skip-tags opendev-validation","accounts_in_message":[],"_revision_number":3},{"id":"e83c4e78e5c4fb18325bdce05622aec8d2b65d71","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 17:14:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b7248f0/ : SUCCESS in 3h 16m 35s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/9679e87/ : SUCCESS in 20m 22s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/947c39b/ : SUCCESS in 18m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"80355ba4fc6077becb17ce3598d97e8e405ca848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 23:11:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e46c6a15c3c541858c1cda2e5471ddeb : SUCCESS in 4m 59s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/8207537a5dfe464094b5403119f10e74 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5eab77238e340e2a4df156eadaffd08 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b004d506a234aa39f37ad618a0351a6 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cea85e3f40b04b249bf9fe10c54ca387 : SUCCESS in 2m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a0964bc09d84191965d726bf965d248 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd874d6d82644b9ab11544b83b350694 : SUCCESS in 4m 28s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/c3509abe513944e791f23428d9d877d4 : SUCCESS in 34m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bd53c6478d1f40d79636706d78f9bcd0 : TIMED_OUT in 4h 00m 09s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c4346167ffd040df8b84378795a5a933 : SUCCESS in 2h 45m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b2da731ecd0746d0b2c80510a2fe0e0f : SUCCESS in 3h 06m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/94cdbf91f4be46e3a88d8edd9732ed42 : SUCCESS in 3h 45m 12s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/e83665c538a24cf4a7986505150c7b8a : SUCCESS in 2h 36m 35s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/9844a5141fde4ee1b0afcf52b9b330a3 : FAILURE in 2h 38m 13s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/764010ad81e549228899e412a6140a0c : FAILURE in 2h 55m 28s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d37a370748184813bc8e4c6e12b7cc13 : SUCCESS in 3h 01m 32s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/85bb8a606b0341949cd44a199c619fe4 : SUCCESS in 2h 33m 26s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/805be02cc1a646858a5015e7449e548b : SUCCESS in 1h 45m 29s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ea64a89f9c84494f9e243fa76af459fe : SUCCESS in 2h 38m 48s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/1b15620100ce428bbdecb66af994d70e : SUCCESS in 2h 13m 36s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/1d8005ca2d91492aad64c6d5cb8779e8 : FAILURE in 2h 23m 48s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bae24125b98f4388986e59966d971784 : TIMED_OUT in 3h 48m 17s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/478f9c46bc9f45368fb462871b71487a : FAILURE in 2h 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"049467aafe4666ce0127249d7dd4faa453c8f6d7","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-09-27 06:00:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0d1f90ed24ec69ed5b585ffa713ee5d6aa18a860","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2019-09-27 08:18:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"355efade48eff18c3f80d9ef3a253bcbc3d80056","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 09:12:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/795da8b/ : SUCCESS in 3h 11m 31s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/a51df2d/ : SUCCESS in 20m 48s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/a70f99b/ : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"faf0af95ecb3f13e1128343283e73f2cd3d5fb19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:09:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acc0d7dd97c14ab2961942faa58e56af : SUCCESS in 4m 16s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/dbefd1ab78064d6c9468c64059753f4e : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fcefa7961e14ab18140c6633c6fc613 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3c29d60cbdf41dc94e49f77ab05eebd : FAILURE in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2465f63d489b45bfadd14a14fcd52970 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d25d728ef90b4885b25d744d2dff32d6 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa571c54fc14b2d98d8334ed49f965d : SUCCESS in 4m 43s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/b546a57cc3d740c786b56bfd3772f4a2 : SUCCESS in 27m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/95503704a3454501bc44ceadb9e841f4 : SUCCESS in 2h 23m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/0347766e3c9a4560821edf410b092b4b : SUCCESS in 2h 31m 17s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/25fd04e1b52b429db5e0d0cfa6efe6a7 : SUCCESS in 2h 23m 07s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c58928d451d83681300923775bd8e68813ed9a6f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-09-27 10:23:41.000000000","message":"Patch Set 3:\n\nrecheck\n\ntox-py27 should be passing... At least it\u0027s all green locally.","accounts_in_message":[],"_revision_number":3},{"id":"7002bbce8d1b94c5a702e862646d819c3c22060f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 13:44:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9e2aaec06ef4d69b24be85ed04a9e0d : SUCCESS in 5m 12s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/96ba3cf53c4e4a93bca39ea2ec399cb3 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4128535412142eaa345288e435909b9 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/005fc4b433fe40f898b24dfb8db9ff75 : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80f739f427264c2c95e59009b3b773e6 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9a136881c0747c1890b8871e0b45852 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0762b4179b084e5aa4e0984a087dafc5 : SUCCESS in 4m 46s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/c75bd61ec993479186333bf0aa2c23f4 : SUCCESS in 22m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/16f08e6548004aadb0949755b398fd30 : SUCCESS in 2h 29m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/f300eee240294d1daec33bc84f160de0 : SUCCESS in 2h 39m 30s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/ad54221ec96c4791886ec471269cb1a9 : SUCCESS in 2h 16m 20s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d27a35979659e36340c2a811404e366b169254e8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-27 13:54:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f113764/ : FAILURE in 3h 29m 10s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/1b1e411/ : SUCCESS in 20m 41s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/685063/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/5473d20/ : SUCCESS in 17m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"67d7dc3aa0723634ed14a78abd68b408986e8529","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-09-28 10:45:40.000000000","message":"Patch Set 3:\n\n\u003e recheck\n \u003e \n \u003e tox-py27 should be passing... At least it\u0027s all green locally.\n\nit fails for me locally as well in the same way it does in CI (fyi)","accounts_in_message":[],"_revision_number":3},{"id":"f1fda6c67afe6dae005b046ed7cb87a104ef602b","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-09-30 21:48:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8f75e2fa9ddaecef3a7a63d4b8d5795b7287d329","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-01 01:27:52.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"57d07f50abc0ddd0a3b198e41c12097dc274b7ae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 05:10:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/685063/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cf8b2a8/ : TIMED_OUT in 3h 41m 59s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/685063/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/2b75860/ : SUCCESS in 21m 05s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/685063/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/01b1117/ : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"45baa04249cd3dce36a05640e1ffe2780c07fb72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 05:16:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2deaa8d4cb154af095eabdca774ea8c0 : SUCCESS in 6m 02s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/a0beabce0f7b4d88a8aa44fd9fa0b468 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73ffbc267f534394be9b045360b47c3a : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e4259066ee54dbe9c4d1fdfa62dc178 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8935c81954e4cc189537b41b0600e4f : SUCCESS in 8m 00s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/068501754d974446851b4551c29c7355 : SUCCESS in 34m 04s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9dad77b314344ed0a970bea5a718a33c : TIMED_OUT in 3h 33m 02s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/632969bf59ef4a16b24f9e039ba40c73 : SUCCESS in 2h 07m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/45815b10a6ac4fbd88d72b340caefe3b : SUCCESS in 3h 09m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/84c763e75d214b09b01d92271e5a59d2 : SUCCESS in 3h 10m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/287351e948cf44489d568ed8be9268db : SUCCESS in 2h 27m 32s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5c824bb7e71d4d3291eb31f349c9b34a : FAILURE in 2h 00m 22s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e3b6563076984f92bfa728aa8430e5c6 : FAILURE in 2h 27m 00s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d7ef9f0f44184661a7dcb328f507a0f8 : SUCCESS in 2h 52m 15s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/964a9458cbd447e2b16853ea660b444a : SUCCESS in 2h 39m 41s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/4cc2031eed7f4107b8811f492a547218 : SUCCESS in 1h 59m 34s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1abd2301370c44518e87282c171cd7b7 : SUCCESS in 2h 47m 47s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/069d0bdf93844373b7381fdace14426d : SUCCESS in 1h 50m 31s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/7a7af07e62bb42ba98df9451bbafa62f : FAILURE in 2h 12m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b21e2925d8841869d4961bbcb31ea99 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53858a98ab6f477080983a69b45201aa : FAILURE in 3m 30s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4fdbba051e5b42079f6cbaa677000f70 : SUCCESS in 3h 30m 08s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/ad7af4eb8538475daaf1f3c0b67622b0 : FAILURE in 1h 56m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9216b0a2083e9d16560d871f8da734a8363c12c9","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-10-01 05:19:17.000000000","message":"Patch Set 4:\n\nrecheck\n\nWTF ?!","accounts_in_message":[],"_revision_number":4},{"id":"1a50b0d6149b6677f580933d44d017f33346f82e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-01 08:32:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/63/685063/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/08789fe/ : SUCCESS in 3h 12m 22s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/63/685063/4/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/9c0c504/ : SUCCESS in 19m 53s (non-voting)\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/63/685063/4/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/8597014/ : SUCCESS in 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9343633e954fbec931c9efd53a09d371a0afaf2c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-10-01 08:46:29.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"477eddb4236aec21936c220ab30dbdea48223a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:47:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2425f8bd12e4cac9476130e365d8bda : SUCCESS in 4m 45s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/882c6ea2f88146c6b4ea24e88c1d5bf8 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7f7c730e8e044b0ac1a36ea43e00b79 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f60b8b2ea81f43ffb40d57205ed885ae : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/519064ed0c58413e9e50e354eb297420 : SUCCESS in 4m 54s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/f5fc980cd9cd4a0baf68423347a41857 : SUCCESS in 22m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/fb49963cf05e42b8a6d0d8aa46bff174 : SUCCESS in 2h 54m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/234a6123a5cb41aabffa5896ca0dc560 : SUCCESS in 1h 49m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/796daac6316044268404c74a2a2b217f : SUCCESS in 3h 09m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a2711ba0202b4fb6bb17a62b072c9214 : SUCCESS in 2h 51m 07s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ac7281942fad47d1af036309f70253df : SUCCESS in 2h 08m 28s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/77fc620349874f85a8317689e8578c0d : FAILURE in 2h 08m 43s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/51f5100335224548a707cd5cd29fb0f8 : FAILURE in 1h 49m 43s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d4fa0fb0088f45d19f2a19fdd44d4a9b : SUCCESS in 2h 49m 19s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/807b1852b4934b4eb2886890f2ba449d : SUCCESS in 2h 26m 17s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/29acc295568840e8ad326793e4c7093f : SUCCESS in 1h 59m 44s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/bf68d10bee4d448b8287d5ca97ca06a4 : SUCCESS in 2h 39m 44s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2f3731b530934e72b35f10aac4c506bc : SUCCESS in 1h 34m 45s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/4694239cbc1548da8d2829833face9e3 : FAILURE in 1h 26m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d2df3554284864ba228e9186bf59da : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/256a2966e00d4e208dd09a35008b8eba : FAILURE in 3m 12s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bdf07a207d664bc3bd1be28250544e2b : SUCCESS in 3h 18m 08s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/571f655ec3244c1881e458e76dc46c17 : FAILURE in 2h 06m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b1b1f785fb984800c9a665caefd5b0c4e431e3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 08:47:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4a2464819f419edd67092dc37824f795f86e899b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:55:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"bc92a2ddf0685946fabdd56797918a52e62aa9a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:55:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1279ee97dae4ab8891d8b91e21ae269 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3945bd542e1429390904654a9e42223 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/058084ff74f6446288238e666954bdf7 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc6048803934a4f9746445a4ee955fb : SUCCESS in 5m 28s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/e65e4741cd0546858dc5b80cf946ebc6 : SUCCESS in 26m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5717865316fe4088a126187696b56e52 : SUCCESS in 3h 13m 48s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/1723ad2ad87641c8a894be38a5c808cf : SUCCESS in 2h 27m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/75a5eacd2e8544b2a2ce2bdd3e960bc4 : SUCCESS in 2h 34m 43s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c88c5a67705742eb81a61a49469fb912 : SUCCESS in 2h 07m 54s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4d8bc74d30504e3cb9b55d0ca98a4c25 : SUCCESS in 2h 27m 56s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/9a375b6c322342cbb29e073c71effad5 : SUCCESS in 2h 16m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cb6765ac3a514da2b00ba7a97fe58bee : SUCCESS in 1h 55m 34s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/62e0d3c829bd4dd89979508c7ed18fe4 : SUCCESS in 2h 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a70b0405b14be18eee92efa74b9c02 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":4},{"id":"d13917887f4884d6e67d8973bca37939882d3aad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:57:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c8a22b46f747e3b4cb6bdeb8997433 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3b9041bb45627abacf6daf820ca01550769839ee","revisions":{"64413c01e73510f9ca2c717d770cdcd764dc5941":{"kind":"REWORK","_number":1,"created":"2019-09-26 12:55:28.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/63/685063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/63/685063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/1"}}},"commit":{"parents":[{"commit":"782cd28ca98adf0733fde190930532a146a879db","subject":"Merge \"Add ovn_dbs_vip to export_data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/782cd28ca98adf0733fde190930532a146a879db"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-26 12:45:26.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-26 12:55:17.000000000","tz":-240},"subject":"(actually) disable the inflight validations by default","message":"(actually) disable the inflight validations by default\n\n- undercloud: make the logic easier to understand (validatons vs\n  no_validations), False by default. (no_validations\n  was False by default.\n\n- tripleo_deploy (undercloud + standalone): add missing extra_args when\n  running ansible.\n\nCloses-Bug: #1843555\nChange-Id: I95b8d7abc632b190ac6731393bd490bfa3aedcca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/64413c01e73510f9ca2c717d770cdcd764dc5941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/64413c01e73510f9ca2c717d770cdcd764dc5941"}]},"branch":"refs/heads/master"},"c21dda528feec908cebc412762db22219169a2bf":{"kind":"REWORK","_number":2,"created":"2019-09-26 13:15:46.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/63/685063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/63/685063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/2"}}},"commit":{"parents":[{"commit":"782cd28ca98adf0733fde190930532a146a879db","subject":"Merge \"Add ovn_dbs_vip to export_data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/782cd28ca98adf0733fde190930532a146a879db"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-26 12:45:26.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-26 13:11:44.000000000","tz":-240},"subject":"(actually) disable the inflight validations by default","message":"(actually) disable the inflight validations by default\n\n- undercloud_config: set no_validations to True by default, since we want\n  to disable them by default now.\n\n- tripleo_deploy (undercloud + standalone): add missing extra_args when\n  running ansible.\n\nCloses-Bug: #1843555\nChange-Id: I95b8d7abc632b190ac6731393bd490bfa3aedcca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c21dda528feec908cebc412762db22219169a2bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c21dda528feec908cebc412762db22219169a2bf"}]},"branch":"refs/heads/master"},"c74990e8398c52cd3e3753c9a183906ca56a12e5":{"kind":"REWORK","_number":3,"created":"2019-09-26 13:57:25.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/63/685063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/63/685063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/3"}}},"commit":{"parents":[{"commit":"782cd28ca98adf0733fde190930532a146a879db","subject":"Merge \"Add ovn_dbs_vip to export_data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/782cd28ca98adf0733fde190930532a146a879db"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-26 12:45:26.000000000","tz":-240},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-09-26 13:55:49.000000000","tz":120},"subject":"(actually) disable the inflight validations by default","message":"(actually) disable the inflight validations by default\n\n- undercloud_config: set no_validations to True by default, since we want\n  to disable them by default now.\n\n  Introduce --inflight-validations in order to get those particular\n  validations independently.\n\n- tripleo_deploy (undercloud + standalone): add missing extra_args when\n  running ansible.\n\nCo-Authored-By: Cédric Jeaneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1843555\nChange-Id: I95b8d7abc632b190ac6731393bd490bfa3aedcca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c74990e8398c52cd3e3753c9a183906ca56a12e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c74990e8398c52cd3e3753c9a183906ca56a12e5"}]},"branch":"refs/heads/master"},"3b9041bb45627abacf6daf820ca01550769839ee":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-01 01:27:52.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/63/685063/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/63/685063/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/63/685063/4"}}},"commit":{"parents":[{"commit":"d31fbf2e4cac5f30443723667b22974cf098d0c5","subject":"Disable voting on openstack-tox-py27","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d31fbf2e4cac5f30443723667b22974cf098d0c5"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-09-26 12:45:26.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-10-01 01:27:52.000000000","tz":0},"subject":"(actually) disable the inflight validations by default","message":"(actually) disable the inflight validations by default\n\n- undercloud_config: set no_validations to True by default, since we want\n  to disable them by default now.\n\n  Introduce --inflight-validations in order to get those particular\n  validations independently.\n\n- tripleo_deploy (undercloud + standalone): add missing extra_args when\n  running ansible.\n\nCo-Authored-By: Cédric Jeaneret \u003ccjeanner@redhat.com\u003e\nCloses-Bug: #1843555\nChange-Id: I95b8d7abc632b190ac6731393bd490bfa3aedcca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3b9041bb45627abacf6daf820ca01550769839ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3b9041bb45627abacf6daf820ca01550769839ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
