)]}'
{"id":"openstack%2Fpython-tripleoclient~691961","triplet_id":"openstack%2Fpython-tripleoclient~master~I27e98fa77966213fd14e9a5e3187fd4e02db0a10","project":"openstack/python-tripleoclient","branch":"master","topic":"691963","hashtags":[],"change_id":"I27e98fa77966213fd14e9a5e3187fd4e02db0a10","subject":"Write minimal paste configuration for heat-all.","status":"MERGED","created":"2019-10-29 20:09:14.000000000","updated":"2019-12-06 02:12:57.000000000","submitted":"2019-12-06 02:10:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691961-1575598244987-a858fa3e","meta_rev_id":"3259362ad09363478bd1d17df157482860af164d","_number":691961,"virtual_id_number":691961,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 02:10:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-05 21:28:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-12-05 19:42:37.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-05 19:41:18.000000000","_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2019-12-05 22:59:32.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 05:04:38.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-12-05 21:28:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-05 22:59:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-06 02:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1986ce8f35a2b4474f7650662229473767901de3","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-10-29 20:09:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad3af11183fd698b0ac8fba15f146340c55d3785","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-29 23:47:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/61/691961/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cbb7c3d/ : SUCCESS in 3h 01m 12s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/61/691961/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/75de6a8/ : FAILURE in 17m 02s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/61/691961/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/32cfe9d/ : SUCCESS in 18m 50s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/61/691961/1/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/0b58a85/ : SUCCESS in 1h 30m 12s","accounts_in_message":[],"_revision_number":1},{"id":"98b4f504c2f22a2513c02139793710308360ce5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 00:23:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/051463c49ac543e8b130471bf9662fae : SUCCESS in 4m 46s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/61724b795d0d4cf3a8e2ef36032e5a83 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/755f7156c3514705b517398f164023aa : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cdb9c60994a4cf59639c1b6ac29beeb : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1366f2488754434da30d849ef8eaecba : SUCCESS in 5m 01s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/4b9518e92fb640368189989226c2d06b : SUCCESS in 20m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/913d097a660e4f6e95abb016fb7114f0 : SUCCESS in 2h 30m 12s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bd9b7e4ca3b84934b09f0d27bf87bdc7 : SUCCESS in 1h 32m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/692571aad1914a2c8006f1da50e41f7c : SUCCESS in 2h 07m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/469d61c80ae0492c81470f33f8a476f2 : SUCCESS in 2h 09m 54s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fe6279ef2006469fabffb32033c3a85c : SUCCESS in 1h 50m 30s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/48b1757928a94e13a300434ba16c5813 : FAILURE in 1h 59m 04s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/06237f149ce34b2abbf3e74881bbf7f5 : FAILURE in 1h 43m 23s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2086479c8ccd456d832e635af43939aa : SUCCESS in 1h 50m 33s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5fcabf93d96b4ba68542fff8c39916bc : SUCCESS in 1h 37m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/987b56bbd5f1475195a7aa14eb5435bd : SUCCESS in 1h 18m 26s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d43921c0df4a4f66b235cf7e70e8535d : SUCCESS in 1h 54m 04s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/2fc18b252f144fc1905f27a6329bf14f : SUCCESS in 1h 37m 46s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/87028ba3ceaa428d9b7d020d02abc6aa : FAILURE in 59m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67d609a6a8445b4a0fb14a173dc735a : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9d5e5c21505444e8744c919ee90e31c : FAILURE in 5m 30s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d692cacd869c428eb61df33943dfb996 : SUCCESS in 2h 55m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"506a3c7105d7a4a09966d3042859c5a7f6090e57","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-11-08 18:37:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0cf610d3203129582d92fde006ff925afa857b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 21:48:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08ba9a5da5f84ad0bce2c246a9682ee0 : SUCCESS in 4m 49s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d3531461c55e4768adb9780ccc090d94 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e38439283ce497b95b225f0966f4863 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ca0097c318547b68834dd5eda8212d5 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c72fc1e8fd42cdb8ec0e8728e456e4 : SUCCESS in 4m 38s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/108a538cafa54ab68a7bda88e57b4672 : SUCCESS in 25m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/460a1f697a8143859b682fb58862b353 : SUCCESS in 2h 59m 58s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0d4bb4d9ab8046ad93a7c84593eb2774 : SUCCESS in 2h 01m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a48189e5b159480ca34664cdf48abf16 : FAILURE in 30m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/66ef13335a7b415fadf269f2cb65d1d6 : FAILURE in 28m 57s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/76016a4aad4740b38164cbcca2bb42ac : SUCCESS in 1h 51m 42s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/eb64c2e54c1e40ff85271d5654e494c6 : SUCCESS in 2h 33m 03s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c4520b2f83e44d34b1cc67adaddcabae : FAILURE in 2h 11m 31s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b45746313aae459e8bf78a249bfe19df : SUCCESS in 1h 53m 14s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/c92f56af752e4e35a179a22bebd9e3e6 : SUCCESS in 1h 42m 37s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/03c8497791d6426eb467618532e0966c : SUCCESS in 1h 30m 22s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/58dbaeff196f40cca6271cf7650c6893 : SUCCESS in 2h 12m 21s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/0b9bb1dda033446a947c468583d8389e : SUCCESS in 1h 33m 13s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/29148085616d4d898606939231ce164c : FAILURE in 1h 46m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2d20512cfc46319420992359d65dca : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd4ba36a001a4b9caa3275af2e89bd97 : FAILURE in 3m 40s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b1c664525f424589850b36babcb9ecae : SUCCESS in 2h 58m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05b18945eb4c64536201304e9e74fec51801d43f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-08 22:54:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4e8f416/ : FAILURE in 4h 15m 37s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/a98882e/ : SUCCESS in 24m 27s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/5ed4e36/ : SUCCESS in 20m 56s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/bc9d0c0/ : FAILURE in 41m 57s","accounts_in_message":[],"_revision_number":2},{"id":"853fa605813f54d87941ad3639384b1ce90ea431","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-11-11 20:21:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dcfa199df55eccee4c726062850be6efd39aab66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 23:22:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6e8966d22f44baea369f9055455ab33 : SUCCESS in 4m 10s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d2d4164fa0bf4fa69efb30a98bcafef0 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95d90159d2bb49aba52f9cef57c68f15 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c9f33478b18454399c23b7b85c62241 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d6b9200b894416bf8458ce63a18058 : SUCCESS in 4m 57s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/399c76442d664ac3b592dc21deb89df4 : SUCCESS in 26m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1c17ff04f3034525b1eaa785c4b4ef77 : SUCCESS in 2h 48m 56s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8d9250e7e2774c4897dd7eef7bf04ad7 : SUCCESS in 1h 43m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/15a99737a1e84f7392b9a031c5b03ab1 : SUCCESS in 2h 19m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f5ac093ce1b548058905dfa594b3dd43 : SUCCESS in 2h 35m 03s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/ec8d7a3d74954f8bb410bc7a83ac7c84 : SUCCESS in 2h 01m 09s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6e964680f0134a53a1070c08ca6de521 : SUCCESS in 2h 24m 25s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f85dfdafda174c46976e2ffd4eeca04d : FAILURE in 2h 14m 08s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4effa874fa974cb2a76f17af4da369b5 : SUCCESS in 2h 21m 06s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/dd921cc3a42f4e7f83df0904b2c1e86d : SUCCESS in 1h 48m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/056f47bd84a94d1bad73bd2f8d6b87af : SUCCESS in 1h 26m 08s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c8a82d1af17d478a95fe1d27b9a46641 : SUCCESS in 1h 57m 23s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/84e604c7798d44d8b4ac74f13adc31a0 : SUCCESS in 1h 39m 17s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/ed44bc3bcea241ed975fdb265ab400a2 : FAILURE in 1h 48m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c1e5163d964bbe860ea9feb4e08e7e : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b9591f34654482a91f24e11bfb875af : FAILURE in 3m 05s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e89a75d136434fffb492d2bdf87d964f : SUCCESS in 2h 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"769fda20af180b7aa41e6dc175f87f00d050a2f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-11 23:24:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fc8d353/ : SUCCESS in 3h 01m 55s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/565551f/ : FAILURE in 17m 17s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/cbf9846/ : SUCCESS in 18m 28s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/1aa4d34/ : SUCCESS in 1h 43m 55s","accounts_in_message":[],"_revision_number":2},{"id":"bff22df621a82918f6f747d965d27a4fa3de2496","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-11-14 21:45:20.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"82ccf6891fad5208546cee93505bfaa5f09e97e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-15 01:02:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f4b2f20/ : SUCCESS in 3h 15m 31s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/2378958/ : FAILURE in 19m 22s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/766419c/ : SUCCESS in 18m 34s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/61/691961/2/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/94be491/ : SUCCESS in 1h 26m 19s","accounts_in_message":[],"_revision_number":2},{"id":"7de879a731986d223d77a05fd64edc7a9a97eca9","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2019-12-04 05:04:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"676816ccd2743075169156a8d669530206852064","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-12-05 19:41:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"13fc7aec618ffbcc9ea78ee3d2c102a5699702b6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-05 19:42:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"aebb5d51ffb347403a30d7e43e0cf1e1abb5dd8a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-05 21:28:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/61/691961/3/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/f9bf877/ : FAILURE in 18m 54s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/61/691961/3/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/2cead5a/ : SUCCESS in 18m 45s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/61/691961/3/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/7f4dcca/ : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":3},{"id":"bec619a34b200cb28a9b800171f34bb911bf5bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 22:58:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f31d6132a0bf46f9983d3002b44ebe12 : SUCCESS in 6m 17s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/17d9e4541ea041f19965ac3ef6ea9c91 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5836619d9bc94437acbfa1e74e05cdb5 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05732d5e33b048f3a6c85d64e085ab1e : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beee0cbcf49c42e397a9f26cc06bda52 : SUCCESS in 6m 37s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/c40e718182e649cd85ecfa4cf4938f35 : SUCCESS in 29m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/645f250f89d145f08ebc94331adec0f1 : SUCCESS in 3h 07m 29s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7b1db283d4014c33abf9449ee5c3fd77 : SUCCESS in 2h 33m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/011936df2aa04852aff64dfe8e6fbee2 : SUCCESS in 2h 08m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b10470c4023d4992a62749295a24b68a : FAILURE in 1h 48m 10s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/119d744105df4f56b1d464a541a575ef : SUCCESS in 2h 01m 13s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/bf07737b64664d0eb3aa9c91587869fb : SUCCESS in 1h 54m 44s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/5f25eaf52d1f4f04bb8c3969d786bc7b : FAILURE in 2h 10m 42s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/aa3fed1439ee4810bfc3eef9edbc96cc : SUCCESS in 1h 48m 01s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f63543dd1d194d459af7e7f60fd9c811 : SUCCESS in 1h 54m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e3e971f485734037b0a3abeebc243d9d : SUCCESS in 1h 30m 31s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/92ca467871f44b5088fb11b79aaf1326 : SUCCESS in 2h 15m 18s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/da0c1c102ac8450f835de802de7886e7 : SUCCESS in 1h 34m 02s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/b2938f43e37e4e279e48752021986227 : FAILURE in 1h 43m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b673d02d6c4972a023bf2bce10de1d : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1592a978727443ffa0357716c07307eb : FAILURE in 4m 20s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/021a95c933634946ac4ac865cd2ed4ad : SUCCESS in 2h 49m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"42fa62327adc78d7a35d4c649b244f1c4ff5c03d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-05 22:59:32.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ab4b91247272bdd05aa806d5890f2a3ad5183fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 22:59:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a7a39a781d25a04c228f2975330a5a4fca73b060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 02:10:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/269d6881902d44ae820767ef0dc8a5bf : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dad16b85ac814e65922ad55f00d3d453 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59731dfbc73b4f91b59337c1f06a00d8 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac0ab8b02194949ae55c6a795d680d0 : SUCCESS in 6m 18s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/08ee4339001a41d084c86ea8fc397851 : SUCCESS in 23m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/68ddf0a861184e6ca81b8469b1b266a5 : SUCCESS in 2h 47m 53s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/fec500e9f75f4b6384f261d3530d3870 : SUCCESS in 2h 05m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8a2ead8ab43a4b94bea601b62d571912 : SUCCESS in 2h 17m 18s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c849a5f978ae4c31af2f33c75d392e44 : SUCCESS in 1h 59m 31s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2612a61534444b5fad57977b30c07c98 : SUCCESS in 1h 52m 32s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/f5b585d5d4e146d89bccbc3a80bb116b : SUCCESS in 1h 39m 15s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5fb8b72be8b948c899e31fb308d009cd : SUCCESS in 1h 18m 46s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d4aa13a2f5cc400ea620057161f8b1ce : SUCCESS in 2h 01m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1865ad7580244279557c6b9ea8b10cd : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"78ec9bab86aec03864bbec572e8f42ce48036adf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 02:10:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"27dadb946178f5808738f1447fe49747202b886d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 02:12:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fbf2aeb13ea40be83fae983a98f33fc : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"43896a6bc45b158fad8145a3c8333c97a666df9d","revisions":{"02de0df0c3469aa2a3ca2d471a5d0ebcf63c5559":{"kind":"REWORK","_number":1,"created":"2019-10-29 20:09:14.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/61/691961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/61/691961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/1"}}},"commit":{"parents":[{"commit":"3934edf116782222c5ef531277d744006dd4bb92","subject":"Configurable heat launcher directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/3934edf116782222c5ef531277d744006dd4bb92"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-29 19:38:15.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-29 19:38:15.000000000","tz":-240},"subject":"Write minimal paste configuration for heat-all.","message":"Write minimal paste configuration for heat-all.\n\nInstead of relying on the system\u0027s paste configuration for heat api, we\ncan write a minimal configuration that is needed by the heat-all\nprocess. This works in a similar way that we generate a minimal\nconfiguration for heat.conf.\n\nIt also ensures that the configuration can be reprodcued if needed\nwithout having to rely on what may have been configured on an individual\nsystem.\n\nChange-Id: I27e98fa77966213fd14e9a5e3187fd4e02db0a10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/02de0df0c3469aa2a3ca2d471a5d0ebcf63c5559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/02de0df0c3469aa2a3ca2d471a5d0ebcf63c5559"}]},"branch":"refs/heads/master"},"e0c71e890c8fb99f368bb01cf2612270c0121cdc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-08 18:37:54.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/61/691961/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/61/691961/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/2"}}},"commit":{"parents":[{"commit":"6c120aed1668e26a52459ec8f7c53700a97a4fcf","subject":"Configurable heat launcher directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6c120aed1668e26a52459ec8f7c53700a97a4fcf"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-29 19:38:15.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-11-08 18:37:48.000000000","tz":-300},"subject":"Write minimal paste configuration for heat-all.","message":"Write minimal paste configuration for heat-all.\n\nInstead of relying on the system\u0027s paste configuration for heat api, we\ncan write a minimal configuration that is needed by the heat-all\nprocess. This works in a similar way that we generate a minimal\nconfiguration for heat.conf.\n\nIt also ensures that the configuration can be reprodcued if needed\nwithout having to rely on what may have been configured on an individual\nsystem.\n\nChange-Id: I27e98fa77966213fd14e9a5e3187fd4e02db0a10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e0c71e890c8fb99f368bb01cf2612270c0121cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e0c71e890c8fb99f368bb01cf2612270c0121cdc"}]},"branch":"refs/heads/master"},"43896a6bc45b158fad8145a3c8333c97a666df9d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-05 19:41:18.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/61/691961/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/61/691961/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/61/691961/3"}}},"commit":{"parents":[{"commit":"a4bf4dbad56dd321236922e1faf9a3c15f48380a","subject":"Configurable heat launcher directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a4bf4dbad56dd321236922e1faf9a3c15f48380a"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-10-29 19:38:15.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2019-12-05 19:41:13.000000000","tz":-300},"subject":"Write minimal paste configuration for heat-all.","message":"Write minimal paste configuration for heat-all.\n\nInstead of relying on the system\u0027s paste configuration for heat api, we\ncan write a minimal configuration that is needed by the heat-all\nprocess. This works in a similar way that we generate a minimal\nconfiguration for heat.conf.\n\nIt also ensures that the configuration can be reprodcued if needed\nwithout having to rely on what may have been configured on an individual\nsystem.\n\nChange-Id: I27e98fa77966213fd14e9a5e3187fd4e02db0a10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/43896a6bc45b158fad8145a3c8333c97a666df9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/43896a6bc45b158fad8145a3c8333c97a666df9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
