)]}'
{"id":"openstack%2Fpython-tripleoclient~697493","triplet_id":"openstack%2Fpython-tripleoclient~master~Iba87e39857269441cad76a33b7a52fcae30dddba","project":"openstack/python-tripleoclient","branch":"master","topic":"rootstack","hashtags":[],"change_id":"Iba87e39857269441cad76a33b7a52fcae30dddba","subject":"tripleo_deploy: set RootStackName","status":"MERGED","created":"2019-12-05 15:05:37.000000000","updated":"2020-02-10 14:02:37.000000000","submitted":"2019-12-07 04:08:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697493-1575691699023-aff7a9a8","meta_rev_id":"54843acd1a74914e2b37b9cf133503708ff9e9b5","_number":697493,"virtual_id_number":697493,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"date":"2019-12-05 20:17:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-07 04:08:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-05 17:41:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"all":[{"value":2,"date":"2019-12-06 11:07:09.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-12-06 18:26:38.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-12-06 18:26:38.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 17:41:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-06 11:07:09.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-12-06 18:26:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-07 04:08:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c36db00f132da6536922fd2dbc8f230e356903d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-05 15:05:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c016646ba59586b606487912b1512d522976405d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-05 17:41:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze06.softwarefactory-project.io:7979/b69cd9aad5af46f8b78eb3c2ebf449e8 : RETRY_LIMIT in 53m 23s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/93/697493/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/6e92774/ : FAILURE in 19m 56s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/93/697493/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/100e08b/ : SUCCESS in 19m 11s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/93/697493/1/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/5681f0b/ : SUCCESS in 1h 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"09c19072c3c5d60c3537dce09597a1560a0459ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 18:23:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/448a3958091a4391b270c8bc046a937c : RETRY_LIMIT in 2m 56s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/08625ea3cc5345c988c1a2476fea7e9f : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a811522bdd9452d9e63f74a70c0e128 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2448783ecaa24d94a63eab6988d3ea23 : RETRY_LIMIT in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496c1a218473481084a23a78a4c47680 : SUCCESS in 7m 07s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/06995ec13d434e11933299f998ba8ab5 : SUCCESS in 25m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2aa2cd13d42141a2a6b4b67c7a86db8a : SUCCESS in 2h 11m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/dc31501f564b4dd6a9262936f7abe7f5 : SUCCESS in 2h 27m 18s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/41c0383f5a2f406dbb899fcc5b1b5837 : SUCCESS in 1h 42m 08s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47eeeb59fb944f06bbdcb59c83c88850 : RETRY_LIMIT in 2m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/665149764401405d901c46976d5ffa13 : RETRY_LIMIT in 2m 56s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9254ed1d1b2281ec8d075f61a54c2cfbc4019bfe","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-05 20:17:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76d8d845b65265908354f66ac3372219dea26c0f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-05 21:50:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-rhel-8-buildimage-overcloud-full tripleo-rhel-8-buildimage-overcloud-full : NODE_FAILURE in 0s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/93/697493/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/4115ecd/ : SUCCESS in 19m 00s (non-voting)\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/93/697493/1/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/65384d9/ : SUCCESS in 1h 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7999e62e900c790f01b8a771ecaea43eee6f7368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 23:46:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/513ff396020946a1b1e6d96cdd8406cd : SUCCESS in 5m 01s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4f8de6ef0b9d4445842123a0bc145476 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa080173f5c94655bc78001db610386b : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0aa39a8623649feb70e77b59e8c6cb3 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca746470488d4e23b40413a3fb876b65 : SUCCESS in 7m 16s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/4f5a0659c5fc499bbd043ee442c97346 : SUCCESS in 26m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/923dff2af1d04c0082ca6fa941e701d0 : SUCCESS in 2h 38m 29s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cc4437305e674f80b6292efca8b01d70 : SUCCESS in 2h 30m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ba74546409ef4d4b8b024bd5643555bb : SUCCESS in 2h 15m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/fa2878979e72466681f2de436b0cb7cf : FAILURE in 2h 04m 51s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/470b971f6b2149ee98854022d5989559 : SUCCESS in 2h 11m 19s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0c16b396b3a24fed9b5ea58bbcab8b87 : SUCCESS in 2h 05m 13s (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4e220ad2f38c4183b73406b7279ea000 : FAILURE in 2h 55m 46s (non-voting)\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c69cdc2a7c9a4732bff3a08590a2e368 : SUCCESS in 2h 01m 42s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/5f81af6add23468cacfa0ceb192edca0 : SUCCESS in 2h 01m 30s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/6a56df9d75fc47e992cfa60378e316a0 : SUCCESS in 1h 28m 11s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/d9c55968276848e99eb7b06447ac4ee2 : SUCCESS in 2h 33m 55s\n- tripleo-ci-centos-7-scenario010-standalone https://zuul.opendev.org/t/openstack/build/d2e425ccca47490cb7fd0462e273e3b4 : SUCCESS in 2h 07m 53s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/6ea6e074cc0d43c49b672d515e61a925 : FAILURE in 1h 38m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93ae1da422e9443dabda373a328b8e8a : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fa0e0366f6c4306814708f2c28b7b59 : FAILURE in 3m 36s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2bfaeeba97b6400392fd2cd1998aec46 : SUCCESS in 3h 14m 49s (non-voting)\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/f2ac0f21a01a40ba9c264c0d89007d87 : FAILURE in 1h 41m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be63e7b81483991b5d356e9c11811b7d45f0edc8","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-12-06 11:07:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9c51a72c1a7d6cc05b44d010ae260e665ba81e8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-06 18:26:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e4a4f078f5099e9efeeeca32541888320f0a51e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:26:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4383127c2398ae55eeba27b51ec9ec83760601c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:08:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e50cdbed0d644adbb32abfb521d58f94 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/344f67fffc704f6782862ef729420215 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05f166d78d6644f99f3c4f587f15e426 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3236ddb0211a4b609aed922ca2b86443 : SUCCESS in 4m 33s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/bda680e71c33447281c08e2c4d1fe15a : SUCCESS in 22m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b95b8165dda0485b96ace540f25f2ca7 : SUCCESS in 2h 13m 19s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/d0cd5eaf0a4346c8a6144e3890178ec6 : SUCCESS in 1h 42m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6ae3055299db493db7a1bf4076b82d13 : SUCCESS in 1h 52m 50s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fc3ce25990414a168f00ea518f624760 : SUCCESS in 1h 54m 03s\n- tripleo-ci-centos-7-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c496ab66967247be86bdc3201dcb3104 : SUCCESS in 2h 00m 55s\n- tripleo-ci-centos-7-scenario002-standalone https://zuul.opendev.org/t/openstack/build/4cfced23ff104cdeb84ba590c2660d5c : SUCCESS in 1h 33m 00s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8f5f83b38d8549a7b451b78c61faf7e0 : SUCCESS in 1h 25m 15s\n- tripleo-ci-centos-7-scenario004-standalone https://zuul.opendev.org/t/openstack/build/5c3eaeeb83534b089df5385c5abe6c42 : SUCCESS in 2h 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bffce5664d242c5b1ee698dbdf250dd : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"85118d8993a0b771655906769de4d034af3ab024","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:08:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1b3afbbf2f37d36467f34ef633e5d7ac5bc01692","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:09:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc7a6cd6eea4c50a9371c28053abf80 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2d6376f9819c1405ae6eeb6b69770d54054ff102","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-10 14:02:37.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit eae3421939e07d7aa95ee6d1c177eb83c4253112","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f494dc61ba0f595da846c768bf54eff11d782d27","revisions":{"f494dc61ba0f595da846c768bf54eff11d782d27":{"kind":"REWORK","_number":1,"created":"2019-12-05 15:05:37.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/93/697493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/93/697493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/93/697493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/93/697493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/93/697493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/93/697493/1"}}},"commit":{"parents":[{"commit":"778351abe687fbaaae59450631a8e461c238e7fe","subject":"Use configured passwords","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/778351abe687fbaaae59450631a8e461c238e7fe"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-12-05 15:05:07.000000000","tz":-300},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-12-05 15:05:07.000000000","tz":-300},"subject":"tripleo_deploy: set RootStackName","message":"tripleo_deploy: set RootStackName\n\nThis parameter is useful for setting up the /etc/openstack/clouds.yaml.\n\nChange-Id: Iba87e39857269441cad76a33b7a52fcae30dddba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f494dc61ba0f595da846c768bf54eff11d782d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f494dc61ba0f595da846c768bf54eff11d782d27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
