)]}'
{"id":"openstack%2Fpython-tripleoclient~697615","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Ftrain~I2d661682b3ee4d82d9f445a04e930c2fb328ef80","project":"openstack/python-tripleoclient","branch":"stable/train","topic":"bug/1855184-stable/train","hashtags":[],"change_id":"I2d661682b3ee4d82d9f445a04e930c2fb328ef80","subject":"Add locks to container image commands","status":"MERGED","created":"2019-12-06 03:40:48.000000000","updated":"2019-12-07 04:55:46.000000000","submitted":"2019-12-07 04:54:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697615-1575694487510-656d7ee0","meta_rev_id":"2fef9a4a0d4ce6fe9e43700d96acecca278c5187","_number":697615,"virtual_id_number":697615,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-12-07 00:30:32.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-07 04:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-06 07:11:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-12-06 14:35:24.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-12-06 14:35:24.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-06 07:11:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-06 14:35:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-07 00:30:32.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-12-07 04:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b3f37a915a862305c9ac0136d7c347ef2ccaf14","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-06 03:40:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9ab84eaa312ec9ef5b4ec146e9960b50e2826b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 06:34:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69a8e1fab54140c1a968b905bde108d0 : SUCCESS in 5m 42s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ba785fc80a5c4343bc0a04fb7b8be41b : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc1fac4885d46e5b21130d59731f821 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07491f3335b346779d69177313b040cf : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b59af9fe951446f9039649bd713d50a : SUCCESS in 4m 48s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/73875bf2e12948dea9aaf11e8a991bb9 : SUCCESS in 27m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6fc7cb1f84084b8cb3afdcc9f462a833 : SUCCESS in 51m 47s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/4dba3c6b74ea42b1a28bf5a5217ccad1 : SUCCESS in 1h 05m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6f44fd59f5aa47b988762146c7bd7c80 : SUCCESS in 2h 40m 23s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/16e1ee1f38144725b46584f83cccf394 : SUCCESS in 1h 59m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e17ca9de7d254f139544254a93847744 : SUCCESS in 2h 15m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ed874b136fd04ccd98014921f54f3171 : SUCCESS in 2h 23m 11s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/54779a8e9bad4efdb8b5e1113abaddb2 : SUCCESS in 2h 10m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ca9f93aef4df4a32a4c8eb5f31a0e102 : SUCCESS in 2h 33m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85762c4062d8487e8dc1978196b9a6b8 : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/816c1b1e98cd4852b64ab448ac91ef1c : FAILURE in 3m 31s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2292cf18d5db47db84dcaca7a32d1760 : SUCCESS in 2h 30m 42s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0b015250137d49c082572559eaef04d1 : FAILURE in 1h 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"abcccee6798986b1baa5bd0f559976f1bbd2d3b8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-06 07:11:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze05.softwarefactory-project.io:7979/f6f35e29d0dc4b7e919660ddc924f943 : RETRY_LIMIT in 1h 05m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7326bbb0d348c3ea82b74139b3db11701fc2c5ad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-06 14:35:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3f217b05bf524c527d63eb01ea7443499c5869a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 14:35:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11120febea822134a378ed1785f9615021ba8cee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:50:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6460fce3d85249399f8f60ca9769e11a : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/044ad41ad6a84b3a9c902ac2704d4f7c : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf90efd695f646f3b1af2fe1ca61615c : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66dfc2b2fb2a4c69a5c48470941f3bf2 : SUCCESS in 5m 23s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/f47ff5651e394794b7edf91bcd2f7158 : SUCCESS in 21m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b535b16468d74506847b77d4bf3ec2bc : SUCCESS in 57m 43s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/db0fa185e2bf46c8a647bc71569b90f4 : SUCCESS in 1h 17m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0acc4be34f714e28bad3577537171e18 : TIMED_OUT in 3h 35m 54s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/81b0392ecfaf430db1e93cb8b87eef5c : SUCCESS in 2h 19m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/54e3e5dee712464592b93bd2e0a7f322 : SUCCESS in 2h 04m 16s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ced0add468c94f5e9a268e72b873e373 : SUCCESS in 1h 41m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748c1976107049d4bcd240d2860393f0 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a7b9e399c4f14b8f386a1729199bb12bc2ab06e7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-06 22:07:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d655ca97d8dc75113ad41464690baf7f1575c86","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-12-07 00:30:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9ab5e405c20f65fb991b270ccdfb48e5376c95dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-07 00:41:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze04.softwarefactory-project.io:7979/8fb9a32118ff440eaa14aa05f280bb32 : RETRY_LIMIT in 1h 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cc97c128847b4c7e8aa5c18bd57e4bb4a04a0d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 00:57:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e2c86e127244c9899f7ac841f94805a : SUCCESS in 6m 28s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5748aa4ae33b442d848f3fee4c367a0e : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fc3f60ca05c4838af0151ab518eed7e : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ec17fe932d84c0abc37e20401a9082c : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7853f3e566841d28851841efc9fd4a3 : SUCCESS in 5m 00s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/0a192c3d29864fbe803d44242793a0f4 : SUCCESS in 22m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7603a3bb67fe47a1928700f34afbf43b : SUCCESS in 51m 56s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/7f5a2ff611d84d269eca68a44b8500fd : SUCCESS in 1h 00m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cacd25bb02af48f5ae58279fa0c2fece : SUCCESS in 2h 25m 17s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e1049e97227c4cc3ae77f7705b968d97 : SUCCESS in 1h 49m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5233558bf02143be93918f92aeda1920 : SUCCESS in 2h 15m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/fbde04a8a013447c984b5e422e860d75 : SUCCESS in 1h 57m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/c9887b44821b4410955fda8f1c717bb6 : SUCCESS in 2h 14m 05s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4b52247a7c7b4eeab97dd65290b9dbfa : FAILURE in 2h 34m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa99af2ee2f4498da2057abebc19db6d : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f10f720e4a5742a0a6403071a769fe1f : FAILURE in 4m 08s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/70e7063bba9d48ffabc3d8776ba446e9 : SUCCESS in 2h 35m 42s (non-voting)\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/34bea9e229a4433e8f2835435b00b63b : FAILURE in 1h 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbf8f8bf87e9b1f7d10dd5ba84b03ff5a4a3b6f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 00:57:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"badd94a31e36feb616119642d57828697c6f62ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:54:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"38fcf298ec9a1f9cad71bae1b3c2c8c6f9bd3bca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:54:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aececd5d5d344ccb9efc2cce571a15bb : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cad6ebbebc74d648e57d486ed63daf4 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/212a4457816a427dac8d63666475b071 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c622e07b6f4fd7b43143c33c323021 : SUCCESS in 4m 32s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/b7ddd52151044c61a8c7c0e7518099de : SUCCESS in 21m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ddeb7ad716374d9291c16d5e83af85a2 : SUCCESS in 49m 34s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/16fbcc132220410cbe69283946d5d789 : SUCCESS in 1h 19m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6b36089e4d514696b6cd371c96240ce6 : SUCCESS in 2h 24m 11s\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/de641871053040b5bb13789901cfcd3c : SUCCESS in 2h 19m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/117cc2ebb6b7426e8b9fd0294bccf0bc : SUCCESS in 2h 01m 44s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c3de95043979428eb6b8726244465081 : SUCCESS in 1h 41m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e0d5ce42f04b538beeb258a3e2afbc : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"84f6d85c5ec1dedc66ef611083396e14e3bb3b82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 04:55:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927b8b1bf325455baffa625fad17cdd6 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a49259357a58900dd4c9e620d7e0dacbfab2f07","revisions":{"1a49259357a58900dd4c9e620d7e0dacbfab2f07":{"kind":"REWORK","_number":1,"created":"2019-12-06 03:40:48.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/15/697615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/15/697615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/697615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/697615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/697615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/15/697615/1"}}},"commit":{"parents":[{"commit":"d24b159f9b9a601cee3819e0f2a65c96214dfd6f","subject":"Use configured passwords","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d24b159f9b9a601cee3819e0f2a65c96214dfd6f"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-12-04 21:27:42.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2019-12-06 03:40:48.000000000","tz":0},"subject":"Add locks to container image commands","message":"Add locks to container image commands\n\nIe5ef4045b7e22c06551e886f9f9b6f22c8d4bd21 added some additional\nprocessing when uploading images. This unfortunately broke the\n\u0027openstack tripleo container image push\u0027 command because it assumes a\nlock is always available. This change adds a process lock to the\nImageManagers used by the cli to ensure we are locking correctly when\ndoing any multiprocessing\n\nChange-Id: I2d661682b3ee4d82d9f445a04e930c2fb328ef80\nCloses-Bug: #1855184\n(cherry picked from commit 94f8ef1df8d2a642c89fcb2838d0b8666aac0b08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a49259357a58900dd4c9e620d7e0dacbfab2f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a49259357a58900dd4c9e620d7e0dacbfab2f07"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
