)]}'
{"id":"openstack%2Fpython-tripleoclient~697860","triplet_id":"openstack%2Fpython-tripleoclient~master~I1f8d9fa4357692c94dae7c63dbce1e63ed5df0b7","project":"openstack/python-tripleoclient","branch":"master","hashtags":[],"change_id":"I1f8d9fa4357692c94dae7c63dbce1e63ed5df0b7","subject":"[WIP] Add minified output callback plugin","status":"ABANDONED","created":"2019-12-07 22:48:41.000000000","updated":"2019-12-09 15:11:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24cf0df3b24b56b5cfda50877a6f8b8f335da714","_number":697860,"virtual_id_number":697860,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-08 00:04:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-12-07 23:22:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-07 23:22:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-08 00:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6721fd325783833adbcf4f27dbdf76fc69ba7c34","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-12-07 22:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44447d1447f0a3156595c1a48b20ff151b6b2e99","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-07 23:22:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/697860/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1fed5da/ : FAILURE in 23m 20s\n- tripleo-rhel-8-buildimage-overcloud-full http://logs.rdoproject.org/60/697860/1/openstack-check/tripleo-rhel-8-buildimage-overcloud-full/f630ce1/ : FAILURE in 17m 05s\n- tripleo-rhel-8-buildimage-ironic-python-agent http://logs.rdoproject.org/60/697860/1/openstack-check/tripleo-rhel-8-buildimage-ironic-python-agent/79bcd1b/ : SUCCESS in 18m 17s (non-voting)\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/60/697860/1/openstack-check/tripleo-build-containers-rhel-8-master/eda155d/ : SUCCESS in 32m 33s\n- tripleo-ci-rhel-8-scenario001-standalone-rdo http://logs.rdoproject.org/60/697860/1/openstack-check/tripleo-ci-rhel-8-scenario001-standalone-rdo/96e1c06/ : FAILURE in 21m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7774268a95b3eab27a5d1f940f72fc88cde686d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 00:04:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/06ab0214ca6c485eaea185a7576406bf : SUCCESS in 2m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f2b642856e14cb5a4165427ca2d844e : SUCCESS in 4m 16s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/19e3579a9b244e578fbea765d996ecd3 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1226360b5fd84c2497cecc8bb4dd0df8 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9735c2da720f46348a10335a2a3a58ed : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2627b0c0f0b747888dd403fe64ec015e : SUCCESS in 4m 31s\n- tripleo-buildimage-overcloud-full-centos-7 https://zuul.opendev.org/t/openstack/build/4bda070ef7ba401c8d954c9d5184fe41 : SUCCESS in 24m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/771ffabe1df1454db19142fb429d0491 : SUCCESS in 48m 21s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/5bcd22c05825442a964705b2db3ab741 : SUCCESS in 1h 10m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ec8ed79fda864498b0ef792d889af00d : FAILURE in 26m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-standalone https://zuul.opendev.org/t/openstack/build/46b652d820b84483ab2f1df5653ec6c2 : FAILURE in 25m 13s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0fd1171d6640638fc8b9c964424af8 : FAILURE in 2m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e159f1aaf7d94f1baea99bf78eaa539e : FAILURE in 2m 50s (non-voting)\n- tripleo-ci-centos-7-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24cf0df3b24b56b5cfda50877a6f8b8f335da714","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-12-09 15:11:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a440540485934c4cfc4e3e023fc20e2f602a11e9","revisions":{"a440540485934c4cfc4e3e023fc20e2f602a11e9":{"kind":"REWORK","_number":1,"created":"2019-12-07 22:48:41.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/60/697860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/60/697860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/60/697860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/60/697860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/60/697860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/60/697860/1"}}},"commit":{"parents":[{"commit":"82f07e94945b4920a5606352434994c3354d399a","subject":"Replace ansible shell with python runner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/82f07e94945b4920a5606352434994c3354d399a"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-07 21:13:57.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2019-12-07 22:48:24.000000000","tz":-360},"subject":"[WIP] Add minified output callback plugin","message":"[WIP] Add minified output callback plugin\n\nThis change adds a minified output callback plugin that ansible,\nvia ansible-runner, will use to simplify the output we see in\nstdout when running ansible playbooks. This will create a much\nbetter user experience when running playbooks while not impacting\nthe verbosity of standard ansible logging.\n\nChange-Id: I1f8d9fa4357692c94dae7c63dbce1e63ed5df0b7\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a440540485934c4cfc4e3e023fc20e2f602a11e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a440540485934c4cfc4e3e023fc20e2f602a11e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
