)]}'
{"id":"openstack%2Fpython-tripleoclient~714629","triplet_id":"openstack%2Fpython-tripleoclient~master~I0443da140020540917396cdfe00da8ac29d20297","project":"openstack/python-tripleoclient","branch":"master","topic":"validations/libs/cli","hashtags":[],"change_id":"I0443da140020540917396cdfe00da8ac29d20297","subject":"Implement validations-libs framework for ALL validator commands","status":"MERGED","created":"2020-03-24 12:55:53.000000000","updated":"2020-06-10 12:29:49.000000000","submitted":"2020-06-10 12:28:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714629-1591792104943-418bdb2a","meta_rev_id":"04c5ff2e946ca2418ef1c0a94758ffd3dd3341db","_number":714629,"virtual_id_number":714629,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 12:28:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-10 00:27:53.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2020-06-10 05:28:54.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-10 09:46:56.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 08:53:41.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2020-05-27 12:26:52.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-28 14:10:42.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-06-01 19:59:38.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2020-06-10 00:27:53.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-10 05:28:54.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2020-06-10 09:46:56.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2020-06-10 12:28:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75f98b7ec02259cd89ffa39ef614fad9828922c5","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-24 12:55:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f19cfbc4bb027de3b796c610aac6f541c912542","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-24 13:43:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d8bbb93138f8a1e47e84fab6d20ce4ee615c4e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:37:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7efc8540c4544cf6a82b987947dc8ba7 : FAILURE in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/620c98d6ae774cc38d409b20a41f3e52 : FAILURE in 4m 07s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/6d73db065b0b42bc993c3ef3f8daf7dd : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16bbe4ba87bd4cffaeaf4f50c00e7509 : FAILURE in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/850e6fec476849629e5bd154f924151e : FAILURE in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3f1c9d78678489e8c1465bef6ceb344 : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b00d1239a5d94e8d9285d8a736280dcd : FAILURE in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f7543ea85e4aa5aa748ea0c2a893a7 : FAILURE in 6m 00s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/1f1e3ae83ea44aa792232f7295370295 : RETRY_LIMIT in 17m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/cec027c5622a480080638eff2517c07f : FAILURE in 13m 14s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a7f5771863db4df083c02b5b2f1837e4 : FAILURE in 34m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/438f60b503b44165969d890da8209a0c : FAILURE in 29m 56s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4ef4b1370e50439f95322fe490762f07 : FAILURE in 24m 31s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"120ff72c2c9a005ee24adefa2d700c1939a8c3d1","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-24 15:36:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ac2937b08341efa08fefef0a4f1094685839de81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:59:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/912cfc866f3640e3b78ac09042a2162e : FAILURE in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d8e87be81254c709084f91f780bbfca : FAILURE in 6m 06s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/87959365261b469c8e53633256ab6956 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ae23f0894e4cb8a3a9513086a60e50 : FAILURE in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb6698f078204989bcd85e2e6c91102d : FAILURE in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d26e637e94e34d4a9cee92a4cf179572 : FAILURE in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/248db28202d94598b1a03ca6a1d04fe4 : FAILURE in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed9073691574c51b3618acb2ecf9990 : FAILURE in 5m 49s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/006a0601af0c4f6f8a0755a5667e38ea : RETRY_LIMIT in 16m 51s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/712ddc83eaaa4cd5b5c3ed918901e908 : FAILURE in 17m 41s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/65b4335169a24f3b9fc0e0f209b605d4 : FAILURE in 36m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0041490e7c3d4be983cb4c8d838908e9 : FAILURE in 37m 28s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/02e01585730547ef91ec36c326d42905 : FAILURE in 33m 16s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d35305a0b83915d4244208b29d97a7302d21c727","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-27 08:45:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1f86df239fe18f9ecfd3214e80207d01bac9ac34","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-27 08:46:05.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"8dc7b188aad45a96cb55d1a51ca8ba62ab72abcc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-27 09:12:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-buildimage-overcloud-full-centos-8 https://review.rdoproject.org/zuul/build/375da1b6f58c4b1bb2002847b324d45f : RETRY_LIMIT in 6m 18s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/39739dc0dca54f5985514caaab25109b : FAILURE in 26m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9ae694034236b3ebf3dea2496ca52e059a51c72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 09:50:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b40abbc66c8f40048f2b1a281f7af310 : FAILURE in 3m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cb2a1cb823a4d59b3a56daadd5d2fa5 : FAILURE in 3m 45s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9bef27be8e114ec793b154361e869aa8 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f94eed67304c5ead77d6e9acd9ad92 : FAILURE in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f7ecc33fce447da85086e5568c9ee2f : FAILURE in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d9a29ade9264d7c95de60ed662d99e8 : FAILURE in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77dd95aed98e494c9c3dff6791fd2e05 : FAILURE in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ec019781cf4ccf9f205bdfb3c16823 : FAILURE in 4m 08s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/39bc1effadd141358b5ec4583d18738e : RETRY_LIMIT in 11m 15s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/99da580f0749494f8708119a325924ff : FAILURE in 15m 22s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9b08c355224f4a549c5647017ef1a7cf : FAILURE in 35m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f1af52dee1f14849917f93c10afca797 : FAILURE in 26m 22s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/63de8fe0b10349fbae850c49da58e026 : FAILURE in 31m 24s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0f7baa3561b22951c8d60bf0ff94310b997d78a9","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-31 08:28:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"92f5a993a6afa1247f9042dffbe0e993a49b012f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-31 08:37:08.000000000","message":"Patch Set 5:\n\nis there a new change for the tripleoclient distgit in order to add the new dependency?","accounts_in_message":[],"_revision_number":5},{"id":"052fc22b842a0fb4cd388ad535ac47a72b82943d","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-31 08:40:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aeec75ca537231fe4b62232218ed60a1ec5c2a34","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-31 09:10:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/29a13099f2f04c85ac37e06369186f85 : FAILURE in 28m 14s","accounts_in_message":[],"_revision_number":6},{"id":"a090c9bc5b9141ae738293a972aa91850024d6d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 09:32:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/01cd271dc5e34f2e9a19682a3cd9d7bf : FAILURE in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d145d375e28c4f6da9444ea1763b6a37 : FAILURE in 3m 43s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ca16f26b0be24e2ea3a11f9c4d05a6d1 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79935e7ad904805aa01e7abeb90f4b6 : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2ae973449154b2f892c507b87b8dacb : FAILURE in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a770991c84504807a2777b31c41ebc5f : FAILURE in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77369eea8166418cb7a7e6d215d28cae : FAILURE in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98b2365d21b4e6ca492353e42c5d9a9 : FAILURE in 5m 42s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/4b3cc6b955bc46d2a5e14a54c5601c06 : RETRY_LIMIT in 11m 00s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d1096f951a614f79adeb1c65cf8d31c3 : FAILURE in 11m 46s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fece5191bb1c45c9b21a8207e02ff5ab : FAILURE in 30m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3c0a84a5a7ae4cc1a7f048ba31224dcd : FAILURE in 29m 27s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ea9ec446d77247e88b0f25ade160a622 : FAILURE in 25m 21s\n- tripleo-ci-centos-7-scenario012-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1595e60ede38bbca1a21e723ab6fe6bc214fab46","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-31 09:47:29.000000000","message":"Patch Set 6:\n\nyou will probably need https://review.rdoproject.org/r/26177","accounts_in_message":[],"_revision_number":6},{"id":"92ea32f792fe8b022d268698b5e903299d36f1de","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-31 09:49:13.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"295f61dda5443fbeacf26c25def1e0db2b3b4baa","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-06 07:26:02.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"05ff7da99cce8e31ae676eebe5ec80b87a818cfa","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-06 07:46:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7bc48da96f6044395afcdf26d6624117ac48da08","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-06 08:17:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bc089c618db04e4d99e2cd02b5785585 : FAILURE in 29m 57s","accounts_in_message":[],"_revision_number":8},{"id":"ad7ef1506fa1e20d3da362149f893472eca18f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 08:53:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7004b22840dd4b93ac8999b6e70efb8d : FAILURE in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d35485b37bd428ca155fe48890d1e65 : FAILURE in 3m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/17e4b6fc8fdf42bca705d611edab1564 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0244807c25e04cc6a4759c6bf29706dd : FAILURE in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/951a08fb8e024a8798e30b6db4da70a9 : FAILURE in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8db747b0057466a812b52abfe85617d : FAILURE in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/793a62f51b524b52b2b425f3cad9c8d6 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d9e830fcb64f918fd705c712e07051 : FAILURE in 7m 38s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/cdc0632b682145f98b8559901e4904c1 : RETRY_LIMIT in 13m 32s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/eabd1567bc184b479e09d22a8345d526 : FAILURE in 15m 17s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0cd5ceadae7d4575ae1079d9ab327359 : FAILURE in 32m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/99faf790e7194d02bba240e29d39c59f : FAILURE in 29m 37s","accounts_in_message":[],"_revision_number":8},{"id":"4373c708c26fae7a866be1eb4fdf82c92c84d465","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-22 19:46:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"09b639afeb2069f0404ee4665f6b8bbcac243101","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-22 20:19:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"78592c36bb599350a32d5e83a201b9d19dee0ab1","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-22 20:32:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0c62f3f1194e1ef7e23fb44b86197c0010e9e463","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-22 20:43:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b7540cf5558199351f8d1bc9a354079e7438cb58","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-22 20:45:23.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"4ae3621c10cfc8bbaf033466764a982f0673f717","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 21:29:25.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8dc62618482f48d4a44a19e42446ae18 : FAILURE in 40m 45s","accounts_in_message":[],"_revision_number":13},{"id":"34035534eec674b61f8287232d320f78ead7a5c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:11:45.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4ab4109aa6848e98b7d8c01a333a2f2 : FAILURE in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1acbfae93aab471fa4f796ef7fb913ff : SUCCESS in 8m 29s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/fa9afec9d69b47d5b40c51daebe622f1 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6025c397f627401398bf710810a37357 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d44686b973f444b3b076a043db268dd1 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e03e88fac95a42a8a636eb2169c1fbcd : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3cb9db9213c48569d795548cddfcf33 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ae31daee9b4090b1b7b0ea91686b7c : SUCCESS in 5m 48s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/9269d636fc034858813607411114a087 : SUCCESS in 30m 36s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/ee581caee2ab4058ba4fa194e7b096db : SUCCESS in 23m 40s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/84638517e879450dad8abbc98432720b : SUCCESS in 47m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f76ed6f2ba7342b0aa39cd46db418b54 : FAILURE in 42m 11s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/21c9e63bf8d04da095b03de83abf8ee2 : FAILURE in 39m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8ecfd4166240435a9c1086176b759f40 : POST_FAILURE in 43m 26s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/168442b6099e4e9bbc7d8c3294cf3676 : FAILURE in 35m 02s","accounts_in_message":[],"_revision_number":13},{"id":"f628c7ec33dacae9c6cbfd1a71dabb420c58d248","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-23 12:12:52.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5709a17a890b92a1dea2edb8fddee255028a3a5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 15:17:53.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9c2d9164f71141f295922e0c3d1fd691 : FAILURE in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e851ab8e88141898cd3842ae45c54b4 : SUCCESS in 5m 49s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0751f1389bfb48a1b77b2caa50a318e0 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413859b1c03e4175bc05bb7d174a9081 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/830bc80522fa4755afcb3799c192564c : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd8e6721894445a8a35c9e8fa1c381b2 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26d599f17eb041c6a59e6c6bbcf9b5bc : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc39df287ea74aaab1ba5cf4478207b2 : SUCCESS in 8m 37s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/fc8dc5d710a3472a9822457daf159fc3 : SUCCESS in 27m 27s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/e1ddfc2bed144debb21f7561207769db : SUCCESS in 26m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/bb1177f489384bba866bf12d43ee40d0 : SUCCESS in 52m 42s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/22038fbb9f6244a78b829cfa922c26e6 : SUCCESS in 2h 39m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9207e83f1c164df4962b7c0f09dec55f : SUCCESS in 2h 24m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3244258f59ad4e84a3d4c20d1f2fdf5e : POST_FAILURE in 2h 47m 59s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a8c5023ec33640a8b61ab993c3aa8cf6 : SUCCESS in 1h 41m 07s","accounts_in_message":[],"_revision_number":14},{"id":"d836e95246d283b22f7f6b07ab74bf71dad86095","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-23 15:25:07.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ce6868aa2dfc49bb8a8d2179266e9afe : SUCCESS in 3h 08m 03s","accounts_in_message":[],"_revision_number":14},{"id":"016b67530976e9a1fb74f1994925b7c0ad66f902","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-28 15:02:28.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"5afd532954dcc313833f8c4be8429229e1adf25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 21:39:48.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2623f1756f5b4a3189113fb1c9522007 : FAILURE in 2m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6ba16556e2c452eb9416af0f69d9e2d : SUCCESS in 8m 38s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/6367068f2fc94f51bad9e54f3000e08c : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdca78461be4429aaf5da876fcb04adf : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acc1968f09744ebbb09615ce4fb19e34 : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd1f239dce5546f8a7d245bca3c1342e : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/183442652b3d4911803ab8b7d3a318fe : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2389cbc19d044abf949cb72aeabda5f5 : SUCCESS in 9m 37s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/24f799b4e1d54ec58e1469c48660d8c4 : SUCCESS in 27m 05s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/3725d882e28846d1af50c408911cf415 : SUCCESS in 25m 47s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/a22f3f54bc054e71b61a1237ef56c14f : SUCCESS in 21m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/8e4c1231a0db4569ba749bbc26354a73 : SUCCESS in 48m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bc648c9134bb4769bf332d69eb3f4e38 : SUCCESS in 2h 42m 31s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a2e68236226c404cac67830762d85bcf : SUCCESS in 2h 28m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/71e5751a29e04bbc84171d5feae92032 : FAILURE in 25m 20s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ed521581dd5d495690328bb83702a74b : FAILURE in 26m 45s","accounts_in_message":[],"_revision_number":15},{"id":"eac46300cfd05415bef18ceb3b948656171ef2c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-28 22:03:03.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/17f78883134f46a78a64e00c52745564 : SUCCESS in 3h 19m 54s","accounts_in_message":[],"_revision_number":15},{"id":"b003d11ae2d180b33d593313e4cf5beffe664686","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-29 06:15:01.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"a369fe6c980a29610fd8ee5e7e90aa4032ef8526","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-29 09:10:52.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/761b230f94524faf8cea727a1e315b0c : SUCCESS in 2h 53m 13s","accounts_in_message":[],"_revision_number":15},{"id":"c8c6b6d237f3d609db97013a3e67446bc7996ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 10:49:59.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d0d6ab73c1754115b5c0fca49cb1d8ed : FAILURE in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d85d89432b4d4825a49f5c511d9010b0 : SUCCESS in 7m 55s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/30f46fbf1f224830ab9ac893e82ec538 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b551e717aa643048dd90ab42b2a8f10 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d259a0a2a524485582c14aedf9f04fdb : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a09947e0eff48d394878dd7ddcd35f6 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78e636a863ee409a85bbd096b8418842 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1ee425bb5943a9874d62949c08f8ba : SUCCESS in 9m 11s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/e94915aeabfa4206b551db8b29d6bdc9 : SUCCESS in 32m 06s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/21bd2c8134cf49c6be09278fc5e59e8b : SUCCESS in 29m 57s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/5c5d7798734c4a8cb6ea961e46d4ef46 : SUCCESS in 29m 26s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/cfdc18813dc144309509b05a3644ccae : SUCCESS in 50m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/758742d95e514092a4aaa52c88f98880 : SUCCESS in 2h 52m 50s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5cb99e47683e4a219f3a4ebdc6fa0f40 : SUCCESS in 2h 15m 33s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a0afa3d3d7014dd98f73cfad789a7fdf : SUCCESS in 2h 05m 59s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/71fd690ce2df449da9fca336450c670c : SUCCESS in 1h 59m 10s","accounts_in_message":[],"_revision_number":15},{"id":"473eac5b9e2276f7770becf5005d6041ecaa6ca7","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-04-29 11:36:11.000000000","message":"Patch Set 15:\n\nguess it needs something in another repository:\nERROR: Requirement \u0027{Requirement(package\u003d\u0027validations-libs\u0027, location\u003d\u0027\u0027, specifiers\u003d\u0027\u003e\u003d1.0.0\u0027, markers\u003d\u0027\u0027, comment\u003d\u0027\u0027, extras\u003dfrozenset())}\u0027 not in openstack/requirements","accounts_in_message":[],"_revision_number":15},{"id":"40fdd274c73bc2c2b1ef70fa36b863a13484bf9a","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-29 11:37:34.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"bd97731c3e3da7fd5401401ae0cd766842f39803","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-04-29 11:38:40.000000000","message":"Patch Set 16:\n\n@Cédric: Nop it\u0027s merged yesterday night so it should be ok.\n\nI rebase the review to see...","accounts_in_message":[],"_revision_number":16},{"id":"46c70baa1e9f5a90ed30c9eca8c78b95eae45fdf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-29 14:36:58.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8f08fcd4a689433ea49a13f9310f6274 : SUCCESS in 2h 57m 40s","accounts_in_message":[],"_revision_number":16},{"id":"7c746ae5696f6edb4c6b84c7faa93f4a765b0d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:34:10.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/902191276fec4318b96e9a8b911065b0 : FAILURE in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9577b11f28e746378bb191791ee2b6a7 : SUCCESS in 6m 43s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/2138096ea9f6451b888cc36e5f5e7d91 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99251eeb5e1f4150a8c497ab42039e0e : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9642ee3a01c04eeca48d22141e9e057c : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0750ed632959440ea4de12c1953ad3bc : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fe09bdb21b142eda9b7d6df87334263 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8fd8871d004a338806040e40585eeb : SUCCESS in 7m 20s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/fa963a902a90403b97418af7c9cf223e : SUCCESS in 28m 58s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/4b4cbeac4e374d0ab43d1c001c002f4b : SUCCESS in 30m 43s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/09f0431e2b8d4f33bf2b8a213985c14f : SUCCESS in 25m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7c61f1a99a0a4e25b5c2f390c0d0af3e : SUCCESS in 43m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a62de3c9b12646d4935507afcd4fba7c : SUCCESS in 2h 30m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2e94f5aa0af64056a7efea179e47bc2d : SUCCESS in 2h 44m 13s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8481d2e7e93243b4a48cfaa06430b04f : FAILURE in 2h 06m 17s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/baa51917c0564815bb83528596bc618d : SUCCESS in 1h 55m 17s","accounts_in_message":[],"_revision_number":16},{"id":"eafdc60b4d67abe71504cbee57ee21aebbff07bc","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-04 07:40:37.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"0506d44b54c91595aa946598effb8136fed518b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 10:16:18.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6546486d5f5342689384dbd63965081b : RETRY_LIMIT in 2m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17f2e0b9ecce4b1ab271d67e13ed4576 : SUCCESS in 5m 42s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/adaceb6860bb48df82facb061dbd7ba9 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21200180fb9f42b7b0746bd9594d1b3e : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f4e6dfd1ffe4143836bc92ad7d2b8be : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dce5a052c04e4c73ab793f395584d4e7 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63a3c3dc13214c61802eef534a8dfbe5 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f78a8b5000474bb38f9e53094e19ce : SUCCESS in 13m 21s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/d5bc1c7af76e46db88214e7c6721723e : SUCCESS in 29m 32s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/3a79f025072b4ba8b692858bba7f13bd : RETRY_LIMIT in 10m 45s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/9635e09d6fc74d1c955d0c9c883e9166 : SUCCESS in 16m 41s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7d6eac34c4c6497facb50498d8db7467 : SUCCESS in 42m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e223e99f77aa421392517c8ded31bd57 : SUCCESS in 2h 17m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/111c013b136d49b2b852be0fd5605da1 : SUCCESS in 2h 04m 54s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/15b2d04ebae348f2840dd020a92b032f : SUCCESS in 1h 58m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/43c86752ea014486be339a9f66420869 : RETRY_LIMIT in 3m 15s","accounts_in_message":[],"_revision_number":17},{"id":"d9e7722b49179fa6ce27e017e2a2b097c260dac0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-04 10:32:21.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/431468cdda874408a3d3c1036c3b90cb : SUCCESS in 2h 50m 34s","accounts_in_message":[],"_revision_number":17},{"id":"9bd491db21ed898f2b60a3fbb2f1d4b43c19e271","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-04 12:01:16.000000000","message":"Patch Set 17:\n\nrecheck\n\nmultiple unrelated CI issues, apparently.","accounts_in_message":[],"_revision_number":17},{"id":"bdf89bd8256df0cb8ada7bfd4316b5e27ec48b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:23:07.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f874a9f3ffa841b8a3072b033fc65df8 : FAILURE in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3854c7f8beee4553929a236929e9d623 : SUCCESS in 6m 04s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/a21c12b70b9a4ab8b4e93ee23425ca89 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8755d6a0864426af10cc592fc19c15 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfe5ff3458fa479b8cc71dea12203567 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b98675e968d4444c9ce86cc6ae543543 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef525de253ce416193d15f28449733ae : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5400a97c623c4aa4a4cca6a654d7ed4b : SUCCESS in 8m 38s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/33844f7a20d4487392e9996ad66796b7 : FAILURE in 17m 38s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/3f67eb974d5f4c4b989b56196d273b5f : RETRY_LIMIT in 6m 48s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/77ff8aab9584437aa464c09866ada572 : SUCCESS in 27m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2750277c47c14939b997ebfda6e8630d : SUCCESS in 44m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c7599b702dd64ba68e9c23a54438e66c : FAILURE in 1h 49m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e879ff66e82d45f1910157b64f134060 : SUCCESS in 2h 10m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/80306ec9e767492bacfb9732011adf9b : SUCCESS in 1h 54m 55s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/22be18c0977d4f3790a835a467a26f67 : SUCCESS in 1h 46m 05s","accounts_in_message":[],"_revision_number":17},{"id":"db5bc316587dc28b4f8497fd03993bffae04ba59","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-04 16:17:43.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/eb9c504dcbd140c69c8dc447e4e413ab : SUCCESS in 2h 51m 36s","accounts_in_message":[],"_revision_number":17},{"id":"151a8664a54427fff7dbfe1f0dc991db2c6ec89f","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-06 20:04:18.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"31de7bb72402f406b5786297b1246aa855ec3695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:37:08.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/942280f1de5948d2b3a211be026d8804 : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55218a75baf54b52932e646146a52433 : SUCCESS in 6m 53s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/13ea2ae9ccc34332972fa5e5f69194d1 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bba94d7c8e4e72a889bc32f45f5723 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0a5aabaa3f845fa9ddacde6595cad4a : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3136f413567d4832a0a9b1dee9ff3083 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82f7dce45f7441be809f931c10218dde : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e07be146a4448cabae6f07898b2355 : SUCCESS in 7m 40s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/6927d40544f84666ac3edb1ba11205c9 : SUCCESS in 30m 50s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/93d623b3fa054915b8642f92b6fcb07a : SUCCESS in 30m 12s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/2cc1c11b4c504ee7bd0f19e57aa24576 : SUCCESS in 19m 39s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3391d4c5dfb04a7bae94368e9bd9d97d : SUCCESS in 43m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d2b705ec655341538006e1279b37500f : SUCCESS in 2h 21m 48s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9dbf6ec8ca6f436cb1bffeb7ab0d42c6 : SUCCESS in 2h 08m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f1215587ef3a4421b8896b751b88a769 : SUCCESS in 1h 54m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2e066f9fbbb8421c9e2e8f5181949476 : SUCCESS in 1h 40m 46s","accounts_in_message":[],"_revision_number":17},{"id":"56dbff65814076d073d82a981ad5f34fcb10bd32","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-07 00:12:06.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/67fb5fd6d55d42878148f2ba4d228b6b : SUCCESS in 2h 45m 34s","accounts_in_message":[],"_revision_number":17},{"id":"f46bdb57ecd2610084c67da06b2b3daea0b6e68a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-07 05:45:54.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"2170bb8e51375e0480352e2ddc4aa1087f471c7d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2020-05-12 12:23:47.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)\n\nJust to be cleaner","accounts_in_message":[],"_revision_number":17},{"id":"7ff72d68c698f12214d96e04a8627e9739986144","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-12 12:56:17.000000000","message":"Patch Set 17: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"bec10604f502922f18394cfe28f93da87d6767df","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-12 14:45:49.000000000","message":"Patch Set 17: Code-Review-1\n\napparently there are some weird issue at some point when using this patch on an env deployed against master:\n\n(undercloud) [CentOS-8.1 - stack@undercloud ~]$ openstack tripleo validator run --validation validate-selinux\nException occured while running the command\nTraceback (most recent call last):\n  File \"/usr/lib/python3.6/site-packages/tripleoclient/command.py\", line 34, in run\n    super(Command, self).run(parsed_args)\n  File \"/usr/lib/python3.6/site-packages/osc_lib/command/command.py\", line 41, in run\n    return super(Command, self).run(parsed_args)\n  File \"/usr/lib/python3.6/site-packages/cliff/command.py\", line 187, in run\n    return_code \u003d self.take_action(parsed_args) or 0\n  File \"/usr/lib/python3.6/site-packages/tripleoclient/v1/tripleo_validator.py\", line 372, in take_action\n    self._run_validator_run(parsed_args)\n  File \"/usr/lib/python3.6/site-packages/tripleoclient/v1/tripleo_validator.py\", line 350, in _run_validator_run\n    t.field_names \u003d results[0].keys()\nIndexError: list index out of range\nlist index out of range","accounts_in_message":[],"_revision_number":17},{"id":"9b11c87ce585f95035a0d563375de0598b9ef23d","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-12 20:50:24.000000000","message":"Patch Set 17:\n\nHmm I think Cedric it\u0027s because you don\u0027t use the callback (validation_json.py) from validations-common.\nI will make a patch for tripleo-validations to update the small change in the callback. Because we will have to use tripleo-validations for at least 16.1.","accounts_in_message":[],"_revision_number":17},{"id":"ec364c09163b1bd8a5e2c430ce1a2816c4431ce4","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-12 20:58:36.000000000","message":"Patch Set 17:\n\nthe tripleo-validations callback review here: https://review.opendev.org/727452","accounts_in_message":[],"_revision_number":17},{"id":"c2630e5946a3c9b2157a5d73df162a74bae75fbf","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-13 05:37:00.000000000","message":"Removed Code-Review-1 by Cédric Jeanneret (Tengu) \u003ccjeanner@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":17},{"id":"cf1b2ec0024080bc3adb10f452f398e3b4fd384c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-13 11:27:46.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"90efd0a76a23879c7b05968f88c6c33e8a2acc41","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-13 11:30:55.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"be4a1f3b0dd57df3c40fa0c657f0d9de22b0aef3","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-18 13:07:15.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"04656334178a5784fcfa585440311188746797ac","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-18 13:08:42.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"9bbdde49ae0aa126d6c6e1d12966f411ad1b07a5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-18 13:14:19.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f7b2f608989766e25a140b0043f6bf66353b2851","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-18 13:17:02.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"c8960a3d23750da4066ee75d284b633b1e0943a3","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-18 13:19:07.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2c39a1aceb67aae1ed0687dd80c02bf4a391fff6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-18 15:57:25.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/61726d954df343f68806de280f3a8651 : SUCCESS in 2h 38m 35s","accounts_in_message":[],"_revision_number":20},{"id":"c6f74c14fc198c8a584f54fc0e114f637ff180ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:00:34.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca408c3ff6df4ff6bcc49572a2044ef9 : SUCCESS in 2m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfff1a9195f249d48465024dcfe18c77 : SUCCESS in 7m 08s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/a8b6e385dd044f80a25a04c7add3e2bc : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7dda2491804a23baee4164ed1bfb9c : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b15a536975c4e9d906e33e0142775c8 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf32021ad1aa4df28060212801445b81 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/164938e8296d4d3da814756ff73403c7 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce734095271474e8afbb6af55a6ee29 : SUCCESS in 9m 23s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/8b943dd34c154b90a2bc362c79c3768d : SUCCESS in 25m 01s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/c127d968d3704575809997547fa72262 : SUCCESS in 26m 25s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/9403a619f1f943969c66bbc03f325be6 : SUCCESS in 20m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c5b29acc0393481f94fc49277171d3ed : SUCCESS in 49m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ebecf39128a643f089804431f8172652 : SUCCESS in 2h 30m 28s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/17b0aa0933e94ceba51bae504be5b696 : SUCCESS in 1h 54m 22s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/469b9c59d6f44dcf82e0bc7950e7ca73 : SUCCESS in 1h 40m 41s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c8236d14a9cc4d4eabc310035c3206d5 : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":20},{"id":"9687a75dba8f7f26743219394d61a64ff6aa295a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-05-19 14:30:40.000000000","message":"Patch Set 20: Code-Review+2\n\nTested on my lab, it seems to work as expected :). GG!","accounts_in_message":[],"_revision_number":20},{"id":"b9af31c571c6eea8e6ad0e0013de827d834e19b1","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-05-22 22:21:03.000000000","message":"Patch Set 20: Code-Review+2\n\nlooks fine for me, but does https://review.opendev.org/#/c/714629/20/tripleoclient/v1/tripleo_validator.py@72 need to be adjusted ?","accounts_in_message":[],"_revision_number":20},{"id":"32014ad3b082b41bfb0726274ecec621c44127c7","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2020-05-27 08:53:41.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"7e13b15f95cc0da4a268405f88fd47af05e2c246","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-27 12:26:52.000000000","message":"Patch Set 20: Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"43d0160977651758712cb1b15ca4c04259fafdc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 12:27:19.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"5132ff9a3acb2dd573ceb24fc5e3fc1484d51d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 22:58:01.000000000","message":"Patch Set 20: Verified-2\n\n(1 comment)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a64d9db0d0c0406ba88f848532e3420c : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bc99a0cc710476f8c5d33d2967133eb : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e136f4a546b34d9e86ccf8ba1f6ac4a6 : FAILURE in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62f7de880c82405687dcf5939704e3f3 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d934cabdd6b44f2f8e44b9ee37648d73 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d2d58cde7b41cf9d4626ecfa2a8e70 : SUCCESS in 8m 02s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/cc97827403c2411497b9d7ff47aef6e0 : SUCCESS in 34m 19s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/6b67d9363952420e9ea85bf043cb3b4c : SUCCESS in 38m 33s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/bb5087d1a56f403e93a8f575e28c8a8a : SUCCESS in 25m 02s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/83fc1dd719154cc1805692c425156d31 : SUCCESS in 30m 20s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/0da86f5d20bb46348748124daabd442f : SUCCESS in 58m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/daf1b5f8e73e4719abc7d1f012550eab : SUCCESS in 2h 41m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cedf031fc6d34e3ba399d802a9d69af9 : SUCCESS in 1h 57m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/72b4d8917b41467dbbf05267eb546a31 : SUCCESS in 2h 27m 14s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/73e2035146c04a7f8ad4ecfcaafe12ce : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":20},{"id":"0b67a72eccdc8b8efecc9afa727b1bb757c78b32","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-05-28 06:09:15.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0de1282816955e4e272325b1dc1179bc0d7fce4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:28:39.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f1c2960e2dfa4c42a5d57bada6cb3e99 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf5343319a5c40bba97c8eed8d28e851 : SUCCESS in 6m 15s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/15c85e0736594598b3e5407db7d93d35 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d616e7d90224cf3adeba8c9a9348d34 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aaed4b15f8942a5b45b296a23cd682c : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09f7e524aa004de882b32743630fd43b : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c33f82c0c18f4ad8bd0b4a6d2cf66f82 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20721a4d119649538a86943d8dff8ab7 : SUCCESS in 9m 11s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/ef1d1d8af1594b6ebed8d12b4c3d4ab9 : SUCCESS in 31m 11s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/1c8784c441a14207a93f8a139ac09ea7 : SUCCESS in 36m 23s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/ec6b190c38f14cdda227240a61021be8 : SUCCESS in 33m 26s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/93bfb16290e94fe8a39e648c8d1964b7 : SUCCESS in 21m 45s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7bc1a707c5534b46856e04a5ac5ee7a2 : SUCCESS in 47m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3ada311be1644544985867ddab1b2e58 : SUCCESS in 2h 16m 06s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/196b1e4b1d5b4607ae6355397e86bfba : FAILURE in 2h 28m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1cef5c407b3b4f569bb86972c5a6b0fa : SUCCESS in 1h 41m 12s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d109e74bc05e441095d81e1b098c4361 : SUCCESS in 1h 30m 10s","accounts_in_message":[],"_revision_number":21},{"id":"02deb619f4cdfffcaedaa0b4661559a8c79ae20d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-28 10:46:16.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8ce005edc853457f9a666367b18f4c79 : SUCCESS in 2h 48m 57s","accounts_in_message":[],"_revision_number":21},{"id":"923e11275b7c9e9d9b7eaf7e212a06b9f2a3e94c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-05-28 14:10:42.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"52862e2b7ef85e472f7cd3dbf08121df8f3119fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 19:31:23.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f9fc3d829304481906353f83d91e409 : SUCCESS in 2m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b2d586ae14840269e6d58deb43067ec : SUCCESS in 7m 15s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/99db6f5484714a41a203b649e0ef843f : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa3f277853544929aa0435f8d7c659c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/678a91c49c0244b190ada71890e616ac : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c493f07165bd45ee963c7b9c00cfd534 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd78e4cf05c402b9e70c3b366502cd2 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec641e43e4447c89599ed33a78c99b3 : SUCCESS in 6m 12s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/2b3e36d2b01e4f2a972a27cdefa26e2c : SUCCESS in 38m 04s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/0468d17403714647a87e6abc290ae63c : SUCCESS in 39m 46s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/2390d874d39d4429853af225bf8f4baf : SUCCESS in 40m 48s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/e9f260a95aff45d986d105d74032f07a : SUCCESS in 23m 48s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a36e2b5a919541ed8ce43fb00ca2459c : SUCCESS in 52m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/17f9c37e25bd43fcab2ac12b1eef4804 : SUCCESS in 2h 35m 04s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c24c9c76dcc047df80705e13fa0aa847 : SUCCESS in 1h 59m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5b505bdc54d84f73863dfdf5c8db0066 : SUCCESS in 2h 03m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5d696982c41c40c1a64af1bb1f88f5f7 : SUCCESS in 1h 39m 07s","accounts_in_message":[],"_revision_number":21},{"id":"730c6d7b900051086d81ef0a286bfb211d106a05","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-28 21:49:43.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4c22c3aeb88d4874b0b279672659abc0 : SUCCESS in 2h 52m 00s","accounts_in_message":[],"_revision_number":21},{"id":"ac5c9cee6b16a7b90c203cef99f4f18766891f47","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-06-01 06:10:23.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"0aeb450e6bf724bed3b57c3f5e63bd2df7aa83e2","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-06-01 19:59:38.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"78fb8cd1b8324ecc1591a383851e9701ffdd5443","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-06-09 13:47:06.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"4c87a9083b187fce658a167bf4a83380baacad9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:08:06.000000000","message":"Patch Set 22: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aa6b5e7cc7a44cdc98213f717dbdee4f : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/227fe8c1105f4561ae724cb3cafe5841 : SUCCESS in 6m 22s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3eee2117fc99482981be56075d61ca49 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/369b7d95569a4a93a7283e6736ecf25c : FAILURE in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31bd8cc54da74745a0ee198554610747 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3158d1941674d4d9d1615a88f2d18d6 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58db362fa2a443ffbdd34651c508472f : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0448a49ef5d5449aa1e17b8cc8be13b1 : SUCCESS in 8m 47s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/d4fc9c4a712444278cc5808cc7fd4976 : SUCCESS in 32m 21s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/b5e94e0d9e914ed88e37c680c3a31036 : SUCCESS in 33m 46s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/8a3f5d652ee64b0288b962921b919099 : SUCCESS in 29m 49s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/311354884c234f1bac6813d0d2eb7e5b : SUCCESS in 28m 48s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/35796d62a6564fe5949c1f128c0b10b6 : SUCCESS in 47m 37s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c9748bb302ff464c8d696e16be2407a0 : SUCCESS in 1h 55m 34s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cfbdd47edb56482da9bb3bbcc54565da : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":22},{"id":"9ca61edb2dbf809b0314c812d2184d0436bdde24","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-06-09 17:13:59.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"7007ce4c3c53cb3e0a62ec2a70e708b669a386b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 21:44:40.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/31476cf70a0d431095367fd0419a797e : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63424d124be044a48b3c5754f4ac040d : SUCCESS in 7m 22s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/bf4055c041b14105a777b696b17ae310 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5befa497084a41b484ffe73950d363c4 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4223dec835464030a7693f82d5be5bb5 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2388813f0364973b9a6a1f957c3d3a0 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bd9a3ba67644d8980abf097d07433e4 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e667488b6a044aa59ac2a7837bf9a290 : SUCCESS in 8m 42s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/4f193688915042488262e0d832db28d1 : SUCCESS in 30m 07s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/d0c5a73f6f7740c681c3dba89921da97 : SUCCESS in 32m 54s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/242edacba5914fd1b63ec231f32a9d75 : SUCCESS in 25m 14s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/d7ca5e96f7994c0b8679f09190178ab1 : SUCCESS in 19m 44s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c61653979c9248ae989a731b5d6803b3 : SUCCESS in 48m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2133d33d49584705836eca42a2c4b95a : SUCCESS in 3h 07m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7853db2636744acc9e6c007fd33752d8 : SUCCESS in 2h 03m 44s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/92508d693e3e44fca591273f7602ac72 : SUCCESS in 1h 48m 58s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/264a5c2dc6b048229491c6cb19868447 : SUCCESS in 1h 36m 04s","accounts_in_message":[],"_revision_number":23},{"id":"a791da32f58f11303d7cf01407cc3a62a41de43a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-10 00:27:53.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/aff960ab0f4f494baa847910b46959b9 : SUCCESS in 2h 47m 44s","accounts_in_message":[],"_revision_number":23},{"id":"55a9e3c016970388352acd89dc51864c359f8c71","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-06-10 05:28:54.000000000","message":"Patch Set 23: Code-Review+2\n\nusing core - this was accepted already but failed gate. We need it asap in train.","accounts_in_message":[],"_revision_number":23},{"id":"33766057a65177d6eec233b2ee52d58e9d90e86a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-06-10 09:46:56.000000000","message":"Patch Set 23: Workflow+1","accounts_in_message":[],"_revision_number":23},{"id":"f4e0e4a056e7788bb3834fd1236d95ad621886c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 09:47:08.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"ff86560dbe3043af56746cb22cff9f5febf1e538","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 12:28:24.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf8d7f3d9c6f435b95ae0fca51e51e07 : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45e5e83957ec41d0a3566bccf4a01a08 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4519f0b0eb4b48a57a71b133a48d9c : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8e214641b2540b999208a16d407c342 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cceee57407bf407f9158399f55262cee : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5186c0d4156484fb984388baeb04ca3 : SUCCESS in 6m 39s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/1f0124dcf8fb4d7cae4eea3b6b54d833 : SUCCESS in 1h 06m 03s\n- tripleo-buildimage-overcloud-full-centos-7-train https://zuul.opendev.org/t/openstack/build/126cfb9558c84e0eb47155483deeec25 : SUCCESS in 54m 12s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/205ef28c40df45c9823de544f6ba729f : SUCCESS in 24m 32s\n- tripleo-buildimage-overcloud-full-centos-8-ussuri https://zuul.opendev.org/t/openstack/build/73106750f1b0419a919241a693695cd8 : SUCCESS in 1h 02m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ff0806154c544133a84078cc7c41269d : SUCCESS in 41m 17s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d1beb4e35a824ab9b2180e78e07440af : SUCCESS in 2h 30m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/bdd00ea421a0494ebe08240fddbb7457 : SUCCESS in 1h 36m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d6ebdf07e7f0474281d62a29b864c731 : SUCCESS in 2h 39m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bdf407d449464de0a3b9c9d8a546924e : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":23},{"id":"a8973aa285767f8506f97f62abfd829563615c0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 12:28:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23},{"id":"9b63d1269411b36cb934c635b351f4291ddce85f","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-06-10 12:29:45.000000000","message":"Patch Set 23: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit b8f4d43bba98fffad38806e641059a5568345ceb","accounts_in_message":[],"_revision_number":23},{"id":"bffadea9a1e8d53851ba7d6819cf0544e3bec8b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 12:29:49.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0c0e3f7a474d5991cc504da0d5229f : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"6dee93a9e7833cea4582385c09ada5655a2b167a","revisions":{"9c04f5bc3df597004f68916600b9131c2779d109":{"kind":"REWORK","_number":1,"created":"2020-03-24 12:55:53.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/1"}}},"commit":{"parents":[{"commit":"becee11a5ad9568d79c5dc77620ea34bbf691f12","subject":"Use ansible playbook for plan deploy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/becee11a5ad9568d79c5dc77620ea34bbf691f12"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c04f5bc3df597004f68916600b9131c2779d109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/9c04f5bc3df597004f68916600b9131c2779d109"}]},"branch":"refs/heads/master"},"f41b37c039956a4042c794233943d06cc6cfb711":{"kind":"REWORK","_number":2,"created":"2020-03-24 13:43:03.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/2"}}},"commit":{"parents":[{"commit":"becee11a5ad9568d79c5dc77620ea34bbf691f12","subject":"Use ansible playbook for plan deploy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/becee11a5ad9568d79c5dc77620ea34bbf691f12"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 13:42:49.000000000","tz":60},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f41b37c039956a4042c794233943d06cc6cfb711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f41b37c039956a4042c794233943d06cc6cfb711"}]},"branch":"refs/heads/master"},"e5d6c7ef5da4c57bb98a2ce8c5271db5fb4eca99":{"kind":"REWORK","_number":3,"created":"2020-03-24 15:36:15.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/3"}}},"commit":{"parents":[{"commit":"becee11a5ad9568d79c5dc77620ea34bbf691f12","subject":"Use ansible playbook for plan deploy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/becee11a5ad9568d79c5dc77620ea34bbf691f12"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 15:36:03.000000000","tz":60},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e5d6c7ef5da4c57bb98a2ce8c5271db5fb4eca99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e5d6c7ef5da4c57bb98a2ce8c5271db5fb4eca99"}]},"branch":"refs/heads/master"},"f29a8214859131002953c8bec7d55f25f5526d80":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-27 08:45:59.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/4"}}},"commit":{"parents":[{"commit":"08567fb088ba90e93393afa92d898265b46e59f2","subject":"Create *-messages container when setting deployment status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/08567fb088ba90e93393afa92d898265b46e59f2"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-03-27 08:45:59.000000000","tz":0},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f29a8214859131002953c8bec7d55f25f5526d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f29a8214859131002953c8bec7d55f25f5526d80"}]},"branch":"refs/heads/master"},"4b9ee2ea8ea94e2b450f2e015978cc3953c34e0f":{"kind":"REWORK","_number":5,"created":"2020-03-31 08:28:43.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/5"}}},"commit":{"parents":[{"commit":"becee11a5ad9568d79c5dc77620ea34bbf691f12","subject":"Use ansible playbook for plan deploy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/becee11a5ad9568d79c5dc77620ea34bbf691f12"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-31 08:28:30.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4b9ee2ea8ea94e2b450f2e015978cc3953c34e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4b9ee2ea8ea94e2b450f2e015978cc3953c34e0f"}]},"branch":"refs/heads/master"},"5e0eea3f671380418d57119f16692d8a569f315d":{"kind":"REWORK","_number":6,"created":"2020-03-31 08:40:29.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/6"}}},"commit":{"parents":[{"commit":"becee11a5ad9568d79c5dc77620ea34bbf691f12","subject":"Use ansible playbook for plan deploy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/becee11a5ad9568d79c5dc77620ea34bbf691f12"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-31 08:40:18.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5e0eea3f671380418d57119f16692d8a569f315d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5e0eea3f671380418d57119f16692d8a569f315d"}]},"branch":"refs/heads/master"},"e6585adff129bec4465da776872f6e59156b3df9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-04-06 07:26:02.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/7"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-04-06 07:26:02.000000000","tz":0},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e6585adff129bec4465da776872f6e59156b3df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e6585adff129bec4465da776872f6e59156b3df9"}]},"branch":"refs/heads/master"},"4a19ce139599cc5a1ccf7c9f704a51b5823baf16":{"kind":"REWORK","_number":8,"created":"2020-04-06 07:46:51.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/8"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-06 07:46:38.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4a19ce139599cc5a1ccf7c9f704a51b5823baf16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/4a19ce139599cc5a1ccf7c9f704a51b5823baf16"}]},"branch":"refs/heads/master"},"c2869f048786ad89c0b14f84cdc146c8f1631a97":{"kind":"REWORK","_number":9,"created":"2020-04-22 19:46:51.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/9"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-22 19:46:39.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c2869f048786ad89c0b14f84cdc146c8f1631a97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c2869f048786ad89c0b14f84cdc146c8f1631a97"}]},"branch":"refs/heads/master"},"dcf68bbda65d92b9c897fded1a64feb10c13984e":{"kind":"REWORK","_number":10,"created":"2020-04-22 20:19:35.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/10"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-22 20:19:21.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/dcf68bbda65d92b9c897fded1a64feb10c13984e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/dcf68bbda65d92b9c897fded1a64feb10c13984e"}]},"branch":"refs/heads/master"},"05214158231fb6c2ab8fabf5167d7e704723049f":{"kind":"REWORK","_number":11,"created":"2020-04-22 20:32:26.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/11"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-22 20:32:14.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/05214158231fb6c2ab8fabf5167d7e704723049f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/05214158231fb6c2ab8fabf5167d7e704723049f"}]},"branch":"refs/heads/master"},"a6b9577b262c557b6d7eedb10e3b1b845346ba98":{"kind":"REWORK","_number":12,"created":"2020-04-22 20:43:42.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/12"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-22 20:43:30.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a6b9577b262c557b6d7eedb10e3b1b845346ba98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a6b9577b262c557b6d7eedb10e3b1b845346ba98"}]},"branch":"refs/heads/master"},"f95107fddbb030a85dec081b9495696bcbd0cfd3":{"kind":"REWORK","_number":13,"created":"2020-04-22 20:45:23.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/13"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-22 20:45:13.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f95107fddbb030a85dec081b9495696bcbd0cfd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f95107fddbb030a85dec081b9495696bcbd0cfd3"}]},"branch":"refs/heads/master"},"02f32478db8c017fa435f7b3c271f9701c879e99":{"kind":"REWORK","_number":14,"created":"2020-04-23 12:12:52.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/14"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-23 12:12:40.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/02f32478db8c017fa435f7b3c271f9701c879e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/02f32478db8c017fa435f7b3c271f9701c879e99"}]},"branch":"refs/heads/master"},"e2b452f62bfa83a5ed644cbb20ad5ea6b7f44779":{"kind":"REWORK","_number":15,"created":"2020-04-28 15:02:28.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/15"}}},"commit":{"parents":[{"commit":"624a61f206724771f1903db6130a21ade3a6f598","subject":"Update verbosity so it is consistently set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/624a61f206724771f1903db6130a21ade3a6f598"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-04-28 15:02:06.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e2b452f62bfa83a5ed644cbb20ad5ea6b7f44779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e2b452f62bfa83a5ed644cbb20ad5ea6b7f44779"}]},"branch":"refs/heads/master"},"36255fd7321ab4b949f19d25b2f691b83c577008":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-04-29 11:37:34.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/16"}}},"commit":{"parents":[{"commit":"c28e805b637bf4fa3f2a6cce8e32649c306a591d","subject":"Merge \"Remove workflow_engine from client setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c28e805b637bf4fa3f2a6cce8e32649c306a591d"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-04-29 11:37:34.000000000","tz":0},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/36255fd7321ab4b949f19d25b2f691b83c577008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/36255fd7321ab4b949f19d25b2f691b83c577008"}]},"branch":"refs/heads/master"},"418dc1c40d258a3f2ea96a1f7eadb37c14f6c97f":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-05-04 07:40:37.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/17"}}},"commit":{"parents":[{"commit":"5951a20650e64c5ed06eb553a2caf35449d05ce8","subject":"Merge \"Fix stack create verbosity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5951a20650e64c5ed06eb553a2caf35449d05ce8"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"mathieu bultel","email":"mat.bultel@gmail.com","date":"2020-05-04 07:40:37.000000000","tz":0},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/418dc1c40d258a3f2ea96a1f7eadb37c14f6c97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/418dc1c40d258a3f2ea96a1f7eadb37c14f6c97f"}]},"branch":"refs/heads/master"},"511a1df314b3cddad196a9d80adb7ad0a69dadc6":{"kind":"REWORK","_number":18,"created":"2020-05-18 13:07:15.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/18"}}},"commit":{"parents":[{"commit":"5951a20650e64c5ed06eb553a2caf35449d05ce8","subject":"Merge \"Fix stack create verbosity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5951a20650e64c5ed06eb553a2caf35449d05ce8"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-18 13:07:04.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/511a1df314b3cddad196a9d80adb7ad0a69dadc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/511a1df314b3cddad196a9d80adb7ad0a69dadc6"}]},"branch":"refs/heads/master"},"c09c31662a8a750e974cdaf5af76561b99153c46":{"kind":"REWORK","_number":19,"created":"2020-05-18 13:08:42.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/19"}}},"commit":{"parents":[{"commit":"5951a20650e64c5ed06eb553a2caf35449d05ce8","subject":"Merge \"Fix stack create verbosity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5951a20650e64c5ed06eb553a2caf35449d05ce8"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-18 13:08:31.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c09c31662a8a750e974cdaf5af76561b99153c46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c09c31662a8a750e974cdaf5af76561b99153c46"}]},"branch":"refs/heads/master"},"40e910e49fec1724ebb8b09d9aeec16735d26944":{"kind":"REWORK","_number":20,"created":"2020-05-18 13:17:02.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/20"}}},"commit":{"parents":[{"commit":"5951a20650e64c5ed06eb553a2caf35449d05ce8","subject":"Merge \"Fix stack create verbosity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5951a20650e64c5ed06eb553a2caf35449d05ce8"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-18 13:16:52.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/40e910e49fec1724ebb8b09d9aeec16735d26944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/40e910e49fec1724ebb8b09d9aeec16735d26944"}]},"branch":"refs/heads/master"},"e24b553b417cf2134c435990563d967d2a79b4b9":{"kind":"REWORK","_number":21,"created":"2020-05-28 06:09:15.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/21"}}},"commit":{"parents":[{"commit":"5951a20650e64c5ed06eb553a2caf35449d05ce8","subject":"Merge \"Fix stack create verbosity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5951a20650e64c5ed06eb553a2caf35449d05ce8"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-05-28 06:09:02.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e24b553b417cf2134c435990563d967d2a79b4b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e24b553b417cf2134c435990563d967d2a79b4b9"}]},"branch":"refs/heads/master"},"7f82a4d593fbac4f3d914323c05d2ca99736ec06":{"kind":"REWORK","_number":22,"created":"2020-06-09 13:47:06.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/22"}}},"commit":{"parents":[{"commit":"f66b42757b3141a4cbbd42cc7d2937874b7b78f0","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f66b42757b3141a4cbbd42cc7d2937874b7b78f0"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-06-09 13:46:47.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7f82a4d593fbac4f3d914323c05d2ca99736ec06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/7f82a4d593fbac4f3d914323c05d2ca99736ec06"}]},"branch":"refs/heads/master"},"6dee93a9e7833cea4582385c09ada5655a2b167a":{"kind":"REWORK","_number":23,"created":"2020-06-09 17:13:59.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/29/714629/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/29/714629/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/29/714629/23"}}},"commit":{"parents":[{"commit":"f66b42757b3141a4cbbd42cc7d2937874b7b78f0","subject":"Merge \"Switch to newer openstackdocstheme and reno versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f66b42757b3141a4cbbd42cc7d2937874b7b78f0"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-03-24 12:55:04.000000000","tz":60},"committer":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-06-09 17:13:47.000000000","tz":120},"subject":"Implement validations-libs framework for ALL validator commands","message":"Implement validations-libs framework for ALL validator commands\n\nImplement All command with validations libs framework\n\nChange-Id: I0443da140020540917396cdfe00da8ac29d20297\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6dee93a9e7833cea4582385c09ada5655a2b167a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6dee93a9e7833cea4582385c09ada5655a2b167a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
