)]}'
{"id":"openstack%2Fpython-tripleoclient~760351","triplet_id":"openstack%2Fpython-tripleoclient~master~I32c1e006d10ceb541aadb486befd8aa8fc6aa045","project":"openstack/python-tripleoclient","branch":"master","topic":"validator/no-auth","hashtags":[],"change_id":"I32c1e006d10ceb541aadb486befd8aa8fc6aa045","subject":"Do not require authentication for the validator CLI","status":"MERGED","created":"2020-10-29 12:20:24.000000000","updated":"2020-11-02 12:22:35.000000000","submitted":"2020-11-02 12:05:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760351-1604318739031-8dccae41","meta_rev_id":"4405eba13e63e199708052a53052ddb57946209b","_number":760351,"virtual_id_number":760351,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"date":"2020-10-31 20:39:50.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-02 12:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2020-10-30 13:05:12.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2020-10-30 13:21:33.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2020-11-02 08:03:33.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-30 17:52:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2020-11-02 08:03:33.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"date":"2020-11-02 06:28:36.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 13:06:10.000000000","updated_by":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"reviewer":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"state":"REVIEWER"},{"updated":"2020-10-30 08:31:54.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2020-10-30 08:31:58.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-10-30 08:32:00.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2020-10-30 13:05:12.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-10-30 13:21:33.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2020-10-30 17:52:42.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2020-10-31 20:39:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-02 06:28:36.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2020-11-02 08:03:33.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2020-11-02 12:05:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e2d4d409a5c860547ad93491595288ef134cb39","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-29 12:20:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9d1eb6c91f4a5e2636d35d49ef95f8dcad9fe4a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-29 12:20:54.000000000","message":"Patch Set 1: Workflow-1\n\nNeed to figure out how to properly handle the \"run\" subcommand","accounts_in_message":[],"_revision_number":1},{"id":"44c1feea425168e81bb373e1f738180c12d5961f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 21:46:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a4b81673ff44d8d9489932e64cb343c : SUCCESS in 6m 11s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3e5ee87024bc49abb9028c02881606df : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43d85651f634fd6960bb99867285016 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a63a7d257071418aa3a1f36015aa2203 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba86b1b174a64723a57b01bfea4553c4 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91d98c710d484327a663049862572f94 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de56ecdb041b40c7a44119f7419f01a4 : SUCCESS in 5m 57s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/83799df42ef24e79af2cb4641db6c149 : SUCCESS in 59m 47s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/60fd95a52bb5488b907462e976c99349 : SUCCESS in 33m 52s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/d960d3057a20456c8e795e6754e3582e : SUCCESS in 37m 24s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c79b1ff4dbe049998f5fceee40a550ff : SUCCESS in 3h 42m 55s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4471b9f803d949589c98deb6dea66cda : SUCCESS in 1h 55m 29s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/3dcc233d9b4b476e944e10dd3f161dc6 : SUCCESS in 1h 01m 10s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/55599f8d7aa0496582115a29654373de : SUCCESS in 2h 14m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9944882befa24a89a57f367a863e5e1a : SUCCESS in 1h 58m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8c6e0481ed7f45beb492a79092cac901 : SUCCESS in 2h 13m 46s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c93c548e3d0f47e1a91d596e2eff658f : SUCCESS in 1h 41m 53s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/94e8b3db92c04984a6c34811d664d7a1 : SUCCESS in 1h 24m 29s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f1be777b932d46e79d7dd5d607a76346 : FAILURE in 27m 32s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b59b988258514846a9b4a6d49f947eec : FAILURE in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edfe153d6765f32d280808a256d8ba4018dab74a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-30 08:09:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e6f3a1348430d40deb274bc5c89afadcdaf286a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-30 08:22:47.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b79cd09fdb36fea3a3118f700465fb51b58d86df","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-30 09:33:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1c66433a8c583398d44a7aeccbb2c081e550553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 12:29:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b403931c1974befa31972ea3d446459 : SUCCESS in 6m 17s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f4827d4b00014ae59cccaf8598db1bec : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da863020b6c4430db662239b511344ba : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d53a8a577d01464fa5665e5eb677f59d : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bad39f0154644dd89420e681bff3c67 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7b27b9818314e16948cd983986a3730 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33f86632c024b1797d1d919db81e59b : SUCCESS in 5m 08s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/ec51bcf5bf7f430791803fb0f2ecf899 : SUCCESS in 1h 02m 50s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/33cac6ccd8d84973bcf0bf7e6eac2081 : SUCCESS in 35m 45s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/038a677e37f4400591d609e378239387 : SUCCESS in 47m 36s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/05374714380b4e529bffac774b117634 : SUCCESS in 4h 01m 41s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0866acfdc921414b9eea3869314baef9 : SUCCESS in 2h 15m 24s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0cb89e88093d41518390b73dd95253de : SUCCESS in 44m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f335280c4e254e5d9d763a3f904600a6 : SUCCESS in 2h 13m 28s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c3d6cdc24a9845188c33e64723d13b47 : SUCCESS in 2h 04m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/406e8240e702499782197e361a0ef725 : SUCCESS in 2h 13m 16s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/93d17cfa1fcf4f348d049b88d13f7971 : SUCCESS in 1h 35m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d9522034966e448ba931cde607a07a64 : SUCCESS in 1h 23m 14s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7ebe219836ef4f5981ad39821ea41389 : FAILURE in 28m 35s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0ae5307631c04d888580294de3fe111d : FAILURE in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"86e082c69d59522320ba2712cab10c4bf597a546","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-30 12:31:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8083a716809720e1d82c6c271d5a232c7df55c99","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-30 13:03:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6ee63f35263668b62ac643f7583cbe9909a60ba8","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-10-30 13:04:45.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4921a11fa1e3c689156a101b227b4807b25c0049","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-30 13:05:12.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d1dbaff62d0e7ab58f92cc40985eb945c848eabc","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-10-30 13:05:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"14f1eb98b0a430b90931af7d74eff6698ca2b871","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2020-10-30 13:21:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"693036b2267aa708666de4233f02f0e0e63429d1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2020-10-30 17:52:42.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"06a476798d06e0f4aa019a02d9e0ef3d06ae004c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 22:30:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b276661a9c7d456197d8b8bb6dc5cd35 : SUCCESS in 8m 45s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/93e5874f660c4d258d178d698e9f7023 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16515cb71bf140b3babf7cf95eeac42c : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a30cdc4b95ca4fb58e576499a11d6c64 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85f1204b9e2d43a482d46471a15e05ca : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16c2a55719a14b2e869535f13d9194b4 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30cd7016b3a24b22925496c5d59b3ce8 : SUCCESS in 5m 00s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/88602703131f4aac900cfa567484926b : SUCCESS in 1h 03m 47s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/e40e361ee3654ba8b17cec078269e50b : SUCCESS in 42m 29s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/2f88e3608599409881be389e0c7b9b77 : SUCCESS in 50m 39s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9ec9068e034b48b3a247419f3e18c4ad : FAILURE in 2h 24m 59s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/688c91aeb5fd45e795c2a1f17bcee40b : SUCCESS in 2h 29m 54s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6bc9672a22234c54915ee5c1c6fc442f : SUCCESS in 52m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"26fe5c3dfff6d0fe1762f8da7fc1668478f2e8fb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-31 20:39:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c82b8841b9952df39f8900b7ca5c8a805102edd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 00:27:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf86a5a034a643979cfa423e44745d1d : SUCCESS in 7m 24s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ccaad06862f04fe7ad941f30cc69ec37 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ec5a8f7ca94896952f34b6eeef286c : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18fc16ecf0c04e109eb28970bde9f045 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bf9750b18174e37ad52d8a471d23f47 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59deab88e33f43c19bc9677fd06c420b : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9125d8bfd9524becb481b268fd7afe72 : SUCCESS in 7m 25s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/e5671b753b3a44b69fa29e7d749eb56e : SUCCESS in 1h 00m 09s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/1d835b27e21647bf9d10efad31c03635 : SUCCESS in 32m 02s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/feeba919ab96487f859950718b16f606 : SUCCESS in 37m 59s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2de0991186d04dc78afef96ead5952b3 : SUCCESS in 3h 41m 41s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/240855283dbe47a2bc516bae9fc89dd7 : SUCCESS in 2h 00m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ecb129bcd235471dbfbf85090239de90 : SUCCESS in 2h 04m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b37597ce8c7c4e99b26d652e1bc33ba1 : SUCCESS in 1h 58m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/89f48c618e894886a8d229524842f69e : SUCCESS in 2h 10m 01s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8e2256e9e43a45a0bbf1c024caec2d8d : SUCCESS in 1h 51m 04s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a10857b72de9438aac4b358e5a467e8a : SUCCESS in 1h 27m 46s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/ecdb6229cd2e43e7b61ce4d9d78b3df7 : FAILURE in 28m 15s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/ac694ce377244707acde3c8f4b1b2b59 : FAILURE in 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3ebb07e889dcab6b4eed61d4d173928ea619cdd2","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2020-11-02 07:45:22.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6ee1b483a6452488fc16e7300ff5a40ce53df676","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2020-11-02 08:03:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1d4306c5b2199f9c54814df8cac1e5b7ca0e8a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 08:03:48.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3a2638fe286ec0ec53571f5c61896665fe33e8e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 12:05:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d15979984fee4d92860e79b036217ec4 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afeb04cc32e04c3bba2ea21843eddce9 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa68aa34ed8a4f8c97aadede82fc6da0 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/434ce0edbde8490ba9d2c552b252e846 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab62632ec4c84911876c54ee462861eb : SUCCESS in 5m 51s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/6149a6e51900406989e1f2ea876fca1d : SUCCESS in 52m 46s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/fea0a53a334e4c3a9a6246156b3cf2b7 : SUCCESS in 29m 05s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/7781d5fae4cc49ac9048f14534b9e42e : SUCCESS in 29m 46s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e32c9b3575cb4e97857e4e7dee375e7a : SUCCESS in 3h 02m 44s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/391417ff69af413093f81e0780cbb979 : SUCCESS in 57m 25s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6a930f35517b43469e8c19ca4a81b17c : SUCCESS in 50m 30s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8305c37c2ecb42429e02432a52255ea8 : SUCCESS in 1h 55m 13s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0bc4bcce2c5c4a3cb3d6cd807543106b : SUCCESS in 1h 26m 32s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8acfe6b496c948efbaff1d17544a6d7a : SUCCESS in 1h 45m 29s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2bc2e4cef74841a0b3ad9c102dc55c2f : SUCCESS in 1h 18m 09s","accounts_in_message":[],"_revision_number":5},{"id":"ee7661a0d30f142aa372b7d75d660c609be7056f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 12:05:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c98a3ee649df60d392e431abb6c3011ff3644895","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 12:07:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ed07e3d6d341eeb458f43771f5647c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5},{"id":"f288acbcea1352e8d27e39ac15f539db9adc4b5c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-11-02 12:21:58.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit f9a16b143ccd26e4f2dbe8606c2cc987869d83ac","accounts_in_message":[],"_revision_number":5},{"id":"a30472124baeac64252843778a610ebdd3b980c0","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-11-02 12:22:35.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit b209a75b6f8725ba763831428eaddb1bb16f1e0b","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aafd981948b28b7786c9c8d5f9b0e7e1b3f9371f","revisions":{"f94c92a84a3d634150a378d1baae2425669649ec":{"kind":"REWORK","_number":1,"created":"2020-10-29 12:20:24.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/51/760351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/51/760351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/1"}}},"commit":{"parents":[{"commit":"03da56c94997049083e2e1dc8901eeb6517f8979","subject":"Merge \"Remove upgrade_cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/03da56c94997049083e2e1dc8901eeb6517f8979"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-29 12:13:11.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-29 12:17:15.000000000","tz":60},"subject":"WIP - Do not require authentication for the validator CLI","message":"WIP - Do not require authentication for the validator CLI\n\nSince most of the actions of the validator is based on the filesystem,\nwe really don\u0027t need to require auth.\nThe only subcommand that might require it is the \"run\", since it might\ngo through the inventory generation.\n\nResolves: rhbz#1892323\nCloses-Bug: #1902052\n\nChange-Id: I32c1e006d10ceb541aadb486befd8aa8fc6aa045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f94c92a84a3d634150a378d1baae2425669649ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/f94c92a84a3d634150a378d1baae2425669649ec"}]},"branch":"refs/heads/master"},"801558f106c7cea97875e2527c24e8e2064201f4":{"kind":"REWORK","_number":2,"created":"2020-10-30 08:09:13.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/51/760351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/51/760351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/2"}}},"commit":{"parents":[{"commit":"03da56c94997049083e2e1dc8901eeb6517f8979","subject":"Merge \"Remove upgrade_cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/03da56c94997049083e2e1dc8901eeb6517f8979"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-29 12:13:11.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-30 08:08:05.000000000","tz":60},"subject":"WIP - Do not require authentication for the validator CLI","message":"WIP - Do not require authentication for the validator CLI\n\nSince most of the actions of the validator is based on the filesystem,\nwe really don\u0027t need to require auth.\nThe only subcommand that might require it is the \"run\", since it might\ngo through the inventory generation.\n\nAlso, pass the \"--os-cloud\" parameter down to the\ntripleo-ansible-inventory script. Value is hardcoded, since it is an\nentry name we can\u0027t set (yet). It allows to authenticate only for the\nrelevant part of the validation run.\n\nResolves: rhbz#1892323\nCloses-Bug: #1902052\n\nChange-Id: I32c1e006d10ceb541aadb486befd8aa8fc6aa045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/801558f106c7cea97875e2527c24e8e2064201f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/801558f106c7cea97875e2527c24e8e2064201f4"}]},"branch":"refs/heads/master"},"138c5082340e53d99e88cc4deb81a1b8e11d756c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-30 08:22:47.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/51/760351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/51/760351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/3"}}},"commit":{"parents":[{"commit":"03da56c94997049083e2e1dc8901eeb6517f8979","subject":"Merge \"Remove upgrade_cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/03da56c94997049083e2e1dc8901eeb6517f8979"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-29 12:13:11.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-30 08:22:34.000000000","tz":60},"subject":"Do not require authentication for the validator CLI","message":"Do not require authentication for the validator CLI\n\nSince most of the actions of the validator is based on the filesystem,\nwe really don\u0027t need to require auth.\nThe only subcommand that might require it is the \"run\", since it might\ngo through the inventory generation.\n\nAlso, pass the \"--os-cloud\" parameter down to the\ntripleo-ansible-inventory script. Value is hardcoded, since it is an\nentry name we can\u0027t set (yet). It allows to authenticate only for the\nrelevant part of the validation run.\n\nResolves: rhbz#1892323\nCloses-Bug: #1902052\n\nChange-Id: I32c1e006d10ceb541aadb486befd8aa8fc6aa045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/138c5082340e53d99e88cc4deb81a1b8e11d756c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/138c5082340e53d99e88cc4deb81a1b8e11d756c"}]},"branch":"refs/heads/master"},"a626509dfa4e099e1b89df62218aaadcc9a03036":{"kind":"REWORK","_number":4,"created":"2020-10-30 13:03:05.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/51/760351/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/51/760351/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/4"}}},"commit":{"parents":[{"commit":"03da56c94997049083e2e1dc8901eeb6517f8979","subject":"Merge \"Remove upgrade_cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/03da56c94997049083e2e1dc8901eeb6517f8979"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-29 12:13:11.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-30 13:02:53.000000000","tz":60},"subject":"Do not require authentication for the validator CLI","message":"Do not require authentication for the validator CLI\n\nSince most of the actions of the validator is based on the filesystem,\nwe really don\u0027t need to require auth.\nThe only subcommand that might require it is the \"run\", since it might\ngo through the inventory generation.\n\nAlso, pass the \"--os-cloud\" parameter down to the\ntripleo-ansible-inventory script. Value is hardcoded, since it is an\nentry name we can\u0027t set (yet). It allows to authenticate only for the\nrelevant part of the validation run.\n\nResolves: rhbz#1892323\nCloses-Bug: #1902052\n\nChange-Id: I32c1e006d10ceb541aadb486befd8aa8fc6aa045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a626509dfa4e099e1b89df62218aaadcc9a03036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a626509dfa4e099e1b89df62218aaadcc9a03036"}]},"branch":"refs/heads/master"},"aafd981948b28b7786c9c8d5f9b0e7e1b3f9371f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-10-30 13:04:45.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/51/760351/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/51/760351/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/51/760351/5"}}},"commit":{"parents":[{"commit":"5face64ed5183e6104cc4018717535d3fffea89e","subject":"Merge \"Remove undercloud_update_packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/5face64ed5183e6104cc4018717535d3fffea89e"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-10-29 12:13:11.000000000","tz":60},"committer":{"name":"Cédric Jeanneret (Tengu)","email":"cjeanner@redhat.com","date":"2020-10-30 13:04:45.000000000","tz":0},"subject":"Do not require authentication for the validator CLI","message":"Do not require authentication for the validator CLI\n\nSince most of the actions of the validator is based on the filesystem,\nwe really don\u0027t need to require auth.\nThe only subcommand that might require it is the \"run\", since it might\ngo through the inventory generation.\n\nAlso, pass the \"--os-cloud\" parameter down to the\ntripleo-ansible-inventory script. Value is hardcoded, since it is an\nentry name we can\u0027t set (yet). It allows to authenticate only for the\nrelevant part of the validation run.\n\nResolves: rhbz#1892323\nCloses-Bug: #1902052\n\nChange-Id: I32c1e006d10ceb541aadb486befd8aa8fc6aa045\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/aafd981948b28b7786c9c8d5f9b0e7e1b3f9371f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/aafd981948b28b7786c9c8d5f9b0e7e1b3f9371f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
