)]}'
{"id":"openstack%2Fpython-tripleoclient~761702","triplet_id":"openstack%2Fpython-tripleoclient~master~I55f4b8cf2ad971e196fbd06db743d07ce3135ea8","project":"openstack/python-tripleoclient","branch":"master","topic":"bug/1902846","hashtags":[],"change_id":"I55f4b8cf2ad971e196fbd06db743d07ce3135ea8","subject":"Map build host repos to distro.repos.d instead of yum.repos.d","status":"MERGED","created":"2020-11-06 06:29:04.000000000","updated":"2020-12-16 06:50:13.000000000","submitted":"2020-11-30 14:43:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761702-bug/1902846","meta_rev_id":"19d61af871a84ba2ef6484b30fe135d51e32cb9a","_number":761702,"virtual_id_number":761702,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2020-11-24 16:46:57.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-30 14:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-24 22:56:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2020-11-20 14:19:48.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-19 11:25:26.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2020-11-19 11:25:44.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2020-11-20 14:19:48.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 13:55:39.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2020-11-19 11:25:26.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2020-11-19 11:25:44.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-11-19 20:31:22.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-11-20 03:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 14:19:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-24 22:56:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09568c97a1aaf49b8bd8503dc6291459419a4198","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-06 06:29:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30e76173ebe7bc3b3a4edc10fa99173830dc1ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 10:53:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9593cabf19ee4b9684bcb557dceb3e41 : SUCCESS in 6m 32s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e026565dbe4944f5862f9763d341c029 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b8e28b3bf54ce8a4279474674fb9e9 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9fbb3ac851349c2bc0c569868d7375d : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e8ed26b2898417d9e34a306801c80c2 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/636969bfeca349e8823276ce82dd3d2a : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcde681b6b3f49bc811e2bdf9084a64d : SUCCESS in 5m 29s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/a3e8a9cdd51c477e9325fbad11a95a82 : SUCCESS in 1h 08m 07s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/530014518b864bd99f56889090de3c59 : SUCCESS in 32m 39s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/f7005f7641d446f9bfb7c5008fbaf636 : SUCCESS in 41m 49s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/af672e4fb68a4308a707a7b8c44f44d9 : FAILURE in 26m 44s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c2da892bd8644d1e957a2809de3f2467 : SUCCESS in 3h 14m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/1e13779b01f64e1c9f11abb5010d9f8a : SUCCESS in 1h 39m 17s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/bd6ea13bb30f4f03a1ab0c09b48d4dfc : SUCCESS in 57m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/84ff532b5dcd43babe3a3df2cb7f6d47 : SUCCESS in 1h 54m 37s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1e79eea75c2947c98a76b9b0747f9d0f : SUCCESS in 1h 53m 48s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/851e3413663c4038863021938e8d8f0c : SUCCESS in 1h 58m 11s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d508d02cbdf841e2b22a5822254c03da : SUCCESS in 1h 27m 52s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/61856856092a49edbf9c447d2058545b : SUCCESS in 1h 17m 27s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/eed84f33d8da4108ab90c090f4838b2b : FAILURE in 22m 27s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/02a9de195a08452ebb66112dd7dfa131 : FAILURE in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8052831374b8ea8751d6f7dfbb608ea593367604","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-06 14:46:44.000000000","message":"Patch Set 1:\n\nthat\u0027s quite elegant!","accounts_in_message":[],"_revision_number":1},{"id":"1f080ad866238e4c0398bbaa0876fbdeac1c227f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-06 14:48:10.000000000","message":"Patch Set 1: Code-Review+1\n\nwhere did it fail https://a6170f8838bfb2ea8967-dcc4c568f423fe000e95109bdebe1978.ssl.cf1.rackcdn.com/761702/1/check/tripleo-build-containers-ubi-8/af672e4/job-output.txt ?..","accounts_in_message":[],"_revision_number":1},{"id":"6db77c3e8ee88676a9d980fa3ae1ec4e0b8bd0cb","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-06 14:52:18.000000000","message":"Patch Set 1:\n\n\u003e where did it fail https://a6170f8838bfb2ea8967-dcc4c568f423fe000e95109bdebe1978.ssl.cf1.rackcdn.com/761702/1/check/tripleo-build-containers-ubi-8/af672e4/job-output.txt\n \u003e ?..\n\nException information: \u0027int\u0027 object is not callable\n\nhttps://a6170f8838bfb2ea8967-dcc4c568f423fe000e95109bdebe1978.ssl.cf1.rackcdn.com/761702/1/check/tripleo-build-containers-ubi-8/af672e4/logs/containers-build-errors.log\n\nI didn\u0027t get why it\u0027s failing, failed in other dependent patch so looks real issue.","accounts_in_message":[],"_revision_number":1},{"id":"8f47ee51ab44594a45970e5f58079a066972c0f8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-09 07:17:13.000000000","message":"Patch Set 1:\n\n\u003e \u003e where did it fail https://a6170f8838bfb2ea8967-dcc4c568f423fe000e95109bdebe1978.ssl.cf1.rackcdn.com/761702/1/check/tripleo-build-containers-ubi-8/af672e4/job-output.txt\n \u003e \u003e ?..\n \u003e \n \u003e Exception information: \u0027int\u0027 object is not callable\n \u003e \n \u003e https://a6170f8838bfb2ea8967-dcc4c568f423fe000e95109bdebe1978.ssl.cf1.rackcdn.com/761702/1/check/tripleo-build-containers-ubi-8/af672e4/logs/containers-build-errors.log\n \u003e \n \u003e I didn\u0027t get why it\u0027s failing, failed in other dependent patch so\n \u003e looks real issue.\n\nOk got it, it failed at building rsyslog container but error didg\u0027t got reported right,\nhttps://a6170f8838bfb2ea8967-dcc4c568f423fe000e95109bdebe1978.ssl.cf1.rackcdn.com/761702/1/check/tripleo-build-containers-ubi-8/af672e4/logs/buildah-builds/container-builds/0f71a32e-8552-459f-b407-4673f1894025/base/rsyslog/rsyslog-build.log\n\nIt\u0027s failing as rsyslog subpackages like rsyslog-elasticsearch,rsyslog-gnutls, rsyslog-mmjsonparse, rsyslog-mmnormalize not available in ubi8 repo:- https://cdn-ubi.redhat.com/content/public/ubi/dist/ubi8/8/x86_64/appstream/os/Packages/r/, The issue should not occur if rsyslog is removed from the list(it will be auto pulled as dependency by other subpackages) as dnf then will pick based from centos-stream repo or centos8.3 is out or nobest option is used.","accounts_in_message":[],"_revision_number":1},{"id":"5f0b299fead5c905ccaedab98571fb281e882777","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-12 18:15:33.000000000","message":"Patch Set 1:\n\nok after fixing the int error, the actual issue is that it can\u0027t install rsyslog.\n\n2020-11-12 18:14:46.779 401722 ERROR openstack Stderr: \u0027error building at STEP \"RUN dnf -y install rsyslog rsyslog-elasticsearch rsyslog-gnutls rsyslog-mmjsonparse rsyslog-mmnormalize \u0026\u0026 dnf clean all \u0026\u0026 rm -rf /var/cache/dnf\": error while running runtime: exit status 1\\n\u0027","accounts_in_message":[],"_revision_number":1},{"id":"f433969bd5610b9cc3f878cb0a556bb1f5b4cccb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-18 13:50:57.000000000","message":"Patch Set 1:\n\n[root@ccdbcaa72a65 /]# yum list rsyslog rsyslog-elasticsearch rsyslog-gnutls rsyslog-mmjsonparse rsyslog-mmnormalize\nLast metadata expiration check: 0:06:05 ago on Wed Nov 18 13:44:39 2020.\nAvailable Packages\nrsyslog.x86_64                                                                                       8.1911.0-6.el8                                                                          ubi-8-appstream\nrsyslog-elasticsearch.x86_64                                                                         8.1911.0-3.el8                                                                          AppStream      \nrsyslog-gnutls.x86_64                                                                                8.1911.0-3.el8                                                                          AppStream      \nrsyslog-mmjsonparse.x86_64                                                                           8.1911.0-3.el8                                                                          AppStream      \nrsyslog-mmnormalize.x86_64","accounts_in_message":[],"_revision_number":1},{"id":"42cf0183c2b4ef8d4fd5817befb9e00d3dbd162e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-18 13:52:47.000000000","message":"Patch Set 1: Code-Review+2\n\nSo we need AppStream repo from centos8 to make this working. For example tripleo-repos tool could support that like https://review.opendev.org/#/c/762520/\n\nPerhaps you could similarly do to the repo setup scripts used in CI?\n\nAnyway, we need this patch, it works very well and allows \"mergin\" both ubi8 and host-bind-mounted repos","accounts_in_message":[],"_revision_number":1},{"id":"9654cc7e345872ef7b41aae9c8493c518f077661","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-18 15:03:35.000000000","message":"Patch Set 1: Code-Review+1\n\nsadly, ubi-8-appstream provides rsyslog 8.1911.0-6.el8, while Centos AppStream prodvides dependencies for 8.1911.0-3.el8, and this conflicts.\n\nSo we still have a problem here. Maybe there should be --nobest option for TCIB?..","accounts_in_message":[],"_revision_number":1},{"id":"1eced1bd9f73a972306eb0cde606634effa3cabb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-18 15:32:02.000000000","message":"Patch Set 1:\n\nshould work with https://review.opendev.org/#/c/763203/","accounts_in_message":[],"_revision_number":1},{"id":"be4e9d75a2b04a9928a233f26be62f1a0dc1b92f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-18 15:36:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a9a6f29d6fe546cdf6e175099e566bdebcdf0158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 03:44:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c159ca8ceff9477abbf59dc354d531b4 : SUCCESS in 5m 36s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9d6cb5b04435450db9b4fabc1f0d8db1 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b45b817c694f4ae09c07898a78fd18a7 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39fd518aa444b17910b4ab8c5f4e84e : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f40fd9c5697492eb7c47db1df7a208e : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45122c2bc2fa4f149d853781019b85ea : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e0b456c549481f976aeff809fad970 : SUCCESS in 5m 14s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/76e08629116147a4a96378a08eb48f7d : SUCCESS in 50m 53s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/d79af5caf7d54be8989ac3a54b4b24b5 : SUCCESS in 25m 36s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/7db408851b234f6da946de9b010f3131 : SUCCESS in 39m 46s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5325f0fe15a64bd3b65ec07a69fe62b4 : SUCCESS in 4h 05m 53s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/c0f3f3b58f7d4ee4be074002c2e58778 : SUCCESS in 4h 06m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2a118a9d09f344d7bdc73bcd62545228 : SUCCESS in 2h 32m 48s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2f704f4f1bf04e8bb3a98b024da44485 : TIMED_OUT in 3h 08m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a3fcede325874fdf8cbc14fb97d3a7ed : SUCCESS in 2h 04m 27s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0cadbd1a63194ad1b0ff2661a7856d9a : SUCCESS in 1h 59m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ebd9a4294185445f89fa7937c7a5fdbf : SUCCESS in 2h 06m 49s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8d74a4d8518e400fbf8d280f2ea49bc3 : SUCCESS in 1h 33m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c09aba1de7f8416e94a681b30bfccee2 : SUCCESS in 1h 28m 33s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c008d536f0d24756ad511fa8c85b5d3c : SUCCESS in 1h 59m 09s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6685d975c07b463daa46095e26201eab : SUCCESS in 2h 46m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f32e727e44dc9c820cc93055db0e23274f91a2b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-19 11:25:21.000000000","message":"Patch Set 2:\n\nrecheck almost perfection","accounts_in_message":[],"_revision_number":2},{"id":"49d099297046cf8b735043b73c74864edcf2d2b7","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-19 11:25:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4dcea07df82be1ba8cadaca978ccef1a563dd9a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 16:30:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/865a5b9172e04027af08a1e9261f93b8 : SUCCESS in 8m 16s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9b9f8e053a0a4edb84a16608481d5a64 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9883350d25214722ab821a9456f50d9d : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b25a1636bb0245f78d051e283ff6778f : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dffaf32b183f475eb0b1ae410112c223 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c98d14360155468eb936a1c5ba5465f2 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fead4e9c88d54a10b5e826965398c924 : SUCCESS in 18m 30s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/699a1cbf595f493fbef8835e080945b9 : SUCCESS in 1h 16m 29s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/b7e80344ab564e859222b595949e31ca : SUCCESS in 25m 34s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/56b055b75aa14acfab519f1a3dcce239 : SUCCESS in 33m 26s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e7ef12e01d46491b84158a4a7b54d25d : SUCCESS in 3h 27m 21s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/3c65a2cd62c04da3ab118770df9ff84a : SUCCESS in 3h 25m 12s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f661476b127c4e809bff7ea383209a0f : SUCCESS in 1h 03m 15s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/89ce09d7832f48d3a599e7eb8fd38026 : TIMED_OUT in 3h 19m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d4a22a3f563c4cd0a9ed08871f6a6b86 : SUCCESS in 2h 19m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b23a09aa68241b9b9c490d8948019ac : SUCCESS in 1h 50m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/cf91adaf9db64621a8dc0b67d5c8ecb1 : SUCCESS in 2h 09m 41s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5a48b8d932a14318b16d5aff0ff0e484 : SUCCESS in 1h 49m 49s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2d794ecf02f343c9af3f1b9234448d89 : SUCCESS in 1h 16m 49s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/37f9848652fa4763b71b7d6a0c8ace5a : FAILURE in 2h 09m 49s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/8a686e2a679c4f0188c16ea8446c3da4 : FAILURE in 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"676871bd0f86557cbc2bdc72bbad6b7631691c98","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-11-19 20:31:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8421135ae473c9a7024c4fe367327e3e9159cd3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 03:30:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cb2f7dbf6604738bbebe2525052717f : SUCCESS in 5m 55s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/a0970eb2d8204d7083cff4bc64fcae97 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d022b405dfcb41b893cb4abe27ced295 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/862264d772a941cab9db4ea5b89007da : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/119ce07f56e649bd84fbba93c7b74a91 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76e6512424db4bf0a204d3223c51075d : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d19a94f1bfce42ebb9be9ab3c7893fe6 : SUCCESS in 6m 34s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/756cc22854f14e9687c363fd20767474 : SUCCESS in 58m 02s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/481e7283c27d4bd2aa4aa9a09d8382eb : SUCCESS in 47m 13s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/dbf89b33382f434182c6b1e006237f73 : SUCCESS in 40m 25s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7d145bf183d9454d807fb7df89b8370c : SUCCESS in 3h 59m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/49d0d3a5cb4d4fbf9866358eadd14dc4 : SUCCESS in 4h 08m 23s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0099a23dd7734eaeacce72e6659ed51e : SUCCESS in 1h 05m 18s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/df115e77d1344ad58c3a9fc090f0b5c7 : SUCCESS in 59m 37s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/95f2855fdd8e462396eb1fcea662de2b : SUCCESS in 2h 13m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8f337439cff04dfda1a750f9ad6622f1 : SUCCESS in 1h 48m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/8ab4bd36546d43fdaad9931966b77c1c : FAILURE in 18m 30s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/567328664bbe4f3a814e97c035b255ce : SUCCESS in 1h 28m 26s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5a1bdfdd810d4a3f9cec7012cfa00743 : SUCCESS in 1h 15m 24s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/41fa01155ef64d008218392dbf3e98b7 : SUCCESS in 1h 47m 45s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/84dd353f6bb4445aa929220a1672b7de : SUCCESS in 2h 47m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a84fcce0512d48c0b3881237c6b79b33f1571cfc","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-11-20 09:50:02.000000000","message":"Patch Set 2:\n\nsup\ncare to merge that?","accounts_in_message":[],"_revision_number":2},{"id":"4590b21d9bf0f0c879ac6c50d143c775850e698d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-20 14:19:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25a19aaf96b1e1273b40fce4fafc8eca15a17478","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-11-24 16:46:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6487af82db6a6f564f2d15968985a9a01a8b15fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-24 22:56:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/855c5a441a09493488847372ab1694f8 : SUCCESS in 2h 45m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29f0b1269901a230c9497f3f9b495c7093d59676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 00:10:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1e492f7792a02fdcbb4da272e513367e81eceb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 04:02:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4719db07e3f9485898d1803101fe5a10 : SUCCESS in 6m 04s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b8c63466f6ff470e860e5059df6d8178 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559e3e8e037f4e74ab4e4c1ecede687b : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4500a44c15c1457dbded6b778a05f7ad : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73ed10b9c18b4b9ca23b677c765504c5 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e52fff6017cf42f2aaeb90bf1dfbea18 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af20a2f63e4642479bc5b46d37d63b75 : SUCCESS in 6m 56s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/141ec69a3ddc4f119bf51984bd91f23d : SUCCESS in 1h 16m 56s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/e2526cab3f4b45a18c72373c54cf9e34 : SUCCESS in 35m 32s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/4885d9cdc4484a9b9b4229255ee91001 : SUCCESS in 45m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e2fbad0a856f479dbc18e1dd04ae4c2e : SUCCESS in 4h 55m 08s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/92db88641b1d4efc8824d4f9ff8fdf1d : SUCCESS in 4h 56m 01s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3d308810a8304406966d7efb5ba5a728 : SUCCESS in 54m 09s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/acb1dad0bfc3435f83a69023b35baca4 : SUCCESS in 38m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/41abac9889064fac9361dfb457061aa8 : SUCCESS in 2h 19m 59s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/066de4d26ef54f398c8d185b4eb3a109 : SUCCESS in 2h 06m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/54e402fdcecb471aa3d09654ae9b9204 : SUCCESS in 2h 20m 26s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d6b2c72b95904ec48eb234beee127182 : SUCCESS in 1h 48m 09s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4eadf8df9ee34565841c502e1911d6ae : SUCCESS in 1h 36m 39s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6f6b099028da4c72b486d1f0bcf4a898 : SUCCESS in 2h 08m 05s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/08ed99a33b6a4b9db09eca4e5533d1ec : SUCCESS in 2h 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fccc891406c4a0b1e25433454097ed9ec417f30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 15:50:16.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b9ccfc8c3e827e447f25a615b66f115e7f5a16d2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-25 15:55:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"539721105329b2e2edb83b2453a8c7dfe8fb7e52","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-25 21:15:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/eda37f43e97b4964a907f8daafce0969 : SUCCESS in 2h 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89041da69ae42ed9a19161776b42c9bcca7ae752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 01:45:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06b593a0b93544549173f1e53ae1860d : SUCCESS in 6m 04s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ca1bf393a2d6416f86684b8b5d6a5fce : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a49a08a979f141deaa1860f73fe0f977 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56fd87f096a64554aa2d661a5dfa5e0c : SUCCESS in 9m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5591adda55d428d806a500bd4ed99f6 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/849535e10fb4491ba2a17f93adf206a7 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75d8feb99d249e0bb7e0f0093dbae08 : SUCCESS in 6m 25s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/8055494fd239405298fc5775717b08d5 : SUCCESS in 54m 09s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/b9871cb55dd346cebab3026e615f8fc6 : SUCCESS in 25m 50s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/644d2802b7d74316b8c40c0242fd4abd : SUCCESS in 39m 53s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/73dd71b9faef4f80baaf6050ecb88047 : SUCCESS in 4h 08m 48s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/63d6c753e49b4361813fe7969ff45252 : SUCCESS in 4h 08m 49s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/826949642c974617ae2c27defd01baa5 : SUCCESS in 53m 28s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d16c5a65c90541b0b3d161f3699f2875 : SUCCESS in 47m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e34a23c56b63471e8ea549edab2b21f4 : SUCCESS in 2h 10m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e8863aac35e340d7957dde76a9f7af0f : SUCCESS in 1h 56m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/df1ab24a74094d0186e09f30c9050da0 : SUCCESS in 2h 04m 07s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8caeba141c334fe086caaa5d19dbdca3 : SUCCESS in 1h 39m 40s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/68e93f57e32d4d9db16d91b86408592e : SUCCESS in 1h 21m 00s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/030892a9c4204574a64d6b348607ccdf : SUCCESS in 2h 04m 55s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a4f06981a9004d9b9e24b583e7487864 : SUCCESS in 2h 37m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"935f1b721fb82bbd05fcd5932712857ec1a5c9cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 08:32:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e996f18d51ed6c8c7c70753c59b203f51903705","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 14:43:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad09656d32884600ba7bed9c54eeeb51 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ebe420e50c41d2b83ecf57d95ae49d : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/602d1f8c62e04707be147a788accf25d : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30157012ff9e4c5c8716ac5de3344af7 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae7943088d841aba21fd3fc070df2c6 : SUCCESS in 5m 46s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/3e650cc602a8406681a6ceb7aad3daa3 : SUCCESS in 1h 01m 03s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/57ee8c2651f04319af82779a09e1b80e : SUCCESS in 34m 03s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/46d817149b0d42cbbd9ef82e4ea8e400 : SUCCESS in 38m 43s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/48487508b8144977811a5d128480c0b2 : SUCCESS in 3h 35m 26s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/4f8f3bf1d38a40faacf68b7c97e680e5 : SUCCESS in 1h 03m 28s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b75cee5119ed4fb28b1825ffcf3ee1cc : SUCCESS in 1h 03m 53s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a33fdfdb96ae4a2b9840e4ac8a259bf4 : SUCCESS in 57m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4fafdfae320e4cc388569e15e4464a28 : SUCCESS in 2h 12m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/03af77b461f2492d827c22bfb8646508 : SUCCESS in 1h 37m 58s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/87cc594da331435e8e319eab7141dd87 : SUCCESS in 1h 55m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f3b550fb9c34f61803bf817320b296c : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9a05d2212efe1af3ed3c691fa04a8dd0f99c1706","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 14:43:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"28099d501cdb77e8c59cbef14e45265a41a471dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 14:45:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7aa4fa2ddf4ea2894f88882b9f15f1 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2},{"id":"19d61af871a84ba2ef6484b30fe135d51e32cb9a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-12-16 06:50:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a4fdf46d83a91e8cbe8f40abc79d94a1c31dc592","revisions":{"78ac34aa39d7ea329a967b9a03a3e5dfdcb59272":{"kind":"REWORK","_number":1,"created":"2020-11-06 06:29:04.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/02/761702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/02/761702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/1"}}},"commit":{"parents":[{"commit":"bec4e6fcb12d6724e8d04b47aebc06c4597c4f68","subject":"Merge \"[Validator Show Run] Display simplified results by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bec4e6fcb12d6724e8d04b47aebc06c4597c4f68"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-11-06 06:28:19.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-11-06 06:28:23.000000000","tz":330},"subject":"Map build host repos to distro.repos.d instead of yum.repos.d","message":"Map build host repos to distro.repos.d instead of yum.repos.d\n\nCurrently, when building container images, we are overriding the ubi\nimage repos with the repos configured in the host. Effectively, this is\nremoving ubi repo config from the image build environment which most of\nthe times is not a problem as centos is on sync with ubi, but in some\ncases as after RHEL 8.3 release, ubi is ahead of CentOS and need\npackages from the ubi repo itself.\n\nThis patch is bind mounting the repos config from build host in\ndistro.repos.d instead of yum.repos.d. In that way, both the repos\nconfigured by default in the image (ubi repos) and the ones in the host\nare available.\n\nCloses-Bug: #1902846\nChange-Id: I55f4b8cf2ad971e196fbd06db743d07ce3135ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/78ac34aa39d7ea329a967b9a03a3e5dfdcb59272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/78ac34aa39d7ea329a967b9a03a3e5dfdcb59272"}]},"branch":"refs/heads/master"},"a4fdf46d83a91e8cbe8f40abc79d94a1c31dc592":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-18 15:36:23.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/02/761702/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/02/761702/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/02/761702/2"}}},"commit":{"parents":[{"commit":"bec4e6fcb12d6724e8d04b47aebc06c4597c4f68","subject":"Merge \"[Validator Show Run] Display simplified results by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bec4e6fcb12d6724e8d04b47aebc06c4597c4f68"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-11-06 06:28:19.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-11-18 15:36:18.000000000","tz":0},"subject":"Map build host repos to distro.repos.d instead of yum.repos.d","message":"Map build host repos to distro.repos.d instead of yum.repos.d\n\nCurrently, when building container images, we are overriding the ubi\nimage repos with the repos configured in the host. Effectively, this is\nremoving ubi repo config from the image build environment which most of\nthe times is not a problem as centos is on sync with ubi, but in some\ncases as after RHEL 8.3 release, ubi is ahead of CentOS and need\npackages from the ubi repo itself.\n\nThis patch is bind mounting the repos config from build host in\ndistro.repos.d instead of yum.repos.d. In that way, both the repos\nconfigured by default in the image (ubi repos) and the ones in the host\nare available.\n\nDepends-On: https://review.opendev.org/#/c/763203\nCloses-Bug: #1902846\nChange-Id: I55f4b8cf2ad971e196fbd06db743d07ce3135ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a4fdf46d83a91e8cbe8f40abc79d94a1c31dc592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a4fdf46d83a91e8cbe8f40abc79d94a1c31dc592"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
