)]}'
{"id":"openstack%2Fpython-tripleoclient~772701","triplet_id":"openstack%2Fpython-tripleoclient~master~I3c0466aeed8cd7d2248c0d769b24821110a0e0a5","project":"openstack/python-tripleoclient","branch":"master","hashtags":[],"change_id":"I3c0466aeed8cd7d2248c0d769b24821110a0e0a5","subject":"Add new option for `setup-only`","status":"MERGED","created":"2021-01-27 14:42:00.000000000","updated":"2021-03-04 20:07:53.000000000","submitted":"2021-03-04 20:06:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"772701","meta_rev_id":"e73d34be3b2b771f6e7afd325b2e9bd2cb7c2550","_number":772701,"virtual_id_number":772701,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 20:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-28 21:42:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2021-03-03 22:15:53.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-02-15 22:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2021-01-29 21:38:26.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-03-04 14:34:56.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 16:57:40.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-01-27 16:57:40.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2021-01-27 17:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 19:08:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 21:38:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-03 22:15:53.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"0442f6722bf7dcd7a8ee069d20e358d966da2e60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-01-27 14:42:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a7b936c72503731c3eeb9ec12c8bb8feeefc63b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-01-27 16:57:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"eccc52ba6af8884fdb6e2e258a56eb1347dd1247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 17:48:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/619573c140ba4b3eb2ff8088464a0901 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38b930a25834784af79383dbbd97bad : FAILURE in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e58f20447444763a60bf049eb6a022b : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6da3bb67a99949478dfbc3b9ccbcc8c5 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19e2c2ab34234d4eb3dd6c8e90584045 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b3e409e6834d4784a8a6d8f5a55f10 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bf3a5e2b79a4c88947330dec3ab04f9 : SUCCESS in 3m 54s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/37e59889bfcd40f080e64e972d356d91 : SUCCESS in 55m 10s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/06a780147dfb4f6db799b88fec672790 : SUCCESS in 27m 15s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/bc39f6b97b4d49dc8ecc79d1fd9da5bf : SUCCESS in 37m 21s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d846c99c92024afbad1b192b969366c8 : SUCCESS in 1h 14m 48s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7e50da215592b20718f52470985d9e365181985","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-27 19:08:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/99098c0d9bb148cbafb7946499960fa7 : SUCCESS in 3h 22m 46s","accounts_in_message":[],"_revision_number":1},{"id":"99155fa46ff28a6d8f229135937a1cf4adb67e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-01-27 19:10:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da5c2d79528f2e76168253c6ca67b8a740fdb04f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-27 21:54:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d7bc384c819140678dd2f0c157c32a50 : SUCCESS in 2h 42m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b744125d19b9b724b1a802d24166b57c10a3abf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 00:48:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/7a825771e7d642c194892d073e0c9b45 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b19e1b705edd4dab91ddf107cc0b136a : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c447aee2aaa4cceb3a39cb1b06af4a7 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b934e630e40427cb7eaf6a7994ec0c4 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8a6c5ffbd644ae989d0c5d618663cc8 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/975b7c2dd6274fe18261dd034c5bdc92 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4933d24e424b450091198864225f7ba6 : SUCCESS in 3m 42s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/a3c5968cb7184084a19c3e7c3918fce3 : SUCCESS in 56m 01s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/60d5ee9d8c704bc989076a7aa3773d4d : SUCCESS in 29m 00s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/9c69000bf6694606b4c48cdcc4450b5f : SUCCESS in 39m 40s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/416db84b541744d7b549320087046705 : SUCCESS in 3h 50m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/11153d0970b349bda7899a288f5c5261 : SUCCESS in 1h 59m 07s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0a4c5816eacd401eac7dce4b72774d46 : SUCCESS in 1h 30m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/23c5279118b746d6b7cb4e017fc9f4d9 : SUCCESS in 1h 11m 40s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/0616ca60ac3648408fe00204e5ead987 : SUCCESS in 4h 18m 59s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/5fcc1aa4204d490986ac3e54e5712654 : SUCCESS in 2h 26m 24s (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/8f13b09237e243cfae9eb43a58ef694d : SUCCESS in 1h 39m 23s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/491603367da54da3a440a59f0cbee4f6 : SUCCESS in 1h 57m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/677536936cfc42ff979de8cd15c3e9bc : SUCCESS in 2h 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf5673b40aabfa09a7393b857eb52537517bf65b","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-01-28 18:31:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\ni don\u0027t completely understand the intent with this one. can we instead breakout the deployment more cleanly into the needed phases?:\n\nstack\nsetup (config-download generate, ssh, etc)\nconfig-download execute\n\ninstead, the code reads a little backwards with this change in that we start to go into the config-download workflow, but then cut it short if --setup-only.\n\nThere also seems to be a fair amount of overlap between \"overcloud config download\" and \"overcloud admin authorize\".\n\nInstead I\u0027d recommend going ahead and reordering the code and workflows to match what we want going forward, and then guarding the execution of those pieces with the cli args (and to match the legacy behavior of --stack-only/--config-download-only) as needed.","accounts_in_message":[],"_revision_number":2},{"id":"be32a39a0575d9fe3bd5cd4c6cbe2c3db370356d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-01-28 18:40:07.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aec728b58248edd31bda16cfd0e7a18e1fd1146f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-01-28 18:48:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb30e81d936a8790d3b9344fefe138c96134a7db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-01-28 18:49:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9738c1962639f8512b4b16aa31042fc032f6fe49","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-28 21:42:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5048cf80808144f087d18b526fbc338d : SUCCESS in 2h 51m 14s","accounts_in_message":[],"_revision_number":3},{"id":"42c4dea5281a0d21517970edcffefeff1c1d8a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 03:57:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3be2e19d9a914ed38974aae80adb510e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2505f50642b543408b847b5a93b5d906 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae774c64d7c540db91ca1165b75a1636 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc5f57676b8d4d7598c9875935897630 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7cb7c6f72804edbbd28fc24dd5e9e5c : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef3f412cf3d45eda686008fcc59dcf5 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e4f8d9875b4aad82efe6d9f0f65791 : SUCCESS in 4m 46s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/8caea0b3e8e84f0cadd2650b55d0f261 : SUCCESS in 1h 06m 58s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/643f3040be164d768562ef8fccca76a5 : SUCCESS in 28m 30s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/5f3908adff464f948f6018fea8b1bbbb : SUCCESS in 34m 03s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9d5a534a7a0b42e096f50dec733f2537 : SUCCESS in 3h 40m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3dd38050a7384b498de64e94ed9971ef : SUCCESS in 2h 08m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2eb63b1a4f834ea09744c4843ba3d2d8 : SUCCESS in 1h 39m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/51f94787ba094643bfb2fc5cc28310d6 : SUCCESS in 1h 31m 41s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/65f05ff85111441581d858994a9500bf : SUCCESS in 3h 52m 31s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/605442e054754b7e93d6ec274f7ed92a : SUCCESS in 2h 38m 20s (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/42053424d27a4264bfa0c1d424ae3167 : SUCCESS in 1h 53m 24s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/28c20262e58944a3b6dfae29d3e9c806 : SUCCESS in 1h 51m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/69a4474c9c6c4542b297772c0a0b6d52 : SUCCESS in 2h 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8c4520ec94895794746ac7967750cb235513e29f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-29 21:38:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"93301b49d4b619a89f6aa3746098415ca785241d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-15 22:19:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9af42de7a89d973d10ec1fe248fa77dd3ed13c2a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-03-03 22:15:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)\n\nA couple of code-style nits in case you do another iteration.","accounts_in_message":[],"_revision_number":3},{"id":"9ad852f07b1ca5cafd10e396517905b8ad7f39db","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-03-04 14:34:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a145c9a6e17a5f4ed1b51313eddaa109741f7f0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:35:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ad189c27ec70f0c81c9eda79e738bfd0e0c6d14e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:06:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31abb8a6ec164880b16895c46e487623 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc36795d2c81464b9910119023ad8a5d : SUCCESS in 8m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af538558794c4d85aa520ae4fb237431 : SUCCESS in 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aead71e3b4884e41a46b6058eed186ff : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1340d8337e2413abd704faf211eac09 : SUCCESS in 5m 53s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7919bd3556db444e8cd15e48f8b93c39 : SUCCESS in 3h 17m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4127ffa25f77428c98bf7b076eb53628 : SUCCESS in 1h 36m 29s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a45110c3a13046bca637f75cacf1e7fb : SUCCESS in 1h 04m 54s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2d3fa86191064550b682a09054c41ddf : SUCCESS in 1h 39m 32s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7f4d4bc93d0d4674803105b6622216ff : SUCCESS in 2h 12m 15s","accounts_in_message":[],"_revision_number":3},{"id":"375731bd0a69c62d6dd884b7d8230cfb3f153329","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:06:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e73d34be3b2b771f6e7afd325b2e9bd2cb7c2550","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:07:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f11e9e24e6b4ee59e2f0073aaef60c2 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd9e925c958f48a7984f34276524d909 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"be580c669065c59789601f28bfbaee8318e67c0d","revisions":{"15975e0479db678048bc78a678f9fb65d1d6d0ce":{"kind":"REWORK","_number":1,"created":"2021-01-27 14:42:00.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/01/772701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/01/772701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/1"}}},"commit":{"parents":[{"commit":"39cfed499a0aa0aa90d1e9430b785c3f8fa1bec8","subject":"Merge \"Drop --update-plan-only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/39cfed499a0aa0aa90d1e9430b785c3f8fa1bec8"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-01-27 14:12:18.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-01-27 14:41:37.000000000","tz":-360},"subject":"Add new option for `setup-only`","message":"Add new option for `setup-only`\n\nThis change implements the ability to run all of the required setup\ntasks to enable manual playbook execution. When the option is invoked,\nthe following workflows will be run:\n\n* Generate the new Ansible content\n* Downlad the Ansible content\n* SSH key validation and application to overcloud nodes\n\nThis new option allows us to retain the current logic of stack-only\nwhich will ensure legacy workflows are not interupted while also\nenabling a more ansible centric user experience.\n\nrhbz: 1873246\nChange-Id: I3c0466aeed8cd7d2248c0d769b24821110a0e0a5\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/15975e0479db678048bc78a678f9fb65d1d6d0ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/15975e0479db678048bc78a678f9fb65d1d6d0ce"}]},"branch":"refs/heads/master"},"03b08a173b51b92616d89b633e0409519cf838e3":{"kind":"REWORK","_number":2,"created":"2021-01-27 19:10:29.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/01/772701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/01/772701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/2"}}},"commit":{"parents":[{"commit":"39cfed499a0aa0aa90d1e9430b785c3f8fa1bec8","subject":"Merge \"Drop --update-plan-only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/39cfed499a0aa0aa90d1e9430b785c3f8fa1bec8"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-01-27 14:12:18.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-01-27 19:10:18.000000000","tz":-360},"subject":"Add new option for `setup-only`","message":"Add new option for `setup-only`\n\nThis change implements the ability to run all of the required setup\ntasks to enable manual playbook execution. When the option is invoked,\nthe following workflows will be run:\n\n* Generate the new Ansible content\n* Downlad the Ansible content\n* SSH key validation and application to overcloud nodes\n\nThis new option allows us to retain the current logic of stack-only\nwhich will ensure legacy workflows are not interupted while also\nenabling a more ansible centric user experience.\n\nrhbz: 1873246\nChange-Id: I3c0466aeed8cd7d2248c0d769b24821110a0e0a5\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/03b08a173b51b92616d89b633e0409519cf838e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/03b08a173b51b92616d89b633e0409519cf838e3"}]},"branch":"refs/heads/master"},"be580c669065c59789601f28bfbaee8318e67c0d":{"kind":"REWORK","_number":3,"created":"2021-01-28 18:49:59.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/01/772701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/01/772701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/01/772701/3"}}},"commit":{"parents":[{"commit":"39cfed499a0aa0aa90d1e9430b785c3f8fa1bec8","subject":"Merge \"Drop --update-plan-only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/39cfed499a0aa0aa90d1e9430b785c3f8fa1bec8"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-01-27 14:12:18.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-01-28 18:49:48.000000000","tz":-360},"subject":"Add new option for `setup-only`","message":"Add new option for `setup-only`\n\nThis change implements the ability to run all of the required setup\ntasks to enable manual playbook execution. When the option is invoked,\nthe following workflows will be run:\n\n* Generate the new Ansible content\n* Downlad the Ansible content\n* SSH key validation and application to overcloud nodes\n\nThis new option allows us to retain the current logic of stack-only\nwhich will ensure legacy workflows are not interupted while also\nenabling a more ansible centric user experience.\n\nrhbz: 1873246\nChange-Id: I3c0466aeed8cd7d2248c0d769b24821110a0e0a5\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/be580c669065c59789601f28bfbaee8318e67c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/be580c669065c59789601f28bfbaee8318e67c0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
