)]}'
{"id":"openstack%2Fpython-tripleoclient~785604","triplet_id":"openstack%2Fpython-tripleoclient~stable%2Ftrain~Id853b1e80d690734ac9c32a1b914e49608f6cc43","project":"openstack/python-tripleoclient","branch":"stable/train","topic":"undercloud/preflight-stable/ussuri-stable/train","hashtags":[],"change_id":"Id853b1e80d690734ac9c32a1b914e49608f6cc43","subject":"Use validation framework for running undercloud checkdisk","status":"MERGED","created":"2021-04-12 05:54:27.000000000","updated":"2021-04-27 16:56:47.000000000","submitted":"2021-04-27 16:50:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785604-undercloud/preflight-stable/ussuri-stable/train","cherry_pick_of_change":762132,"cherry_pick_of_patch_set":1,"meta_rev_id":"83410e9787aee513a87932fe273e092d116f9088","_number":785604,"virtual_id_number":785604,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 16:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-04-20 07:48:24.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-04-15 10:34:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":2,"date":"2021-04-16 05:42:26.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2021-04-21 07:38:04.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-12 06:36:35.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-04-12 09:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 10:34:11.000000000","updated_by":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-04-16 05:42:17.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"CC"},{"updated":"2021-04-16 05:42:26.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-04-19 06:48:42.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-04-19 06:48:42.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"}],"messages":[{"id":"5280ef1127c9b60f4df8989b2e189bff6fcec2c5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-12 05:54:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* tripleoclient/v1/undercloud_preflight.py","accounts_in_message":[],"_revision_number":1},{"id":"dd18d1e14fffd255e8e5b2e82f70e92bb8cb4144","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-12 05:56:18.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"d0b27787f07a32fd1f76604be43265a751e7e642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-12 05:56:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"46e94035b044806570a6bb4db86610d7a0162d56","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-04-12 06:36:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c12b2c3d8a3cfed9730fd8f2928e4c582420256e","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-12 06:46:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a56d272801b585f482e957dbd31503b97adbb1d6","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-04-12 07:00:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03e4b1ce17a47fd0c4c6fe2b611364d331903e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 09:30:02.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b9277548abb4b89bd0258845fd30900 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ac443c0b0054246bf72217df668ca46 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bcd21ebb8b847308a5cb32b8eefd642 : SUCCESS in 8m 50s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/1f5d9888e671461fa970cb74931443c3 : SUCCESS in 3h 31m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/eddade0dfbf6451f951b993da65bd58c : SUCCESS in 2h 11m 43s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9df9e2ff0044d12b4326d9d8d5a5272 : FAILURE in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/078d0276eef64a53b47bbf4249d488a7 : FAILURE in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"167a3213e7173e04f4b3e7d6f7a73373112cf21e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-12 10:28:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a1b4a357e6ddf1156e205e1776eb6b7dfa5c30d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 14:19:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6aecf63aeecb4a29909e1310ee7a2ccb : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffb893b154a04dfd8a57d044e41c008e : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a4383a80704293b8c6900cdaeb7174 : SUCCESS in 5m 47s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/21ea80fbbea343f9816f255ce1820e02 : SUCCESS in 2h 51m 25s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/d118a7a535c846ea9399f3aff8984b3f : SUCCESS in 3h 50m 19s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/005912a865a64820b70544402e61920c : SUCCESS in 1h 42m 16s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d676ef63686c46a59cc9889c5e02c09c : SUCCESS in 2h 23m 01s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ad7aaf0abd0242ed896de2791ddc4588 : FAILURE in 29m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/415a557cbea244d59d00a067740bfa32 : SUCCESS in 1h 04m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/677b187b2b40490da72c259f8cc90868 : FAILURE in 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b0422ff18594c2b9ebb9c5e0506af97 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8bad069a19148369ddd40cd4e2675b7 : FAILURE in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92001d80a7b719c32607d7093e346c7246d25216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-13 07:19:12.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2419f5dac2c744a35631d954ea0783f61f361032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 10:48:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aaa5e66704542b1b30984d41acfc192 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ca3439934b140028dc594d78c34a32e : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580cbaa0897f49c2b0fe8002f86891d3 : SUCCESS in 7m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f232d93133c94b909758247beb6b8d27 : SUCCESS in 3h 11m 02s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/30d6225fcbf9400580c47ade123dc36a : SUCCESS in 3h 26m 39s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d06c4279d16341139c925b5897eaaa4c : SUCCESS in 2h 03m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ed968a8244884f0d8cc872042a89ca09 : SUCCESS in 2h 08m 33s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4dadc5fc1f014f1996fba1b68f40d0b9 : SUCCESS in 1h 36m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/84ab271cf1094f9a9753d26562989ec1 : SUCCESS in 1h 06m 30s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/de5cc20cbbba447a9b8925059026b7a9 : SUCCESS in 1h 57m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b43a50d2a14421899499a297326b3ec : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dff8a30b85ef41d5862f0e4d84f1f2c7 : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a6d60d399d36637d2dcfc69a03054fecb2b2269b","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2021-04-15 10:34:11.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b315ebf6811fdcb21f1b6582995402875860ab2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-04-16 05:42:17.000000000","message":"Patch Set 4:\n\nTODO comment removal patch on master [1] is on his way to be merged.\n\nYes better having neat backport!\n\n[1] https://review.opendev.org/c/openstack/python-tripleoclient/+/785819","accounts_in_message":[],"_revision_number":4},{"id":"84ddd91a59215b05172a60f4aa2a158425c1c789","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-04-16 05:42:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"71337821b2f8a8a3efce21d569b7f7e692ea3d26","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-04-20 07:48:24.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nclean cherrypick but holding on the inline discussion","accounts_in_message":[],"_revision_number":4},{"id":"474678b3e8e109313de0760020f4d49f2221b55f","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-20 12:38:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d1bc9e214c639a3ab3c2e6e587b159f26f27f937","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-04-21 07:38:04.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)\n\nthanks lgtm then let\u0027s go :)","accounts_in_message":[],"_revision_number":4},{"id":"3e8e3d4c05e598796826dc2aef2af8541c20812d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 07:38:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9e1193612410fc9d817c6ce0cd2ea13fa3250ceb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:39:23.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00f45ea31a7d4536b8e791251a088df4 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83cab2630c364b429c0d7730deb8fb7a : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b8c6fa23c34f9396d401f0875a39f7 : SUCCESS in 6m 24s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/825e8bad495d46c0be24609a1b034ea4 : SUCCESS in 2h 59m 13s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/9c053bdb99554d84b87aec4fd823dd24 : FAILURE in 13m 37s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/60ec25a4bce941c9b9901136ccd13bd5 : SUCCESS in 1h 56m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6e02c032d0b84b199a4f7ec8eb71f1b8 : SUCCESS in 1h 39m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d20bb1cb57524a449702828242550f2f : SUCCESS in 1h 10m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/ac0c0872440d4abe972e4c4ddd6528c8 : SUCCESS in 1h 47m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e1c2e2a957c486a9dd036ab36b55ae4 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":4},{"id":"8a42a759e7acefa34c17621874e104c4c290b84e","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-04-27 07:52:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"80126569d03f7bf7c8a8b85b955abbaf652a74bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:55:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/697e27bf4bd042a6bc8518f19fdaf2e6 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a75a3b9ae4c24bb9a7fa11c784c77bcd : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef247e55db44ccca0827e4678de241f : SUCCESS in 6m 16s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/946d179246344694baa09d1edb58800a : SUCCESS in 3h 27m 58s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/d80f495c67404e3694fdc49b63c18d4e : SUCCESS in 4h 00m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/51275693ecc3478a8127d20bcf971154 : SUCCESS in 2h 06m 16s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b1de3687b1aa4fe1b8c31dbc9711980c : SUCCESS in 2h 15m 23s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/09562e7548f14c1baceb7f6701f1f472 : SUCCESS in 1h 19m 58s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8cdd64dfffe044128aca5116826d8133 : SUCCESS in 1h 04m 18s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/be8d2e65cd0f437aa17ee5637d255cd6 : SUCCESS in 1h 44m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c3a47d8bee4176a9ff02f6e36756f7 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e111f3f40c34e57aac46ddc96f927fe : FAILURE in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"001d5e8e7eac26d2a98b02ba99d4d1b9a3df4867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:55:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"42ac2f43f9bc4051ec5dd692fb7a6e1fd9611fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:50:19.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f83aff7d0cc74e909a3344eb140e4e2f : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73ca204ba9104bd38e089e40d7a91758 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/094b562609034fe3ab86a59e2a462a0b : SUCCESS in 7m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/114983a3c8274fa3a543cc086b217cdf : SUCCESS in 3h 06m 21s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/cfbb4150b4b2402e902aec093595c359 : SUCCESS in 4h 06m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8b6c7279016540df82ae7988519d322c : SUCCESS in 1h 55m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4216d492627a4aff87fc797d9ff2e21e : SUCCESS in 2h 14m 55s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b960be6115074e49855d266805912c5a : SUCCESS in 1h 29m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d8536658d60c4222b6c40a1139d85f86 : SUCCESS in 1h 08m 35s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/543e146e2a2d4fdebe7cd0ab530766dd : SUCCESS in 1h 51m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614c4b23cb384873b4eb4dc62723aa01 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"b4ac8044c09c8352114a0ef9c3b631041d9165ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:50:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"83410e9787aee513a87932fe273e092d116f9088","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:56:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a17ad958d1946778d764096a5f51078 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"429169bde6975afd9606b2d410d81e6f32ba2a31","revisions":{"b152405517ba5aa028f54539992ddd51e660618d":{"kind":"REWORK","_number":1,"created":"2021-04-12 05:54:27.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/04/785604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/04/785604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/1"}}},"commit":{"parents":[{"commit":"1a64962d0b821a3edc163fb6d2442b33fb4ee342","subject":"Move static-inventory option into parser group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a64962d0b821a3edc163fb6d2442b33fb4ee342"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-09-23 14:33:06.000000000","tz":120},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2021-04-12 05:54:27.000000000","tz":0},"subject":"Use validation framework for running undercloud checkdisk","message":"Use validation framework for running undercloud checkdisk\n\nThe validations playbook is own and moved under the VF\nansible dir.\nThe undercloud preflight check should use VF for running it.\n\nChange-Id: Id853b1e80d690734ac9c32a1b914e49608f6cc43\n(cherry picked from commit 3dd2c83883960bd279e82b9fb0d4a3cb16c5e2cb)\n(cherry picked from commit 8c7942ed2555dad8c34a7dada88fff5854db1151)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b152405517ba5aa028f54539992ddd51e660618d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b152405517ba5aa028f54539992ddd51e660618d"}]},"branch":"refs/heads/stable/train"},"33f3de074d7a6d3c6004e7b8b96b91d441c6047d":{"kind":"REWORK","_number":2,"created":"2021-04-12 05:56:23.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/04/785604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/04/785604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/2"}}},"commit":{"parents":[{"commit":"1a64962d0b821a3edc163fb6d2442b33fb4ee342","subject":"Move static-inventory option into parser group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a64962d0b821a3edc163fb6d2442b33fb4ee342"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-09-23 14:33:06.000000000","tz":120},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2021-04-12 05:56:07.000000000","tz":0},"subject":"Use validation framework for running undercloud checkdisk","message":"Use validation framework for running undercloud checkdisk\n\nThe validations playbook is own and moved under the VF\nansible dir.\nThe undercloud preflight check should use VF for running it.\n\nChange-Id: Id853b1e80d690734ac9c32a1b914e49608f6cc43\n(cherry picked from commit 3dd2c83883960bd279e82b9fb0d4a3cb16c5e2cb)\n(cherry picked from commit 8c7942ed2555dad8c34a7dada88fff5854db1151)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/33f3de074d7a6d3c6004e7b8b96b91d441c6047d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/33f3de074d7a6d3c6004e7b8b96b91d441c6047d"}]},"branch":"refs/heads/stable/train"},"20ee93fbf2ed5cb9c9a63b2a635fbb4d912d7da1":{"kind":"REWORK","_number":3,"created":"2021-04-12 10:28:17.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/04/785604/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/04/785604/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/3"}}},"commit":{"parents":[{"commit":"1a64962d0b821a3edc163fb6d2442b33fb4ee342","subject":"Move static-inventory option into parser group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a64962d0b821a3edc163fb6d2442b33fb4ee342"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-09-23 14:33:06.000000000","tz":120},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2021-04-12 10:28:07.000000000","tz":0},"subject":"Use validation framework for running undercloud checkdisk","message":"Use validation framework for running undercloud checkdisk\n\nThe validations playbook is own and moved under the VF\nansible dir.\nThe undercloud preflight check should use VF for running it.\n\nChange-Id: Id853b1e80d690734ac9c32a1b914e49608f6cc43\n(cherry picked from commit 3dd2c83883960bd279e82b9fb0d4a3cb16c5e2cb)\n(cherry picked from commit 8c7942ed2555dad8c34a7dada88fff5854db1151)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/20ee93fbf2ed5cb9c9a63b2a635fbb4d912d7da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/20ee93fbf2ed5cb9c9a63b2a635fbb4d912d7da1"}]},"branch":"refs/heads/stable/train"},"429169bde6975afd9606b2d410d81e6f32ba2a31":{"kind":"REWORK","_number":4,"created":"2021-04-13 07:19:12.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/04/785604/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/04/785604/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/04/785604/4"}}},"commit":{"parents":[{"commit":"1a64962d0b821a3edc163fb6d2442b33fb4ee342","subject":"Move static-inventory option into parser group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/1a64962d0b821a3edc163fb6d2442b33fb4ee342"}]}],"author":{"name":"Mathieu Bultel","email":"mat.bultel@gmail.com","date":"2020-09-23 14:33:06.000000000","tz":120},"committer":{"name":"mbu","email":"mat.bultel@gmail.com","date":"2021-04-13 07:19:05.000000000","tz":0},"subject":"Use validation framework for running undercloud checkdisk","message":"Use validation framework for running undercloud checkdisk\n\nThe validations playbook is own and moved under the VF\nansible dir.\nThe undercloud preflight check should use VF for running it.\n\nChange-Id: Id853b1e80d690734ac9c32a1b914e49608f6cc43\n(cherry picked from commit 3dd2c83883960bd279e82b9fb0d4a3cb16c5e2cb)\n(cherry picked from commit 8c7942ed2555dad8c34a7dada88fff5854db1151)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/429169bde6975afd9606b2d410d81e6f32ba2a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/429169bde6975afd9606b2d410d81e6f32ba2a31"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
