)]}'
{"id":"openstack%2Fpython-tripleoclient~788200","triplet_id":"openstack%2Fpython-tripleoclient~master~I9d9dea09a36914cffaef1d251c49f3c4fea4fa68","project":"openstack/python-tripleoclient","branch":"master","topic":"bug/1926314","hashtags":[],"change_id":"I9d9dea09a36914cffaef1d251c49f3c4fea4fa68","subject":"Remove leftover for undercloud keepalived service","status":"MERGED","created":"2021-04-27 08:10:41.000000000","updated":"2021-04-28 18:38:41.000000000","submitted":"2021-04-28 18:36:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788200-bug/1926314","meta_rev_id":"f3c819a2f5019b02a48a3ea886a2e4a7c398561d","_number":788200,"virtual_id_number":788200,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 18:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-28 13:35:08.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-28 13:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":2,"date":"2021-04-28 09:22:21.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-04-28 13:35:08.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 12:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 09:22:21.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-04-28 11:48:37.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"state":"CC"},{"updated":"2021-04-28 11:59:31.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-28 13:10:06.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"}],"messages":[{"id":"0571d116017d6b4f77347a991f6d7df4541f6c7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 08:10:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b71f526525f3e6b08c9141a00be3932b17a45eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 08:11:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1009414d3ebf34dc0809dfac909f5dd230bdada0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:31:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c0faa66cb7b2480286ac5c448eb8d008 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27ec5720d0a4bc69c1cb3db15b4b94c : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ece88251f9146d499b6fbdbc6959927 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89698aa581684232b01e5b27ac9e8f45 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/374c77d45af6460aaa7a165432a8b1d9 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32855fec11df4effa73dd07f2e114d12 : SUCCESS in 6m 00s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e84e057e86994a6d92ca7d9c0ddcfb12 : SUCCESS in 3h 45m 06s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/952f8a90199643c9ab22f226de345671 : SUCCESS in 1h 49m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cc551feb4acf442b86cea87571f37d02 : SUCCESS in 1h 15m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/da25db7f50dc46b7a2e51191205b8063 : SUCCESS in 1h 29m 51s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cf08f4ff4b684696b216bf7a78fe20cf : FAILURE in 38m 41s","accounts_in_message":[],"_revision_number":2},{"id":"43ace005eff41e19324437962172c4f887e11927","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 15:35:23.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f5b0183211b18d1a1b34ef7464135b983693ab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 16:05:46.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"68ab3cd926bf203b937d84aa7a1ea9dd4d083b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 19:13:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e5edb1d51686494393109726c113c1d2 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40c40135fcc74d709977a62c6c49b071 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/092c4e156bb34f6b8692afdc646197da : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25ae1d5a12d84ce2bab64554be251656 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7baad6d8b091425f836687d40ccbd7ec : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c21f0f573b47b5b02775c69f4ae5e1 : SUCCESS in 14m 28s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/85dca0a8b03041d6b6ac7d43a603d1f6 : SUCCESS in 2h 58m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f87445fa04774a3b91252e19fd667d9c : SUCCESS in 1h 45m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9586cd8b01d54141824ec1caf90f7844 : SUCCESS in 1h 20m 58s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/685dddd3b21940d2bacb309ea4a6f0d5 : SUCCESS in 1h 19m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/192488213cea4e20ad1e1cc27edc0951 : SUCCESS in 1h 33m 24s","accounts_in_message":[],"_revision_number":3},{"id":"9a3e3ee93a04c90841aba197e88ea51826e5fcbc","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-28 09:22:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"90d1fe0b592b517d63f4e07ecc4ae8a782466a74","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2021-04-28 11:48:37.000000000","message":"Patch Set 3:\n\nDo we delete keepalived when we do upgrade?","accounts_in_message":[],"_revision_number":3},{"id":"fdc912cae9ed0fe676bf31bdeea78d751ec7b54c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-28 11:59:31.000000000","message":"Patch Set 3:\n\nthat\u0027s a great question.\nAFAIK, neither of the services removed after deprecation have such cleanup tasks","accounts_in_message":[],"_revision_number":3},{"id":"d1b629d88b89239cd93678fb928957d6b27cbfc4","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-28 12:00:31.000000000","message":"Patch Set 3:\n\n...aside of basic systemd units management I think, which had been implemented for upgrades from non-containerized services?","accounts_in_message":[],"_revision_number":3},{"id":"b835ce938c53443427abdf991968dd6dc5679093","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2021-04-28 13:10:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3c5cc479a734e06b1bc6de95f7940389004d9aed","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-28 13:17:28.000000000","message":"Patch Set 3:\n\n@Sergii, the upgrade path concerns aswered here https://review.opendev.org/c/openstack/tripleo-heat-templates/+/788095","accounts_in_message":[],"_revision_number":3},{"id":"0c59cc4f54b0935908c7b0706478cbd4872b85b3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-28 13:35:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2227f2d0e0acce6fbb807ac7916f560c0c3a7e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 13:35:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e6980bcc40cbcebeee8c0a922c1efab0737ad1f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:36:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdb6fcd3cb77454aa60a1ceae840bbd5 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b3b8aef21345acb12d47a0d3b7cb11 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/116672702e984435b3149659c73514a6 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f13a2c4d38c48089674abf286fe27fb : SUCCESS in 7m 01s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/814aea4073b346a684fdeb28c55e07eb : SUCCESS in 3h 15m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d337d3e0ad564877ae5fe6cc7f1d4189 : SUCCESS in 1h 58m 03s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b57430f416644657a72731fed22459ab : SUCCESS in 1h 05m 22s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c342454f1b8f4fd7ba0772bebb8d30cf : SUCCESS in 1h 25m 01s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9d5c318327ea4ddaae3ddbd00792bdc0 : SUCCESS in 1h 52m 19s","accounts_in_message":[],"_revision_number":3},{"id":"5407a6ba7ff24010c9fb1499e62a88818747cc99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:36:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f3c819a2f5019b02a48a3ea886a2e4a7c398561d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:38:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c17e5e38856044bba65749ac7e603d30 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b02acaa6e17e8c3d1ab29ab30fa21e8b65ba8b43","revisions":{"2f94b3371da5a8134950594af524626fddc5b8e0":{"kind":"REWORK","_number":1,"created":"2021-04-27 08:10:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/788200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/00/788200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/1"}}},"commit":{"parents":[{"commit":"6fad903be9608e5354dc8dbfd5fcea1c23ceb44a","subject":"Merge \"Base for improvement of validator test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6fad903be9608e5354dc8dbfd5fcea1c23ceb44a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 08:07:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 08:07:52.000000000","tz":540},"subject":"Remove Keepalived service in undercloud","message":"Remove Keepalived service in undercloud\n\nKeepalived service was deprecated in Train, and is no longer required\nsince VIPs are now deployed as a part of network deployment.\n\nChange-Id: I9d9dea09a36914cffaef1d251c49f3c4fea4fa68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2f94b3371da5a8134950594af524626fddc5b8e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/2f94b3371da5a8134950594af524626fddc5b8e0"}]},"branch":"refs/heads/master"},"8f3e1a5eef94711da0849820f15a78784e74a8e7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-27 08:11:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/788200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/00/788200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/2"}}},"commit":{"parents":[{"commit":"6fad903be9608e5354dc8dbfd5fcea1c23ceb44a","subject":"Merge \"Base for improvement of validator test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6fad903be9608e5354dc8dbfd5fcea1c23ceb44a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 08:07:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 08:11:03.000000000","tz":540},"subject":"Remove Keepalived service in undercloud","message":"Remove Keepalived service in undercloud\n\nKeepalived service was deprecated in Train, and is no longer required\nsince VIPs are now deployed by network deployment process.\n\nChange-Id: I9d9dea09a36914cffaef1d251c49f3c4fea4fa68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8f3e1a5eef94711da0849820f15a78784e74a8e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8f3e1a5eef94711da0849820f15a78784e74a8e7"}]},"branch":"refs/heads/master"},"b02acaa6e17e8c3d1ab29ab30fa21e8b65ba8b43":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-04-27 16:05:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/788200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/00/788200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/00/788200/3"}}},"commit":{"parents":[{"commit":"6fad903be9608e5354dc8dbfd5fcea1c23ceb44a","subject":"Merge \"Base for improvement of validator test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6fad903be9608e5354dc8dbfd5fcea1c23ceb44a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 08:07:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-27 16:02:04.000000000","tz":540},"subject":"Remove leftover for undercloud keepalived service","message":"Remove leftover for undercloud keepalived service\n\nCurrently environments/services/undercloud-keepalived.yaml is included\nwhen undercloud SSL is enabled but this file is no longer effective\nsince the Keepalived service was removed from undercloud role data[1].\n\n[1] c712355e4bae4ef2fc1b83e5603c0364dbd50a78\n\nBecause VIPs are managed by os-net-config now and we don\u0027t expect any\nsituation where users requires keepalived running in undercloud,\nthis change simply removes all remaining implementation for undercloud\nkeepalived service.\n\nCloses-Bug: #1926314\nChange-Id: I9d9dea09a36914cffaef1d251c49f3c4fea4fa68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b02acaa6e17e8c3d1ab29ab30fa21e8b65ba8b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b02acaa6e17e8c3d1ab29ab30fa21e8b65ba8b43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
