)]}'
{"id":"openstack%2Fpython-tripleoclient~794577","triplet_id":"openstack%2Fpython-tripleoclient~master~I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295","project":"openstack/python-tripleoclient","branch":"master","topic":"validationlogpath/override","hashtags":[],"change_id":"I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295","subject":"Overriding the log path argument for tripleoclient cli","status":"MERGED","created":"2021-06-03 12:33:33.000000000","updated":"2021-08-23 19:55:32.000000000","submitted":"2021-08-23 19:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"794577-validationlogpath/override","meta_rev_id":"c760f7bf351b3d9b81f588062db980e9c538b2d3","_number":794577,"virtual_id_number":794577,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 19:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":2,"date":"2021-08-23 14:25:58.000000000","_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":2,"date":"2021-08-23 19:55:32.000000000","post_submit":true,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":2,"date":"2021-08-23 14:25:56.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":1,"date":"2021-08-23 14:25:58.000000000","_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 12:36:06.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-06-03 12:36:06.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-06-03 12:36:06.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-06-03 12:36:06.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-06-03 12:36:06.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-06-03 12:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-09 07:05:23.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-08-09 07:05:23.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"b67d2ea11e8930e9b2b6e5892a4c9f527a1791f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-03 12:33:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9274682e935fc1ad9539c65eb582028408ed773c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 12:47:21.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"8aebccbce29d12e7baf50e72b145ec3527bf5dc6","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-03 13:08:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3af894bf1eb8cb893a2867274cf424fb4765bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 16:05:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/6ca8ef04fe624b879caf0a40f867bbdf : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b58693a777346ef8e19621db060ec1f : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c36ea2e096544d59675b7d5b7475eb4 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06a66dac79ec41f9a608f10b87b9d714 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae4b050299244fb7baeac8483b60b638 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c333821504f4b70a1e3b2aabc6407c8 : SUCCESS in 6m 32s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3a1459479c1d47fba06743af9440cd2f : SUCCESS in 2h 44m 42s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4596d47e9e3a4f7b8a25e6e362df9eea : FAILURE in 1h 33m 42s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fa6380b462c64c18a6d095fa4fa3aacb : SUCCESS in 56m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a745c04c663f4fddb376d3c4d95f1a67 : SUCCESS in 1h 15m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/72f3061775504f0fbd7288c14ee8f4d9 : SUCCESS in 1h 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"aa330d97d89b25b25320c9fb0c2bac33188005af","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-06-04 10:04:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"aaf07f533a4271ac80c355f5b0b95bcc6c910703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-04 10:24:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"34028c6c8af1a6c0045d5e864f4e1d949f83f1cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 10:44:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/6359cd0ea68e4013b16fcc6b03933909 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9eb5ef39b84040b762a271f5f6e5ed : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/363f3afb490f4d9d8fcdc9fd7f697072 : FAILURE in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b0db02b387145c8b34197ad17d85842 : FAILURE in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de408d34ab5d4bad9b048d3b0155954e : FAILURE in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2209fb5e068c4583919c12d822d89598 : SUCCESS in 5m 56s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"b40f4abd2c005eeb1e3272ec57dd40e8ecd9503f","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-04 12:14:13.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6db8f2ad877d87a93ef922901daede10e459e7dc","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-07 07:36:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fc69a7bb0c6b90c238f6077bad4cc091e2c45e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 07:56:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e4412023283d49529f8129bf3107c5d3 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e423004429472a8fc751fec15eea9e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12feeaa309fb460ab139c21d3a16ab63 : FAILURE in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9323291f9ad4859a67797426d8097e2 : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c272044136784450a84210f72c987972 : FAILURE in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0b34fd5f6140559e4d0a3c549a412e : SUCCESS in 6m 53s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"fdd3c78a2ea7dd133e3010b1c5c4566f55f22cdc","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-06-09 15:21:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"778113760f5b9a63120c7f78b35debac91206c0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-10 06:14:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"927264c404992945d9010e2f96886e9ee2728f2c","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-10 06:14:51.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e6687e2212fde7fb6f452095c67aaa08921157fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 07:18:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/13765f85f1ba4b58bdd70de5e5a95a39 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3780507b4ce4d8992fb92e5cfaef4b7 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99b391910a7c4649809e50c8b5571a80 : FAILURE in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/074599e62ae84afe840bab72d25bc985 : FAILURE in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2979c49a92fd44f5bb6b823c828f2cd0 : FAILURE in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4bd92ecb72454cb9d50458badb75b3 : SUCCESS in 7m 25s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":3},{"id":"c41ad332db77ef119ff5ef8957d0e85c89d03da4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-14 07:44:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"44e8b75898f394e338acef9e4f175cec11662446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 07:55:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3892418b6624400d948c2a9a47247673 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fb7bab604694e7f814ff651f3883638 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14f91a4695eb4a219a815c6a1df34575 : FAILURE in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c85b0120c28b4210a7c528ef7154898f : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1fb4d22e6d141f595891cfd2e773d8b : FAILURE in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57eeefe63b384da6b586f3a3775dcd17 : SUCCESS in 8m 46s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":4},{"id":"174348bdc9d5c966e9198ec044df61637378bb6c","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-15 10:59:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5460c99fd42a991fe35dc681470e90d8df4f046c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:12:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f376b816d2964501b7c327fa1748629a : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb7887732b254600bc357803ea492396 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d050cd273db4224bd1a56da4f5042b1 : FAILURE in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9f41b056ff44b0fa1adff7c974578dc : FAILURE in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c491080ac4914518a9b1a67e2b665fdf : FAILURE in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb050e512e144e21abc1efdc0991e8c7 : SUCCESS in 6m 24s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":4},{"id":"96c591f31062674c8c420584a08452ace59c24f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-21 15:09:40.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d444d0a4f5b5182bffcf2e69648ed11932072ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 15:19:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/588cd0030eb442c2a8fafe7d41048edd : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4995554e99c41a59c91013e1a1fab07 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7774ce192484815b3c935156912d684 : FAILURE in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e7115c3f1e1494dae6481958a335979 : FAILURE in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a75d6a745df4abf8c4b6d74d4c69922 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1863d576c04ca3b03bf57510b203f6 : SUCCESS in 7m 36s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":5},{"id":"989df45db638bd76d2b8d7ccce5ec9a80ecb5a4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-24 09:51:01.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5f593883a417255ca6f0c3730c10a9c7e996e7d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-24 09:53:26.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"5367fc903aecca32c6ad1227f40b57163c5bb782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 10:17:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/93849afc1c4c4faabd1a1ef778b73ea4 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9932bb2f12f642398544f447fb811824 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a914663232e4422bbab1a7155431f13 : FAILURE in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c7d1fadc5ed456d9a59c8265802c9d3 : FAILURE in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f723bc2cb00b4a628511d2f0712cb29f : FAILURE in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c41bcf432b7431ba7ba74b15f1c9b0d : SUCCESS in 8m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":7},{"id":"f7da48effe3cc5d8958ec10726632bfbc81bd5ae","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-07-02 06:17:06.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"45df909373015137147c2e03d56dfa4f699db295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 07:35:46.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/98812eefede94a399449ae5b799471d7 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6840bf6908e40219341f770448614b2 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/609519810ce446b39f8bd373e8c2e913 : FAILURE in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2299c8e11362441ba6d7a6d1453b20a0 : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3427c93cf6b24574b7ce7d8524f00e65 : FAILURE in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d412199c2b0045368aebdff5035b43e9 : SUCCESS in 6m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":7},{"id":"b2ca7a46a30e1bbda455c682da2392e5915e1e0e","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-07-21 07:22:54.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"cf5487bd48be60319a0edd79a90c02e6bd0b846c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 08:32:02.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/2417568f46f6428e85c1bc5648a94065 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bed24492cd6443f814b8f2b2c4b693c : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09e017e64d641c6aa3f5e1331929f2b : FAILURE in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/843b3376012a4cf7b9f0e1910ae5fc96 : FAILURE in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ec26a136b344117b770fdfee6acb310 : FAILURE in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05267b6669d4c83984676fbc096c833 : SUCCESS in 7m 25s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/67951385e36c465e94ed1c1d375d1a78 : SUCCESS in 1h 07m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":7},{"id":"c46fa09ad21253de975a5794b78e1c563cda4925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-07-26 11:34:35.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"04edab9a04df741e7620f4c00f6772813a8b6c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 13:01:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ddb2300942cc40b78a4bb1aa57fe5223 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8395f57fa7ca421988b15b6c90baafbe : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4543fc78163343ad89160b07fdfab7a5 : FAILURE in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7be3b50f86bf497f8e5107a198c38b08 : FAILURE in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29db1ff8b0b446db99dbc0831c6968e3 : FAILURE in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd57b2dff7aa4a04a69e05c7cebe345a : FAILURE in 6m 12s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5aec6decb391479d8df8cae2b0b70ef9 : SUCCESS in 1h 20m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":8},{"id":"57649a805e1113127dcc7f404a1898f742e8378c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-07-30 08:00:36.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"983ae4dcf6c106e5be090f9ba3c022f1ce41f0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 09:25:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/ec3d14db76f1462286ddf8a9ccc0b993 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/968bab346e5e4415bb211278dc7a7049 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e81893ad0ea40779dad420ec49b6953 : FAILURE in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebf6c38d058743b9ae0ece749341b108 : FAILURE in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/966819ad65a5477e8884df06ef0a47be : FAILURE in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db6e71aaddc4bbaa6d3ede9b21014f6 : SUCCESS in 10m 47s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9d3d571c70c54f5f8b5de27150f95dc5 : SUCCESS in 1h 22m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":9},{"id":"21e8c5898896ed40a4791de2fa0c8d82dfecff13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-03 12:43:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"96a7fc842dcdfc85fe4fa70f0c38f5fa3e337479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 13:56:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b1043965af243df9c2effbc471377ce : SUCCESS in 3m 46s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3987858c5ee7478189f0d70bcd4ae105 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617c2b9cc13a4ee9b40c9028c3c33e32 : FAILURE in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc7aa85f7c34573b868a3f872306611 : FAILURE in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53cd6f06658a4c7a992e880d84cb3421 : FAILURE in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76db1f9771c047dba64f2cfd1aa5648d : FAILURE in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7b5d1fe3ab4595af062240b20c027b : FAILURE in 5m 22s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/13aa3eb031ea4af4a9d54d99a45ecfbc : SUCCESS in 1h 10m 47s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":10},{"id":"258f3bea67d06abcaa406e06e150cbd0fc5119cb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-04 07:28:27.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d896fba2527e7ae524b5a400936e9dd7eaf10136","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-04 08:23:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"88286f7ba82be86743ccad69f679f96b6438ae87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-06 11:07:30.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"6fa58c4f321a5b62340eedb278635f43cd09cc1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-06 12:05:14.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c12e468ebd2ac917597432d562a58669e22d9832","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-08-06 13:16:04.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"72d60847fc712f65c2c2feedbd00fcae7bf3fdbb","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-06 13:58:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fb564fcea88299be49bfb9fa5276841340f82427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-06 14:05:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"65b38cd839d851a191de6279b71128ca8e620049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-06 14:30:43.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"0f4f0c6a50da29bb0db71718987b532246528171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 17:35:47.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4e8cd1791acb437fba4581bcf85dec82 : SUCCESS in 3m 58s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5d54c7dc4ab6474b969c439c0bb6b938 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a63fc7e02d34c238f745208ca44deef : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ae8ea9f89314bfab1f3c3af98532d46 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f7736b826584b928e6cfbefc936d903 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/875947b106b24cf2a9c48b669afbecda : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ceba237c88409b9a06dbefb4dd5660 : SUCCESS in 8m 07s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/21f0d0c984754fb8bb7f5e345ab069ca : SUCCESS in 2h 43m 06s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f2987a70cdfa47bc97dd96695fad097d : SUCCESS in 1h 33m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e7390694cd3a434491e8c549fe4e8622 : SUCCESS in 1h 01m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/88a785da098f4f0fb67e72199e2b3b2d : SUCCESS in 1h 14m 03s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/b1be8528b725426fa98de73971008fb7 : SUCCESS in 2h 45m 06s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e12b877e27824846a4990ce880324647 : SUCCESS in 1h 24m 04s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/8cd2ff95bbaa40c4974470946072633e : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":14},{"id":"b696f89350dde944a02f7a7dffd88775ace2c785","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-09 14:38:37.000000000","message":"Patch Set 14: Code-Review-1\n\nplease make sure the version requirements are met in the rpm spec before landing this","accounts_in_message":[],"_revision_number":14},{"id":"8698b7629f91c74559376ee7d3f08f1c6ab4e6ea","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-09 15:07:35.000000000","message":"Patch Set 14: Workflow-1\n\n\u003e Patch Set 14: Code-Review-1\n\u003e \n\u003e please make sure the version requirements are met in the rpm spec before landing this\n\nGood point. Putting -workflow until they are merged.\nhttps://review.rdoproject.org/r/q/topic:%22VLVERSIONINC%22+(status:open%20OR%20status:merged)","accounts_in_message":[],"_revision_number":14},{"id":"64a70f640c644e5c37feaeeeeef220394d8c342d","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-16 08:09:10.000000000","message":"Patch Set 14: -Workflow\n\nhttps://review.rdoproject.org/r/c/openstack/tripleoclient-distgit/+/34871 was merged today. Updating the version in the tripleoclient-distgit.","accounts_in_message":[],"_revision_number":14},{"id":"a177f93c2b735d9d5fca88c831017afaf1e5a7e0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-16 13:54:33.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"a1f4e5e1a3ec44888ded5e3c6be01e855064e525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-08-23 08:51:33.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"6200f348a81aa80e9314b7fb8ba9af949d81bb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 12:05:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a290fe849e1e480fbd296853f7ce9c84 : SUCCESS in 3m 13s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9e49531653284d1e8ce111b6e10428eb : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97284898db544fa09e2db0b3801c051e : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f69dda5e2e245739f287ee6a45998fe : SUCCESS in 8m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abfdd7101e8f482db33a5c3470095bb7 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cc77aab08904ad9a80300044f8b43f8 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb753dbc6b5e4db4bcebda741f2b0516 : SUCCESS in 6m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2350a23984044d81a2d0038095842831 : SUCCESS in 2h 55m 38s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/172a8401093c47039113c6bd815c7763 : SUCCESS in 1h 36m 51s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/77bf127538f04029b3eb50177b489fd5 : SUCCESS in 46m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/060b8aa407b84591bad119bce729bf33 : SUCCESS in 1h 13m 32s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3507019e03534f8d865b275fb7cea6b9 : SUCCESS in 2h 38m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/067d119655a840989ed34cf9dda61f25 : SUCCESS in 1h 28m 49s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7e36781e50b5417c9b7ad47a6220bbfa : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":15},{"id":"4fb8d69cc61f2d65785d37c722e09e1d8354f8ac","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-08-23 14:25:56.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"aa7a0f2bd03cbee1c8687f9295268e4f298b2f2c","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2021-08-23 14:25:58.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"8895b160f21edd2bb2be061e1a7a09f57955d090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 14:26:46.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"1e56f03956f27611098f5a8022e28ddf29f4c864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:53:58.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff7982c8e3714282abc55a0a2fa263d2 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d772167a1b24bdab10c49cdbe39510b : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99a6756b63f940a9aa69e370fda8943a : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a23ff55319840b18565d237d992b558 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d14c33d1d4ae4fad97b97aec41590c04 : SUCCESS in 7m 55s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7b1f604f2440488b9f16562d6d9be1ed : SUCCESS in 2h 52m 41s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fc0b8d81b13943b2a91090104bcf2b72 : SUCCESS in 1h 30m 41s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a452275693eb4bb780f25b078908350e : SUCCESS in 45m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/69d22c0195be468abfd2f09556185eb0 : SUCCESS in 1h 16m 33s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f775f2b176ea4442b5f7d6ccf1728452 : SUCCESS in 2h 33m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5ce2f909a9e44fa98fab87196f485909 : SUCCESS in 1h 23m 02s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/40391f09ce374b258f05bfecfdb9df8a : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":15},{"id":"d598faf44b06c883da6bd710b4f6959c5af234c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:53:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"c07abbdb7fe4ebdf7c2faa2168030da0beb79485","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:55:32.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f990f0bc17de4da1a42a29b6bee7cd6f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"812681dc6bdfb212076dd3b6dfce2eeabffeb0f1","revisions":{"6c250b4934c708932a6cc819529a59af64df80b5":{"kind":"REWORK","_number":1,"created":"2021-06-03 12:33:33.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/1"}}},"commit":{"parents":[{"commit":"fe22d67515154f819ba6fd75d673668df04038b5","subject":"Merge \"Fix check_service_vips_migrated_to_service()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/fe22d67515154f819ba6fd75d673668df04038b5"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:32:57.000000000","tz":120},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/validations-libs/+/789770\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6c250b4934c708932a6cc819529a59af64df80b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6c250b4934c708932a6cc819529a59af64df80b5"}]},"branch":"refs/heads/master"},"8aad06fde29b14d950982013a1850f49f2f93604":{"kind":"REWORK","_number":2,"created":"2021-06-04 10:24:06.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/2"}}},"commit":{"parents":[{"commit":"fe22d67515154f819ba6fd75d673668df04038b5","subject":"Merge \"Fix check_service_vips_migrated_to_service()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/fe22d67515154f819ba6fd75d673668df04038b5"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-04 10:22:05.000000000","tz":120},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/validations-libs/+/789770\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8aad06fde29b14d950982013a1850f49f2f93604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8aad06fde29b14d950982013a1850f49f2f93604"}]},"branch":"refs/heads/master"},"e43572993c85cc1027bbd4b2316e9cd1c51b8321":{"kind":"REWORK","_number":3,"created":"2021-06-10 06:14:23.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/3"}}},"commit":{"parents":[{"commit":"fe22d67515154f819ba6fd75d673668df04038b5","subject":"Merge \"Fix check_service_vips_migrated_to_service()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/fe22d67515154f819ba6fd75d673668df04038b5"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-10 06:14:15.000000000","tz":120},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/validations-libs/+/789770\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e43572993c85cc1027bbd4b2316e9cd1c51b8321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e43572993c85cc1027bbd4b2316e9cd1c51b8321"}]},"branch":"refs/heads/master"},"b9e53a1edc053405bc802999c64d1a2b49586c1a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-14 07:44:39.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/4"}}},"commit":{"parents":[{"commit":"030e9f0b570e3ab119c68b7ab7e0220a09dd48a1","subject":"Use ephemeral Heat by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/030e9f0b570e3ab119c68b7ab7e0220a09dd48a1"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-14 07:44:39.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/validations-libs/+/789770\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b9e53a1edc053405bc802999c64d1a2b49586c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b9e53a1edc053405bc802999c64d1a2b49586c1a"}]},"branch":"refs/heads/master","description":"Rebase"},"8bfbeda96a76bfb22331bf34067689dd546b2e3a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-21 15:09:40.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/5"}}},"commit":{"parents":[{"commit":"6cdab86ed3c2dd4b93a41c1bf49aec6cfd055868","subject":"Merge \"overcloud deploy - net ports with baremetal deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/6cdab86ed3c2dd4b93a41c1bf49aec6cfd055868"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-21 15:09:40.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/validations-libs/+/789770\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8bfbeda96a76bfb22331bf34067689dd546b2e3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8bfbeda96a76bfb22331bf34067689dd546b2e3a"}]},"branch":"refs/heads/master","description":"Rebase"},"a8388babd860ae6d7a1e7eb17ba461960840526c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-06-24 09:51:01.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/6"}}},"commit":{"parents":[{"commit":"e8dae35edd219712a42c33914f9ef3806a4ac0a6","subject":"Merge \"Default Heat images to container image prepare defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e8dae35edd219712a42c33914f9ef3806a4ac0a6"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-24 09:51:01.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/validations-libs/+/789770\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a8388babd860ae6d7a1e7eb17ba461960840526c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/a8388babd860ae6d7a1e7eb17ba461960840526c"}]},"branch":"refs/heads/master","description":"Rebase"},"8a95fb24dbd0225645376eedc208d1416f55a47a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-06-24 09:53:26.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/7"}}},"commit":{"parents":[{"commit":"e8dae35edd219712a42c33914f9ef3806a4ac0a6","subject":"Merge \"Default Heat images to container image prepare defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e8dae35edd219712a42c33914f9ef3806a4ac0a6"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-24 09:53:21.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a95fb24dbd0225645376eedc208d1416f55a47a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8a95fb24dbd0225645376eedc208d1416f55a47a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"34ad46bc96cd8e7978a519e0a89572d7c5310114":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-07-26 11:34:35.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/8"}}},"commit":{"parents":[{"commit":"d7e8bfb0539bf14a5ee4c670164d2b8ad78db657","subject":"Enabling pylint no-else-return","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/d7e8bfb0539bf14a5ee4c670164d2b8ad78db657"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-07-26 11:34:35.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/34ad46bc96cd8e7978a519e0a89572d7c5310114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/34ad46bc96cd8e7978a519e0a89572d7c5310114"}]},"branch":"refs/heads/master","description":"Rebase"},"b8505a236ac6cb88290b1dad058273b4736faeaf":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-07-30 08:00:36.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/9"}}},"commit":{"parents":[{"commit":"0adc5bb9dff517cbe298d97d5bcee698b2a2e77c","subject":"Fix --stack-only to download ansible playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/0adc5bb9dff517cbe298d97d5bcee698b2a2e77c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-07-30 08:00:36.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b8505a236ac6cb88290b1dad058273b4736faeaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/b8505a236ac6cb88290b1dad058273b4736faeaf"}]},"branch":"refs/heads/master","description":"Rebase"},"ff91bea4d320d059a024ea7f588d0bd6eb1d76d9":{"kind":"REWORK","_number":10,"created":"2021-08-03 12:43:12.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/10"}}},"commit":{"parents":[{"commit":"525b2bdb71cd53db74efc30d68a922d5c32950e9","subject":"Merge \"make net_config_override of undercloud.conf work with yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/525b2bdb71cd53db74efc30d68a922d5c32950e9"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-08-03 12:42:28.000000000","tz":120},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ff91bea4d320d059a024ea7f588d0bd6eb1d76d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/ff91bea4d320d059a024ea7f588d0bd6eb1d76d9"}]},"branch":"refs/heads/master"},"e6fb794f4a6403f6da5464415961f65990ed1a7f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-08-06 11:07:30.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/11"}}},"commit":{"parents":[{"commit":"8031acb16bfccb1608e5dd3d8505d7932df37473","subject":"Merge \"Fail the overcloud deploy is ceph-ansible.yaml is included\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8031acb16bfccb1608e5dd3d8505d7932df37473"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-08-06 11:07:30.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e6fb794f4a6403f6da5464415961f65990ed1a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e6fb794f4a6403f6da5464415961f65990ed1a7f"}]},"branch":"refs/heads/master","description":"Rebase"},"22ed3eba23bc1102a89279d2d65c0fadedd3a452":{"kind":"REWORK","_number":12,"created":"2021-08-06 12:05:14.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/12"}}},"commit":{"parents":[{"commit":"8031acb16bfccb1608e5dd3d8505d7932df37473","subject":"Merge \"Fail the overcloud deploy is ceph-ansible.yaml is included\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8031acb16bfccb1608e5dd3d8505d7932df37473"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-08-06 12:04:39.000000000","tz":120},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/803736\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/22ed3eba23bc1102a89279d2d65c0fadedd3a452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/22ed3eba23bc1102a89279d2d65c0fadedd3a452"}]},"branch":"refs/heads/master"},"e904134fb09c52ff39cbc6caa509882691d49241":{"kind":"REWORK","_number":13,"created":"2021-08-06 14:05:27.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/13"}}},"commit":{"parents":[{"commit":"8031acb16bfccb1608e5dd3d8505d7932df37473","subject":"Merge \"Fail the overcloud deploy is ceph-ansible.yaml is included\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8031acb16bfccb1608e5dd3d8505d7932df37473"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-08-06 14:05:15.000000000","tz":120},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/803736\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e904134fb09c52ff39cbc6caa509882691d49241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/e904134fb09c52ff39cbc6caa509882691d49241"}]},"branch":"refs/heads/master"},"af183cc12aa024f0810bb7353d6958dbdb908622":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2021-08-06 14:30:43.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/14"}}},"commit":{"parents":[{"commit":"8031acb16bfccb1608e5dd3d8505d7932df37473","subject":"Merge \"Fail the overcloud deploy is ceph-ansible.yaml is included\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/8031acb16bfccb1608e5dd3d8505d7932df37473"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-08-06 14:28:20.000000000","tz":120},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/803761\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/af183cc12aa024f0810bb7353d6958dbdb908622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/af183cc12aa024f0810bb7353d6958dbdb908622"}]},"branch":"refs/heads/master"},"812681dc6bdfb212076dd3b6dfce2eeabffeb0f1":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-08-23 08:51:33.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/794577/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/77/794577/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/77/794577/15"}}},"commit":{"parents":[{"commit":"72cba8790dcb6829d1cf0ba2b118ee47ea24907c","subject":"Merge \"Remove reference to --network-ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/72cba8790dcb6829d1cf0ba2b118ee47ea24907c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-06-03 12:29:21.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-08-23 08:51:33.000000000","tz":0},"subject":"Overriding the log path argument for tripleoclient cli","message":"Overriding the log path argument for tripleoclient cli\n\nDepends-On: https://review.opendev.org/c/openstack/cliff/+/794792\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/803761\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I0d1f5f7fbd35d69ec9f0e44aab0831e0de875295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/812681dc6bdfb212076dd3b6dfce2eeabffeb0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/812681dc6bdfb212076dd3b6dfce2eeabffeb0f1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
