)]}'
{"id":"openstack%2Fpython-tripleoclient~844636","triplet_id":"openstack%2Fpython-tripleoclient~master~I9610ce51213ca2ad191d60914398131f11016453","project":"openstack/python-tripleoclient","branch":"master","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-07-13 20:57:33.000000000","reason":"Change was submitted"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-07-13 20:57:33.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-13 14:33:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2022-07-13 20:57:33.000000000","reason":"Change was submitted"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-07-01 14:50:56.000000000","reason":"\u003cGERRIT_ACCOUNT_6926\u003e replied on the change","reason_account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2022-07-13 20:57:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9610ce51213ca2ad191d60914398131f11016453","subject":"Update cell export for ephemeral heat","status":"MERGED","created":"2022-06-03 18:51:28.000000000","updated":"2022-08-02 12:06:13.000000000","submitted":"2022-07-13 20:57:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844636","meta_rev_id":"88f4ab7072cd79be654d572627f12ab84f2b486c","_number":844636,"virtual_id_number":844636,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 20:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"all":[{"value":1,"date":"2022-06-07 06:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":2,"date":"2022-07-13 14:35:46.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-06-06 12:03:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-06 14:22:21.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":1,"date":"2022-07-13 14:35:46.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-03 20:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-06 06:00:05.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"CC"},{"updated":"2022-06-06 12:03:27.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2022-06-06 12:03:27.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-06-07 06:22:29.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2022-06-14 12:09:47.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2022-07-13 14:13:54.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-07-13 14:13:54.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-07-13 14:13:54.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"119d3de50e5c4c6e0e49583f0e75b5dd7d4eb9b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-06-03 18:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f164d0fb5b23739d408839bdbf78fe72a90056a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 20:06:31.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cab3901770ec479d8b22a66654e30bd1 : SUCCESS in 5m 04s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/f67704e4443f48de86e4004fac81a5f5 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b7fbb6f4d6447495d24c954888262b : FAILURE in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9c4c987f05f4a48a0a4ba66f2e8ef88 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eb3061cc02d4d7ba01c6a2a53e81650 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a0e017ee32b4a9082ded5ab1a546970 : FAILURE in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db15cc29ebf544af9b6373f3717a4d0c : SUCCESS in 6m 07s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/49ebf16a978343caabad8ce6fac2827b : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/4e85bec85ade4444b4e969d0079236ee : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/593487c65b074251a3b2a1012634b16b : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/6d1e4b32665d4f02a84a23c9c12efb1b : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/abc432d613f744528b316eb68d86d2d5 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/54c25eface7148428b6af1a78424521d : SUCCESS in 1h 08m 39s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a85f265422324547884094f11bdc9ec4 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/918283dc0f4a45fe8fe610448778a022 : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"4e207ed512effe9d45b039b2f3c96ab9feb3b194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-06-03 21:00:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92a2cc062003a79bd89b4a7d8f2855e9a3d74b71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 21:24:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/311e1998d9e34d74ba8ac47ebb4677d5 : SUCCESS in 4m 49s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/0fbcd2e2c9cf4ab3a64cd64d157ab97f : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcaddbcee01247dea0b5be9a6f951aba : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65494595fb9f44c6ac9fd8cbc00855a5 : FAILURE in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9dca6e41c134430a1531dd360ddf2a8 : FAILURE in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c950754106f04b24be2ff55110eac82d : FAILURE in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e762e749d0354e7b875ddfb4016c9fd7 : SUCCESS in 6m 39s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8905e06906b24bc6b3159f83341287bc : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f318d38a83aa4c2ca255933061f84d70 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/11dbaef5869a44619fda338e19e99ca8 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/7fb2f11328c94a7a90b73f4deeb54b49 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a20bec2bc6f4419c87b4899c3af4e855 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/815080aa34314768b4b3a596c808a641 : FAILURE in 18m 59s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6afabf008a144b499b24113ba0d3a55d : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/09f0b6363ec242c9bc3b1fe516a6cbf2 : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"54b1d0bc4d7cf3c68966742554025f9cee714286","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-06-06 06:00:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb6c83f2f99f64a35de130a16dda88fef916105b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 09:25:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecba0382e93c424e9e168b4edfb70ad0 : SUCCESS in 4m 48s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/c510bc2bcf03416da04bce882b069fe0 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9942e17aef904ecca157d2b0afd7466c : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/570e7d5951d449b28688f290d5896989 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2966043ff5644b2b6f76e8925021921 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4290f8c476724c3783dff23f7a7312a9 : FAILURE in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce99430dc3d422dad2233f4ee250361 : SUCCESS in 4m 52s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5f780b379fec40b0b6bfaed637069321 : SUCCESS in 3h 14m 15s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/68ac5a3b08084dad8464a0b4576a55d5 : SUCCESS in 2h 13m 05s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/aad0642cfb5f41db8d6bd0c399346714 : SUCCESS in 2h 06m 15s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f96f254aeb20467ca27b13bc06bf2c78 : SUCCESS in 42m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/e1cc675755ba419dbbe8d7f422a3d364 : SUCCESS in 1h 17m 09s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d04d17fdd5794b33b22476b99ea1785c : SUCCESS in 2h 40m 46s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9b89ee7a829941f6852d6c00ec300627 : SUCCESS in 1h 13m 21s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c3ffaa600d1a4dea95887c52ae810a40 : SUCCESS in 1h 27m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5a15f17e7c11133843faf2a2e1dc143b761e3e2e","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-06-06 12:03:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5eac9686f6d7a915bbc56cc10ee2b0b40dee7437","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2022-06-07 06:22:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fbe118156d93b91234fdd4cafa7574c010052757","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-06-08 17:16:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bab5725a83f26210590261a96d5866da52c46075","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-06-14 12:09:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6bc4037af9237d50f3dc2a1118f9ee8cb43fa47","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-06-14 19:50:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d865f2bcb6452f66722fbdca30c16d253628469f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 14:50:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81bfdc49a802086febab08dfe37a2d2301caea66","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 14:51:33.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9992cba7d7cf4cf565b5e0ef5276370a11f9c78a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-06 14:22:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ef967bd82910fbe70cb55cad76b4277299297143","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-06 14:22:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ebff72150a4810c458d60c6c41504b6db5de7b9","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-13 14:13:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bf558cd1d8b4ed5b6e338bdb076327b67358e8c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-13 14:33:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ba9ccbc1953188209ddb3a61d5412af3bdde30d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-13 14:35:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5aac80eb93df6a22724cdf8ebbe5840b2b0b6cdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 14:36:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07100c2b9ba6deb2cc99b7661f3b56033b05315f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:57:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5aa99b5846d48c6848c909c00de7fc4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82713630993540aa822b26afc1189a77 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b365832e43043e5ac6559a90ecbc7ac : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08383bbfefb74c0e89ee9d5bc2cc5eb4 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede892c794e24d73a8450b928d068ef4 : SUCCESS in 6m 01s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4e419c154f99451e8781996f94b03382 : SUCCESS in 3h 16m 53s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7ffcecb9e1e44c4d99fe988d7544676d : SUCCESS in 1h 54m 08s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/5a2bffa2f49e4eb8bb5b3a6ae7be82bb : SUCCESS in 55m 11s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/d188a60056fb4259b8e474fad31707b7 : SUCCESS in 1h 27m 30s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/424d22f93bd0416394e5f07838f12f67 : SUCCESS in 2h 35m 17s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a593574e1cb6468db3a4edba39b1b31c : SUCCESS in 1h 11m 47s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6291e6d0dada427cbbb1e6557f5da798 : SUCCESS in 1h 44m 43s","accounts_in_message":[],"_revision_number":2},{"id":"77b5ae3e69cd56e54f01202363089b4baa45e8f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:57:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1d5cb0c4fa72391e50e64c09549a59b19d3bc960","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:59:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26ae42d70f8245c8b4517cd8a73f9879\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b05b11f26ab24a9aac9caa0e20e01af6 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"88f4ab7072cd79be654d572627f12ab84f2b486c","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-08-02 12:06:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c11228c569747a0789ef38a1e873d2749ed0321c","revisions":{"bf9d73d507c8901226de92e682a48894e84b392b":{"kind":"REWORK","_number":1,"created":"2022-06-03 18:51:28.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/36/844636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/36/844636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/1"}}},"commit":{"parents":[{"commit":"da971fb02a3171c1ffb2529c2420eb02897fd594","subject":"Merge \"Enable snapshots on the CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/da971fb02a3171c1ffb2529c2420eb02897fd594"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2022-06-03 18:50:39.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2022-06-03 18:50:39.000000000","tz":-240},"subject":"Update cell export for ephemeral heat","message":"Update cell export for ephemeral heat\n\nopenstack overcloud cell export was never updated to account for\nephemeral Heat changes. This patch removes the dependency on using heat\nclient, and updates it to use the working directory instead.\n\nChange-Id: I9610ce51213ca2ad191d60914398131f11016453\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bf9d73d507c8901226de92e682a48894e84b392b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/bf9d73d507c8901226de92e682a48894e84b392b"}]},"branch":"refs/heads/master"},"c11228c569747a0789ef38a1e873d2749ed0321c":{"kind":"REWORK","_number":2,"created":"2022-06-03 21:00:20.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/36/844636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-tripleoclient","ref":"refs/changes/36/844636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-tripleoclient refs/changes/36/844636/2"}}},"commit":{"parents":[{"commit":"da971fb02a3171c1ffb2529c2420eb02897fd594","subject":"Merge \"Enable snapshots on the CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/da971fb02a3171c1ffb2529c2420eb02897fd594"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2022-06-03 18:50:39.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2022-06-03 21:00:16.000000000","tz":-240},"subject":"Update cell export for ephemeral heat","message":"Update cell export for ephemeral heat\n\nopenstack overcloud cell export was never updated to account for\nephemeral Heat changes. This patch removes the dependency on using heat\nclient, and updates it to use the working directory instead.\n\nChange-Id: I9610ce51213ca2ad191d60914398131f11016453\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c11228c569747a0789ef38a1e873d2749ed0321c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-tripleoclient/commit/c11228c569747a0789ef38a1e873d2749ed0321c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
