)]}'
{"id":"openstack%2Fpython-troveclient~1003258","triplet_id":"openstack%2Fpython-troveclient~master~I64c433e0346ec50526335f9c2e72ff5e0b80df2d","project":"openstack/python-troveclient","branch":"master","hashtags":[],"change_id":"I64c433e0346ec50526335f9c2e72ff5e0b80df2d","subject":"Add an option to append allowed CIDRs","status":"NEW","created":"2026-09-01 11:33:05.000000000","updated":"2026-09-02 08:53:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"34bc92f344fd960865fc32efa155ebce4483daf0","_number":1003258,"virtual_id_number":1003258,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-02 08:53:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-09-01 11:40:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89aacb496c8cb7fef2e68c4e8c0ad6b8a9da1835","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-01 11:33:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"711dbc0c3d0295ad72a76e040bdcbb8bf19b0815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 11:40:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff2196e3d904e51b55812cf2aeb12cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd240ddc0c704c56a0921b204f865a8f : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2e1ed6306bd47bfa76e4b2fe7f0eff7 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/139580ddcd0b4a3d81438d083a303188 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/247d551fefe04ed79abdabcd426a0bed : FAILURE in 5m 46s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/72fd3a5b25824b519a291560b197817e : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783ea5bb08a3488a8a11c7164176d3e2 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887a06c537f245c6a7a49582c70826c9 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1411a74fe43997d1ac10f18dfb5e5312d277d465","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-01 11:44:44.000000000","message":"Topic allowed-cidrs-sync-for-replicas removed","accounts_in_message":[],"_revision_number":1},{"id":"b8b99d5de83236c1c4fc08ad41990b98c5946d86","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-02 08:37:20.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"34bc92f344fd960865fc32efa155ebce4483daf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:53:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a09f8a12244c3d808fb89384e0d3e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579be797a234466d89def50bad7c4d8e : SUCCESS in 1m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a4b6771e59347da80a07fe5faf3b2b0 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7cd448efe904465795166cf4577343cb : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c73c8ec2421a4a63a8a890cd317ef71f : FAILURE in 4m 56s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/dc532df4316a4f6a91ccb30d3378cd30 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936d954347e442c3a00225cb662f7d08 : SUCCESS in 2m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f813fd5745df4ec5bd58282943e39a33 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"93410bc2f58c6298c8c117a6e4f267670e6a8775","revisions":{"91f2063a098abc6ffa368161874493216cb6457c":{"kind":"REWORK","_number":1,"created":"2026-09-01 11:33:05.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/58/1003258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/58/1003258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/1"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-01 11:31:19.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-01 11:31:19.000000000","tz":300},"subject":"Add an option to append allowed CIDRs","message":"Add an option to append allowed CIDRs\n\nThe existing --allowed-cidr option replaces the current list of allowed\nCIDRs. Add --add-allowed-cidr to preserve the current list and append one\nor more CIDRs without duplicates.\n\nMake the two options mutually exclusive.\n\nChange-Id: I64c433e0346ec50526335f9c2e72ff5e0b80df2d\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/91f2063a098abc6ffa368161874493216cb6457c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/91f2063a098abc6ffa368161874493216cb6457c"}]},"branch":"refs/heads/master"},"93410bc2f58c6298c8c117a6e4f267670e6a8775":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-02 08:37:20.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/58/1003258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/58/1003258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/58/1003258/2"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-01 11:31:19.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-02 08:37:20.000000000","tz":0},"subject":"Add an option to append allowed CIDRs","message":"Add an option to append allowed CIDRs\n\nThe existing --allowed-cidr option replaces the current list of allowed\nCIDRs. Add --add-allowed-cidr to preserve the current list and append\none or more CIDRs without duplicates.\n\nMake the two options mutually exclusive.\n\nChange-Id: I64c433e0346ec50526335f9c2e72ff5e0b80df2d\nSigned-off-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/93410bc2f58c6298c8c117a6e4f267670e6a8775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/93410bc2f58c6298c8c117a6e4f267670e6a8775"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
