)]}'
{"id":"openstack%2Fpython-troveclient~998328","triplet_id":"openstack%2Fpython-troveclient~master~I587b0d4d6eeba8e94621ff0392d876f7d712d50e","project":"openstack/python-troveclient","branch":"master","topic":"api-validation-rules","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-28 11:46:17.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-28 11:59:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I587b0d4d6eeba8e94621ff0392d876f7d712d50e","subject":"Add datastore volume type association support","status":"MERGED","created":"2026-07-22 15:51:43.000000000","updated":"2026-09-28 12:01:31.000000000","submitted":"2026-09-28 11:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998328-api-validation-rules","meta_rev_id":"3860b0c5b6d1cbf3e59743af239be38028049d5b","_number":998328,"virtual_id_number":998328,"owner":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 11:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-28 11:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-28 11:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-22 16:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 08:04:04.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-28 04:23:21.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"f631ee838dc98f7f7b7c156b767f5bfb5cab3a62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-22 15:51:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a1517937a7d862f6b495d9c3d184cfed0416e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:06:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6582d2da42ba4fbb8e4d570493aff381\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b28b014d384c0cbb34671450b0ef8d : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1c7e941ab9a41fe8ae5aca75be5642a : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bac894b044f747999f057c284d5908b8 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9029ebb11da247efa1498aeff45060d5 : FAILURE in 6m 06s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/5735eda900c047fda77fef15725db3bf : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa6ec4bbcc54891b852d38bda0d03db : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9619930343184aac89591012895ca07b : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"df675a6960902b27c63dc0784c41a9e61efa147f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-07-24 15:43:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a137bed406e164040c70c00aebb7175ed6ad5065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 15:58:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/900939bb847b4be59ba479e96e64ec5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b807f8d6ba7427099a0beb72a5cd102 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/654413cbff794e68bda3b54474dafc8f : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/400d8c2c9a9345ce902aaee31c8617c9 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/35037345e1e04f98b4f6e708d72edf89 : FAILURE in 5m 21s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/d649a4ff373143c6ba9ddaa18c43e6ca : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd9d8e6782045c9a431fd03026beac4 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567df5ec3f9c4cd7be88ba16b00f1f1c : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1287e23234898bd99601970b74f98134daf94d60","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 08:04:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95902eab0d416c499e4a9dfc6a686c7d9ce5cd2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-17 09:06:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1e8927652936c216fd83b122a2375ab20f8a6553","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-17 09:07:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a82ada2602668f1ed58d287cc1b0530f91dca9a6","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-17 09:11:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2a7093cf2c5b2467a886c9123a0ab1659d007888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:23:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f76a2f2173ea48a3b4bb238a5cedce02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d106ae87dd54667b71c7507b6fdd036 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce2c6d21a45b44cfb6d29176546fd68b : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5512949e398346688f75d3f6be7e7af9 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bf7c4a49dc9a4709ab8e3421fb4e084f : FAILURE in 5m 50s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/3a8ffb3ca0864426944d8ea09e1cfc78 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9890d6b4e254a54b69027ce6eae5d8d : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7b65c9670ba4b3c86ad77675ecac63a : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5c74e7307c760af7280f98fffc48ba8e54010fc7","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-28 04:23:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a902e088a4d12c21dcbe2523435dde19ac02429b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 04:23:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"de412459e492c5b35e7545eb9fafb53a263cb49e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 04:32:46.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc4a7716d10d48f389a5baffa5bf06d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f7da0486164b3282534e7df4b29263 : SUCCESS in 2m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f810f979f1d4ff0869b2d6c5b1dc8f5 : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/915c7ab5459f4356a39b2dea91a971c4 : FAILURE in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61bae1652fdf49f0808c649e2d6cfbb6 : SUCCESS in 2m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b75655ebd5d64bc1922559c1fc1f8c3b : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":4},{"id":"96e4b5dd74d5ba05a667f6cbb5bd544c704574c4","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-28 04:34:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"79ea1bfb22ff11dc1814659cdcb868bce026a13b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 04:44:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccd994bbd8d1495891f1bc520526602e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fe37eac74b84c5baa4c074315ed4b07 : SUCCESS in 1m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a358f161a4a44d819708bdb3bdd3a46d : SUCCESS in 2m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c235f55e9fc24735b44c122fbbb85f56 : FAILURE in 3m 47s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/6c9b4876e84c4ba9934c921ebc900470 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17347b6f422498094c072370d33a40a : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b400c4065ca426c90685a773b7384f7 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":4},{"id":"ded6fae8efe472cb1ce46bf5abe57ed35ab737a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-28 10:45:51.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"dc8199fabaefadd4440579b0bd1422a30a7f08e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:25:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad09b44dc225479ca8005e1bcac7c2cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d0dee54684d4f6eb8c1d7a03ff46032 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15062c8aabf34764b356db01909d1a11 : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4d4c1044f43e4b2ca89b5f606b172501 : SUCCESS in 6m 44s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/dfc15bf8f8774873b2f8b59cbab78830 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf422a67d0a4fd0ae5be62756ee614c : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d1cd5b683e345e589dcd1668b2b1913 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":5},{"id":"17eda86e06e3b7529d36b448e8ecf50b3a4ec9da","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-28 11:46:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f856d35957cd150ad59420555a38049fbe119314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:46:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f84f232833a63f79056dd000a54a30ee2442a97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:59:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf8c4b3982a44ff396ee84e93c05e5ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23fec0a203c14e42b60e6a43c0ba23f9 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4fb176e225d461db14417b3a312cf55 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b05e97882b624f67802a3694aab6be51 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7923ca586be4840978d5c97c85dbf42 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab89276e208d4818abfdcfdeeeb5148c : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":5},{"id":"8a729c993fe5900cb47b7ca4f8c4be87c047b49f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:59:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3860b0c5b6d1cbf3e59743af239be38028049d5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:01:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2a42961b72f46e887927510992e73c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7951ed059ed94b328e15092ae0874bf3 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/329ff6ca35e343bc83113b00348ef3a2 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e2c48b3dacf4162677d51a758bbfd905784f5de5","revisions":{"d394a1ed1179c70ccbfd2c1a9a548098a9a59a66":{"kind":"REWORK","_number":1,"created":"2026-07-22 15:51:43.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/28/998328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/28/998328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/1"}}},"commit":{"parents":[{"commit":"8e27a00e588cd324e99cdca04bac156e724b4de5","subject":"Merge \"Use unversioned environment by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/8e27a00e588cd324e99cdca04bac156e724b4de5"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-22 15:44:31.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-22 15:46:29.000000000","tz":300},"subject":"Add datastore version volume type association support","message":"Add datastore version volume type association support\n\nAdd --volume-types and --remove-volume-types options to the openstack\ndatastore version set command.\n\nExpose management client methods for updating datastore version volume\ntype associations through the Trove API.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/997966\nChange-Id: I587b0d4d6eeba8e94621ff0392d876f7d712d50e\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/d394a1ed1179c70ccbfd2c1a9a548098a9a59a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/d394a1ed1179c70ccbfd2c1a9a548098a9a59a66"}]},"branch":"refs/heads/master"},"13f54803f8b9cacc4f935cef1c61d0b0a4340371":{"kind":"REWORK","_number":2,"created":"2026-07-24 15:43:22.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/28/998328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/28/998328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/2"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-22 15:44:31.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-24 15:42:10.000000000","tz":300},"subject":"Add datastore volume type association support","message":"Add datastore volume type association support\n\nTrove currently supports associating volume types with datastore versions\nthrough trove-manage, which requires direct access to the Trove control\nplane. This is unsuitable for deployments where datastore version\nmaintenance is automated through API workflows or control plane access\nis restricted.\n\nAdd --volume-types and --remove-volume-types options to the openstack\ndatastore version set command. These options allow operators to manage\ndatastore version volume type associations through the Trove management\nAPI.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/997966\nChange-Id: I587b0d4d6eeba8e94621ff0392d876f7d712d50e\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/13f54803f8b9cacc4f935cef1c61d0b0a4340371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/13f54803f8b9cacc4f935cef1c61d0b0a4340371"}]},"branch":"refs/heads/master"},"ee776cabc4f0a9b2791a95f479116df9bf3ac581":{"kind":"REWORK","_number":3,"created":"2026-09-17 09:06:49.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/28/998328/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/28/998328/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/3"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-22 15:44:31.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-17 09:06:45.000000000","tz":300},"subject":"Add datastore volume type association support","message":"Add datastore volume type association support\n\nTrove currently supports associating volume types with datastore versions\nthrough trove-manage, which requires direct access to the Trove control\nplane. This is unsuitable for deployments where datastore version\nmaintenance is automated through API workflows or control plane access\nis restricted.\n\nAdd --volume-types and --remove-volume-types options to the openstack\ndatastore version set command. These options allow operators to manage\ndatastore version volume type associations through the Trove management\nAPI.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/997966\nChange-Id: I587b0d4d6eeba8e94621ff0392d876f7d712d50e\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/ee776cabc4f0a9b2791a95f479116df9bf3ac581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/ee776cabc4f0a9b2791a95f479116df9bf3ac581"}]},"branch":"refs/heads/master"},"66ab9ad70ba1d806f5b1a0cc21d63477e5c07793":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-17 09:07:53.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/28/998328/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/28/998328/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/4"}}},"commit":{"parents":[{"commit":"2548c2dc0901aba272bdd1201dc8e852217ec0a2","subject":"Test partial instance access requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/2548c2dc0901aba272bdd1201dc8e852217ec0a2"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-22 15:44:31.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-17 09:07:35.000000000","tz":300},"subject":"Add datastore volume type association support","message":"Add datastore volume type association support\n\nTrove currently supports associating volume types with datastore versions\nthrough trove-manage, which requires direct access to the Trove control\nplane. This is unsuitable for deployments where datastore version\nmaintenance is automated through API workflows or control plane access\nis restricted.\n\nAdd --volume-types and --remove-volume-types options to the openstack\ndatastore version set command. These options allow operators to manage\ndatastore version volume type associations through the Trove management\nAPI.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/997966\nChange-Id: I587b0d4d6eeba8e94621ff0392d876f7d712d50e\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/66ab9ad70ba1d806f5b1a0cc21d63477e5c07793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/66ab9ad70ba1d806f5b1a0cc21d63477e5c07793"}]},"branch":"refs/heads/master"},"e2c48b3dacf4162677d51a758bbfd905784f5de5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-28 10:45:51.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/28/998328/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/28/998328/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/28/998328/5"}}},"commit":{"parents":[{"commit":"3a8ea594cd0b37faeb66769c88e2a22393f16777","subject":"Fix CLI help formatting on Python 3.14","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/3a8ea594cd0b37faeb66769c88e2a22393f16777"}]}],"author":{"name":"Olessya Khussainova","email":"freckled.cellofun@gmail.com","date":"2026-07-22 15:44:31.000000000","tz":300},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-28 10:45:26.000000000","tz":300},"subject":"Add datastore volume type association support","message":"Add datastore volume type association support\n\nTrove currently supports associating volume types with datastore versions\nthrough trove-manage, which requires direct access to the Trove control\nplane. This is unsuitable for deployments where datastore version\nmaintenance is automated through API workflows or control plane access\nis restricted.\n\nAdd --volume-types and --remove-volume-types options to the openstack\ndatastore version set command. These options allow operators to manage\ndatastore version volume type associations through the Trove management\nAPI.\n\nDepends-On: https://review.opendev.org/c/openstack/trove/+/997966\nChange-Id: I587b0d4d6eeba8e94621ff0392d876f7d712d50e\nSigned-off-by: Olessya Khussainova \u003cfreckled.cellofun@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e2c48b3dacf4162677d51a758bbfd905784f5de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e2c48b3dacf4162677d51a758bbfd905784f5de5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
