)]}'
{"id":"openstack%2Fpython-troveclient~998755","triplet_id":"openstack%2Fpython-troveclient~master~I228038323b39ed4ee3766d4da92f9e1bfa1cf7bb","project":"openstack/python-troveclient","branch":"master","topic":"support-special-chars","attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-08-19 15:11:07.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"removed_from_attention_set":{"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-08-19 15:11:07.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"hashtags":[],"change_id":"I228038323b39ed4ee3766d4da92f9e1bfa1cf7bb","subject":"Add new API support to Trove client users and databases","status":"NEW","created":"2026-07-27 07:43:52.000000000","updated":"2026-08-19 15:11:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01543deeed08bc5828a4127abd7dd6d45cd82aba","_number":998755,"virtual_id_number":998755,"owner":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-19 13:59:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"all":[{"value":1,"date":"2026-08-19 15:11:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-27 08:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-18 11:54:11.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"CC"},{"updated":"2026-08-19 15:11:07.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"REVIEWER"}],"messages":[{"id":"6e9b86769725a0aa8bda2693641e59b989c677a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 07:43:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c717594f14493825ae3b43996f056df22f92dd09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 08:00:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/189632aa24fb46f595fe269120b7821d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01db2c62ced646f0b85044ca0af35a39 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61609967569b4d01bfa0c47fbbc24f23 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d75ce772cc92459aa3bd6370fd9541f9 : SUCCESS in 2m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54818ad13d83474282049f8cc086f3ea : FAILURE in 7m 02s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/65fc40d49b73456da675ca9dc66d8807 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70fe8f976d9649489bd08f8377dbc39d : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88aaf9cac2d24f71991f93f09f454c0f : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8490c47720e50307f1c8ee903f08bcff3d38c64d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 11:32:55.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"52eb35c94c3bfcadc83bca679670884a204f99e4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-07-27 11:33:21.000000000","message":"Topic support-special-chars removed","accounts_in_message":[],"_revision_number":2},{"id":"8acf5e34b292be0893b9dedf98bba4bdf32a6029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 11:44:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e214779685fd4677886fe20b83e447c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5f2aff9cb8b47a69924ee1c1ec4c46a : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d7d5863abd4478eaf51be42766e938f : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43da90331abb44739a3c50bed54b6022 : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b87ad0d4b6754ab683a9640d148951aa : FAILURE in 5m 22s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/e87e66f8e5954716813c1cb44f2bd2eb : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1712a9565769424792b1696ecbe1f2c0 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/083282fb2f4b4aac9a98b2216cbe3401 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e8210bb5f708a1c329eb1fd3670e493c8dd80893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-11 14:38:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7625eca685cbaa914a05af3bb2e114f9720f4843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 14:54:35.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a99c8c02f3e64334a62dc0cc51062da3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8c699fbc814b468d1bac34fcdf916c : FAILURE in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b27d4b0236554f6abe73a7a3aa37408a : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9fc3c76331b4a5484174d8b929c5fd6 : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/869dc7d25ee64828b8bc78a2c107740b : FAILURE in 5m 32s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/1a2a05e3ea8e45878b803d03585c2c6a : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090b75aceb734eeba5b5a9ff3a260ec0 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/102e41f2f2114b6ba080ac85672c608c : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"1ef43cbcb7f630768c23117661c9de93febd42d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-11 14:58:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"84544a476d51b13b143cdbddd635f4016f173985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 15:15:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cef0c48d929b4160af33df5256569be2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1510a2d733fd431f92c21beff7d31b7b : SUCCESS in 1m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6c5573fc5a24bbba500b4f8a0770f37 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/191f92b137ec4d6e98b68f7669ca83d7 : SUCCESS in 3m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8dbef38141834404a06b1bfe9561c2d0 : FAILURE in 3m 23s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/04cdfd1a969042bc934fcc3c6bef626e : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5dd40f0c0a486e970afc98b08d344e : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb96dcd6f6bd41c1a70aeb3fffc45bd3 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":4},{"id":"07a79891576dbdea16b6bdbb591251dd45caf285","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-18 11:54:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0aaddd893d7347ef05d6fcf45e1bbbc872397481","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-19 13:46:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"05bdfc702fcbe75810b426ea08bcdcad34f4bea5","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-08-19 13:49:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5614132400b59a0f472ddb8c8fbcd59f144f7233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 13:59:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0912e164e2435aa165547b929551e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ea3ca92dcc4480d818948eaa6a1a40e : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ef0d508bde5458c92dab4133f8115cf : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/878ffe6344484ebfa6918adf9c744063 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b02a01d472840c7a516dddd73366e46 : FAILURE in 5m 19s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/27b4a109644a4428b1b9774d35f00235 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9b5487376048cfba0508a9f4b8e2e8 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b97c632e54147c6959995e3df483682 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":5},{"id":"01543deeed08bc5828a4127abd7dd6d45cd82aba","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-08-19 15:11:07.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b17c653ebeafb6f6812688d5c0e95c7b12740a20","revisions":{"e703d88148ba3ac6f8ca30c38b3c77b3b1a26adf":{"kind":"REWORK","_number":1,"created":"2026-07-27 07:43:52.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/55/998755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/55/998755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/1"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:42:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:42:07.000000000","tz":300},"subject":"Support special chars in db names","message":"Support special chars in db names\n\nAdd escaping for special chars and double escaping for\n\"/\" and \".\"\n\nChange-Id: I228038323b39ed4ee3766d4da92f9e1bfa1cf7bb\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e703d88148ba3ac6f8ca30c38b3c77b3b1a26adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e703d88148ba3ac6f8ca30c38b3c77b3b1a26adf"}]},"branch":"refs/heads/master"},"720f802f38dc31750ec41c859ee9adc31964fa5d":{"kind":"NO_CHANGE","_number":2,"created":"2026-07-27 11:32:55.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/55/998755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/55/998755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/2"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:42:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 11:32:35.000000000","tz":300},"subject":"Support special chars in db names","message":"Support special chars in db names\n\nAdd escaping for special chars and double escaping for\n\"/\" and \".\"\n\nChange-Id: I228038323b39ed4ee3766d4da92f9e1bfa1cf7bb\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/720f802f38dc31750ec41c859ee9adc31964fa5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/720f802f38dc31750ec41c859ee9adc31964fa5d"}]},"branch":"refs/heads/master"},"9d40ccd1077e5336d4c2d616e01abcb2de929066":{"kind":"REWORK","_number":3,"created":"2026-08-11 14:38:27.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/55/998755/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/55/998755/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/3"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:42:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-11 14:34:56.000000000","tz":300},"subject":"Add new API support to Trove client users and databases","message":"Add new API support to Trove client users and databases\n\nAdd support for the new action-based API while preserving\nbackward compatibility with the legacy endpoints. Extend\nunit tests to cover the new API and fallback behavior.\n\nChange-Id: I228038323b39ed4ee3766d4da92f9e1bfa1cf7bb\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/9d40ccd1077e5336d4c2d616e01abcb2de929066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/9d40ccd1077e5336d4c2d616e01abcb2de929066"}]},"branch":"refs/heads/master"},"bb5c65f2364e89a930c37cb5066e866be2b90ac5":{"kind":"REWORK","_number":4,"created":"2026-08-11 14:58:25.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/55/998755/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/55/998755/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/4"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:42:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-11 14:58:10.000000000","tz":300},"subject":"Add new API support to Trove client users and databases","message":"Add new API support to Trove client users and databases\n\nAdd support for the new action-based API while preserving\nbackward compatibility with the legacy endpoints. Extend\nunit tests to cover the new API and fallback behavior.\n\nChange-Id: I228038323b39ed4ee3766d4da92f9e1bfa1cf7bb\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/bb5c65f2364e89a930c37cb5066e866be2b90ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/bb5c65f2364e89a930c37cb5066e866be2b90ac5"}]},"branch":"refs/heads/master"},"b17c653ebeafb6f6812688d5c0e95c7b12740a20":{"kind":"REWORK","_number":5,"created":"2026-08-19 13:46:18.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/55/998755/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-troveclient","ref":"refs/changes/55/998755/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-troveclient refs/changes/55/998755/5"}}},"commit":{"parents":[{"commit":"e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2","subject":"Merge \"SSL management feature support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/e186e4d1176d9235a0f4c6e1bcd0e603ba3474e2"}]}],"author":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-07-27 07:42:07.000000000","tz":300},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-08-19 13:45:52.000000000","tz":300},"subject":"Add new API support to Trove client users and databases","message":"Add new API support to Trove client users and databases\n\nAdd support for the new action-based API while preserving\nbackward compatibility with the legacy endpoints. Extend\nunit tests to cover the new API and fallback behavior.\n\nChange-Id: I228038323b39ed4ee3766d4da92f9e1bfa1cf7bb\nSigned-off-by: Erkin Mussurmankulov \u003cmangust404@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/b17c653ebeafb6f6812688d5c0e95c7b12740a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-troveclient/commit/b17c653ebeafb6f6812688d5c0e95c7b12740a20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
