)]}'
{"id":"openstack%2Fpython-zaqarclient~1005480","triplet_id":"openstack%2Fpython-zaqarclient~master~I65e1279eaedeebf6312d39716a7d34f51206a1fb","project":"openstack/python-zaqarclient","branch":"master","hashtags":[],"change_id":"I65e1279eaedeebf6312d39716a7d34f51206a1fb","subject":"Resolve warnings from setuptools","status":"MERGED","created":"2026-09-14 03:48:58.000000000","updated":"2026-09-14 05:09:06.000000000","submitted":"2026-09-14 05:08:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005480","meta_rev_id":"34a15059346698ca0dd89d6ac53070c40d331ebb","_number":1005480,"virtual_id_number":1005480,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 05:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-14 04:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-09-14 04:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-14 04:03:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b2f002e00b4df36a47b33fd676e458e636eebee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 03:48:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5beb64ab95d5f1a9c455d298da2da56c77bf0662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 04:03:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87991b71ee8240ddb0c81d17dbcb73cd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/acd5b3245a9e43468b480276a97ebad9 : SUCCESS in 1m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc05f6bfc9d2470f9e687860ff7c172e : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e577d2bf9e5745ac81809b93bb6852f0 : SUCCESS in 2m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/123f90d9ea9a4d9e9a441b76eab49a08 : SUCCESS in 2m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1afaf144a3b1407fbb092556abe1ebef : SUCCESS in 2m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a9034800ce264b3e9f2c22442c144b7e : SUCCESS in 3m 34s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d18f55865653426a92954bc79eb2e98f : SUCCESS in 3m 38s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/9d8f0dce67c146d8838660934c24caa5 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e86c51c8325a4cea964162a6f6ff90bc : SUCCESS in 2m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afa96ee75f6b42719250ea15a34bae58 : SUCCESS in 3m 40s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/5017023806724085a5864b6e656ed3aa : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1409a592dfc1abeb59ab993ccd245a62c5257703","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-14 04:52:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d59910a410adbe755ba713e44984964868e3eb93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 04:53:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66df354b134af9abebfac7bffa0ce24a78a5ee5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 05:08:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87b6117665a74d33b87c56b2c45499ce\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/acd681d52b0f4d31a739bea715a996c1 : SUCCESS in 3m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/71604444285f427195b8508ee4138bd2 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb953df59214781b711293a4dd70848 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13cf6931e56847c38aab7fc4e91ab265 : SUCCESS in 2m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26cc2c84ff4040a3bd2996cab282c3aa : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df0c93fdfed44389afa74216310a6322 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09e29b9f27545b798da63e843e870a4 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdb2d89ac1f941568eba81e11545d17a : SUCCESS in 3m 01s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/6d4936cda6a246f4a8931f884d91d842 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fb3b7320d1e5732573a24bf964e9a57eaa7253f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 05:08:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"34a15059346698ca0dd89d6ac53070c40d331ebb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 05:09:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/720ddce0ce3f494ebdb5f01f11829e51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1170f30231d24ec0927ac9eaa438cd05 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b49578eb1f9bd0854eaab2b89b173f6d393b233","revisions":{"9b49578eb1f9bd0854eaab2b89b173f6d393b233":{"kind":"REWORK","_number":1,"created":"2026-09-14 03:48:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/1005480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/80/1005480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/80/1005480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/80/1005480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/80/1005480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/80/1005480/1"}}},"commit":{"parents":[{"commit":"b12ea0cd1a29676641f6bf8707314df8c9439df0","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/b12ea0cd1a29676641f6bf8707314df8c9439df0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 03:48:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-14 03:48:34.000000000","tz":540},"subject":"Resolve warnings from setuptools","message":"Resolve warnings from setuptools\n\nChange-Id: I65e1279eaedeebf6312d39716a7d34f51206a1fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/9b49578eb1f9bd0854eaab2b89b173f6d393b233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/9b49578eb1f9bd0854eaab2b89b173f6d393b233"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
