)]}'
{"id":"openstack%2Fpython-zaqarclient~1007634","triplet_id":"openstack%2Fpython-zaqarclient~master~I586378071283989a77b68b07f351701d750d5935","project":"openstack/python-zaqarclient","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-27 15:20:25.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I586378071283989a77b68b07f351701d750d5935","subject":"Enable more functional tests","status":"NEW","created":"2026-09-27 14:42:46.000000000","updated":"2026-09-28 05:10:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"334e109a8a745c81171abd1b8a160e3e235d746e","_number":1007634,"virtual_id_number":1007634,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-28 05:10:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-27 15:20:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78e43c175d8ef68a6852277e943b1c4fad4493a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-27 14:42:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83bec12ebc7ffcb7b601f0238caadb1f1f389758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-27 15:20:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df785f967c834071a37aee0196224c7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e39c438f0734a4abb9f68653d48a314 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4644887420b948caad7bd084fb9ac66d : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c96a5290a7ab401a8072e63aef8957e0 : SUCCESS in 4m 32s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/166b9efcc99b4a05b2be8d85fbfbdc8d : SUCCESS in 4m 12s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/6055c5286deb47dcac7db0c1b3b674f4 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293f7436a218487db7a10d528d21c248 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e678e6340cd44971a1aa58f409231fd2 : SUCCESS in 2m 28s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/2d2167e9ea4847d38d43adca3114e638 : FAILURE in 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ea3a2164aec8f73063c1549e45cd3f85da475155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-27 15:31:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3c4c6b94193e7a3af8a1b3504f264bac03c3d758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-27 16:07:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c358b4078ed84c24ac3cdfb4fd1c3ca8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e65ba35089440d18ad4c4b28dd32777 : FAILURE in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e58a828107a44fac9b8c4b652fd99661 : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2102ed0cdbba4093b8fe55ad96c7aa27 : SUCCESS in 4m 33s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/63526db3abf04990b0fe2d3763f53ada : SUCCESS in 6m 04s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/b66a7d7217d8498790b6bd98b08556c3 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f6c02eab1246ce8d601699aea37d39 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f39252ece6147edb9949af176e3968f : SUCCESS in 5m 05s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/34039c3344d14118bef5e44c1e0ac052 : FAILURE in 27m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9c6bda4a39dd0cf2ff1bdd01b15f3bef472ab8b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-27 16:39:12.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b7878538c631fc03c8ef70cbb527f84771307cf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-27 16:39:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1cf67120cfef974a870ff2b9d079e9d7a6afbd60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-27 16:54:10.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ee51433f2f917adc0fa6cb84092c4f101e443bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-27 17:13:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d970301476e4f828b3c6f06813f52c7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c39f161249224bef9de5ca312a4b00c8 : SUCCESS in 1m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd22f3735fb14f0384f88be01d0f1957 : SUCCESS in 1m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32bc16292f8049668e5ba1b474a0ddee : SUCCESS in 5m 26s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4ae274ce05c94ed2a2d1a3b3a30ec4ce : SUCCESS in 5m 30s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/060aaf5cc9ec4b62a0d8e39dae88ad8a : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5beb785b1e5e4fdaaf846483bb4bd970 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a884ec9445d43898cfcae327dd4441c : SUCCESS in 3m 43s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/1ae5857b1ca44d19bd865acd34a68324 : FAILURE in 16m 15s","accounts_in_message":[],"_revision_number":5},{"id":"dfade8509d4afca5cb1c1ed194c27c9d4efc82cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 01:33:53.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"6d44bd4833c71a5daafc09391316972eff27f559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 01:50:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce741ab6161f4770a09d25995f0df8b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4cd021d71a4ba4a5f4ac1170004983 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/891795ea22534e7697665f1c0155beb1 : SUCCESS in 2m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4af4e512e6d24575ac26a0996e32a820 : SUCCESS in 6m 28s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/62e701c8279242fbb80886fe09fee8b8 : SUCCESS in 5m 57s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/85d60b0738614723b250c43e481397cb : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c28896a5704880b89976a9031fdebb : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6455b63ffda84cfbb9e392d5f9a3fe12 : SUCCESS in 7m 18s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/6d885ae2647e44afbe7b4c260cbe7219 : FAILURE in 14m 14s","accounts_in_message":[],"_revision_number":6},{"id":"23d40d2161ba99cd030fb11de393ffb4d07ed790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 02:43:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"7317a822e72a323f725a65392a00e7fba80f4661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 03:47:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bc6c35e3d54409f97a04e5779c8e13f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6764f9c5f2ba4cbf9f452d60491c1e4f : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1632135841574f53838ecddce04c1b8c : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af98ed2733744813b29f4539588c5074 : SUCCESS in 6m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/30d81159369c4d32aa86068873194af7 : SUCCESS in 4m 33s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/23d81d1a517241eca79ac9518e40d82d : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab34041cbcb541e4a4e4dbeaad83ff5f : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ece8f8fe51ed4fc080016456dd8c86df : SUCCESS in 4m 09s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/9ba2615251b048edb36983d719092080 : FAILURE in 27m 52s","accounts_in_message":[],"_revision_number":7},{"id":"d3215bd8323df586e8f9f3a90de1077bf1796ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 04:51:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"334e109a8a745c81171abd1b8a160e3e235d746e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 05:10:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33bd5e1ea69b4624b82d6e28f943bf15\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4938760246c482bb504eb480a886d78 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3247c6961eb54935ad0be96403f27251 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c3237f2ab7124b50b1f737a74d28d5fa : SUCCESS in 5m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9b770385f68a4902b8e260e5c8659463 : SUCCESS in 4m 07s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/376c7698ef034d48aced260a2ccc8ddf : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf4d7697da646839cafda5aa492b8e5 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/899f003669d74b428e3f005a3800144f : SUCCESS in 3m 58s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/a109c674f45741c0afbd36ec86c85d6e : FAILURE in 14m 21s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f92146844a17e76f87d480cccc3a47fffae562cf","revisions":{"ad4e15f447ebdfd802c8c9332f9efabae2f67796":{"kind":"REWORK","_number":1,"created":"2026-09-27 14:42:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/1"}}},"commit":{"parents":[{"commit":"58e52118f832c6980b7a4e97ed7e8292dc1ddfe9","subject":"Merge \"Resolve warnings from setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/58e52118f832c6980b7a4e97ed7e8292dc1ddfe9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/ad4e15f447ebdfd802c8c9332f9efabae2f67796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/ad4e15f447ebdfd802c8c9332f9efabae2f67796"}]},"branch":"refs/heads/master"},"e24eb43d6df8422603b5358ca309e8a99ef91fc3":{"kind":"REWORK","_number":2,"created":"2026-09-27 15:31:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/2"}}},"commit":{"parents":[{"commit":"58e52118f832c6980b7a4e97ed7e8292dc1ddfe9","subject":"Merge \"Resolve warnings from setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/58e52118f832c6980b7a4e97ed7e8292dc1ddfe9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 15:30:51.000000000","tz":540},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/e24eb43d6df8422603b5358ca309e8a99ef91fc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/e24eb43d6df8422603b5358ca309e8a99ef91fc3"}]},"branch":"refs/heads/master"},"f325595e2ee20f81f0a6210053128e764126a175":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-27 16:39:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/3"}}},"commit":{"parents":[{"commit":"e029940364b535f4306f87ad1348d7530e730fa4","subject":"Fix queue.exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/e029940364b535f4306f87ad1348d7530e730fa4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 16:39:01.000000000","tz":540},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/f325595e2ee20f81f0a6210053128e764126a175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/f325595e2ee20f81f0a6210053128e764126a175"}]},"branch":"refs/heads/master"},"87127c3f1df91d97ab7f3afa15374c2c9ed65831":{"kind":"NO_CHANGE","_number":4,"created":"2026-09-27 16:39:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/4"}}},"commit":{"parents":[{"commit":"93b958ae630b284c4ee5eb75e98dbb2509616611","subject":"Fix queue.exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/93b958ae630b284c4ee5eb75e98dbb2509616611"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 16:39:37.000000000","tz":0},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/87127c3f1df91d97ab7f3afa15374c2c9ed65831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/87127c3f1df91d97ab7f3afa15374c2c9ed65831"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e029940364b535f4306f87ad1348d7530e730fa4","ours":"f325595e2ee20f81f0a6210053128e764126a175","theirs":"93b958ae630b284c4ee5eb75e98dbb2509616611","merge_strategy":"recursive","contains_conflicts":false}},"acb9e2b4f303f395fda0d0d0cd7edf17e18d1e8f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-27 16:54:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/5"}}},"commit":{"parents":[{"commit":"f13568279d5a1104eda2ec271b6b3694a1233e4a","subject":"Fix queue.exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/f13568279d5a1104eda2ec271b6b3694a1233e4a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 16:54:02.000000000","tz":540},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/acb9e2b4f303f395fda0d0d0cd7edf17e18d1e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/acb9e2b4f303f395fda0d0d0cd7edf17e18d1e8f"}]},"branch":"refs/heads/master"},"f54cc88386007b0a62cd2855f4bcc96b255a17dd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-09-28 01:33:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/6"}}},"commit":{"parents":[{"commit":"dee89766f06ee539430e9f003d245983e336abc2","subject":"Fix queue.exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/dee89766f06ee539430e9f003d245983e336abc2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-28 01:33:53.000000000","tz":0},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/f54cc88386007b0a62cd2855f4bcc96b255a17dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/f54cc88386007b0a62cd2855f4bcc96b255a17dd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f13568279d5a1104eda2ec271b6b3694a1233e4a","ours":"acb9e2b4f303f395fda0d0d0cd7edf17e18d1e8f","theirs":"dee89766f06ee539430e9f003d245983e336abc2","merge_strategy":"recursive","contains_conflicts":false}},"e33827838a6eca75508eb17b58b3505b3cc36046":{"kind":"REWORK","_number":7,"created":"2026-09-28 02:43:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/7"}}},"commit":{"parents":[{"commit":"d5f242fd01fb19267c1e5446251a565be45e0a06","subject":"Fix queue.exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/d5f242fd01fb19267c1e5446251a565be45e0a06"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-28 02:43:21.000000000","tz":540},"subject":"Enable more functional tests","message":"Enable more functional tests\n\nNote that some tests are still disabled due to problems in CI.\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/e33827838a6eca75508eb17b58b3505b3cc36046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/e33827838a6eca75508eb17b58b3505b3cc36046"}]},"branch":"refs/heads/master"},"f92146844a17e76f87d480cccc3a47fffae562cf":{"kind":"REWORK","_number":8,"created":"2026-09-28 04:51:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/1007634/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/34/1007634/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/34/1007634/8"}}},"commit":{"parents":[{"commit":"d5f242fd01fb19267c1e5446251a565be45e0a06","subject":"Fix queue.exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/d5f242fd01fb19267c1e5446251a565be45e0a06"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-27 14:42:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-28 04:51:32.000000000","tz":540},"subject":"Enable more functional tests","message":"Enable more functional tests\n\nNote that some tests are still disabled due to problems in CI.\n\nChange-Id: I586378071283989a77b68b07f351701d750d5935\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/f92146844a17e76f87d480cccc3a47fffae562cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/f92146844a17e76f87d480cccc3a47fffae562cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
