)]}'
{"id":"openstack%2Fpython-zaqarclient~964887","triplet_id":"openstack%2Fpython-zaqarclient~master~Ia99193d3a385f760bc8d97aa603929b384291e1c","project":"openstack/python-zaqarclient","branch":"master","topic":"pyproject","hashtags":[],"change_id":"Ia99193d3a385f760bc8d97aa603929b384291e1c","subject":"Migrate setup configuration to pyproject.toml","status":"NEW","created":"2025-10-27 15:17:03.000000000","updated":"2026-07-19 16:12:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6bcad6a6090cb73b0e68e83bf5fb60e1ea9fa64e","_number":964887,"virtual_id_number":964887,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-19 16:12:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 15:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b4ac927c4b299181fefe150533c7412b377978f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-27 15:17:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3870a1b6dd394b3d733a336afcad233e629587cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:46:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad1eb12f5a8640ab9d0f6c0067a902c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dfe261d86119448e91e241ee7c1a0e5d : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6831d27b7140e5bf7840d768ae9346 : SUCCESS in 1m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f42c1458b8b144d2a765993c0d976bae : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5aa5cec0a2ed4a33b1fb697e42a4d8b3 : SUCCESS in 5m 25s\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4302eee3906d4cc2b7d04e6ade8a891e : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d14a131aa1946679c3466bf79ebde99 : SUCCESS in 3m 29s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/0316a92a5a3342c6ad791eadd60efb29 : SUCCESS in 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9e24c25872650cb5bff518e82a75c87664043e7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-18 06:52:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c9d41c4835e743b9a50c27545004dc8a754c65c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 07:05:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0269d1a496741d99655254757489a05\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e8d1f11a754f476086cef60090ca69ec : SUCCESS in 2m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/07aa1fd24f414e07ac8c6288039023c4 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/491892a2839b4ae68d437941aff044a9 : SUCCESS in 1m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad8a8fe8585a4cf1aea7b7073d4984e4 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0e2e9379d60407fa0f55d4ec02c216f : SUCCESS in 5m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8021336653904b729076b98bb013d7de : SUCCESS in 3m 15s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/7919f96a313f49d8a934fb8dcbe610cd : SUCCESS in 4m 54s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/4b725f92e65a4106b9ab1a102729ec1c : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935d4c665ca54c65b56538ee0a3c9e72 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f3872b4c4664ecdb6c171e4a095e061 : SUCCESS in 4m 09s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/06069474725e45cc8e1bfd020eba285c : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c2b317efb439e8a32e8ef6fd3705e57c5be06ac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 15:38:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"6bcad6a6090cb73b0e68e83bf5fb60e1ea9fa64e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 16:12:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0413b90a7a5f4035abee05eec68f61d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a0f61886a2524ecaab27fc8f65c6adef : SUCCESS in 3m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1637b16fb9bc429d99e55b695db3c03a : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25e5f9ad7d84d089ce5a271061c6cd4 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39a40f27459a4c01a162c41f03bb72ca : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c39656dfa5484319b50abe2f46398198 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a56717e12e63495e91592bd472a333f5 : SUCCESS in 4m 25s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/af2b9d6151154e5da8d0fea8b791ea1b : SUCCESS in 6m 04s (non-voting)\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/17ad2e4f75254aa99768b9905a7eb6c9 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1276255a0aaa4d57a051644f410657db : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b19d17aeda148d4b984590157272f22 : SUCCESS in 3m 01s\n- python-zaqarclient-functional https://zuul.opendev.org/t/openstack/build/097fd4b86fe74e8ca219dbc73614cd07 : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b12ea0cd1a29676641f6bf8707314df8c9439df0","revisions":{"fce1f826ff9086335e044a6b58ab911e02946d4d":{"kind":"REWORK","_number":1,"created":"2025-10-27 15:17:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/964887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/87/964887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/1"}}},"commit":{"parents":[{"commit":"43c1571fe7a51cc0c94df30255c480d1a9da1a60","subject":"Use pre-commit for pep8 checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/43c1571fe7a51cc0c94df30255c480d1a9da1a60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-27 15:15:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-27 15:16:45.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Ia99193d3a385f760bc8d97aa603929b384291e1c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/fce1f826ff9086335e044a6b58ab911e02946d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/fce1f826ff9086335e044a6b58ab911e02946d4d"}]},"branch":"refs/heads/master"},"23ee2f30a0ba1b70c1353f41d91b4e11f013b375":{"kind":"REWORK","_number":2,"created":"2026-07-18 06:52:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/964887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/87/964887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/2"}}},"commit":{"parents":[{"commit":"2065a0407b673b2a3f4a37957b1d48663f00caeb","subject":"Add openstack-python3-next-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/2065a0407b673b2a3f4a37957b1d48663f00caeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-27 15:15:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-18 06:51:35.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Ia99193d3a385f760bc8d97aa603929b384291e1c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/23ee2f30a0ba1b70c1353f41d91b4e11f013b375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/23ee2f30a0ba1b70c1353f41d91b4e11f013b375"}]},"branch":"refs/heads/master"},"b12ea0cd1a29676641f6bf8707314df8c9439df0":{"kind":"REWORK","_number":3,"created":"2026-07-19 15:38:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/964887/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zaqarclient","ref":"refs/changes/87/964887/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zaqarclient refs/changes/87/964887/3"}}},"commit":{"parents":[{"commit":"6321adaa8ec49ac1a0467d24ed54cedd507f1afe","subject":"Fix unmanaged optional dependency on websocket-client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/6321adaa8ec49ac1a0467d24ed54cedd507f1afe"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-27 15:15:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 15:38:40.000000000","tz":540},"subject":"Migrate setup configuration to pyproject.toml","message":"Migrate setup configuration to pyproject.toml\n\nChange-Id: Ia99193d3a385f760bc8d97aa603929b384291e1c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/b12ea0cd1a29676641f6bf8707314df8c9439df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zaqarclient/commit/b12ea0cd1a29676641f6bf8707314df8c9439df0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
