)]}'
{"id":"openstack%2Fpython-zunclient~1002897","triplet_id":"openstack%2Fpython-zunclient~master~I18915da23b2a24b9867e37950c70037bbe34bae0","project":"openstack/python-zunclient","branch":"master","hashtags":[],"change_id":"I18915da23b2a24b9867e37950c70037bbe34bae0","subject":"Drop support for Python 3.8, 3.9 and 3.10","status":"MERGED","created":"2026-08-29 22:38:36.000000000","updated":"2026-08-29 23:06:31.000000000","submitted":"2026-08-29 23:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002897","meta_rev_id":"a873aff1301d5343a7ad8cc9b16abde1f2c5d8f5","_number":1002897,"virtual_id_number":1002897,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-29 23:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-29 22:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-29 22:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-29 22:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a8f8a8f5b29b7bc60b08ce0a08ed2feab294236","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:38:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed12c16d45c85d0ad46e4509ab14477a3116f547","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:39:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1f643f96bb8c8300f6101b837a415cf1625258c7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:39:58.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"dadd927607834ae57f11a81bf4afaf5d3a819acc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 22:50:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/333ed446e5754aaa8803a2e7a18a026c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/535af54bc20444238715dd9be55c6a3b : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c753f33ff364225aa73334739a85f47 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373d809f50274934b7a8b5ae92581097 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36003b2213a0426191ccd027b21789ef : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5746e61216134c0ebc5148ecb5140987 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c64effa417ef4ac48430cd7013f40cb4 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312950ae1e3f479abd2eaf1a4f26c0e8 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"4afd18f5780d6a2a6299aa584fc75a95b9f0bb80","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:53:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"615a59328c93ff8a5d7ad765cf3b9ca9c8ee2341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 22:53:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c3992d7dcb6ad16abec86b97264f7dd59d054910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 23:05:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9963d69c0fac4aa8a4caab16e602c1de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9c5134cef3144fdaa91d8b57acdfbfbc : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c2bd0577fe4a8b972c9db64182cb34 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e137483adaa4221b8994b8ef2d54555 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ed2733ad9ad4b7f9ec9114acfecb2e8 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31a375f1a794878a2a6a827965a4eff : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ea931d852dc51dfa66145aa6179774918cc09b4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 23:05:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a873aff1301d5343a7ad8cc9b16abde1f2c5d8f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 23:06:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e4bea5402294c34bd95ff74d3b035b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3272086961204092ab75768a4669a889 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2c1c3d05f857aa0ea87553cf87c4d7955a63fd5","revisions":{"22d1aec4e617e3539183e87b2088ed0b089f50de":{"kind":"REWORK","_number":1,"created":"2026-08-29 22:38:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/97/1002897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/97/1002897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/1"}}},"commit":{"parents":[{"commit":"cf86d6aec765c3939ae2367888d463d198a701f1","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/cf86d6aec765c3939ae2367888d463d198a701f1"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:38:36.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:38:36.000000000","tz":0},"subject":"Drop support for Python 3.8, 3.9 and 3.10","message":"Drop support for Python 3.8, 3.9 and 3.10\n\nAs per latest testing runtime, we stopped testing the\npython 3.8, 3.9 and 3.10 so dropping their support.\n\nChange-Id: I18915da23b2a24b9867e37950c70037bbe34bae0\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/22d1aec4e617e3539183e87b2088ed0b089f50de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/22d1aec4e617e3539183e87b2088ed0b089f50de"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"b2c1c3d05f857aa0ea87553cf87c4d7955a63fd5":{"kind":"REWORK","_number":2,"created":"2026-08-29 22:39:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/97/1002897/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/97/1002897/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/97/1002897/2"}}},"commit":{"parents":[{"commit":"cf86d6aec765c3939ae2367888d463d198a701f1","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/cf86d6aec765c3939ae2367888d463d198a701f1"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:38:36.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:39:52.000000000","tz":0},"subject":"Drop support for Python 3.8, 3.9 and 3.10","message":"Drop support for Python 3.8, 3.9 and 3.10\n\nAs per latest testing runtime, we stopped testing the\npython 3.8, 3.9 and 3.10 so dropping their support.\n\nChange-Id: I18915da23b2a24b9867e37950c70037bbe34bae0\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/b2c1c3d05f857aa0ea87553cf87c4d7955a63fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/b2c1c3d05f857aa0ea87553cf87c4d7955a63fd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
