)]}'
{"id":"openstack%2Fpython-zunclient~633647","triplet_id":"openstack%2Fpython-zunclient~master~I079803111b3956865dfa75be7b1f95edd5623306","project":"openstack/python-zunclient","branch":"master","attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-30 20:13:00.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-08-30 20:13:00.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I079803111b3956865dfa75be7b1f95edd5623306","subject":"Optimization for attaching a running container","status":"NEW","created":"2019-01-29 02:44:43.000000000","updated":"2026-08-30 20:13:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"022b6f444fed1a3050174c08328e3e3c5581b230","_number":633647,"virtual_id_number":633647,"owner":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-30 19:32:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-29 07:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-30 20:13:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-08-30 20:13:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-08-30 20:13:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"}],"messages":[{"id":"ee248698645b8bf2733267d4bd1f7236ac72d94a","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-01-29 02:44:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"397c82b4af04783ab769213d3c30e0deceaf43b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 03:24:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/47/633647/1/check/openstack-tox-cover/6f7372a/cover/ : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/633647/1/check/openstack-tox-lower-constraints/5d03ccf/ : SUCCESS in 7m 05s\n- openstack-tox-pep8 http://logs.openstack.org/47/633647/1/check/openstack-tox-pep8/d5a63e6/ : FAILURE in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/47/633647/1/check/openstack-tox-py27/bf38186/ : SUCCESS in 6m 12s\n- openstack-tox-py35 http://logs.openstack.org/47/633647/1/check/openstack-tox-py35/ace16f3/ : SUCCESS in 6m 57s\n- openstack-tox-py36 http://logs.openstack.org/47/633647/1/check/openstack-tox-py36/c0688a6/ : SUCCESS in 6m 11s\n- openstack-tox-docs http://logs.openstack.org/47/633647/1/check/openstack-tox-docs/37e7008/html/ : SUCCESS in 5m 16s\n- zunclient-devstack-docker-sql http://logs.openstack.org/47/633647/1/check/zunclient-devstack-docker-sql/7778369/ : FAILURE in 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a5d07704662aa679c9afc35a6ac55d62880fa731","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-01-29 03:40:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ee8cc1eebc0d1ee434d1a5b1f3f6d1dd151b90ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 04:16:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/47/633647/2/check/openstack-tox-cover/0992ed2/cover/ : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/633647/2/check/openstack-tox-lower-constraints/4d6e4e0/ : SUCCESS in 6m 42s\n- openstack-tox-pep8 http://logs.openstack.org/47/633647/2/check/openstack-tox-pep8/091e81b/ : FAILURE in 7m 41s\n- openstack-tox-py27 http://logs.openstack.org/47/633647/2/check/openstack-tox-py27/64fdb9c/ : SUCCESS in 5m 49s\n- openstack-tox-py35 http://logs.openstack.org/47/633647/2/check/openstack-tox-py35/7a6c2c0/ : SUCCESS in 7m 51s\n- openstack-tox-py36 http://logs.openstack.org/47/633647/2/check/openstack-tox-py36/372e13c/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/47/633647/2/check/openstack-tox-docs/7a9859e/html/ : SUCCESS in 7m 03s\n- zunclient-devstack-docker-sql http://logs.openstack.org/47/633647/2/check/zunclient-devstack-docker-sql/bdf253d/ : FAILURE in 30m 32s","accounts_in_message":[],"_revision_number":2},{"id":"944aabfd30cc2b6ff3ec295ecfa734bffe1ef2f4","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-01-29 06:08:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d875af70c9280e0aaf7f1cd1fe6c7dbd0537b207","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 07:13:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/47/633647/3/check/openstack-tox-cover/6b88092/cover/ : SUCCESS in 6m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/633647/3/check/openstack-tox-lower-constraints/0bfa601/ : SUCCESS in 8m 02s\n- openstack-tox-pep8 http://logs.openstack.org/47/633647/3/check/openstack-tox-pep8/74dfa83/ : FAILURE in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/47/633647/3/check/openstack-tox-py27/0a9a0b2/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/47/633647/3/check/openstack-tox-py35/3cf463d/ : SUCCESS in 5m 41s\n- openstack-tox-py36 http://logs.openstack.org/47/633647/3/check/openstack-tox-py36/e1b18d0/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/47/633647/3/check/openstack-tox-docs/afeca12/html/ : SUCCESS in 5m 59s\n- zunclient-devstack-docker-sql http://logs.openstack.org/47/633647/3/check/zunclient-devstack-docker-sql/ab6357a/ : FAILURE in 31m 55s","accounts_in_message":[],"_revision_number":3},{"id":"16e82369789185c7fbc8278d82a3de4b0f010f49","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-01-29 07:26:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c1a9a2d78b555a9ab3d1c66933a82e88da9b4a07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 07:59:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/47/633647/4/check/openstack-tox-cover/48ccf27/cover/ : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/633647/4/check/openstack-tox-lower-constraints/8525e9d/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/47/633647/4/check/openstack-tox-pep8/f88bc69/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/47/633647/4/check/openstack-tox-py27/a732b1d/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/47/633647/4/check/openstack-tox-py35/aa650f8/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/47/633647/4/check/openstack-tox-py36/121744f/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/47/633647/4/check/openstack-tox-docs/479de73/html/ : SUCCESS in 4m 21s\n- zunclient-devstack-docker-sql http://logs.openstack.org/47/633647/4/check/zunclient-devstack-docker-sql/67ea350/ : FAILURE in 30m 52s","accounts_in_message":[],"_revision_number":4},{"id":"d721586ffd0ea9fd0b96f601a9342872e39f10c8","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-01-30 01:41:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b30b1801541b5577a6ab8d2070edaaf2256b58c7","author":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"date":"2019-01-31 07:40:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6d37f5ba79470ed616d3b9f9d49bb8cec2b64a0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 08:23:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/47/633647/4/check/openstack-tox-cover/d0f6e6c/cover/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/633647/4/check/openstack-tox-lower-constraints/ca1eb16/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/47/633647/4/check/openstack-tox-pep8/a0f3cb8/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/47/633647/4/check/openstack-tox-py27/fceb012/ : SUCCESS in 4m 48s\n- openstack-tox-py35 http://logs.openstack.org/47/633647/4/check/openstack-tox-py35/39d1e22/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/47/633647/4/check/openstack-tox-py36/e8b0009/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/47/633647/4/check/openstack-tox-docs/359fda3/html/ : SUCCESS in 5m 44s\n- zunclient-devstack-docker-sql http://logs.openstack.org/47/633647/4/check/zunclient-devstack-docker-sql/dafa5ef/ : FAILURE in 35m 01s","accounts_in_message":[],"_revision_number":4},{"id":"6daab9c7627c2c7c6d610fe1953090dbc5493b93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 19:22:56.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b4b6c07215e645b562db352a29b5a9c77ca3b4f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-30 19:23:00.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"dbb3e9f8f244a8771fd759598fabc8ddb4c24db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-30 19:32:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/814497c74e864c1ebe269f2114408504\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11e3e671d9ee42ba9b049bf8e415f2f3 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4429fc35887c489390fcb883b6d02ba6 : SUCCESS in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75368f4d05554c73b82ca2239bb1ddbf : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3a2ebc4f2704694a3c5bda639d4d21f : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/138c611ee857467e9ba0b272dc3d7da6 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc17c74ef5174c08a0a0c36405d82c36 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"11d9020942c73af35aaf1c593fde8f1a82dfd5d9","revisions":{"e0949c0851e62ea9527ae4aa01389a9a45d679bb":{"kind":"REWORK","_number":1,"created":"2019-01-29 02:44:43.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/47/633647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/47/633647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/1"}}},"commit":{"parents":[{"commit":"ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff","subject":"Merge \"Wait for the container to rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 02:49:43.000000000","tz":480},"committer":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 02:49:43.000000000","tz":480},"subject":"Optimization for attaching a running container","message":"Optimization for attaching a running container\n\nChange-Id: I079803111b3956865dfa75be7b1f95edd5623306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/e0949c0851e62ea9527ae4aa01389a9a45d679bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/e0949c0851e62ea9527ae4aa01389a9a45d679bb"}]},"branch":"refs/heads/master"},"6250375ee3b2c76a70a63cfff2252d70df0805da":{"kind":"REWORK","_number":2,"created":"2019-01-29 03:40:21.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/47/633647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/47/633647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/2"}}},"commit":{"parents":[{"commit":"ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff","subject":"Merge \"Wait for the container to rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 02:49:43.000000000","tz":480},"committer":{"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 03:40:09.000000000","tz":0},"subject":"Optimization for attaching a running container","message":"Optimization for attaching a running container\n\nChange-Id: I079803111b3956865dfa75be7b1f95edd5623306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/6250375ee3b2c76a70a63cfff2252d70df0805da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/6250375ee3b2c76a70a63cfff2252d70df0805da"}]},"branch":"refs/heads/master"},"423868bd9f8718410279766fdd6807a4997cbf43":{"kind":"REWORK","_number":3,"created":"2019-01-29 06:08:51.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/47/633647/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/47/633647/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/3"}}},"commit":{"parents":[{"commit":"ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff","subject":"Merge \"Wait for the container to rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 02:49:43.000000000","tz":480},"committer":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 06:16:56.000000000","tz":480},"subject":"Optimization for attaching a running container","message":"Optimization for attaching a running container\n\nChange-Id: I079803111b3956865dfa75be7b1f95edd5623306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/423868bd9f8718410279766fdd6807a4997cbf43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/423868bd9f8718410279766fdd6807a4997cbf43"}]},"branch":"refs/heads/master"},"b141a510a7480831785e3ebcad833a19590304ab":{"kind":"REWORK","_number":4,"created":"2019-01-29 07:26:27.000000000","uploader":{"_account_id":23365,"name":"feng.shengqin","email":"feng.shengqin@zte.com.cn","username":"FengShengqin"},"ref":"refs/changes/47/633647/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/47/633647/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/4"}}},"commit":{"parents":[{"commit":"ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff","subject":"Merge \"Wait for the container to rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 02:49:43.000000000","tz":480},"committer":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 07:34:32.000000000","tz":480},"subject":"Optimization for attaching a running container","message":"Optimization for attaching a running container\n\nChange-Id: I079803111b3956865dfa75be7b1f95edd5623306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/b141a510a7480831785e3ebcad833a19590304ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/b141a510a7480831785e3ebcad833a19590304ab"}]},"branch":"refs/heads/master"},"5e8e200e04e79eae3d55e8e2940d6ee76dae23b3":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-08-30 19:22:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/47/633647/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/47/633647/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/5"}}},"commit":{"parents":[{"commit":"ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff","subject":"Merge \"Wait for the container to rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 02:49:43.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 19:22:56.000000000","tz":0},"subject":"Optimization for attaching a running container","message":"Optimization for attaching a running container\n\nChange-Id: I079803111b3956865dfa75be7b1f95edd5623306\nSigned-off-by: feng.shengqin \u003cfeng.shengqin@zte.com.cn\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/5e8e200e04e79eae3d55e8e2940d6ee76dae23b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/5e8e200e04e79eae3d55e8e2940d6ee76dae23b3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"11d9020942c73af35aaf1c593fde8f1a82dfd5d9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-30 19:23:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/47/633647/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/47/633647/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/47/633647/6"}}},"commit":{"parents":[{"commit":"b60422e25e8a54f6701070be1fd18e6c485380c6","subject":"tox: Drop redundant install_command override","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/b60422e25e8a54f6701070be1fd18e6c485380c6"}]}],"author":{"name":"Feng Shengqin","email":"feng.shengqin@zte.com.cn","date":"2019-01-29 02:49:43.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-30 19:23:00.000000000","tz":0},"subject":"Optimization for attaching a running container","message":"Optimization for attaching a running container\n\nChange-Id: I079803111b3956865dfa75be7b1f95edd5623306\nSigned-off-by: feng.shengqin \u003cfeng.shengqin@zte.com.cn\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/11d9020942c73af35aaf1c593fde8f1a82dfd5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/11d9020942c73af35aaf1c593fde8f1a82dfd5d9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"ed3d2a3ae3dc88ec6697433009df2108c7a1c4ff","ours":"5e8e200e04e79eae3d55e8e2940d6ee76dae23b3","theirs":"b60422e25e8a54f6701070be1fd18e6c485380c6","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
