)]}'
{"id":"openstack%2Fpython-zunclient~934999","triplet_id":"openstack%2Fpython-zunclient~unmaintained%2F2023.1~I3bd980e8da5d1464dee0448557265a21f7fb7480","project":"openstack/python-zunclient","branch":"unmaintained/2023.1","topic":"create-2023.1","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2026-08-31 09:37:10.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-31 09:03:29.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-31 09:16:40.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I3bd980e8da5d1464dee0448557265a21f7fb7480","subject":"Update .gitreview for unmaintained/2023.1","status":"MERGED","created":"2024-11-13 18:55:36.000000000","updated":"2026-08-31 09:38:16.000000000","submitted":"2026-08-31 09:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934999-create-2023.1","meta_rev_id":"712ffdafe56a634ac42af4d2b705472a11adc056","_number":934999,"virtual_id_number":934999,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 09:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-08-31 09:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-29 23:16:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-08-31 09:03:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 19:07:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-05 08:51:50.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"CC"},{"updated":"2025-09-04 10:27:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2026-08-29 22:40:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-08-29 23:16:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2026-08-29 23:16:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-08-29 23:16:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"14f24bf639aeed47f7c850ae9d8d5bb5fe78739f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-11-13 18:55:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec18e0cfee900ae4bed68fb6730404ac5311e753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 19:07:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/755c09c0bfee43e7b2dab89150397dc7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5500632327524df0b5bfe463aed3a667 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fabbb275044470b789080756ba7590 : FAILURE in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b2b1939e5b484282f74e0f8bcc5b49 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e7a073b43016ac48aaa41508714e0fe173133a92","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2024-12-05 08:51:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c006c4fd79cc268beebc99397c5a22a7c724f8c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 09:06:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87e65baca8a2454cb261c25a74836ac4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34bc5baf7af240fe9bf13d3e683a2f39 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0cf4b214e443daaa57ff571659b9ff : FAILURE in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b42237d83474d4bb408e52bc71b3447 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8bbb76b337d7237d585fe50caff22269d8e21bc5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-04 10:27:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71c086b3b798983c342dc9c1db364d8f90f8c182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 10:40:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce7da202da3a44acbe753eb2c0a4b272\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8a8130f2a7f4131a09ae0c5ff3820b4 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc849f671c0741ddb05d88b202ef442c : FAILURE in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03b965ae7cd457ab3704ec3b432574a : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8ded4b9910de7f2cb48d00d92284b52133fcd265","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2026-01-17 03:56:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a75f01e95cfb3a4e38e223d576ec80df71eef0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-17 04:04:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca660261fa3e49bd8366e09984152a4e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27002a04ba684877882cafbf0754e59b : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7df3f85e2940e1a0f2061e1add9514 : FAILURE in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408a58fdc3fc4d2d987acc6b1a679bdd : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c912b8f4d59c54add0f06b07bdd3e7bdca08330f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:40:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d036d346ea1f950492281b0bf90a24ffcf5ed936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 22:46:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9daf216cf9cb4f058b19eef9f1d06554\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75657ffe1ef4c9eb0c8a7e027e9d24b : FAILURE in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d598dca430854710aed8b394649d1fca : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"13380fb75b7e6400b9fa7afd6fe3771ddcc7af05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 22:57:37.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"115fcd96984ccabc3cbc621feae0c62701a3333c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 23:01:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f9f900532a4fcc10870a399d9d14216af796e9fc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 23:01:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7e8efc6ae4086511ef1707f5f359671fb1d068a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 23:10:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489077eeec5f409cb748f1dd5c73208e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dafd8c3a097e4da3817ba2a8854e8f7b : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a8970551dc4b3f8b8f21dee9129478 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3991555b33614d23a2076812d54452e1 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b012448562c478fae6161195ba63e6c : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a1d098758a2cca0b71f2335271af345e7e3afd17","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-29 23:16:48.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3ce6e9aeead262ba5f90c585893523b220daf3b5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-31 09:03:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47855ba0090230ab3e8bc4b2d86960de8d70afc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:03:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bf473d61674e9e5f3d0a16618c9981ac1d30f445","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-31 09:16:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"104de51b2ccd9e22e6fcd5b519100ec4fb7b9fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:37:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d814a05e0f49e395871d45bab4efa9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cccf3b2c3ebb448a92d5cd9b629dddd6 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f100f495659f40d78a9a87d184b00473 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c4740d168544eeaef5632547637132 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":3},{"id":"3ed3d5a21a23f1604dd92f6bb7faa03ffe5613f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:37:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"712ffdafe56a634ac42af4d2b705472a11adc056","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:38:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78250c696c0a4c0d918aa84845244420\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa946804f70646fcbca801978cd1626b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a371d5ba7d149c9ad9c07e5b9ba6538bec0ea200","revisions":{"c15ce04ea2caa70d60c83d9c28eac252d33bb4ed":{"kind":"REWORK","_number":1,"created":"2024-11-13 18:55:36.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/99/934999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/99/934999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/1"}}},"commit":{"parents":[{"commit":"d3993aab037f8e4d31e6b817a538fcbeea6e97b3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/d3993aab037f8e4d31e6b817a538fcbeea6e97b3"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-13 18:55:34.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-13 18:55:34.000000000","tz":0},"subject":"Update .gitreview for unmaintained/2023.1","message":"Update .gitreview for unmaintained/2023.1\n\nChange-Id: I3bd980e8da5d1464dee0448557265a21f7fb7480\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/c15ce04ea2caa70d60c83d9c28eac252d33bb4ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/c15ce04ea2caa70d60c83d9c28eac252d33bb4ed"}]},"branch":"refs/heads/unmaintained/2023.1"},"f71ec580ab626875cf20fb5b29ae62859e18c115":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-29 22:57:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/99/934999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/99/934999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/2"}}},"commit":{"parents":[{"commit":"d3993aab037f8e4d31e6b817a538fcbeea6e97b3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/d3993aab037f8e4d31e6b817a538fcbeea6e97b3"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-13 18:55:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 22:57:37.000000000","tz":0},"subject":"Update .gitreview for unmaintained/2023.1","message":"Update .gitreview for unmaintained/2023.1\n\nChange-Id: I3bd980e8da5d1464dee0448557265a21f7fb7480\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/f71ec580ab626875cf20fb5b29ae62859e18c115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/f71ec580ab626875cf20fb5b29ae62859e18c115"}]},"branch":"refs/heads/unmaintained/2023.1","description":"Edit commit message"},"a371d5ba7d149c9ad9c07e5b9ba6538bec0ea200":{"kind":"REWORK","_number":3,"created":"2026-08-29 23:01:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/99/934999/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/python-zunclient","ref":"refs/changes/99/934999/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/python-zunclient refs/changes/99/934999/3"}}},"commit":{"parents":[{"commit":"d3993aab037f8e4d31e6b817a538fcbeea6e97b3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/d3993aab037f8e4d31e6b817a538fcbeea6e97b3"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-11-13 18:55:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-29 23:01:01.000000000","tz":0},"subject":"Update .gitreview for unmaintained/2023.1","message":"Update .gitreview for unmaintained/2023.1\n\nChange-Id: I3bd980e8da5d1464dee0448557265a21f7fb7480\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/a371d5ba7d149c9ad9c07e5b9ba6538bec0ea200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/python-zunclient/commit/a371d5ba7d149c9ad9c07e5b9ba6538bec0ea200"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
