)]}'
{"id":"openstack%2Fqa-specs~443504","triplet_id":"openstack%2Fqa-specs~master~I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df","project":"openstack/qa-specs","branch":"master","topic":"bp/openstack-dt-patch9","hashtags":[],"change_id":"I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df","subject":"Add spec for high availability testing","status":"ABANDONED","created":"2017-03-09 08:56:28.000000000","updated":"2020-12-03 18:11:43.000000000","total_comment_count":68,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d035bc59cc8c3bde9b07907a63e66f1da2ca0c7b","_number":443504,"virtual_id_number":443504,"owner":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11520,"name":"Jamey McCabe","email":"jamemcc@gmail.com","username":"jamemcc"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2018-07-06 04:56:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"_account_id":14614,"name":"Kyrylo Romanenko","display_name":"K.Romanenko","email":"romanenko-kv@hotmail.com","username":"kromanenko"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":13919,"name":"Evgeny Sikachev","email":"esikachov@gmail.com","username":"esikachev"},{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},{"_account_id":18916,"name":"Maksym Shalamov","email":"mshalamov@ukr.net","username":"mshalamov"},{"_account_id":20772,"name":"Sergei Chipiga","email":"schipiga@mirantis.com","username":"schipiga"},{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},{"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},{"_account_id":15784,"name":"Igor Pukha","email":"ipukha@mirantis.com","username":"ingwarr"},{"_account_id":16425,"name":"Bin Bin Cong","email":"binbin.cong@easystack.cn","username":"stephen_cong"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17517,"name":"Andrew Beekhof","email":"abeekhof@redhat.com","username":"beekhof"},{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"_account_id":17270,"name":"Sergii Turivnyi","email":"sturivnyi@mirantis.com","username":"sturivnyi"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11520,"name":"Jamey McCabe","email":"jamemcc@gmail.com","username":"jamemcc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14614,"name":"Kyrylo Romanenko","display_name":"K.Romanenko","email":"romanenko-kv@hotmail.com","username":"kromanenko"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13919,"name":"Evgeny Sikachev","email":"esikachov@gmail.com","username":"esikachev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18916,"name":"Maksym Shalamov","email":"mshalamov@ukr.net","username":"mshalamov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20772,"name":"Sergei Chipiga","email":"schipiga@mirantis.com","username":"schipiga"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15784,"name":"Igor Pukha","email":"ipukha@mirantis.com","username":"ingwarr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16425,"name":"Bin Bin Cong","email":"binbin.cong@easystack.cn","username":"stephen_cong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17517,"name":"Andrew Beekhof","email":"abeekhof@redhat.com","username":"beekhof"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17270,"name":"Sergii Turivnyi","email":"sturivnyi@mirantis.com","username":"sturivnyi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11520,"name":"Jamey McCabe","email":"jamemcc@gmail.com","username":"jamemcc"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"_account_id":14614,"name":"Kyrylo Romanenko","display_name":"K.Romanenko","email":"romanenko-kv@hotmail.com","username":"kromanenko"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"date":"2018-07-30 01:51:04.000000000","_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":13919,"name":"Evgeny Sikachev","email":"esikachov@gmail.com","username":"esikachev"},{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},{"_account_id":18916,"name":"Maksym Shalamov","email":"mshalamov@ukr.net","username":"mshalamov"},{"_account_id":20772,"name":"Sergei Chipiga","email":"schipiga@mirantis.com","username":"schipiga"},{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},{"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},{"_account_id":15784,"name":"Igor Pukha","email":"ipukha@mirantis.com","username":"ingwarr"},{"_account_id":16425,"name":"Bin Bin Cong","email":"binbin.cong@easystack.cn","username":"stephen_cong"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17517,"name":"Andrew Beekhof","email":"abeekhof@redhat.com","username":"beekhof"},{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"_account_id":17270,"name":"Sergii Turivnyi","email":"sturivnyi@mirantis.com","username":"sturivnyi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":11520,"name":"Jamey McCabe","email":"jamemcc@gmail.com","username":"jamemcc"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13919,"name":"Evgeny Sikachev","email":"esikachov@gmail.com","username":"esikachev"},{"_account_id":14614,"name":"Kyrylo Romanenko","display_name":"K.Romanenko","email":"romanenko-kv@hotmail.com","username":"kromanenko"},{"_account_id":15784,"name":"Igor Pukha","email":"ipukha@mirantis.com","username":"ingwarr"},{"_account_id":16425,"name":"Bin Bin Cong","email":"binbin.cong@easystack.cn","username":"stephen_cong"},{"_account_id":17270,"name":"Sergii Turivnyi","email":"sturivnyi@mirantis.com","username":"sturivnyi"},{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},{"_account_id":17517,"name":"Andrew Beekhof","email":"abeekhof@redhat.com","username":"beekhof"},{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},{"_account_id":18916,"name":"Maksym Shalamov","email":"mshalamov@ukr.net","username":"mshalamov"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20772,"name":"Sergei Chipiga","email":"schipiga@mirantis.com","username":"schipiga"},{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-09 09:41:46.000000000","updated_by":{"_account_id":13919,"name":"Evgeny Sikachev","email":"esikachov@gmail.com","username":"esikachev"},"reviewer":{"_account_id":13919,"name":"Evgeny Sikachev","email":"esikachov@gmail.com","username":"esikachev"},"state":"REVIEWER"},{"updated":"2017-03-09 09:41:56.000000000","updated_by":{"_account_id":15784,"name":"Igor Pukha","email":"ipukha@mirantis.com","username":"ingwarr"},"reviewer":{"_account_id":15784,"name":"Igor Pukha","email":"ipukha@mirantis.com","username":"ingwarr"},"state":"REVIEWER"},{"updated":"2017-03-09 09:42:54.000000000","updated_by":{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},"reviewer":{"_account_id":17716,"name":"Luz Cazares","email":"luz.cazares@intel.com","username":"luzcazares"},"state":"REVIEWER"},{"updated":"2017-03-09 09:43:06.000000000","updated_by":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"reviewer":{"_account_id":6849,"name":"Roman Podoliaka","email":"roman.podoliaka@gmail.com","username":"rpodolyaka"},"state":"REVIEWER"},{"updated":"2017-03-09 09:43:12.000000000","updated_by":{"_account_id":20772,"name":"Sergei Chipiga","email":"schipiga@mirantis.com","username":"schipiga"},"reviewer":{"_account_id":20772,"name":"Sergei Chipiga","email":"schipiga@mirantis.com","username":"schipiga"},"state":"REVIEWER"},{"updated":"2017-03-09 09:44:02.000000000","updated_by":{"_account_id":17270,"name":"Sergii Turivnyi","email":"sturivnyi@mirantis.com","username":"sturivnyi"},"reviewer":{"_account_id":17270,"name":"Sergii Turivnyi","email":"sturivnyi@mirantis.com","username":"sturivnyi"},"state":"REVIEWER"},{"updated":"2017-03-09 09:45:40.000000000","updated_by":{"_account_id":18916,"name":"Maksym Shalamov","email":"mshalamov@ukr.net","username":"mshalamov"},"reviewer":{"_account_id":18916,"name":"Maksym Shalamov","email":"mshalamov@ukr.net","username":"mshalamov"},"state":"REVIEWER"},{"updated":"2017-03-23 18:29:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2017-04-11 13:54:49.000000000","updated_by":{"_account_id":11520,"name":"Jamey McCabe","email":"jamemcc@gmail.com","username":"jamemcc"},"reviewer":{"_account_id":11520,"name":"Jamey McCabe","email":"jamemcc@gmail.com","username":"jamemcc"},"state":"REVIEWER"},{"updated":"2017-04-12 04:16:42.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2017-04-27 02:40:58.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2017-06-27 05:36:30.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2017-07-04 00:55:41.000000000","updated_by":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"reviewer":{"_account_id":8988,"name":"Rikimaru Honjo","email":"honjo.rikimaru@ntt-tx.co.jp","username":"rikimaru.honjo","inactive":true},"state":"REVIEWER"},{"updated":"2017-07-12 10:26:29.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2017-07-14 09:51:11.000000000","updated_by":{"_account_id":14614,"name":"Kyrylo Romanenko","display_name":"K.Romanenko","email":"romanenko-kv@hotmail.com","username":"kromanenko"},"reviewer":{"_account_id":14614,"name":"Kyrylo Romanenko","display_name":"K.Romanenko","email":"romanenko-kv@hotmail.com","username":"kromanenko"},"state":"REVIEWER"},{"updated":"2017-07-31 05:19:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2017-08-31 19:20:51.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2017-08-31 19:28:35.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-09-03 16:42:17.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2017-10-24 12:01:42.000000000","updated_by":{"_account_id":16425,"name":"Bin Bin Cong","email":"binbin.cong@easystack.cn","username":"stephen_cong"},"reviewer":{"_account_id":16425,"name":"Bin Bin Cong","email":"binbin.cong@easystack.cn","username":"stephen_cong"},"state":"REVIEWER"},{"updated":"2018-02-27 16:40:46.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2018-02-28 00:16:31.000000000","updated_by":{"_account_id":17517,"name":"Andrew Beekhof","email":"abeekhof@redhat.com","username":"beekhof"},"reviewer":{"_account_id":17517,"name":"Andrew Beekhof","email":"abeekhof@redhat.com","username":"beekhof"},"state":"REVIEWER"},{"updated":"2018-03-15 15:59:39.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2018-03-16 06:30:23.000000000","updated_by":{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},"reviewer":{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},"state":"REVIEWER"},{"updated":"2018-03-16 11:56:59.000000000","updated_by":{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},"reviewer":{"_account_id":17491,"name":"Lujin Luo","email":"luo.lujin@jp.fujitsu.com","username":"Lujin"},"state":"REVIEWER"},{"updated":"2018-03-30 08:05:56.000000000","updated_by":{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},"reviewer":{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},"state":"REVIEWER"},{"updated":"2018-04-05 10:11:30.000000000","updated_by":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"reviewer":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"state":"REVIEWER"},{"updated":"2018-04-05 17:26:18.000000000","updated_by":{"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},"reviewer":{"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},"state":"REVIEWER"},{"updated":"2018-05-21 14:50:42.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2018-07-06 04:56:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-30 01:51:04.000000000","updated_by":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"reviewer":{"_account_id":17887,"name":"Doug Schveninger","email":"ds6901@att.com","username":"Doug.Schveninger"},"state":"REVIEWER"}],"messages":[{"id":"abc80d9e1279fc919a4241caa2f95f53ff4c869d","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-03-09 08:56:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef3518fa2ed2b21bf8978c9a32b004fb576d9c52","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-03-09 08:56:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d31f1c38fcc76cd202d0509b33b7c1c754b0b2f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-09 08:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-qa-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/04/443504/1/check/gate-qa-specs-docs-ubuntu-xenial/a6bb6d3//doc/build/html/ : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b98d3b80e449c208f395c26fa2f94dc22f6d1df9","author":{"_account_id":13919,"name":"Evgeny Sikachev","email":"esikachov@gmail.com","username":"esikachev"},"date":"2017-03-09 10:29:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0fe856b9beb78e3bfc47d66d6ae99e32831d895","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2017-03-11 01:01:34.000000000","message":"Patch Set 1:\n\nPlease go ahead with this spec instead of the original.","accounts_in_message":[],"_revision_number":1},{"id":"17a0577f4adc55fe21c0584de51a996c9d38ebb7","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-04-11 10:49:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1108a7bc419fec9497bb048cab830094599a2266","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-11 10:51:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-qa-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/04/443504/2/check/gate-qa-specs-docs-ubuntu-xenial/6ce60ad//doc/build/html/ : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f3f368f58d7cf34e36c8625a487988cb069a6b04","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-04-11 11:12:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3d6f605260941f2e5694d2314f0f0a5dc9a33f5","author":{"_account_id":11520,"name":"Jamey McCabe","email":"jamemcc@gmail.com","username":"jamemcc"},"date":"2017-04-11 13:54:49.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nGreat start - will work with you on this.  Please see comment.","accounts_in_message":[],"_revision_number":2},{"id":"aa63a49a1dc863207f63f29083f404be59081b48","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2017-04-27 02:40:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nSorry for my late response, some comments inline.","accounts_in_message":[],"_revision_number":2},{"id":"b55d3e8fe021441698ecf3371fbc3f02f396bcc7","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-06-22 05:48:18.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nThank you for all the comments. Please check the next patch for updates.","accounts_in_message":[],"_revision_number":2},{"id":"a32aecf0682cccafeb7766b768de40556b5c3f76","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-06-22 06:00:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f13fd4a36830423ff41954a869aa69dcebb3978e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-22 06:02:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-qa-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/04/443504/3/check/gate-qa-specs-docs-ubuntu-xenial/c28b6e4//doc/build/html/ : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":3},{"id":"69223c116c8697df8b19ea80570e3c2f128cd02f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2017-06-27 05:36:30.000000000","message":"Patch Set 3:\n\nI think this is good start. I have a couple of questions.\n\n1. Are you thinking about to run this test in the gate? (We should, IMO. But it requires a lot of resources, probably.)\n\n2. Do you already have code for this project?\n I feel this is a bit broad like HA, destructive, etc.. So, I think making a Minimum viable product is better to know issues/problem/good points of this spec.","accounts_in_message":[],"_revision_number":3},{"id":"a5d151f918f1d15762af08b7ddb927ecbd6bfa99","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-06-27 08:17:28.000000000","message":"Patch Set 3:\n\nHi Masayuki,\nThank you for the review. \n# sorry for long reply..\n\n(1) Run this on gate?\n# Mentioned in #L117\nYes we should. However, there are some concerns;\n(a) Testing Duration\n(b) Required resources (as you mentioned..)\n(c) Technical difficulties\nTherefore, I proposed to divide the test scenarios in to 3 groups\nand run light weight tests in the gate and offload heavy tests in\nto 3rd party CIs. Though, I would love to get more feedback or\nsuggestions on this.\n\n(2)Do you already have code for this project?\nWell, we (Masakari team) implemented part of test scenarios with \nstepler and os-fault [1]. We also have 3rd party CI for Masakari.\n However, currently, there are lot of people doing this in \nthere own way. Masakri Team and LCOO [2] are working together\non this topic[3], and currently we are doing gap/overlap analysis\non existing frameworks (in \"Alternatives\" section).\nAim is to find out what we can reuse, and what we should\nre-write to make this framework more feasible.\n\n Totally agree on start with \"minimum viable product\".\n\n[1] https://review.openstack.org/#/q/topic:add-stepler-tests-alt\n[2] https://wiki.openstack.org/wiki/LCOO\n[3] https://openstack-lcoo.atlassian.net/wiki/display/LCOO/Working+Space+for+Exteme+Testing+and+related+topics","accounts_in_message":[],"_revision_number":3},{"id":"598e5835c0cea1b5a44f7dda49401cdb4216ad9f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2017-07-12 10:26:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nThe spec looks good.","accounts_in_message":[],"_revision_number":3},{"id":"0838b8f18c4b897aef6e994f39d6523cab6c0fd8","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-07-12 10:59:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"233be1c974bbccaa46a302dc522b2062a6860856","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-07-13 08:40:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c33feba43699f723e936eaca7f2849af253eddbf","author":{"_account_id":14614,"name":"Kyrylo Romanenko","display_name":"K.Romanenko","email":"romanenko-kv@hotmail.com","username":"kromanenko"},"date":"2017-07-14 09:51:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nHi, Sampath. Here are some comments to this spec.","accounts_in_message":[],"_revision_number":3},{"id":"724b6b151f2148c04136e54f87b0030cec189181","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-07-14 10:48:55.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5bc7aeaee1b15bedf80fdc973861e237ca424cbe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2017-07-31 05:19:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)\n\nThanks Sam for taking this. this is close but need to take care of some comments.","accounts_in_message":[],"_revision_number":3},{"id":"8e56b86c7487a9e06dd1b2418b6f7a317fdb1a3d","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2017-08-02 17:11:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"58fbd8676d04328a61629404a17961437c8cf306","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2017-08-28 23:36:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ef6a10412bc6931529762b2a4a9d5159ce4d7c8a","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2017-08-28 23:51:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bcc3dd79dd13ec737192bb9f67a5a6f5ed38a141","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 02:20:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-qa-specs-docs-ubuntu-xenial http://logs.openstack.org/04/443504/4/check/gate-qa-specs-docs-ubuntu-xenial/92df1f6/ : FAILURE in 3m 01s","accounts_in_message":[],"_revision_number":4},{"id":"7f5dca712c45ccd52d8c1b784c09844219db62f6","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-08-29 14:44:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fbcc9b19d564ad34b801e9d15fbef88893c1a1d8","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-08-29 14:44:51.000000000","message":"Patch Set 3:\n\n(11 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c378398555f975efb7987c947073b2fc6476adce","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-08-29 14:49:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c4ccaaf8adbc2f8eb94a0c7c3f6f169ac6022f03","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-29 15:19:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-qa-specs-docs-ubuntu-xenial http://logs.openstack.org/04/443504/6/check/gate-qa-specs-docs-ubuntu-xenial/d0bb94e/ : FAILURE in 2m 38s","accounts_in_message":[],"_revision_number":6},{"id":"71bfb81857ce79ced66d64181300e885c7c77c28","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2017-08-29 15:30:46.000000000","message":"Patch Set 6:\n\nFix for broken gate is here: https://review.openstack.org/#/c/498665/","accounts_in_message":[],"_revision_number":6},{"id":"334d1c9c23883bea9a764f7608e90314811a00a7","author":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"date":"2017-08-29 16:01:05.000000000","message":"Patch Set 6:\n\nThanks. That works.","accounts_in_message":[],"_revision_number":6},{"id":"746dc6d778e6a044c30e9588bee4c20f1603ce5f","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2017-08-31 19:20:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1fa6ad5c3eeac3fa5e3da670670fadf642ea2226","author":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"date":"2017-08-31 19:20:51.000000000","message":"Patch Set 7:\n\nRebase to pass gate.","accounts_in_message":[],"_revision_number":7},{"id":"e100153bdeb88fcd066606c64ade4c5d84d08046","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-31 19:28:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-qa-specs-docs-ubuntu-xenial http://docs-draft.openstack.org/04/443504/7/check/gate-qa-specs-docs-ubuntu-xenial/75c63da//doc/build/html/ : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":7},{"id":"25925a602772323b10d4e12e4d9d72a711678cc0","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2017-09-03 16:42:17.000000000","message":"Patch Set 7: Code-Review-1\n\n(9 comments)\n\nThank you for the spec.\n\nHigh availability testing is certainly something interesting to consider.\nAs you mentioned in the spec it may be difficult to run all / many of these tests on OpenStack infrastructure, but it\u0027s worth investigating and see if there is something we can integrate in our automated testing infrastructure.\n\nI have a few comments inline. I think it would be nice to be a bit more specific about what services / functionality we\u0027re looking to provide, and build a rough priority list of things to be done.\n\nIs there a PoC available or some existing tests - perhaps in a GitHub repo - that we could use as a starting point?","accounts_in_message":[],"_revision_number":7},{"id":"8f974c4add475f64587e50c321a6ff5e4bc02c51","author":{"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},"date":"2017-09-07 15:37:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7d53811c1a5e6c2655afcc7498c2cedecf5c583e","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2017-09-20 09:00:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"228621ae53644c41a0fdfff0d28d627f3a6216e9","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2017-09-21 23:28:44.000000000","message":"Patch Set 7:\n\nBig +1 for the name Eris - great idea! :-)","accounts_in_message":[],"_revision_number":7},{"id":"eb69d6c1da7a13ebf0964b0c0eb0b2c94c3d01f9","author":{"_account_id":16425,"name":"Bin Bin Cong","email":"binbin.cong@easystack.cn","username":"stephen_cong"},"date":"2017-10-24 12:01:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ca4c827ec1a0bfa39e37e7ec67c906b9c1c794fc","author":{"_account_id":17517,"name":"Andrew Beekhof","email":"abeekhof@redhat.com","username":"beekhof"},"date":"2018-02-28 00:16:31.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ae0e6e563135bb499ddb51968bb17b587aff8a82","author":{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},"date":"2018-03-30 08:05:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d7b45ad85a3d181c95c1a394e2aff8acc2af3b9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-30 08:11:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/04/443504/8/check/build-openstack-sphinx-docs/22a3c76/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":8},{"id":"b2ffdc9789de72fb43a17b557c1361988a94dc0e","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2018-04-05 10:11:30.000000000","message":"Patch Set 8: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b654d327c020e9c6304623ac07798b36f088db0f","author":{"_account_id":26856,"name":"Gautam Divgi","email":"gautamdivgi@gmail.com"},"date":"2018-04-05 17:26:18.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9c7e9cd6de1ce462038d1b8e59179c7d03156c2f","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2018-04-26 13:55:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"942a627c2cd07f4c81088d079de6628489371902","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2018-05-21 14:50:42.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6dd6af577587105802ae95f004748bea4521d2e7","author":{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},"date":"2018-07-06 04:45:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"beeec2620c1edda3a44ca23ca80ac1cd99b1e16c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 04:56:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/04/443504/9/check/build-openstack-sphinx-docs/a05dcee/html/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":9},{"id":"b471e07d69d501c4796394fbb7e0b073fbecbc4a","author":{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},"date":"2018-07-06 08:31:13.000000000","message":"Topic changed from bp/openstack-dt-patch7-patch8 to bp/openstack-dt-patch9","accounts_in_message":[],"_revision_number":9},{"id":"d035bc59cc8c3bde9b07907a63e66f1da2ca0c7b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-03 18:11:43.000000000","message":"Abandoned\n\nno activity on this proposal since many years. Feel free to restore it once we find volunteer to drive this.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3218981bdac788e32a7a23a6eedb49594932a21a","revisions":{"7159ba44a000f5784d9e28c370dae2cadaf0f417":{"kind":"REWORK","_number":1,"created":"2017-03-09 08:56:28.000000000","uploader":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"ref":"refs/changes/04/443504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/1"}}},"commit":{"parents":[{"commit":"860c51df1d0206693b3ce24978ba49ad2f3037d3","subject":"Requirements.txt: block docutils 0.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/860c51df1d0206693b3ce24978ba49ad2f3037d3"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"subject":"Added spec for high availability testing","message":"Added spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/7159ba44a000f5784d9e28c370dae2cadaf0f417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/7159ba44a000f5784d9e28c370dae2cadaf0f417"}]},"branch":"refs/heads/master"},"f0c19d811db7ff8628de64dfd6e9284442b7744b":{"kind":"REWORK","_number":2,"created":"2017-04-11 10:49:46.000000000","uploader":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"ref":"refs/changes/04/443504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/2"}}},"commit":{"parents":[{"commit":"860c51df1d0206693b3ce24978ba49ad2f3037d3","subject":"Requirements.txt: block docutils 0.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/860c51df1d0206693b3ce24978ba49ad2f3037d3"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-04-11 10:49:44.000000000","tz":540},"subject":"Added spec for high availability testing","message":"Added spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/f0c19d811db7ff8628de64dfd6e9284442b7744b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/f0c19d811db7ff8628de64dfd6e9284442b7744b"}]},"branch":"refs/heads/master"},"4c468991a970a5b108469d61fb8fb4b4d8e7b25b":{"kind":"REWORK","_number":3,"created":"2017-06-22 06:00:18.000000000","uploader":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"ref":"refs/changes/04/443504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/3"}}},"commit":{"parents":[{"commit":"860c51df1d0206693b3ce24978ba49ad2f3037d3","subject":"Requirements.txt: block docutils 0.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/860c51df1d0206693b3ce24978ba49ad2f3037d3"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"sampath","email":"sam47priya@gmail.com","date":"2017-06-22 05:59:57.000000000","tz":540},"subject":"Added spec for high availability testing","message":"Added spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/4c468991a970a5b108469d61fb8fb4b4d8e7b25b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/4c468991a970a5b108469d61fb8fb4b4d8e7b25b"}]},"branch":"refs/heads/master"},"b3646a4399f8c0b950d262cc8021e15bf36a9246":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-08-28 23:51:40.000000000","uploader":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"ref":"refs/changes/04/443504/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/4"}}},"commit":{"parents":[{"commit":"a328be844cc7b662972d82b7a049dcae79512aa5","subject":"Merge \"Update document theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/a328be844cc7b662972d82b7a049dcae79512aa5"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"Felipe Monteiro","email":"felipe.monteiro@att.com","date":"2017-08-28 23:51:40.000000000","tz":0},"subject":"Added spec for high availability testing","message":"Added spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/b3646a4399f8c0b950d262cc8021e15bf36a9246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/b3646a4399f8c0b950d262cc8021e15bf36a9246"}]},"branch":"refs/heads/master"},"a2a4238769c219e139571f8cca26594f8c0a4c35":{"kind":"REWORK","_number":5,"created":"2017-08-29 14:44:38.000000000","uploader":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"ref":"refs/changes/04/443504/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/5"}}},"commit":{"parents":[{"commit":"a328be844cc7b662972d82b7a049dcae79512aa5","subject":"Merge \"Update document theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/a328be844cc7b662972d82b7a049dcae79512aa5"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"sampath","email":"sam47priya@gmail.com","date":"2017-08-29 14:44:11.000000000","tz":540},"subject":"Added spec for high availability testing","message":"Added spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nPartially implements blueprint openstack-dt\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/a2a4238769c219e139571f8cca26594f8c0a4c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/a2a4238769c219e139571f8cca26594f8c0a4c35"}]},"branch":"refs/heads/master"},"618a17524a1081b4e661ef9ed2b959ce89a23212":{"kind":"REWORK","_number":6,"created":"2017-08-29 14:49:04.000000000","uploader":{"_account_id":8716,"name":"Sampath Priyankara","email":"sam47priya@gmail.com","username":"sampath"},"ref":"refs/changes/04/443504/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/6"}}},"commit":{"parents":[{"commit":"a328be844cc7b662972d82b7a049dcae79512aa5","subject":"Merge \"Update document theme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/a328be844cc7b662972d82b7a049dcae79512aa5"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"sampath","email":"sam47priya@gmail.com","date":"2017-08-29 14:48:49.000000000","tz":540},"subject":"Added spec for high availability testing","message":"Added spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nPartially implements blueprint openstack-dt\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/618a17524a1081b4e661ef9ed2b959ce89a23212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/618a17524a1081b4e661ef9ed2b959ce89a23212"}]},"branch":"refs/heads/master"},"2c9f26f6b5818ad692433ff121861f4161f8ba8c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2017-08-31 19:20:40.000000000","uploader":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"ref":"refs/changes/04/443504/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/7"}}},"commit":{"parents":[{"commit":"c2fb1f2f00d92613332db60ef98ee18c5662c607","subject":"Sync sphinx version with g-r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/c2fb1f2f00d92613332db60ef98ee18c5662c607"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"Felipe Monteiro","email":"felipe.monteiro@att.com","date":"2017-08-31 19:20:40.000000000","tz":0},"subject":"Added spec for high availability testing","message":"Added spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nPartially implements blueprint openstack-dt\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/2c9f26f6b5818ad692433ff121861f4161f8ba8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/2c9f26f6b5818ad692433ff121861f4161f8ba8c"}]},"branch":"refs/heads/master"},"578571dc303bfe4bb3a904193712d46424582bbb":{"kind":"REWORK","_number":8,"created":"2018-03-30 08:05:56.000000000","uploader":{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},"ref":"refs/changes/04/443504/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/8"}}},"commit":{"parents":[{"commit":"c2fb1f2f00d92613332db60ef98ee18c5662c607","subject":"Sync sphinx version with g-r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/c2fb1f2f00d92613332db60ef98ee18c5662c607"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","date":"2018-03-30 08:03:20.000000000","tz":540},"subject":"Add spec for high availability testing","message":"Add spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nPartially implements blueprint openstack-dt\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/578571dc303bfe4bb3a904193712d46424582bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/578571dc303bfe4bb3a904193712d46424582bbb"}]},"branch":"refs/heads/master"},"3218981bdac788e32a7a23a6eedb49594932a21a":{"kind":"REWORK","_number":9,"created":"2018-07-06 04:45:02.000000000","uploader":{"_account_id":27106,"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","username":"diceken"},"ref":"refs/changes/04/443504/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/04/443504/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/04/443504/9"}}},"commit":{"parents":[{"commit":"c2fb1f2f00d92613332db60ef98ee18c5662c607","subject":"Sync sphinx version with g-r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/c2fb1f2f00d92613332db60ef98ee18c5662c607"}]}],"author":{"name":"Sampath Priyankara","email":"sam47priya@gmail.com","date":"2017-03-09 08:48:09.000000000","tz":540},"committer":{"name":"Daisuke Nakazawa","email":"nakazawa.daisuke@lab.ntt.co.jp","date":"2018-07-06 04:40:16.000000000","tz":540},"subject":"Add spec for high availability testing","message":"Add spec for high availability testing\n\nThis spec propose to add new repository with framework to\nprovide High Availability, destructive, failover, and recovery testing.\nThis work is originally porposed by Timur[1].\n\n[1] https://review.openstack.org/#/c/399618/\n\nPartially implements blueprint openstack-dt\nChange-Id: I4dcbb2307ebe953f92495db6894f9fe6f6d8d1df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/3218981bdac788e32a7a23a6eedb49594932a21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/3218981bdac788e32a7a23a6eedb49594932a21a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
