)]}'
{"id":"openstack%2Fqa-specs~630380","triplet_id":"openstack%2Fqa-specs~master~I16a30c49c3b8c6e562594fef987db58b65f2219d","project":"openstack/qa-specs","branch":"master","hashtags":[],"change_id":"I16a30c49c3b8c6e562594fef987db58b65f2219d","subject":"Spec: Re-factor Account Generator","status":"ABANDONED","created":"2019-01-11 20:06:35.000000000","updated":"2023-07-20 01:41:52.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c83ccecb4ba6a37e7170438ac32b28f02586ce1e","_number":630380,"virtual_id_number":630380,"owner":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-07-25 22:31:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":-1,"date":"2019-08-14 23:27:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":2,"date":"2019-07-25 22:25:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},{"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-11 20:32:11.000000000","updated_by":{"_account_id":28701,"name":"Dustin Specker","email":"v-dspecker@microsoft.com","username":"dustinspecker"},"reviewer":{"_account_id":28701,"name":"Dustin Specker","email":"v-dspecker@microsoft.com","username":"dustinspecker"},"state":"REVIEWER"},{"updated":"2019-01-16 23:48:55.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2019-01-16 23:48:55.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2019-01-16 23:48:55.000000000","updated_by":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"reviewer":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"state":"REVIEWER"},{"updated":"2019-01-16 23:48:55.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-01-16 23:48:55.000000000","updated_by":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"reviewer":{"_account_id":7350,"name":"Jordan Pittier","email":"jordan.pittier@gmail.com","username":"JordanP"},"state":"REVIEWER"},{"updated":"2019-01-16 23:48:55.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2019-01-16 23:48:55.000000000","updated_by":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"reviewer":{"_account_id":23186,"name":"Felipe Monteiro","email":"felipe.carneiro.monteiro@gmail.com","username":"felipe.monteiro"},"state":"REVIEWER"},{"updated":"2019-01-29 19:06:54.000000000","updated_by":{"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},"reviewer":{"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},"state":"REVIEWER"},{"updated":"2019-02-04 17:44:46.000000000","updated_by":{"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"},"reviewer":{"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"},"state":"REVIEWER"},{"updated":"2019-07-08 03:14:37.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-07-25 22:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-14 23:27:30.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-02-17 22:26:06.000000000","updated_by":{"_account_id":28701,"name":"Dustin Specker","email":"v-dspecker@microsoft.com","username":"dustinspecker"},"reviewer":{"_account_id":28701,"name":"Dustin Specker","email":"v-dspecker@microsoft.com","username":"dustinspecker"},"state":"REMOVED"}],"messages":[{"id":"e5cf3e31aeefc3e692587da699f9631f0368c0c5","author":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"date":"2019-01-11 20:06:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d368da952a0163e6ec0229c7285fc5af430e8fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 20:10:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/630380/1/check/openstack-tox-docs/93b545a/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"759103ed2c361342837ebbb8c99d209453434db7","author":{"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"},"date":"2019-01-11 20:23:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d29a38c27ebe3abed746f55e07f2e23d3ee9ee37","author":{"_account_id":28701,"name":"Dustin Specker","email":"v-dspecker@microsoft.com","username":"dustinspecker"},"date":"2019-01-11 20:32:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dab21140e27f535744af0f69eb10f2f194db50a4","author":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"date":"2019-01-29 17:56:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f9cf27a1541a097b871db8f420d20e2db719107","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 18:13:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/630380/1/check/openstack-tox-docs/e453e66/html/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7d50a6aae645cd9f600f660236ebcb151ee5c672","author":{"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},"date":"2019-01-29 18:20:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nlooks good except for a few minor things.","accounts_in_message":[],"_revision_number":1},{"id":"6c9932b8a824c737b56aa723dad1baf1cc9862d5","author":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"date":"2019-01-29 19:05:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"52993ffd8d776c15ae31fcb36abfd4423d971575","author":{"_account_id":20378,"name":"Trevor McCasland","email":"TM2086@att.com","username":"twm2016"},"date":"2019-01-29 19:06:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f42102ada4454c33743faf95ec60cf70a2111301","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 19:21:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/630380/2/check/openstack-tox-docs/052c572/html/ : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"06b2f3bdcc27207d7c9db9c259bde32779c46e8c","author":{"_account_id":29441,"name":"Jeffrey Williams","email":"jw2610@att.com","username":"jw2610"},"date":"2019-02-04 17:44:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0e44d3bd263285c3ec5f360fbc606414656cfd40","author":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"date":"2019-02-15 16:55:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5af7ab98002d3d855a28047f6a13be9f55df377e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-15 17:02:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/630380/3/check/openstack-tox-docs/2ac429f/html/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"fd650f7f50dec576ffbaa225a2be291e54c2ef0e","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-21 01:51:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(3 comments)\n\nLGTM. some comments inline","accounts_in_message":[],"_revision_number":3},{"id":"3120a4d13b7c18ed2f53503d36c1f09882e763a0","author":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"date":"2019-07-03 21:07:20.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5a1c58efdd9e00491d8d3ed07c2e2995c5619946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 21:15:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/630380/4/check/openstack-tox-docs/11efa09/html/ : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":4},{"id":"0d77246bd7184f8e0fc9dbcda0e33c3d2607fd37","author":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"date":"2019-07-05 16:19:13.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3b600ab7bc470f9e0d7b356c03d81e4d96faf9a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 16:34:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/630380/5/check/openstack-tox-docs/0015653/html/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":5},{"id":"65e6dd4b894ab2087b556c152dc84faebf4434e5","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-07-08 03:14:37.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks for updating!","accounts_in_message":[],"_revision_number":5},{"id":"ce7d26dce708a2c0a57bff7e71456aaca313feb4","author":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"date":"2019-07-25 22:25:12.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d64580b11147abf372f6a532656accdc0ad16529","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 22:31:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/630380/6/check/openstack-tox-docs/5a07a8c/html/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":6},{"id":"ed9b056efaab3efe2ff3c87b06630a5d18cf7229","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-08-14 23:27:30.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)\n\nI like the idea and this is really cool.\n\nFew more details we need to add, comment inline. Thanks.","accounts_in_message":[],"_revision_number":6},{"id":"0b71fd2f5f39ccff7f810195f041897e0b9bd6ed","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":28701,"name":"Dustin Specker","email":"v-dspecker@microsoft.com","username":"dustinspecker"},"date":"2021-02-17 22:26:06.000000000","message":"Removed reviewer Dustin Specker.","accounts_in_message":[],"_revision_number":6},{"id":"c83ccecb4ba6a37e7170438ac32b28f02586ce1e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 01:41:52.000000000","message":"Abandoned\n\nno activity since long. feel free to restore it if it is still needed","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"273336fc796cf342b47f8efa11c095bde7c21bb2","revisions":{"4b072040ba72f39299f9f8b2c897b1f1fdcd0261":{"kind":"REWORK","_number":1,"created":"2019-01-11 20:06:35.000000000","uploader":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"ref":"refs/changes/80/630380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/80/630380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/1"}}},"commit":{"parents":[{"commit":"ac9b2333d3c67709c095fb972c19689c5d0f2ab7","subject":"Show also Patrole specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/ac9b2333d3c67709c095fb972c19689c5d0f2ab7"}]}],"author":{"name":"nicolas","email":"nh202b@att.com","date":"2019-01-11 20:01:17.000000000","tz":-480},"committer":{"name":"nicolas","email":"nh202b@att.com","date":"2019-01-11 20:06:10.000000000","tz":-480},"subject":"Spec: Re-factor Account Generator","message":"Spec: Re-factor Account Generator\n\nRefactor the account generator function into two processes that\ncan be run independantly. This will provide more testing flexibility\nand customization without changing the current functionality.\n\nAdd the ability to:\ncreate account_file.yaml without creating resources\ncreate resources from provided files\nverify provided files\n\nChange-Id: I16a30c49c3b8c6e562594fef987db58b65f2219d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/4b072040ba72f39299f9f8b2c897b1f1fdcd0261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/4b072040ba72f39299f9f8b2c897b1f1fdcd0261"}]},"branch":"refs/heads/master"},"0d8405479a34aeb3e82db75edce01dd81c198e33":{"kind":"REWORK","_number":2,"created":"2019-01-29 19:05:19.000000000","uploader":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"ref":"refs/changes/80/630380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/80/630380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/2"}}},"commit":{"parents":[{"commit":"ac9b2333d3c67709c095fb972c19689c5d0f2ab7","subject":"Show also Patrole specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/ac9b2333d3c67709c095fb972c19689c5d0f2ab7"}]}],"author":{"name":"nicolas","email":"nh202b@att.com","date":"2019-01-11 20:01:17.000000000","tz":-480},"committer":{"name":"Nicolas Helgeson","email":"nh202b@att.com","date":"2019-01-29 19:05:16.000000000","tz":0},"subject":"Spec: Re-factor Account Generator","message":"Spec: Re-factor Account Generator\n\nRefactor the account generator function into two processes that\ncan be run independantly. This will provide more testing flexibility\nand customization without changing the current functionality.\n\nAdd the ability to:\ncreate account_file.yaml without creating resources\ncreate resources from provided files\nverify provided files\n\nChange-Id: I16a30c49c3b8c6e562594fef987db58b65f2219d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/0d8405479a34aeb3e82db75edce01dd81c198e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/0d8405479a34aeb3e82db75edce01dd81c198e33"}]},"branch":"refs/heads/master"},"1d76e1642b39ddc71c547033da6108c7c4accf05":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-02-15 16:55:41.000000000","uploader":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"ref":"refs/changes/80/630380/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/80/630380/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/3"}}},"commit":{"parents":[{"commit":"09de20ea6723a1ab7bcc9232d8476988a4f59ad8","subject":"Fix author-email in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/09de20ea6723a1ab7bcc9232d8476988a4f59ad8"}]}],"author":{"name":"nicolas","email":"nh202b@att.com","date":"2019-01-11 20:01:17.000000000","tz":-480},"committer":{"name":"Nicolas Helgeson","email":"nh202b@att.com","date":"2019-02-15 16:55:41.000000000","tz":0},"subject":"Spec: Re-factor Account Generator","message":"Spec: Re-factor Account Generator\n\nRefactor the account generator function into two processes that\ncan be run independantly. This will provide more testing flexibility\nand customization without changing the current functionality.\n\nAdd the ability to:\ncreate account_file.yaml without creating resources\ncreate resources from provided files\nverify provided files\n\nChange-Id: I16a30c49c3b8c6e562594fef987db58b65f2219d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/1d76e1642b39ddc71c547033da6108c7c4accf05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/1d76e1642b39ddc71c547033da6108c7c4accf05"}]},"branch":"refs/heads/master"},"5da0b1e354faa4335df975946b35ff6f147c84b7":{"kind":"REWORK","_number":4,"created":"2019-07-03 21:07:20.000000000","uploader":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"ref":"refs/changes/80/630380/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/80/630380/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/4"}}},"commit":{"parents":[{"commit":"09de20ea6723a1ab7bcc9232d8476988a4f59ad8","subject":"Fix author-email in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/09de20ea6723a1ab7bcc9232d8476988a4f59ad8"}]}],"author":{"name":"nicolas","email":"nh202b@att.com","date":"2019-01-11 20:01:17.000000000","tz":-480},"committer":{"name":"Nicolas Helgeson","email":"nh202b@att.com","date":"2019-07-03 21:07:13.000000000","tz":0},"subject":"Spec: Re-factor Account Generator","message":"Spec: Re-factor Account Generator\n\nRefactor the account generator function into two processes that\ncan be run independantly. This will provide more testing flexibility\nand customization without changing the current functionality.\n\nAdd the ability to:\ncreate account_file.yaml without creating resources\ncreate resources from provided files\nverify provided files\n\nChange-Id: I16a30c49c3b8c6e562594fef987db58b65f2219d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/5da0b1e354faa4335df975946b35ff6f147c84b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/5da0b1e354faa4335df975946b35ff6f147c84b7"}]},"branch":"refs/heads/master"},"598509c2f8affed7d82fa4656b4dd267e1d3fd13":{"kind":"REWORK","_number":5,"created":"2019-07-05 16:19:13.000000000","uploader":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"ref":"refs/changes/80/630380/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/80/630380/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/5"}}},"commit":{"parents":[{"commit":"09de20ea6723a1ab7bcc9232d8476988a4f59ad8","subject":"Fix author-email in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/09de20ea6723a1ab7bcc9232d8476988a4f59ad8"}]}],"author":{"name":"nicolas","email":"nh202b@att.com","date":"2019-01-11 20:01:17.000000000","tz":-480},"committer":{"name":"Nicolas Helgeson","email":"nh202b@att.com","date":"2019-07-05 16:19:06.000000000","tz":0},"subject":"Spec: Re-factor Account Generator","message":"Spec: Re-factor Account Generator\n\nRefactor the account generator function into two processes that\ncan be run independantly. This will provide more testing flexibility\nand customization without changing the current functionality.\n\nAdd the ability to:\ncreate account_file.yaml without creating resources\ncreate resources from provided files\nverify provided files\n\nChange-Id: I16a30c49c3b8c6e562594fef987db58b65f2219d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/598509c2f8affed7d82fa4656b4dd267e1d3fd13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/598509c2f8affed7d82fa4656b4dd267e1d3fd13"}]},"branch":"refs/heads/master"},"273336fc796cf342b47f8efa11c095bde7c21bb2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-25 22:25:12.000000000","uploader":{"_account_id":23625,"name":"Nicolas Helgeson","email":"nh202b@att.com","username":"nhelgeson"},"ref":"refs/changes/80/630380/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/80/630380/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/80/630380/6"}}},"commit":{"parents":[{"commit":"cbe33846857c6dbeda4b4d5882b2b5a055a19de9","subject":"Remove redundancy and classify projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/cbe33846857c6dbeda4b4d5882b2b5a055a19de9"}]}],"author":{"name":"nicolas","email":"nh202b@att.com","date":"2019-01-11 20:01:17.000000000","tz":-480},"committer":{"name":"Nicolas Helgeson","email":"nh202b@att.com","date":"2019-07-25 22:25:12.000000000","tz":0},"subject":"Spec: Re-factor Account Generator","message":"Spec: Re-factor Account Generator\n\nRefactor the account generator function into two processes that\ncan be run independantly. This will provide more testing flexibility\nand customization without changing the current functionality.\n\nAdd the ability to:\ncreate account_file.yaml without creating resources\ncreate resources from provided files\nverify provided files\n\nChange-Id: I16a30c49c3b8c6e562594fef987db58b65f2219d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/273336fc796cf342b47f8efa11c095bde7c21bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/273336fc796cf342b47f8efa11c095bde7c21bb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
