)]}'
{"id":"openstack%2Fqa-specs~656890","triplet_id":"openstack%2Fqa-specs~master~Ic63f35b238d4410a438c99d228a932b3fa3ec98c","project":"openstack/qa-specs","branch":"master","hashtags":[],"change_id":"Ic63f35b238d4410a438c99d228a932b3fa3ec98c","subject":"Whitebox Tempest Plugin","status":"MERGED","created":"2019-05-02 22:35:02.000000000","updated":"2020-03-20 01:58:13.000000000","submitted":"2020-03-20 01:56:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656890-1584669417524-51d1296f","meta_rev_id":"8ed3c530119b9fd3c04d0c15fc339a945c1d8ead","_number":656890,"virtual_id_number":656890,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 01:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":1,"date":"2020-03-05 09:57:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2020-03-20 01:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2020-03-08 03:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-03-20 01:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 07:18:56.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2020-03-05 09:57:06.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-03-08 03:17:29.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-20 01:51:34.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-03-20 01:56:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ad6fd35fe69ad8676cc514b1b2abd921bb2471b","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-02 22:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e4f4d3c74eb9a86144dadabde74b609f7ddcfc7","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-05-02 22:35:33.000000000","message":"Patch Set 1: Workflow-1\n\nJust wanted to throw this up to have a link, nowhere near reviewable.","accounts_in_message":[],"_revision_number":1},{"id":"7d0340c1f92f543d50ca6efeb07a5a864bdbe9bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:43:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/656890/1/check/openstack-tox-docs/29a9d2a/ : FAILURE in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"905ef980e4c5b076492664e6d3977d27ce681a61","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2019-09-17 17:34:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b58593ddf19e4cf3d69f0c5b5801b3367a80590","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:43:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0affe7074ca3478a8d2e87cdfdb38665 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2317da4f61a6759a06aaf08cbfc7ea39a4e6525f","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2020-02-19 07:18:56.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good to me","accounts_in_message":[],"_revision_number":2},{"id":"824d52e4d995209d3edeceb8752904800f8da55a","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-02-25 07:50:37.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)\n\nI found a few typos, but the proposal overall makes sense.","accounts_in_message":[],"_revision_number":2},{"id":"7ff786b0cf4ca339da1fa30f6efda6712c1bb936","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-26 18:04:33.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nit is good to go, wanted to know devstack plugin requirement if it has to install something else on host side.","accounts_in_message":[],"_revision_number":2},{"id":"1b8a7a97c17f307da15ce6ff5a241a65fbd924a5","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-03 18:23:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ecee6d42fd7e7c71463f65f44af3aeb951e9b94d","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-03 18:24:02.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"194014a1bbef92032d3e637f410dfd99a92459f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 18:32:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c099c95447d4ad3835f34b1ef617728 : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5d9ebbb7cb617ef9840ca3e10048b69d02c73d1a","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-03 19:37:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c35b5361f1f15814ae13c4a36d809fbcec8a193f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-03 19:44:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f12292683b44ae46af2dafde86e894fe70ef25c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-03 19:45:41.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nThis will not pass \"tox -e docs\" (run it locally to verify).","accounts_in_message":[],"_revision_number":4},{"id":"711c0fd3deeb887136ceda86f176ec5676d40b6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 19:52:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a6434bc9ae4c71bf1825b21fcdc7b5 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":4},{"id":"46c003b0c6c6a0ece05e3778dde88e38dd604dd7","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-04 13:33:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c1c3dda110b56a62d826616e0518ff0de87c3800","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-04 13:33:41.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fec904ecb862a2b173522d54a278bfb5ef2be117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 13:44:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b26d5ed8a144bb2bf3042ccae0e3ced : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":5},{"id":"4159d250584e656790ba85bb3ad87293de986615","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-04 17:36:02.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks. lgtm.","accounts_in_message":[],"_revision_number":5},{"id":"231492f611de539bf4d9088530087ba0e7920d82","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-03-04 23:27:35.000000000","message":"Patch Set 5: Code-Review+1\n\n(5 comments)\n\nLGTM except for the URL. And how about adding the URL for whitebox tempest plugin\u0027s repo to this spec and/or in the commit message as a reference?\n\n　https://opendev.org/x/whitebox-tempest-plugin","accounts_in_message":[],"_revision_number":5},{"id":"8858bbd57537a9f452cce787bc761ecbcd4853c0","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-05 01:38:29.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0c45b75aba9a69fe42ba169d28b7435379c82d9d","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-05 01:38:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bbb941c9e6ba700ce990347d77444440955e0399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 01:47:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2535987916462e9f0b34dc0794c722 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":6},{"id":"9e4ea46c2a87b3d8fb5a986c29f6d3603390cbad","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-03-05 09:57:06.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"9274d6a51251f8bf8e39d8aba8fec9eea802f001","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-08 03:17:29.000000000","message":"Patch Set 6: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":6},{"id":"ff437ddeb280d3e7219d656e9622df8022019933","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-03-20 01:51:34.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThanks for updating! LGTM","accounts_in_message":[],"_revision_number":6},{"id":"8afe8a18328e34cbfdc2ba8f60aa41d85f0e7ba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 01:51:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fea5f265790740d273dee8173422e0199297a561","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 01:56:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6716cf0618dc47da83e0bedbb67e95d3 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":6},{"id":"64e01ca4d7d55e00d98797695e0e3677a19453d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 01:56:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"8ed3c530119b9fd3c04d0c15fc339a945c1d8ead","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 01:58:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/07bae18c92494ba08ea2cb392ae4b81a : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6c500e1933b35edfe6092f3339ec3ed364bfc7dd","revisions":{"86a2b1a6eb27fdbab8c6c2aa890cf2dc4e9cce5a":{"kind":"REWORK","_number":1,"created":"2019-05-02 22:35:02.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/90/656890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/90/656890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/1"}}},"commit":{"parents":[{"commit":"0e227644bf2843dbb119fb8608cbbe4f2b068fa0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/0e227644bf2843dbb119fb8608cbbe4f2b068fa0"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-05-02 22:34:25.000000000","tz":-360},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-05-02 22:35:01.000000000","tz":-360},"subject":"[Very WIP] Whitebox Templest Plugin","message":"[Very WIP] Whitebox Templest Plugin\n\nChange-Id: Ic63f35b238d4410a438c99d228a932b3fa3ec98c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/86a2b1a6eb27fdbab8c6c2aa890cf2dc4e9cce5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/86a2b1a6eb27fdbab8c6c2aa890cf2dc4e9cce5a"}]},"branch":"refs/heads/master"},"1dd109ca0fa7ecbb8cb3e54bf02ffebd5241a091":{"kind":"REWORK","_number":2,"created":"2019-09-17 17:34:37.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/90/656890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/90/656890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/2"}}},"commit":{"parents":[{"commit":"cbe33846857c6dbeda4b4d5882b2b5a055a19de9","subject":"Remove redundancy and classify projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/cbe33846857c6dbeda4b4d5882b2b5a055a19de9"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-05-02 22:34:25.000000000","tz":-360},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-09-17 17:33:06.000000000","tz":-240},"subject":"Whitebox Templest Plugin","message":"Whitebox Templest Plugin\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nChange-Id: Ic63f35b238d4410a438c99d228a932b3fa3ec98c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/1dd109ca0fa7ecbb8cb3e54bf02ffebd5241a091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/1dd109ca0fa7ecbb8cb3e54bf02ffebd5241a091"}]},"branch":"refs/heads/master"},"2ac200e678b2015b15ef0d1b714cbe69291da27b":{"kind":"REWORK","_number":3,"created":"2020-03-03 18:23:59.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/90/656890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/90/656890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/3"}}},"commit":{"parents":[{"commit":"cbe33846857c6dbeda4b4d5882b2b5a055a19de9","subject":"Remove redundancy and classify projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/cbe33846857c6dbeda4b4d5882b2b5a055a19de9"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-05-02 22:34:25.000000000","tz":-360},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-03 18:23:42.000000000","tz":-300},"subject":"Whitebox Templest Plugin","message":"Whitebox Templest Plugin\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nChange-Id: Ic63f35b238d4410a438c99d228a932b3fa3ec98c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/2ac200e678b2015b15ef0d1b714cbe69291da27b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/2ac200e678b2015b15ef0d1b714cbe69291da27b"}]},"branch":"refs/heads/master"},"4621dbf49f84286eef50353cdeae6dfe41ba2d15":{"kind":"REWORK","_number":4,"created":"2020-03-03 19:37:34.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/90/656890/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/90/656890/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/4"}}},"commit":{"parents":[{"commit":"cbe33846857c6dbeda4b4d5882b2b5a055a19de9","subject":"Remove redundancy and classify projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/cbe33846857c6dbeda4b4d5882b2b5a055a19de9"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-05-02 22:34:25.000000000","tz":-360},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-03 19:37:27.000000000","tz":-300},"subject":"Whitebox Templest Plugin","message":"Whitebox Templest Plugin\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nChange-Id: Ic63f35b238d4410a438c99d228a932b3fa3ec98c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/4621dbf49f84286eef50353cdeae6dfe41ba2d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/4621dbf49f84286eef50353cdeae6dfe41ba2d15"}]},"branch":"refs/heads/master"},"69a6bce70f61563fcf315fabd1db52965558c5e8":{"kind":"REWORK","_number":5,"created":"2020-03-04 13:33:10.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/90/656890/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/90/656890/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/5"}}},"commit":{"parents":[{"commit":"cbe33846857c6dbeda4b4d5882b2b5a055a19de9","subject":"Remove redundancy and classify projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/cbe33846857c6dbeda4b4d5882b2b5a055a19de9"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-05-02 22:34:25.000000000","tz":-360},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-04 13:32:51.000000000","tz":-300},"subject":"Whitebox Tempest Plugin","message":"Whitebox Tempest Plugin\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nChange-Id: Ic63f35b238d4410a438c99d228a932b3fa3ec98c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/69a6bce70f61563fcf315fabd1db52965558c5e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/69a6bce70f61563fcf315fabd1db52965558c5e8"}]},"branch":"refs/heads/master"},"6c500e1933b35edfe6092f3339ec3ed364bfc7dd":{"kind":"REWORK","_number":6,"created":"2020-03-05 01:38:29.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/90/656890/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/90/656890/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/90/656890/6"}}},"commit":{"parents":[{"commit":"cbe33846857c6dbeda4b4d5882b2b5a055a19de9","subject":"Remove redundancy and classify projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/cbe33846857c6dbeda4b4d5882b2b5a055a19de9"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2019-05-02 22:34:25.000000000","tz":-360},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-05 01:38:21.000000000","tz":-300},"subject":"Whitebox Tempest Plugin","message":"Whitebox Tempest Plugin\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nChange-Id: Ic63f35b238d4410a438c99d228a932b3fa3ec98c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/6c500e1933b35edfe6092f3339ec3ed364bfc7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/6c500e1933b35edfe6092f3339ec3ed364bfc7dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
