)]}'
{"id":"openstack%2Fqa-specs~664212","triplet_id":"openstack%2Fqa-specs~master~I3d4e6f862d90c86855085621cfb1557ea38eed89","project":"openstack/qa-specs","branch":"master","topic":"switch-openstackdocstheme","hashtags":[],"change_id":"I3d4e6f862d90c86855085621cfb1557ea38eed89","subject":"Switch from oslosphinx to openstackdocstheme","status":"MERGED","created":"2019-06-10 07:28:38.000000000","updated":"2019-06-13 04:48:58.000000000","submitted":"2019-06-13 04:48:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664212-1560401338372-8e310be5","meta_rev_id":"0ea1621c122b596cbda07c71f9fe75d4d70fa628","_number":664212,"virtual_id_number":664212,"owner":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-06-13 04:48:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2019-06-13 04:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2019-06-13 04:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 04:39:22.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-13 04:48:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54586e57319bea95e52a59311377c13d23eeb374","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-10 07:28:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffeae375d84d53d8945de70e2a6a5308bb91512f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 07:34:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/664212/1/check/openstack-tox-docs/7ca8123/html/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ba172d59f29a1f107d13b663b885a7f713a3d79f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-13 04:39:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nfast approving as this is doc update only and fixing gate.","accounts_in_message":[],"_revision_number":1},{"id":"e3ade9dd8d3117ec85b6ddfcbd688170d4ceada3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 04:40:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b47b788faaf521a7bd667fae9531a243f800f21e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 04:48:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"122edf3d0adb9d2c342da3f266bc99b4ce8aacd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 04:48:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/664212/1/gate/openstack-tox-docs/22958ec/html/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"439a41824b3e1ae5120b0d48195bc72721e51138","revisions":{"439a41824b3e1ae5120b0d48195bc72721e51138":{"kind":"REWORK","_number":1,"created":"2019-06-10 07:28:38.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/12/664212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/12/664212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/12/664212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/12/664212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/12/664212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/12/664212/1"}}},"commit":{"parents":[{"commit":"0e227644bf2843dbb119fb8608cbbe4f2b068fa0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/0e227644bf2843dbb119fb8608cbbe4f2b068fa0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-06-10 07:23:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-06-10 07:23:02.000000000","tz":540},"subject":"Switch from oslosphinx to openstackdocstheme","message":"Switch from oslosphinx to openstackdocstheme\n\nThis patch switches to use openstackdocstheme instead of oslosphinx to\nuse it properly[0].\n\n[0] https://docs.openstack.org/openstackdocstheme/latest/#using-the-theme\n\nChange-Id: I3d4e6f862d90c86855085621cfb1557ea38eed89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/439a41824b3e1ae5120b0d48195bc72721e51138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/439a41824b3e1ae5120b0d48195bc72721e51138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
