)]}'
{"id":"openstack%2Fqa-specs~665078","triplet_id":"openstack%2Fqa-specs~master~I81be42630f2e6661d52777d35ce2db35b8214800","project":"openstack/qa-specs","branch":"master","hashtags":[],"change_id":"I81be42630f2e6661d52777d35ce2db35b8214800","subject":"Move completed spec to \u0027implemented\u0027 folders","status":"MERGED","created":"2019-06-13 06:53:19.000000000","updated":"2019-06-17 04:51:56.000000000","submitted":"2019-06-17 04:51:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665078-1560747116348-e3a0d8e9","meta_rev_id":"d8c4893993034137ba1e5e5d7a64be06956a9b40","_number":665078,"virtual_id_number":665078,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-06-17 04:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2019-06-13 07:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2019-06-13 07:30:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-13 09:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":1,"date":"2019-06-17 04:46:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 07:30:00.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-06-13 09:14:39.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-17 04:46:03.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-06-17 04:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80282d086274628bbde9c91592bbabe5e559da3b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-13 06:53:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf063fe8499adc92b04c659bd71997121bf3094a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 07:02:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/78/665078/1/check/openstack-tox-docs/1f72a38/html/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7d6b9d2cdb0e8d65ac945fe54063b9b801081d4f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-13 07:15:51.000000000","message":"Patch Set 1: Code-Review+2\n\nmake sense, lgtm","accounts_in_message":[],"_revision_number":1},{"id":"51ed4b1700ddf363d40b75724c111f90654f4837","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-06-13 07:30:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7999ee31d2dadebe0051106f7b6db3d66ebbd02a","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-13 09:14:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fbc545a60ee4a2ad9d4bc757b98817247a7a7902","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-17 04:46:03.000000000","message":"Patch Set 1: Workflow+1\n\nFast merging. I think we don\u0027t need to wait that much because this is just a housekeeping patch.","accounts_in_message":[],"_revision_number":1},{"id":"05a7c45721c56ff13070743c515678a960eefa45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 04:46:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c1919987a08a7fdef35dd2253e15d4a0bb6daa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 04:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/78/665078/1/gate/openstack-tox-docs/9fb0b53/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d8c4893993034137ba1e5e5d7a64be06956a9b40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 04:51:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00723dad46f1211e4310efcb500f5a70e9341626","revisions":{"00723dad46f1211e4310efcb500f5a70e9341626":{"kind":"REWORK","_number":1,"created":"2019-06-13 06:53:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/78/665078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/78/665078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/78/665078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/78/665078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/78/665078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/78/665078/1"}}},"commit":{"parents":[{"commit":"439a41824b3e1ae5120b0d48195bc72721e51138","subject":"Switch from oslosphinx to openstackdocstheme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/439a41824b3e1ae5120b0d48195bc72721e51138"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-06-13 06:52:40.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-06-13 06:53:17.000000000","tz":0},"subject":"Move completed spec to \u0027implemented\u0027 folders","message":"Move completed spec to \u0027implemented\u0027 folders\n\nChange-Id: I81be42630f2e6661d52777d35ce2db35b8214800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/00723dad46f1211e4310efcb500f5a70e9341626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/00723dad46f1211e4310efcb500f5a70e9341626"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
