)]}'
{"id":"openstack%2Fqa-specs~750974","triplet_id":"openstack%2Fqa-specs~master~Id93200ae85c25dfb713e96a09d2ee4aea611ed33","project":"openstack/qa-specs","branch":"master","topic":"feature/pdf_doc","hashtags":[],"change_id":"Id93200ae85c25dfb713e96a09d2ee4aea611ed33","subject":"Create tox environment to build pdf documentation","status":"MERGED","created":"2020-09-10 12:37:54.000000000","updated":"2020-10-08 19:31:36.000000000","submitted":"2020-10-08 19:29:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750974-1602185391247-7963eb8c","meta_rev_id":"922370d2b2fb1858f67f263d1323067807c32597","_number":750974,"virtual_id_number":750974,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 19:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2020-10-01 07:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-08 19:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"date":"2020-10-01 09:12:29.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-08 19:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-10-01 09:12:18.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 14:21:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-17 14:21:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-10-01 07:52:50.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-10-01 09:12:18.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-10-01 09:12:29.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-10-08 19:17:21.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-10-08 19:29:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bb88c75535db9f37078abce7c13cbe3d43fecc4","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-09-10 12:37:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffc60a066758d0c2fc47e4d5688af7a4a808096f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 12:52:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29805fb8da25408aad609154476e2a71 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f35b6d324164bfeff1c202dcf904117626ec769d","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-09-17 12:50:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8df1f6f895e28019f3a243e5203f82e6b9abf1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 13:09:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ea8af807814e52a9e6f0ed67036a8b : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":2},{"id":"6d4e2a92106c065bc10fc966eaf64bd993b7d719","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-09-18 12:38:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"242c9a0edbd75cde3c4fa3ccd47964d99664c58e","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-09-18 12:54:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4297d3fc33801de4c1a406d647aab46883f7cd34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:13:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3219615df004fc6a735a99d9000ac28 : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":4},{"id":"071ed300d419e35133261cc75a1698edbfbef605","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-09-18 13:26:42.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"aee3184d02c5786afa45513e80057379a0392afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:44:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062ac9f81de648a0a89508c336213f04 : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":4},{"id":"f5493ea418b8dd0e8255a66b1eae2fb6b3c3e126","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-09-18 13:45:44.000000000","message":"Patch Set 4:\n\nDocumentation: https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_062/750974/4/check/openstack-tox-docs/062ac9f/pdf/doc-qa-specs.pdf","accounts_in_message":[],"_revision_number":4},{"id":"2d4943771f558b5e0654c9b091037346fbf2d32f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-29 20:21:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a6a5140f88fffefda63c171d3c4ce72958d7805e","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-10-01 07:09:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3a140fedb9d1a88d423b002e4f56c8c240267516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 07:18:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a80c3abfa44442da534a38b16bb491b : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":5},{"id":"1348d3c6552d10984cef4e32fcf3310c61491c8b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-01 07:52:50.000000000","message":"Patch Set 5: Code-Review+2\n\nlooks good, thanks","accounts_in_message":[],"_revision_number":5},{"id":"bc21bbf201c6c8ce3ecdfadc9352b23a2acf2cf1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-08 19:17:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a339b8336d656a772882477f633aa763fdf82b24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 19:17:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"21c9c709b3e085fb92eaecdca9efe9e4f48eb1eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 19:29:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"cb7ea8bdc7c5a1cbc8a1a4723bd8182225306afb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 19:29:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e465e908f0f14798949fa59cfd60d29e : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":5},{"id":"8bb7f2f074dccc903c2bd851f13ec801aa83d625","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 19:31:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/170e11416f45450faf5a902ca66a3564 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3349e897b9835c91f9fa75e8a6c9770e8abea890","revisions":{"a85b7fa709ba1ff48b482900d06214eb9359ff7a":{"kind":"REWORK","_number":1,"created":"2020-09-10 12:37:54.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/750974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/74/750974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/1"}}},"commit":{"parents":[{"commit":"be7dc8aa490ae826710dbccaff6ab5640428c64e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/be7dc8aa490ae826710dbccaff6ab5640428c64e"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-10 12:32:06.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-10 12:37:49.000000000","tz":120},"subject":"Create tox environment to build pdf documentation","message":"Create tox environment to build pdf documentation\n\nThis patch enables building .pdf documentation using new \u0027pdf-docs\u0027\ntox environment. The newly created tox environment creates latex\nsource codes using sphinx-build and then builds the pdf\ndocumentation using make.\n\nChange-Id: Id93200ae85c25dfb713e96a09d2ee4aea611ed33\nTask: 35470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/a85b7fa709ba1ff48b482900d06214eb9359ff7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/a85b7fa709ba1ff48b482900d06214eb9359ff7a"}]},"branch":"refs/heads/master"},"62c85ecba7ee8b5cbfe12db4fa0de08cc5f06f5d":{"kind":"REWORK","_number":2,"created":"2020-09-17 12:50:27.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/750974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/74/750974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/2"}}},"commit":{"parents":[{"commit":"be7dc8aa490ae826710dbccaff6ab5640428c64e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/be7dc8aa490ae826710dbccaff6ab5640428c64e"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-10 12:32:06.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-17 12:50:17.000000000","tz":120},"subject":"Create tox environment to build pdf documentation","message":"Create tox environment to build pdf documentation\n\nThis patch enables building .pdf documentation using new \u0027pdf-docs\u0027\ntox environment. The newly created tox environment creates latex\nsource codes using sphinx-build and then builds the pdf\ndocumentation using make.\n\nChange-Id: Id93200ae85c25dfb713e96a09d2ee4aea611ed33\nTask: 35470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/62c85ecba7ee8b5cbfe12db4fa0de08cc5f06f5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/62c85ecba7ee8b5cbfe12db4fa0de08cc5f06f5d"}]},"branch":"refs/heads/master"},"7f934c8594572bd13d593adc7dabcc3484fcc1a9":{"kind":"REWORK","_number":3,"created":"2020-09-18 12:38:38.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/750974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/74/750974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/3"}}},"commit":{"parents":[{"commit":"be7dc8aa490ae826710dbccaff6ab5640428c64e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/be7dc8aa490ae826710dbccaff6ab5640428c64e"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-10 12:32:06.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-18 12:38:20.000000000","tz":120},"subject":"Create tox environment to build pdf documentation","message":"Create tox environment to build pdf documentation\n\nThis patch enables building .pdf documentation using new \u0027pdf-docs\u0027\ntox environment. The newly created tox environment creates latex\nsource codes using sphinx-build and then builds the pdf\ndocumentation using make.\n\nStory: 2006070\nTask: 35470\n\nChange-Id: Id93200ae85c25dfb713e96a09d2ee4aea611ed33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/7f934c8594572bd13d593adc7dabcc3484fcc1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/7f934c8594572bd13d593adc7dabcc3484fcc1a9"}]},"branch":"refs/heads/master"},"2b9a05542156caa3582bdffa55991375b1ce496d":{"kind":"REWORK","_number":4,"created":"2020-09-18 12:54:18.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/750974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/74/750974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/4"}}},"commit":{"parents":[{"commit":"be7dc8aa490ae826710dbccaff6ab5640428c64e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/be7dc8aa490ae826710dbccaff6ab5640428c64e"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-10 12:32:06.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-18 12:54:06.000000000","tz":120},"subject":"Create tox environment to build pdf documentation","message":"Create tox environment to build pdf documentation\n\nThis patch enables building .pdf documentation using new \u0027pdf-docs\u0027\ntox environment. The newly created tox environment creates latex\nsource codes using sphinx-build and then builds the pdf\ndocumentation using make.\n\nStory: 2006070\nTask: 35470\n\nChange-Id: Id93200ae85c25dfb713e96a09d2ee4aea611ed33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/2b9a05542156caa3582bdffa55991375b1ce496d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/2b9a05542156caa3582bdffa55991375b1ce496d"}]},"branch":"refs/heads/master"},"3349e897b9835c91f9fa75e8a6c9770e8abea890":{"kind":"REWORK","_number":5,"created":"2020-10-01 07:09:06.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/74/750974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/74/750974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/74/750974/5"}}},"commit":{"parents":[{"commit":"be7dc8aa490ae826710dbccaff6ab5640428c64e","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/be7dc8aa490ae826710dbccaff6ab5640428c64e"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-09-10 12:32:06.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-10-01 07:06:55.000000000","tz":120},"subject":"Create tox environment to build pdf documentation","message":"Create tox environment to build pdf documentation\n\nThis patch enables building .pdf documentation using new \u0027pdf-docs\u0027\ntox environment. The newly created tox environment creates latex\nsource codes using sphinx-build and then builds the pdf\ndocumentation using make.\n\nStory: 2006070\nTask: 35470\n\nChange-Id: Id93200ae85c25dfb713e96a09d2ee4aea611ed33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/3349e897b9835c91f9fa75e8a6c9770e8abea890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/3349e897b9835c91f9fa75e8a6c9770e8abea890"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
