)]}'
{"id":"openstack%2Fqa-specs~93037","triplet_id":"openstack%2Fqa-specs~master~I28fa0300a99e1cb5f4ea30546be7c44e67b7b985","project":"openstack/qa-specs","branch":"master","topic":"bp/client-checks-success","hashtags":[],"change_id":"I28fa0300a99e1cb5f4ea30546be7c44e67b7b985","subject":"Add spec for checking success response in clients","status":"MERGED","created":"2014-05-09 13:14:57.000000000","updated":"2014-06-12 18:00:36.000000000","submitted":"2014-06-12 18:00:35.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"83622d7307b2b0d0fb94ab50cde199beba2ead8a","_number":93037,"virtual_id_number":93037,"owner":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":2,"date":"2014-06-12 18:00:35.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2014-06-05 18:47:24.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"all":[{"value":2,"date":"2014-06-10 16:48:14.000000000","_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"value":1,"date":"2014-06-10 16:17:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":1,"date":"2014-06-11 05:04:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2014-06-11 21:14:01.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":2,"date":"2014-06-10 22:24:05.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"all":[{"value":0,"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2014-06-12 17:20:21.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-06-10 16:17:41.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2014-06-10 16:48:14.000000000","updated_by":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"reviewer":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"state":"REVIEWER"},{"updated":"2014-06-10 22:24:05.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2014-06-11 05:04:22.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2014-06-12 17:20:21.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2014-06-12 18:00:35.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"82e37d5a2d2e3b73eec5614713f1695366dee303","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-05-09 13:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5267bcb8012db552e27cab7b6f9edab4ec13f078","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-09 13:14:59.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"f0b2ddf9ff8f636e77729507f9dc3c7487af5815","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-09 13:15:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/1/check/gate-qa-specs-docs/07ae728/doc/build/html/ : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"b570462a0621e5c6671ff324ea789e9683f1317f","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-05-09 13:22:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"611ea171b0403b0757dc940bb0d73f63a889c1ec","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-09 13:22:11.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"a4d8b790ed445de05ba877be8a2f1429e0550c4d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-09 13:22:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/2/check/gate-qa-specs-docs/1989534/doc/build/html/ : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"f55c359900a51b18c015139157aa152b5165ec14","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-05-09 17:24:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88359bc1efb328b569a43c78d1d95e740cbc7c07","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-05-19 08:49:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"607d12c2689e55403f421b6e7d294f1d78fa8f95","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-19 08:49:34.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"223659bc6a20adb5343d7ef69c0bccb2795c0322","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-19 08:50:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/3/check/gate-qa-specs-docs/e484889/doc/build/html/ : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3},{"id":"ff3b8442d8e4f34fab2ea47b6f05b62f5470d728","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 06:37:06.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"387fa33410ea495ee3826e873bf0e6397a9bb631","author":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"date":"2014-05-23 06:37:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nHi David, LGTM, Just some questions.","accounts_in_message":[],"_revision_number":3},{"id":"9103eb6dadcb4eb604466fd7dac09c9e953084b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 06:45:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/3/check/gate-qa-specs-docs/48ec8b4/doc/build/html/ : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4a3c4e4ab89e814a88f85c382a7967cf3400c4cd","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-06-02 11:54:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c1436b09fcbbb37e636d8a52d297d6101a05095","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-02 11:54:35.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"15b2d1d8a65fa740bde5f3f4e41371035676f54f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-02 11:55:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/3/check/gate-qa-specs-docs/71d2793/doc/build/html/ : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3},{"id":"4932c9ee07d6ff4f5d75dd483a9b5ab7fec5729a","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-06-05 18:47:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b48123201e0f57ec81fb168b44f3c590cca7c25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-05 18:47:28.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"6c3265c6928b5a7a382c25cc00263f672f5ee4c3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-05 20:01:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/3/check/gate-qa-specs-docs/5c03329/doc/build/html/ : SUCCESS in 32s","accounts_in_message":[],"_revision_number":3},{"id":"301fc8f9eafce66b37385c26a5793b8a652e029d","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2014-06-10 13:31:20.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThanks for this, it\u0027s definitely something we should have. \nI think it would be good to have a little more details on how you plan to implement this.","accounts_in_message":[],"_revision_number":3},{"id":"893a1cc42bc3b591259049a555ad66614985fc30","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 13:31:21.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"be43f7d5ab97351ffb7765d935f71f67c4594356","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 13:39:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/3/check/gate-qa-specs-docs/209b7fa/doc/build/html/ : SUCCESS in 26s","accounts_in_message":[],"_revision_number":3},{"id":"177e1cab65aecc28ec289b6aee4d3903786b7419","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-06-10 16:02:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba600a4cc562da79ab343a08295efc34f98bcfba","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2014-06-10 16:17:41.000000000","message":"Patch Set 3: Code-Review+1\n\nthanks!","accounts_in_message":[],"_revision_number":3},{"id":"7f9de565de31d5084fe34de07ff1492ba227e2f1","author":{"_account_id":7872,"name":"Marc Koderer","email":"marc@koderer.com","username":"mkoderer"},"date":"2014-06-10 16:48:14.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nSince my comment is only a nit: LGTM","accounts_in_message":[],"_revision_number":3},{"id":"dc26e9aeb109ae74a15b0e742dd7687fb0cb4fdf","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-06-10 22:24:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d4b268a7450fd2027c2cb5a8cd2eaf2f9ed0b428","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2014-06-11 05:04:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"46f1c3b01898bf8af73c1460fbf3fa7d02ff5d31","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2014-06-11 21:14:01.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)\n\nLGTM just a couple quick questions and then I\u0027m +A","accounts_in_message":[],"_revision_number":3},{"id":"27816e8c0f129629b6fb55cee9d85f4cbcb3b2da","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-06-12 16:05:22.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6bc4c53fcc850eb722a04e460914563b41395bba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-12 17:20:21.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"eb5361c19244f6902205440fee4c962e87e8b46d","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2014-06-12 17:20:21.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"69a97d76fa715cadc459a0553c782a1513f75fda","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-12 17:59:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/3/check/gate-qa-specs-docs/f98da49/doc/build/html/ : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8854daef94faccbb4d06fcca8545749c406a42cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-12 17:59:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"a5eff96ff2963767dc11c8d000c4e3313def9ffb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-12 18:00:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded.\n\n- gate-qa-specs-docs http://docs-draft.openstack.org/37/93037/3/gate/gate-qa-specs-docs/a1b3515/doc/build/html/ : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3},{"id":"2992ac9ee4a4f27b51d167d8776358df5347940c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-12 18:00:36.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b4a91c6e0f7498d3cbadae5e5e7fa983f345504","revisions":{"3baae4c9eea710729f79b746be26dfc026c4cc65":{"kind":"REWORK","_number":1,"created":"2014-05-09 13:14:57.000000000","uploader":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"ref":"refs/changes/37/93037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/37/93037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/1"}}},"commit":{"parents":[{"commit":"6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624","subject":"add bp:multi-keystone-api-version-tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624"}]}],"author":{"name":"David Kranz","email":"dkranz@redhat.com","date":"2014-05-09 13:13:42.000000000","tz":-240},"committer":{"name":"David Kranz","email":"dkranz@redhat.com","date":"2014-05-09 13:13:42.000000000","tz":-240},"subject":"Add spec for checking success response in clients","message":"Add spec for checking success response in clients\n\nSpecifies bp:client-checks-success\n\nChange-Id: I28fa0300a99e1cb5f4ea30546be7c44e67b7b985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/3baae4c9eea710729f79b746be26dfc026c4cc65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/3baae4c9eea710729f79b746be26dfc026c4cc65"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624","is_merged_in_target_branch":true,"change_id":"Ib9ff5ea56d708db7fd310a4f08dadacfc0b80f54","change_number":81294,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1e4befaaa893e281a87933da43092bf5718407f5":{"kind":"REWORK","_number":2,"created":"2014-05-09 13:22:10.000000000","uploader":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"ref":"refs/changes/37/93037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/37/93037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/2"}}},"commit":{"parents":[{"commit":"6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624","subject":"add bp:multi-keystone-api-version-tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624"}]}],"author":{"name":"David Kranz","email":"dkranz@redhat.com","date":"2014-05-09 13:13:42.000000000","tz":-240},"committer":{"name":"David Kranz","email":"dkranz@redhat.com","date":"2014-05-09 13:22:00.000000000","tz":-240},"subject":"Add spec for checking success response in clients","message":"Add spec for checking success response in clients\n\nSpecifies bp:client-checks-success\n\nChange-Id: I28fa0300a99e1cb5f4ea30546be7c44e67b7b985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/1e4befaaa893e281a87933da43092bf5718407f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/1e4befaaa893e281a87933da43092bf5718407f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624","is_merged_in_target_branch":true,"change_id":"Ib9ff5ea56d708db7fd310a4f08dadacfc0b80f54","change_number":81294,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9b4a91c6e0f7498d3cbadae5e5e7fa983f345504":{"kind":"REWORK","_number":3,"created":"2014-05-19 08:49:33.000000000","uploader":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"ref":"refs/changes/37/93037/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qa-specs","ref":"refs/changes/37/93037/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qa-specs refs/changes/37/93037/3"}}},"commit":{"parents":[{"commit":"6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624","subject":"add bp:multi-keystone-api-version-tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624"}]}],"author":{"name":"David Kranz","email":"dkranz@redhat.com","date":"2014-05-09 13:13:42.000000000","tz":-240},"committer":{"name":"David Kranz","email":"dkranz@redhat.com","date":"2014-05-19 08:49:13.000000000","tz":180},"subject":"Add spec for checking success response in clients","message":"Add spec for checking success response in clients\n\nSpecifies bp:client-checks-success\n\nChange-Id: I28fa0300a99e1cb5f4ea30546be7c44e67b7b985\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/9b4a91c6e0f7498d3cbadae5e5e7fa983f345504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qa-specs/commit/9b4a91c6e0f7498d3cbadae5e5e7fa983f345504"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6465dfdcd46a31f24d5e5f14fbfd4f4945ceb624","is_merged_in_target_branch":true,"change_id":"Ib9ff5ea56d708db7fd310a4f08dadacfc0b80f54","change_number":81294,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
